)]}'
{"id":"openstack%2Fswift~721214","triplet_id":"openstack%2Fswift~stable%2Fussuri~I331b55201839c772cfd521239c16b34aad731bca","project":"openstack/swift","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I331b55201839c772cfd521239c16b34aad731bca","subject":"Update .gitreview for stable/ussuri","status":"MERGED","created":"2020-04-20 09:59:57.000000000","updated":"2020-04-20 17:48:46.000000000","submitted":"2020-04-20 17:46:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721214-1587404806996-26f7b502","meta_rev_id":"7c98d3a1f39abaaec9c562fac62a11ccba7dd862","_number":721214,"virtual_id_number":721214,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-20 17:46:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-20 15:03:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-20 15:03:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 15:03:28.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-04-20 17:46:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1109134e247e41433af66aedad7f366062cb890e","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-20 09:59:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e20eb23245a53bd7d8671c9d87e46909b3ab149b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 12:17:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c413eb4ae540b5825319e626b2716a : SUCCESS in 9m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5013dbc32ab04e1995e3a9d07937769f : SUCCESS in 11m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/be0356bfbf5f479cb46be22bc2b5e8fd : SUCCESS in 12m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/52264d3ebe894c5c8b1ba24ab42fc925 : SUCCESS in 14m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9292a6a8fc444c03b90f9754bb846f71 : SUCCESS in 14m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6f27fa7a2a86459f82f07da816344275 : SUCCESS in 13m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/16b109370df94757b71cb4b5430b52a2 : SUCCESS in 16m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c6bf9c07a7a4f4bbba83c3ed91ceb4b : SUCCESS in 15m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03dcd9b69c82456984d60f4749db141a : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1c06844a1bb59e0cd8a691e1bcedf891057c7731","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-20 15:03:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9c06c521db6da44dcd6170f7ec435d8dd4845bc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 15:04:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f283572c056c1c0eb6026965cc477eac5b03ce53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 17:46:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba5c3a291898453786cc124b78967c44 : SUCCESS in 8m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/23df3e25a69b40098f53fc21d3076cb5 : SUCCESS in 11m 17s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e7ef6d4a4b864ee19e9e968f27d390fd : SUCCESS in 11m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f79d15be0f8d46218344402032513135 : SUCCESS in 13m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c5ebe79195c1435984d994a9d9af8937 : SUCCESS in 13m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ce77ddfa07fd4234bf500f983fcc2b2c : SUCCESS in 22m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5f36c16f71bd43c3a08669d0aef7bde7 : SUCCESS in 13m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ac2dcbebbff94206a1a3c23ea25c2425 : SUCCESS in 11m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3715260f97a0470c8228b4af5dee1e96 : SUCCESS in 13m 36s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c59222e2468943bfaa351314ed08dd4f : SUCCESS in 12m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/187d47d7dc424a82b0d4d76747df4c3e : SUCCESS in 11m 13s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/2898f78bd0e9439c9223803151a9387c : SUCCESS in 15m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/46523b552e8e4adcbeb95b1ffd7bbcf0 : SUCCESS in 10m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b093b2079ed548509bd0b355423b8d14 : SUCCESS in 11m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4c30cf9f9e4b460680bce19117504724 : SUCCESS in 12m 35s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d6c6144d3cf54376ab0b84b0efebca09 : SUCCESS in 17m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91ea7bec8b57449ebc49ab8bfb45771f : SUCCESS in 14m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf99684585d44713b9c55ef28bece64e : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"cff496acb24d4117bb9b2a65c22cb4b5e72ab0c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 17:46:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3a0cf1757e187daa63497d1d4b60dba1a9db0aa0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 17:48:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18a9e70a94574fc5aabb13f84caa779f : SUCCESS in 1m 37s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3b36d39828324fe19c8925ca7e099321 : SUCCESS in 1m 03s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/af6d4507f0304a8886aec8a43ba1bc20 : SUCCESS in 1m 08s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"630fea68371ceaca3b3fb40798c86361fa78909a","revisions":{"630fea68371ceaca3b3fb40798c86361fa78909a":{"kind":"REWORK","_number":1,"created":"2020-04-20 09:59:57.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/14/721214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/721214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/721214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/721214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/721214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/721214/1"}}},"commit":{"parents":[{"commit":"f3543f5346a3d7ae31fe193f1336ef178da7db9b","subject":"Authors/ChangeLog for 2.25.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3543f5346a3d7ae31fe193f1336ef178da7db9b"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-20 09:59:55.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-20 09:59:55.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: I331b55201839c772cfd521239c16b34aad731bca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/630fea68371ceaca3b3fb40798c86361fa78909a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/630fea68371ceaca3b3fb40798c86361fa78909a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
