)]}'
{"id":"openstack%2Fswift~721215","triplet_id":"openstack%2Fswift~stable%2Fussuri~Ibdb29f14eaeb354d792b844edd819a6f80acec6d","project":"openstack/swift","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"Ibdb29f14eaeb354d792b844edd819a6f80acec6d","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-04-20 09:59:59.000000000","updated":"2020-04-22 23:16:36.000000000","submitted":"2020-04-22 23:14:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721215-1587597297474-dfe5b5ea","meta_rev_id":"2beb97a3c74d107d6df5e8fb8b60a44d4e04f5d3","_number":721215,"virtual_id_number":721215,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-22 23:14:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-20 15:05:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-20 15:05:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 15:05:00.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-04-22 23:14:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc74c3c3ebcb469608f14d693a63b55f5daa8a06","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-20 09:59:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4eddb28ae7534dfab351ac67647295951076a978","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-20 15:05:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f23970802ae304a049d5259143cef5bbadb977bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 19:34:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac9642b056db4818af147e375109434c : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c61f28a8679481abdb84388f31ffc0a : SUCCESS in 9m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f710a11481784fa69eab72c90517f234 : SUCCESS in 10m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7da80d59c0374e3b9b382b8e6297b7aa : SUCCESS in 9m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0fc73c16cb8f47808516ae61ff91dd5c : SUCCESS in 12m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0db960b738d6487bb6a0fd4ad7ec4aa0 : SUCCESS in 13m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/796f11c0aae141cbba8c2010177cb01d : SUCCESS in 12m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f693ce96e1d442408e41635dfd6710e0 : SUCCESS in 12m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/614758eef6094b2ea3c66746a6ef07eb : SUCCESS in 10m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eb8a8ea106f74ae3a10ab51f13f3bd5c : SUCCESS in 35m 46s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ab2ba561248a46e69268b89b51cfcb02 : SUCCESS in 9m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4f547dc891d14e40b0a71cd6bc161ef1 : SUCCESS in 20m 10s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3c0ec8c5af224f6b9150043fa1713508 : SUCCESS in 15m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9d09ef1b409245db9b5ef8fc145d8bfc : SUCCESS in 10m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/32cb5007e59b430ab8a950cc2606e649 : SUCCESS in 13m 04s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/6404bb8f4a694b3f997df7ff95f39d3b : SUCCESS in 27m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1d20e40da39845b3b5512f303ace9825 : SUCCESS in 28m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1b7c8036d1d34a18977de9b2cededc63 : SUCCESS in 21m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5d93e4e2618a4a7facf02a166e2ba4ef : SUCCESS in 18m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d5a47dbcb27449669457dcecf9cb4bf2 : SUCCESS in 44m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b314da6e2584457cb5f29a0f8a36d770 : SUCCESS in 14m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f38d5847dc4644a189af8d55ae4871b4 : SUCCESS in 6m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d289f034bea24c4d8d3def4aab41d447 : FAILURE in 50m 52s","accounts_in_message":[],"_revision_number":1},{"id":"93e08759c5d1b7899b4db524dc84fd8ae65b2862","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-20 21:17:09.000000000","message":"Patch Set 1:\n\nrecheck\n\nLooks like some etag-quoter metadata didn\u0027t get cleaned up fully? Or something?","accounts_in_message":[],"_revision_number":1},{"id":"55cfdb41eda9d453859ad811312bb487c0997959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 01:08:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f98b25f73c5457cac949327956d59e5 : SUCCESS in 11m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b8928127c7d4742b209111ebfa9b6ff : SUCCESS in 8m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fdaad478e9ac48798806e1469bb1a835 : SUCCESS in 9m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c5db39d573294db09f888761d65a644c : SUCCESS in 10m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b441d1b9398749dc88397967dafd3274 : SUCCESS in 13m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/91800fddc8cf4cdab94555a85c9e7dd4 : SUCCESS in 16m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/83f5797a1f504c59af7216353aa30924 : SUCCESS in 32m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a50f764332b043f292be99fbc1e3d00f : SUCCESS in 26m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c4bd4b65a27e48988bb86eb202273c0f : SUCCESS in 10m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3309c74923f042de82177dffc4612abd : SUCCESS in 18m 03s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b5464438481d420d8830e30579bfba5b : SUCCESS in 15m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2d76788537da4eacb0c37cb95e268cf8 : SUCCESS in 10m 58s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0b9288b786d04f2eb1f1258e921b5822 : SUCCESS in 17m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c149b07da31a4f148656095c1ef0cade : SUCCESS in 10m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bc013353e44b439bb2b811f44a5e66af : SUCCESS in 13m 27s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/57b10203e25444abb6c2e43332364fb1 : TIMED_OUT in 41m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f3c38b47fa5042eda2200435bc5422ae : SUCCESS in 19m 17s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/9899d7806ba343fb91a5a0cd6682de68 : SUCCESS in 27m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bf5630b1d8b74b66a4ef8f17a58e56ae : SUCCESS in 23m 41s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c3b92ae815aa4466b91cb95c51f6e15d : TIMED_OUT in 1h 01m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58a984c3a33945daac177aa30fcfab36 : SUCCESS in 14m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0f06b464ef942fb861283a5baa4f289 : SUCCESS in 6m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bbbfcbcff20240ef8f3c5ce52f0ca79a : FAILURE in 21m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7f4d8ba3346e509c0f5b176f2a79bc09ea9530e1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-21 01:53:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ba09dd44802c72e15d8168dd4b422a6c3a885c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 02:48:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8041a3f65aa846c1a486b8840c9c725a : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73785ee9b60d4292b31220aea09d9d55 : SUCCESS in 9m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f8f21c188d7d4776a574d6e9396fc70f : SUCCESS in 10m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9cefaba1867242f69d711b30909347f0 : SUCCESS in 10m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1932c03f4549462e86936e681bd7ca79 : SUCCESS in 14m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/899b5c7cf64640f38470c4b815fb7e3a : SUCCESS in 14m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4ffa6322b5c6453388890d22de4ab4c9 : SUCCESS in 14m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0a819c34a8484c9284fd3364148aeb2e : SUCCESS in 28m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/777a6c7f677646af82f87857b5889adb : SUCCESS in 11m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c7539d518f814155a9ee40970c433218 : SUCCESS in 32m 48s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/457895e73b27495db7ac8f57d9064c4b : SUCCESS in 29m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6ce0968dee594affa2a92873614f3bf1 : SUCCESS in 12m 04s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/21bb5470398148bcbfa29e6a7ba92e5a : SUCCESS in 18m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/45c7fcbf43614b8184462c84247671fc : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/51d2796c994245b8b9329309b2f01a35 : SUCCESS in 12m 25s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4a97fc5aff214a7489be799e0b382065 : SUCCESS in 26m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/333715341dad4a5fb0bb898895c57cc2 : SUCCESS in 13m 56s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ff0af5e7778348588cfc65595db7e1a9 : SUCCESS in 35m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2846d3c6e4ec43b0b02c1567d649cc45 : SUCCESS in 18m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e7de579bc5104c36bccad091315f91e8 : SUCCESS in 53m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f266f11fcab48158a6e9211fba7b638 : SUCCESS in 30m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc49a7a8ca9147a6bbe870ac32507867 : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e40651fae1c4adfa3379e0240bb24c6 : FAILURE in 31m 12s","accounts_in_message":[],"_revision_number":1},{"id":"bfb19ae67df55332af0c05ff3878237e3db3b759","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-21 04:15:22.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"7303c4566da9c27a1456cb5b9f7132796e2c1a70","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 04:52:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/62762e6fd5ab40138e6835d9aaa048f1 : SUCCESS in 18m 19s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/42993c5f1dc34fcaae52c3643563c976 : SUCCESS in 11m 15s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/8d808abdbf4e48e8b4587eff90fa073e : SUCCESS in 14m 04s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/f99481728f464879b0b9866c8136802c : SUCCESS in 12m 31s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/ee314eaab7bc42f48995c6552b4a213e : SUCCESS in 15m 35s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/88eded62cae748f7a84a62a0f14d3099 : SUCCESS in 15m 17s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/87f69333a69b485ebd2a26cb6ef5452d : SUCCESS in 16m 44s","accounts_in_message":[],"_revision_number":1},{"id":"73637b41c83422a671efc04999ac3973f37bcae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 07:31:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70eb922692b74578ba906d38a0b74312 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2117d9184fc43329b0032ce60df9aa1 : SUCCESS in 8m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1b7e83bd45cf4282a47e040714d964cc : SUCCESS in 9m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/34b9f8be2da34f23b498f6d61f868882 : SUCCESS in 9m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2e9f51b9e7e94018a4e4946e8a2db060 : SUCCESS in 13m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a31c83d113a4f64828365cb283b0fc2 : SUCCESS in 14m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/25d98be30d16479d88a677bcaf4a166a : SUCCESS in 12m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a7a4dbdd18a749219066772790514779 : SUCCESS in 20m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/20602d71f6594d308fae9e3d684db4da : SUCCESS in 9m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7c69701d4a35439285384f1ce85a0946 : SUCCESS in 14m 33s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/41494171679f45e1b21c64c2ec0d9596 : SUCCESS in 11m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b7bbe20b26504bc38fc56c7f6b14e4a7 : SUCCESS in 10m 32s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f33a143feeee4458afa29f5210ce50bf : SUCCESS in 15m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/962f104d314c49c2a2006fa7b24e5c42 : SUCCESS in 10m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9a59f1c493c6495193fe04dea8c02f67 : SUCCESS in 14m 37s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d6091984a6e64d8fb4134e8bbc4d6a29 : SUCCESS in 10m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/31ac36df5fd34ed58ff60e430a2f3b6c : SUCCESS in 11m 53s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/09ac133ccb8e4d0789f1f84c8a387017 : SUCCESS in 16m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d33846059d2b48d79efbe10a24b07a4a : SUCCESS in 19m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/69c18117fcaa49b38115a2532934170c : SUCCESS in 52m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ab0f2e34c80423cb8765e7036ce7ff4 : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9534af7c4b8463c9a1cf0e305bdf9f2 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a561637fa2854cdb8490f5b9319623b7 : FAILURE in 32m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2578c4402a09d6b77d55aff554ed1e8532bd1912","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-22 14:42:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2762cd3ecdf3eaf8f3dbfc3fa737a403762b4fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 18:05:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27c839c9cc414d07a119e0683cc8b4b9 : SUCCESS in 12m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baa6e64d3c8343a98a90a1bd631a0767 : SUCCESS in 8m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bcb98694b57c4c5d91c4b1944ace1ac4 : SUCCESS in 12m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4f9c6bb3c9be406db2485d2bc0cdf40c : SUCCESS in 12m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/088def0e723c4e27907b125a81a1faa2 : SUCCESS in 14m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d730765ee21f4109865b68a922b9b822 : SUCCESS in 13m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cb8a42729fb449f5af258b80366c25a0 : SUCCESS in 12m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/38e2ac90623d4b9481c5e79248a7cd32 : SUCCESS in 14m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/921b8beebe944626b1f13b76504705ca : SUCCESS in 10m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f3d3da7afaad45919ad0df335c1a9b50 : SUCCESS in 16m 07s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1c2908c5b794452ab87177bb746f8800 : TIMED_OUT in 42m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c6ff46093388454ebc6f29923c7d6ae9 : SUCCESS in 16m 20s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a741f0d760cb4e0fb8713f75ff042c4e : TIMED_OUT in 43m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/efbf1411855c4d6da7d0de03291aab1f : TIMED_OUT in 41m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5ff58bd6c8af4c97a30b0bbaa8aeddd5 : SUCCESS in 18m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d32fbee4a7ce4107b5fe41da68369246 : SUCCESS in 14m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/329a8099decd4ec1bd784871c61362ef : SUCCESS in 13m 04s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c16d967a18ab42a9be1c2d95a09859b9 : SUCCESS in 24m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ac6e354dbc6f4054a43f76d5bd48d935 : SUCCESS in 22m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ad79e564504d4ef19bf5cdb90691bcaf : SUCCESS in 45m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffd4fa2f40be4b4390c68ae4e1f1d7bf : SUCCESS in 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53f07838d77d45558888a99511c162fa : SUCCESS in 6m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/34afa0d9b7604360b2e6857b15d1f9bc : FAILURE in 28m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4846b125bfa30a41737212382f7b082c11cb633","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-22 18:57:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bf4c353b46bade7fe1f417cd6041463f3c9ae1e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 19:47:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4be17b2f6ffe426fa938be7510c8fa45 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10537a981efb4ff69b36dec0c18a1677 : SUCCESS in 9m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/49cc264989454c84ba957b6c549f15a9 : SUCCESS in 10m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e6933e8bfad54e4aafc495a86dd56bf5 : SUCCESS in 11m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dceb8a0b257748d09c8c9c99c4b07165 : SUCCESS in 16m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/307e0a663d3d493ab82d3f8902c068de : SUCCESS in 14m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/76514992e55347cba46eb1e65f4f73b0 : SUCCESS in 17m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9416cfde28fd4608b1bf014fb8fad5ec : SUCCESS in 16m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b8c673a8334b421a86c082b55c7fa7ca : SUCCESS in 13m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/abf52cb73d89445ab0e583d211bd97e4 : SUCCESS in 15m 34s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c0e0a486fc0d4904a8c10e04c75b6b84 : SUCCESS in 12m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9e8a910b1a3548d1bbb695d56c257802 : SUCCESS in 15m 01s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/07e1d327251d489e9989c54c3e45c9f1 : SUCCESS in 17m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1983e4b590b146d8b1bd5e210ca0aebd : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4f5675bd92b542f7acd999b6e698b6ed : SUCCESS in 17m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e552ac819cb14b538c3abaced2ae9c39 : SUCCESS in 30m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e32fdbd205a545d4a78acbc37f06c54e : SUCCESS in 13m 28s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c457319bcda644eab47973b9511a9fb4 : SUCCESS in 20m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8122ed2ffc9543c3b44771530fdf60e1 : SUCCESS in 22m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2ed256297ba8490b9121a444b78b2693 : SUCCESS in 47m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/375fc30908784f0daeba9a2fdadfdd22 : SUCCESS in 15m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60710cea31f14ce48ca06a08a17e87a3 : SUCCESS in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/51a7f25979c144838a76f12c34808433 : FAILURE in 27m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"076cb3d05c685d0bde3404a964b4f617c4dc13e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 19:48:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d715ca786a1ab98fb53ffa0996b498a71b28b265","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 23:14:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47644edfdfd145d3bb7626755502231e : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d24f3550d8a54bb49329850ad27c97a8 : SUCCESS in 8m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/535ba7a8781845af9d29bacf302247d5 : SUCCESS in 10m 28s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e15cb5718b6843a49888d57c678a728a : SUCCESS in 12m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/02f37926b8d44cf39eb8f74cabaa93b4 : SUCCESS in 18m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c2d98773d8fb438ba871c8b6a0097878 : SUCCESS in 19m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f2c542cd2c5f4560bd4ef7dd5f0a25bf : SUCCESS in 17m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/677fb97bb5a244ccaa830626ed4ea8ae : SUCCESS in 16m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/af5a55305e564ef5b4c49d05f268711f : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c30ca7f93fa04e73baacf307773d2387 : SUCCESS in 18m 18s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/0f0421861d784a8abce25069a57ec5a0 : SUCCESS in 38m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9ac271360e8346eea086564c2a5c4fbc : SUCCESS in 16m 15s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a7c149f5749645b88a58ed3dc0f92a24 : SUCCESS in 20m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3664a7f9a0cc4c25b9075d29fc9c134d : SUCCESS in 20m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/537e52c92844422585959a07df165cd5 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/eb4c58897bac4a8c83edbe769763012c : SUCCESS in 13m 07s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6c5ac62e165d458e9d593fe2045cc71b : SUCCESS in 22m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/feaaabe80d19404a838cc7346250649e : SUCCESS in 49m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38f1d8ae828c47a99d04dda219cb3732 : SUCCESS in 18m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee4a64a3363345e6b9a32aa0b16ee7d2 : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7d78892ddf31dfcb632784b31e4ae4c36dc03da7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 23:14:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"074d32ba66ed9534ae8189e61ceffae28bca3b05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 23:16:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45c87e715305462f871619a892f1ac6c : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e872dd5a552a4855998b185dacf74cdf : SUCCESS in 1m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b9a9f929b12c400d94b8121aa6e06388 : SUCCESS in 1m 00s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0429d99519d14f65945eec889cda4b3f : SUCCESS in 55s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f9048fc25deb63b145c0ead4a8326def52447ef","revisions":{"5f9048fc25deb63b145c0ead4a8326def52447ef":{"kind":"REWORK","_number":1,"created":"2020-04-20 09:59:59.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/15/721215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/721215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/721215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/721215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/721215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/721215/1"}}},"commit":{"parents":[{"commit":"630fea68371ceaca3b3fb40798c86361fa78909a","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/630fea68371ceaca3b3fb40798c86361fa78909a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-20 09:59:57.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-20 09:59:57.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ibdb29f14eaeb354d792b844edd819a6f80acec6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f9048fc25deb63b145c0ead4a8326def52447ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f9048fc25deb63b145c0ead4a8326def52447ef"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
