)]}'
{"id":"openstack%2Fswift~721518","triplet_id":"openstack%2Fswift~master~Ifb1d091e58b9331b027edf981f6af5c015b56936","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ifb1d091e58b9331b027edf981f6af5c015b56936","subject":"func tests: Allow test_etag_quoter to be run multiple times","status":"MERGED","created":"2020-04-21 04:33:43.000000000","updated":"2020-04-23 21:33:21.000000000","submitted":"2020-04-23 01:28:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721518-1587605329086-792c2e80","meta_rev_id":"af83ab154853202da10bb370e3f6fd3afe94758f","_number":721518,"virtual_id_number":721518,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 01:28:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-22 23:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-22 23:26:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 01:28:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e942e65e6aca3b24c8c036973d4d663c213ab7e5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-21 04:33:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7274f0599c8029c0b9e72e2ff6e433a0d8353fae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 08:51:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9384ad368c343e4a9fb9049c3712ee0 : SUCCESS in 11m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/03607f50e8574de2a555cedc39669fd3 : SUCCESS in 10m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e24d4eea1e2c42bcac48082c0de04e92 : SUCCESS in 27m 35s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c3e28a15cdbf4d008abaf6e76769b1e1 : SUCCESS in 9m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d0f4b206bafe45c5a3e67587f183ec83 : SUCCESS in 10m 07s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/2f69c3af425c404cbea5a641326f3f3d : SUCCESS in 15m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c310adcbfcf44f0d8e9efb83204bf905 : SUCCESS in 9m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5b0a56af738f443e826966bc451fd0e4 : SUCCESS in 26m 11s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/518587217f0847c385a75d89b79246d9 : SUCCESS in 9m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/04a4023399d34a319352ca3da1cd30e3 : SUCCESS in 10m 41s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1110a34e5a6943778a3cd403036d9293 : SUCCESS in 15m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd7452415f71401b8af45c52065d74a0 : SUCCESS in 20m 04s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1e37fb4780043c688061097f045f75d : SUCCESS in 30m 08s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7e19530030564466a2fa89728096b6c0 : SUCCESS in 28m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dd016a85e8194892a79202db4c58565c : SUCCESS in 41m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/588ff6f7f2334359b29662c88fde1b75 : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ea2b3dd7a7aa4910abbeb50eb2fd6e9f : FAILURE in 34m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b4bd62879f4328f0f596ae9fec9eaa3dfb84bf40","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-22 00:59:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"22d3dc0d9f2a25bf1b43ae90b56d77a8a86bed4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 02:06:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac69d17c10b64d59a760b824783dd0b4 : SUCCESS in 20m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ec581d87cb1c4baa8944d0b09c9b13bb : SUCCESS in 9m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bb98417f7d724d91b25bdfd2c35da184 : TIMED_OUT in 41m 11s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a86a61891236454ca6f278001865a968 : TIMED_OUT in 41m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6f46ed6766e94ce0b7efdbacb3b2574b : SUCCESS in 11m 08s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/29f2db208d7442068db339904e816920 : SUCCESS in 17m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bf21e0af6aac4bb09e8cbf28c9bea405 : TIMED_OUT in 41m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8f0fe0a9c05f4a17b61b70e33e1a6aae : SUCCESS in 13m 07s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/bbc965baefcc40c19e421fc79c16e3c0 : SUCCESS in 18m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/27eb1ef89a7441f3ba9ec480d48131ab : TIMED_OUT in 41m 00s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/3761f76df1e74abaa0d2439ec32ce89c : SUCCESS in 17m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2975fa4120434c9e9f7b666a678ff603 : TIMED_OUT in 41m 21s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3e1d002b7e1c46dba27f93da7733863d : TIMED_OUT in 47m 29s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c6856800e3e2402794a32abc3c2c11e1 : TIMED_OUT in 47m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/124345c682ba42f9bf29a370054cf46e : TIMED_OUT in 47m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3faad5358c44dea95b725d6760edd2c : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/31deda6b121b4f888109b205c822fbf2 : FAILURE in 57m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9ec063e83665b57fba278a0ca5dbeb762195e98","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-22 04:13:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a3bcfb18c09402cc8896fd2da710f61873db342b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 04:56:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12f5516dcc234aabb7ffee7ee19383a6 : SUCCESS in 9m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ca7b928de05b48f6aaa4a5585e0c70be : SUCCESS in 10m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5b34b976e7c2436f88c4b4761d86ed79 : SUCCESS in 12m 18s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f1f98c600d8543a886c3b0de266cfd15 : SUCCESS in 11m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/49b409ec073c443babee5bdba52c36c4 : SUCCESS in 10m 50s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/9c7a951684ef4587910e00219c8178bc : SUCCESS in 16m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e3905d9dd98948d39b16137e12f2f520 : SUCCESS in 9m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c1040cb912874f2da9eb3e9e2c620bb1 : SUCCESS in 12m 03s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ea0d00e86f8d4010a730ce0ceadbe8c2 : SUCCESS in 9m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/09678bb0dd3443b68d6bbfc7bf01a162 : SUCCESS in 13m 10s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/831b309461c94855ae823d1a952d6041 : SUCCESS in 15m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d1f97d672de442119264bf7f2f542ec3 : SUCCESS in 18m 50s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/afc2c344f9904602b4cec2393377c090 : SUCCESS in 31m 40s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/2b05701c07de495893c2e25414d9da02 : SUCCESS in 27m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/96d4212683904ffd8b87058411044f8d : SUCCESS in 31m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/946c55d63f8c43499414c5709f06f94e : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b6b1bd0f58143918247635d3396a940 : FAILURE in 21m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f81cbc4e8aba09f7c4b3adbebbd3edd34103ab2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-22 23:26:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nFunc tests leaving dirty state around is bad, m\u0027kay?\n\n(I should look at fixing the tempurl tests, too....)","accounts_in_message":[],"_revision_number":1},{"id":"28ba0768f1a9e1740158ab1350bd6772361e9592","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 23:27:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47fdafeaea4b7b03e13e7d325e16891132074eed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 01:28:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cad19a12902496cac883f9d737992fd : SUCCESS in 8m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/44987f302be84660b96c1aa5a6d62881 : SUCCESS in 11m 06s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8fd3ebda333e4a33a21db60ec4f1d630 : SUCCESS in 12m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fd34efc385a54c89bc465b73e8370520 : SUCCESS in 14m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2d77b90019cf4e99bbac4254ae4ebb14 : SUCCESS in 14m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a33e02b28fa8444abf83f1dd9c734ccd : SUCCESS in 14m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fc0b52fc60bf41d6822455a8bdcbfc14 : SUCCESS in 14m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/97ccc31e7e9b4df0aa565e9c7a6ece3c : SUCCESS in 11m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f8dc72d0c6c4ea8b915adfe92635cfb : SUCCESS in 15m 48s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a565a67efd7a44a78b68b01661d55447 : SUCCESS in 10m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c31df002624940e880e942bead46527a : SUCCESS in 12m 12s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/8a5ec79acf354998900f7ed8ca949995 : SUCCESS in 16m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5b4634ac2a4c47a09e2f9af289329b2c : SUCCESS in 11m 37s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8feac43c2a684e7cb87cc948478852a8 : SUCCESS in 11m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d18687b4d560490fa9f7ae1425b2a1fc : SUCCESS in 12m 07s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d239186b4f994a36b88a86e27b62d24b : SUCCESS in 18m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8bd4cfb3c2994fa5986bb9a241b82ca1 : SUCCESS in 29m 24s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/09999d5319434fe4b093139f4f240953 : SUCCESS in 29m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d294f513bd3f4e3e8990096f42fa155d : SUCCESS in 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab96a01b8d944f39a0e2706ee8f22213 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c8207a2916cf4e797c21c267c46cf4c7f5d3f39b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 01:28:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"745719c67751b3a780a917ffe74494d1864b78b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 01:30:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c36d0efe25d4df4b39b32942e0c6d48 : SUCCESS in 1m 14s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/404ef3ab51d24b0b8ee54fc6b46a22a6 : SUCCESS in 1m 05s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/7348884c8af042239a01e70febc57fd4 : SUCCESS in 1m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af83ab154853202da10bb370e3f6fd3afe94758f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-23 21:33:21.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 0f4dfe779b9383ffe94f5e812ec55e92ef23ba22","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72a6177520f2f4f1cea6cb4f59b328844b335414","revisions":{"72a6177520f2f4f1cea6cb4f59b328844b335414":{"kind":"REWORK","_number":1,"created":"2020-04-21 04:33:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/721518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/721518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/721518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/721518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/721518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/721518/1"}}},"commit":{"parents":[{"commit":"c7fe823c0aa6c2e55310ba20093d5a4ab686758f","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7fe823c0aa6c2e55310ba20093d5a4ab686758f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-21 04:19:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-21 04:19:45.000000000","tz":-420},"subject":"func tests: Allow test_etag_quoter to be run multiple times","message":"func tests: Allow test_etag_quoter to be run multiple times\n\nChange-Id: Ifb1d091e58b9331b027edf981f6af5c015b56936\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72a6177520f2f4f1cea6cb4f59b328844b335414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72a6177520f2f4f1cea6cb4f59b328844b335414"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
