)]}'
{"id":"openstack%2Fswift~721714","triplet_id":"openstack%2Fswift~master~Ic1a81f5b7262942ea52aa1de87cfcde25a2457ba","project":"openstack/swift","branch":"master","topic":"721518","hashtags":[],"change_id":"Ic1a81f5b7262942ea52aa1de87cfcde25a2457ba","subject":"py3: Make etag-quoter work","status":"MERGED","created":"2020-04-21 19:27:23.000000000","updated":"2020-04-29 19:48:08.000000000","submitted":"2020-04-29 19:45:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721714-1588189556936-a0c18567","meta_rev_id":"0294146609f049b5cdb74f1c98ea944f9de3227f","_number":721714,"virtual_id_number":721714,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 19:45:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-29 18:15:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-29 18:15:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 19:45:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de63917b47a0c2374ae51a5234a83f737622e45c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-21 19:27:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22217c985f09c697552fd32e6af8243206a053fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 21:16:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7716a56c163a4163ac5675c9b297beb1 : SUCCESS in 7m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8ef4f8c98a0147af9ac20a86e795e240 : SUCCESS in 1h 09m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d75717d0c4dd4c0e8a2412f2d1e0b75f : SUCCESS in 49m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bc2190f6de2749fea16782eb9ac89cc5 : SUCCESS in 9m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7c11f9ff03c54e23be5cba4e70caf0a0 : SUCCESS in 9m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f3636c06b0304f34af4ecf4a12995464 : SUCCESS in 13m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef7afb9b581b4b56a8bcd1095f6bef77 : SUCCESS in 12m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9fea0d9fa41c461180485a5f43bd706f : SUCCESS in 21m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1e24a71bc1334d28bf7485a622346beb : SUCCESS in 32m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/902dc1ecbc00486bb034cd4298ea94f7 : SUCCESS in 10m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e168736a77b24d4ca8424293878c7e85 : SUCCESS in 13m 06s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/789b1bc0df5847778ff22149dd8cef01 : SUCCESS in 22m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6227a0289b5f4db49aec95b7dca92fe8 : SUCCESS in 14m 24s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/89a0cdeb24a940e68f15cf5582053257 : SUCCESS in 14m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/663aa29d32eb4946854c57c8bffa7a83 : SUCCESS in 10m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3149872d2f454165a4f5116723b7f3b3 : SUCCESS in 14m 03s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1a3de657d8f54485a2527f87db5c376e : SUCCESS in 10m 42s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/025e7265790b4fbb95925d39c50f421a : SUCCESS in 11m 38s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/7969e003f65445839f10e3ead2dea9d4 : SUCCESS in 21m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/68a03424b74b4823af16014a46b3184a : TIMED_OUT in 41m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/592847d855d545dfaa0fa0fae7e1e0d7 : SUCCESS in 48m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76e7c3536862439bb11faf5ecda66793 : SUCCESS in 28m 19s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/de617e2737ee46ff8bb565d441317191 : SUCCESS in 29m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/454f8654344f44d8a5f306b9ef8164dc : SUCCESS in 29m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efe5f26b1a5e4f71aa686080c826eccb : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52b71f1e4f604be4a98890584e9932b2 : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/206071ce56e04841adb0d22fe6758615 : FAILURE in 19m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6ae5849c33e4c09ba69ed3afb7a9d6b : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":1},{"id":"71d3f1090e204fe6e72e020c21f4f847f890dfff","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-22 00:58:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"71805701bdf536df898e315b3d2a840f5888ab95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 02:16:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7008f9254f44e29a2e535351a94eaa4 : SUCCESS in 8m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9aacdd5b12d643a1adcd51636aa7279b : SUCCESS in 1h 09m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ead13916b8eb49658cd6b44c6c80fcbd : SUCCESS in 52m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d05e5b74db704b11a62312155aaa8164 : SUCCESS in 9m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e6c89505e6c647a691194454b9cda8fa : SUCCESS in 9m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f5e86b06c6614b939a2d78f7ef13c622 : SUCCESS in 11m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/889f07d988074adc816cf9f8f6d6e35b : SUCCESS in 13m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6a6d2146599640a6b0a9f7950c806337 : SUCCESS in 13m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9e27d2763381415e9289f6e2495be458 : SUCCESS in 14m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fa12bf47edc242e39941db697369f685 : SUCCESS in 10m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a4d518853aff49afb22f2a6427dbe14e : SUCCESS in 13m 25s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/680e0d68e1b84a959a33378a87788dc2 : SUCCESS in 10m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/37c891020ee8432b9b2651d5a1e115ba : SUCCESS in 10m 32s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/76ce0595e1004d4ca34bf8a8ec45af73 : TIMED_OUT in 41m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d12431085d0c4848b52091cc4159bed1 : SUCCESS in 10m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ff5d7b72ec5d4003b1f6d1e33d37937f : TIMED_OUT in 41m 06s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3ae9dbe71be3422cb5b21b5dc3a181f8 : TIMED_OUT in 41m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bd82c83a1324463b8eac219f547a643b : TIMED_OUT in 41m 05s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/90e284119e6243fbbe9c150c873e456f : TIMED_OUT in 41m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0a4eee61dfb54e4087ca505d40ff692e : SUCCESS in 24m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/059067319db642d99e21b34969ca7579 : SUCCESS in 42m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d7f871fa146a40db9c926966435e468f : TIMED_OUT in 47m 33s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b1c64c9c0ede42c5843a89c616386244 : SUCCESS in 31m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba52c1f8e8a24df5b7509b108949242a : TIMED_OUT in 47m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7219d3de31ab4ecd9a3efdbfe560ffbd : TIMED_OUT in 41m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20541aa8027544eb968759e3ec1e4eb5 : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/875dd3e55513408bade3037118282f8f : TIMED_OUT in 1h 01m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eed73fb49e4743aba6e84f183c9db643 : SUCCESS in 1h 11m 41s","accounts_in_message":[],"_revision_number":1},{"id":"164c719a0826dc2af216975fd0f055a0875b5907","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-22 23:27:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e9107b4416610526d99aa5ed3388d6c441e04e13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 00:41:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13ddca513bee4f91be08230f6f40f089 : SUCCESS in 17m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a1c67b2529194759a9faf2624a26eb4f : SUCCESS in 1h 11m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/036f2ece7c384a689891e3e89e8920f0 : SUCCESS in 56m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bae044d05a684cc8947f11df93fc12a3 : SUCCESS in 12m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d78d6eaf40d849e5a2a9f8617535d261 : SUCCESS in 12m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2a8920b623f447578987ed88dd059756 : SUCCESS in 40m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/678f41e26e6a41f684a3fa7dff391d4d : SUCCESS in 27m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bb895afa23734b3b84056dff5129c643 : SUCCESS in 28m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/156b3f5ef6d24067b00a97cfa8b00ad7 : SUCCESS in 14m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3e5ae08eefd246ac821bb82d20ac254b : TIMED_OUT in 41m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e41e39d311094da3a7b12980a0bd8eba : TIMED_OUT in 41m 09s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c363aff1498a486f93f0e65ea4a89031 : SUCCESS in 11m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8eb76e4cfdbc4716952841cbe8471ef7 : TIMED_OUT in 41m 03s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0da143f7127240769e2b80f4d93a5f73 : SUCCESS in 16m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/216071ca26fb4066be47399af801eb15 : TIMED_OUT in 41m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c4790f688677467fbcacbe31ef83a17d : SUCCESS in 15m 05s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9d66786d71f9406e9ea2cf197ab4322a : TIMED_OUT in 41m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/21d3889bb4ef452b8fa5e58cdd8ebb14 : SUCCESS in 11m 31s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4c74fb0727e54703bbb86babcaad80f7 : SUCCESS in 17m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8d506979984f4dee8a3f4dc43b4392ac : TIMED_OUT in 41m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/444f5ad2ec664f12872179166526cbd6 : SUCCESS in 42m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a71275644924b93bc18ab7a35b2117b : TIMED_OUT in 48m 03s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f83c285f2bef4c2c9a21b41e70105869 : SUCCESS in 27m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d271f8814e949c4b587ce8ecd2aa700 : SUCCESS in 31m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eae16d2a096243f581b7b5e95fe14c7c : SUCCESS in 23m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2631026a0a644ddeaa727f34798eef55 : SUCCESS in 7m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6335454b023c4440bdda354f09b0d763 : FAILURE in 53m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e356e7dd98e40b29cc3e9d212e5fdae : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0aa5b9a96f575fdb86cb78a8678bda9232d6657f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-23 02:46:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9d6618d027bf6bad45b2a96ad37fa84efa9119af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 03:50:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68c03a4ecdf94b7ca30db1209aae5e2c : SUCCESS in 12m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d023f4fdac9748138b3bb405aa5c060d : SUCCESS in 57m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a42e06f3cdce4ba38b7f79e0847c6b2f : SUCCESS in 1h 02m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2577870b79c74552b6da32e42f44f575 : SUCCESS in 10m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0e867bba08dc43e68d9908066e3cc652 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/143e52c1396343d9b0e04d531eebc5ea : SUCCESS in 15m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3fd78ad4533d4c34967709fdad0a0614 : SUCCESS in 12m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0ed3342ef2b14b83a17e0eeb2de043d1 : SUCCESS in 13m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5e72b8428f7f43a987cd86aa2a2581b1 : SUCCESS in 12m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/83505fff1a9f487ca15dc3155b05fe5e : SUCCESS in 9m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ce33e2d00efd434497b1235534128a1d : SUCCESS in 13m 02s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a7294b82c979484faa3c045785a766da : SUCCESS in 13m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/25234c0cc0bd4e2aa64079ee323776e6 : SUCCESS in 26m 02s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/95e1a2ecafd24f18bcde2b74d6768c51 : SUCCESS in 15m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f28e4305111849ffadaa0bb971305239 : SUCCESS in 25m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/be05564d0da54a95a0dcfe2cb8906791 : SUCCESS in 14m 56s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/72daea3d01fd4f35b0a8d979f6794f72 : SUCCESS in 25m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d64140bc9ad74d4f935cc8a6d27a5f4d : SUCCESS in 17m 07s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c1c1a1992fef4823a225ee60d6e1ff13 : SUCCESS in 35m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c9c72e6731ff409e9870ef6b98e03b69 : SUCCESS in 22m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/50ebd7e4d85a44639d7b7a6a0663e624 : TIMED_OUT in 1h 01m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d6062a3d486e429384e0c1fb98dcd355 : SUCCESS in 33m 45s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b0d19075f4f24de69efa3e9cbd512cde : SUCCESS in 41m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a037e1d731f94652b2e56099f2e69fdc : SUCCESS in 41m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdd5d7f7934d406b802fa5005d2116c2 : SUCCESS in 16m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91e41060c4f242c59cf048f62d2a9e2c : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/86509439306d44638902092cb7787f4a : FAILURE in 22m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d0a30e717e6445490ab436be6d50f96 : SUCCESS in 56m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3eaccc1e32f474ebf2585f3a8fa7231ef36b6720","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-23 16:03:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9cb39f8d0b3af79a243b02fc2c627b569b5d040f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 21:23:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e2cad1a45d14751bc9fa3be879e697f : SUCCESS in 29m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/12adc3b722ef4cbbba46423ce9aa2f20 : SUCCESS in 1h 15m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cf44affbb53e4f0bb9339a00ae91c53e : FAILURE in 1h 48m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/913b2d48b586498c98e2011feff351c7 : SUCCESS in 12m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c9e10504d154482caf53516066a52292 : SUCCESS in 17m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/44bf0a9b321048b5850137a89d17edf0 : SUCCESS in 18m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e265435a17cd4d2db84f2588cb824e78 : TIMED_OUT in 41m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0fdb3ed6c06c473c83785469ff6e0692 : SUCCESS in 19m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0cb8a0343368493ca94e5a3668fd6f86 : TIMED_OUT in 41m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b87e8c6c372b44afaa7d1fbf531df6ea : SUCCESS in 33m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8f7c4129bc77469da5c288452e3b6b3d : TIMED_OUT in 41m 17s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/148bf515054d425d8e0be9c62794dbc2 : SUCCESS in 33m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/560ab1e93e5d450dbf696176fc75bda4 : TIMED_OUT in 41m 18s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/34f308033f654debaea97cfca0286289 : SUCCESS in 20m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cd80056733484dcba796af761f695883 : TIMED_OUT in 41m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5c2d766298cf40e5b97dcd83e0e3bc59 : SUCCESS in 39m 09s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/898ca91e12c64174bb2609754c0ffc1f : TIMED_OUT in 41m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/99e86448bb77480797f3ee7aac8d6edc : SUCCESS in 36m 16s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d0bd3facd659431e9402b70bfcc8e3ca : TIMED_OUT in 41m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f74797829f004b8691b24cdb23f7f7ae : TIMED_OUT in 41m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6e302580cf1745c48b8f0fe5c24ae9cd : FAILURE in 16m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8f0c3b37ec6447b7b8364779793a9aee : SUCCESS in 48m 11s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/aa3d77fb0a4c49d29befa2acfd44075c : TIMED_OUT in 49m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fb83cc84614c45819d72f4b325cc3da6 : SUCCESS in 47m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57cedd5648ed41a481f103bd9951c49b : TIMED_OUT in 41m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/006d880e793546d9bd694534ce6ce2a7 : SUCCESS in 7m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a780890dbe744a708af87a294de87a09 : TIMED_OUT in 1h 02m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dae82c0c87a243f8b77b240a2a1ca9f2 : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4875d7290fb84aa2fe99be225b9ae21312a00b23","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-23 21:30:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d11fa94f7e5a5f56a5b7caeac005de9f5b58fe5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 22:57:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b86b05123493464e97fd2804eb3d54b7 : SUCCESS in 10m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b702f9e26b16410b915a67dc9ea28c3c : SUCCESS in 1h 19m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/08d3d670781c40da9342acf97d1930c7 : SUCCESS in 59m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cbe7b3b0c7014d649fbc37987d410ac0 : SUCCESS in 11m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/368dfb750a544dcdadf602382b4f18c5 : SUCCESS in 10m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f2cd813daa204ecab5968ee79be9806c : SUCCESS in 14m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c6fc40d851c44899a70fc4c4dea09e35 : SUCCESS in 13m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bfa37803ad7b4323908b7605f8008130 : SUCCESS in 15m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5e1b33a5692346fe936fd549b695d122 : SUCCESS in 15m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/56080c3c52b04540a6883d6a90f445ce : SUCCESS in 13m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bef9516c796d4c56885876ea6e4fd7e6 : SUCCESS in 15m 55s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/483eccf29b1143dd9d98018d80b3c9a8 : SUCCESS in 12m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f54deb5b34e945a0b026515506e6fc06 : SUCCESS in 12m 10s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f8d8c48be9f946389719672bbdb02487 : TIMED_OUT in 41m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2387ed4b01dc4076bab70f3999f95865 : SUCCESS in 13m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/90e90f25fe8a4724abad4237eb197513 : SUCCESS in 13m 21s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b197ee41786a48d4b5e44dc4ccf0f5b5 : SUCCESS in 11m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/db8f1ef674e246278f904ec1f25aa3ef : SUCCESS in 12m 32s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1126ef0c1ac5421baad934edba391f01 : SUCCESS in 34m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aae382d82db34313baa442b1e323cf9c : TIMED_OUT in 41m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c0150097a8ef4496be4a1671fb6325f7 : TIMED_OUT in 1h 01m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fee617601ce34afa980f4f499715c087 : TIMED_OUT in 49m 05s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b002a1e609c94932a476986ed78fc019 : TIMED_OUT in 48m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad18d04c6fa44407bf319df5d5fcc650 : SUCCESS in 28m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e5218392bf04073b8c33775fbd24bda : SUCCESS in 33m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/156d3754e56b49518ce36e9f5258499b : SUCCESS in 8m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/923efc1238104e2893db83e7391425ec : FAILURE in 53m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1320965df9fe4acd94bf13f5406c1492 : SUCCESS in 1h 09m 20s","accounts_in_message":[],"_revision_number":1},{"id":"9475a218a6b174b36696bb0c0b089b6cee9480c9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-24 01:02:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"47aff8409dfd0e14477a4cdfa858e0d4565418df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 02:13:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc3894f1d5440a0ad1baaf820400d19 : SUCCESS in 8m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f7c17e9672534c4087fd4a878681a524 : SUCCESS in 1h 09m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/efda1b150c174196888d1395fc8a09cb : SUCCESS in 55m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6667717f36ed4363855ba987c17d26ab : SUCCESS in 10m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6e0c06f5794b40448cd50e520b467746 : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c9b1b5acebbf4efc9dc35ace32652727 : SUCCESS in 16m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1eee61b896484854b0e5d04cbc81a517 : SUCCESS in 16m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/313b0bf5697e4507bce8b6e6689d6841 : SUCCESS in 14m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/520b721969004e19bc8e93b4fb8ceecc : SUCCESS in 16m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c7d3f842fb5540ffa546a7ba926ed295 : SUCCESS in 11m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/49f9dcad677c40208b0389d65e105ba4 : SUCCESS in 14m 28s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d8e7cfadf7234855bfde1ca8d1685b46 : SUCCESS in 11m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cda95327e6b64f61be1b5c8ec56ac974 : SUCCESS in 10m 02s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c3ec44d29f0449d5baa1bf2d4448df4b : SUCCESS in 19m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2650447fd7b74fb18fbdbdf7487e9914 : SUCCESS in 12m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4ad9e54545304811911b5cd79fd49860 : SUCCESS in 12m 10s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/372e11ac532e4bb7b8d1fe1ecd46cf91 : SUCCESS in 10m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/780290255f4a41a795a984b15c65b7ab : SUCCESS in 15m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8940c1c25d85466ea9bffd915a608b29 : SUCCESS in 15m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f98d8bb424d24974ba025f91c269fb66 : SUCCESS in 24m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/37193729f46b4ddebd7a963f4299be45 : SUCCESS in 40m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c7becfd8d7f4e4186abb3fa96ba1230 : SUCCESS in 30m 34s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0051b721fe484ffeb0c8babbd335ec94 : SUCCESS in 25m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5617f635fed74cf1bc9e0886fcfbf562 : SUCCESS in 41m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc77ea0e03eb4b62a22a8a2ac0a6d4fc : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fc34e4fbc7c423db3bdc616cca1312d : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a8b9b9091024d85996c6172b6f2c92c : FAILURE in 19m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65305eff8bdf4d9b84e8966eabc63a4b : SUCCESS in 1h 09m 30s","accounts_in_message":[],"_revision_number":1},{"id":"519bf6c2f760ba48731e41afecf6ee91805c3b37","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-29 18:15:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea8b3a71daf78c692afc4ea57030358d8e57f94d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-29 18:15:46.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit cdc477e816145d392e0968f6152d9a660a44b090","accounts_in_message":[],"_revision_number":1},{"id":"ce8793ebc08cacefc24cf599c1414f89cc8fbbe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 18:15:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3dd465da223535c845e749c68bb8c1e282cf615","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 19:45:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6f3bd306056ae3b5e5548c2ab400f2feb1049c0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 19:45:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6aac7f2e53744728310a7cf31321a70 : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/263e524770e842a3a4d4f056974c1b45 : SUCCESS in 1h 03m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/837413bc3d9b40c0bbb4e1e62312667b : SUCCESS in 53m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5945b7f6c8084f3ab8bd61901268f6e5 : SUCCESS in 12m 48s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b58378c21a1d465bb6f78f4b19284fcc : SUCCESS in 12m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/069ec7d0196c4e98a852699f5033a048 : SUCCESS in 17m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2c63c352378543ac9ecf0a5e3d811bad : SUCCESS in 13m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/126aa9600c5540938704f5e7ba0a96ce : SUCCESS in 14m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6377e68c7cf34a328e99d33fa6d1eea7 : SUCCESS in 14m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/99f85cb4ea46458881a8d17f05eff55e : SUCCESS in 11m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3c2aa10091e745bd8a7186ee0d9f90db : SUCCESS in 25m 56s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/54daa2a5386e42a9ac0cfa33a191a243 : SUCCESS in 23m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/046ebf25598143048366b1164ff6032f : SUCCESS in 21m 15s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/05a2d4f24aaa480994ba006e9b35a269 : SUCCESS in 26m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1d683307f43c4f2e8a5176905b5f4c0e : SUCCESS in 12m 12s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c524612299e94175a3a9358ecef08301 : SUCCESS in 10m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8f00e8561d464f3c8602faa354e90c10 : SUCCESS in 18m 26s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8ebd8237334645108a5dbd3146664954 : SUCCESS in 17m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a5655e927f25442fb142e2bfc5b981d9 : SUCCESS in 48m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/efd3805d5c4c4e0abc9487f0ea35e540 : SUCCESS in 40m 09s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/a20db76c5910452ea0eaf6b243d6d6e7 : SUCCESS in 30m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/662bec49ef454513b15a1fda02832a2b : SUCCESS in 32m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82f002cf2f3b4d5c96f1ae0deff24e8d : SUCCESS in 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3936456513344944b53bb9859b072a64 : SUCCESS in 7m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f826d9d5f9144a8a66a1a62a8056e8d : SUCCESS in 1h 00m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6dd491268b4445c79d384fe02c589284 : SUCCESS in 1h 00m 40s","accounts_in_message":[],"_revision_number":1},{"id":"2b12a852c15ebbe5dac5118dced855f5c8003294","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 19:48:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61ddbba1c0ec4a7f8ea25d1cb57fbf4c : SUCCESS in 1m 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/24b4a6b77dbd44c2895316e37b1271f8 : FAILURE in 1m 07s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/926941814a584bd2af707d0bb1ca6ad8 : FAILURE in 53s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e84d531b954a906caa68cc962c64404dd8c958d3","revisions":{"e84d531b954a906caa68cc962c64404dd8c958d3":{"kind":"REWORK","_number":1,"created":"2020-04-21 19:27:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/721714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/721714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/721714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/721714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/721714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/721714/1"}}},"commit":{"parents":[{"commit":"72a6177520f2f4f1cea6cb4f59b328844b335414","subject":"func tests: Allow test_etag_quoter to be run multiple times","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72a6177520f2f4f1cea6cb4f59b328844b335414"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-21 18:43:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-21 19:27:12.000000000","tz":-420},"subject":"py3: Make etag-quoter work","message":"py3: Make etag-quoter work\n\nPreviously, func tests (might?) complain like\n\n   TypeError: \u0027dict_items\u0027 object does not support item assignment\n\nChange-Id: Ic1a81f5b7262942ea52aa1de87cfcde25a2457ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e84d531b954a906caa68cc962c64404dd8c958d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e84d531b954a906caa68cc962c64404dd8c958d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
