)]}'
{"id":"openstack%2Fswift~722511","triplet_id":"openstack%2Fswift~stable%2Fussuri~Ifb1d091e58b9331b027edf981f6af5c015b56936","project":"openstack/swift","branch":"stable/ussuri","hashtags":[],"change_id":"Ifb1d091e58b9331b027edf981f6af5c015b56936","subject":"func tests: Allow test_etag_quoter to be run multiple times","status":"MERGED","created":"2020-04-23 21:33:21.000000000","updated":"2020-04-24 04:40:43.000000000","submitted":"2020-04-24 04:39:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722511-1587703151703-276f548f","meta_rev_id":"92a66e395b15be03e1b24e69de275e7892068d6d","_number":722511,"virtual_id_number":722511,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 04:39:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-24 04:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-24 04:11:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 04:39:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb0aab9e4a923605f18846c49a316f65eb544da2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-23 21:33:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2cdb6879310df132cefb2216a8c587213144cf97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 22:07:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ebe53c9b26f4bbfbf6f3a188e5ecd7c : SUCCESS in 11m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1c557ada50cc41938094770800dc2488 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/737811f3c2e44213b7fc09e66aa9f345 : SUCCESS in 14m 51s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f8e1c8a8f7b746508fd1a8184cb05bf4 : SUCCESS in 11m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e651839f1c4340dca02cc6e58b18f128 : SUCCESS in 12m 58s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7fb1124afe114cfa93ea36fc11dce437 : SUCCESS in 15m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c8d5a727a2814e39ba79a1bc2181ed82 : SUCCESS in 13m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7e6ff500d3c44a9b8218c74ea819fde5 : SUCCESS in 15m 10s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/19aef300421145c69ff604b57ac184db : SUCCESS in 14m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bddfad9cf62649178993a5631f07efb1 : SUCCESS in 14m 58s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/eb155c847e9147f1b6065542cff9ba23 : SUCCESS in 22m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8654e2f17aa44ea788947e6b65e945b2 : SUCCESS in 24m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d60e52c9ea7456aa688dcf7a3ee8c08 : SUCCESS in 9m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7271e2863fec4b1c9aa6244609cd30e3 : FAILURE in 23m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4bdaab2e386f5685121c27566cd321777c0c662d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-24 04:11:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"29794570488d8013eb7be6d13671ca692231a094","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 04:11:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"187aed8a0839aa83a290f78bd52dc1caee03e954","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 04:39:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9b998a71b5d4f4eabaf3757a431ac61 : SUCCESS in 8m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c81291a8d39d416890c9241da99ccaa9 : SUCCESS in 9m 15s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/558245c2f5aa48339527e297675795d3 : SUCCESS in 9m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/45e19bf513334699a231810ad1df4302 : SUCCESS in 18m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/deffafe5ecc14b1d90b5f422a1a3ff4c : SUCCESS in 12m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f83185a6f7d04492a1d087cf4330eec4 : SUCCESS in 12m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1258353cc3fb4c41babb1c963f3692e6 : SUCCESS in 12m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d6b0972850514d759dc4f155223c71ee : SUCCESS in 12m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/524ec7a2a4944bc68d7895c3368e0fb6 : SUCCESS in 12m 24s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ba518d6ff1db4ccf9835d1812d05252f : SUCCESS in 10m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/09dfcb364b7e4deaa7a9e635ac7b6bb9 : SUCCESS in 26m 42s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a666381752514522bf8f4ef750632ea3 : SUCCESS in 15m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9609567843004ee9b760bfce907383eb : SUCCESS in 12m 21s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/44240078989a4ef2b3e1e9589e698604 : SUCCESS in 9m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/44b2659ccd5f4e9aab1ae1af2b701bec : SUCCESS in 12m 34s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/94ba0c32642748b5b21c28c3a3366169 : SUCCESS in 15m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bbac3836d65430390ad0617c4e9c7ba : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8e24bb46598a17b43622f6e9ff4fa2be99e32a36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 04:39:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"92a66e395b15be03e1b24e69de275e7892068d6d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 04:40:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1064507c8dc347eeb62028ddcd68809e : SUCCESS in 1m 15s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/bf70ae88dfd348a09ed6d147c317ef68 : SUCCESS in 55s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c8a33f50c4d8496f9a271594f85058b4 : SUCCESS in 1m 12s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f4dfe779b9383ffe94f5e812ec55e92ef23ba22","revisions":{"0f4dfe779b9383ffe94f5e812ec55e92ef23ba22":{"kind":"REWORK","_number":1,"created":"2020-04-23 21:33:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/722511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/722511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/722511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/722511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/722511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/722511/1"}}},"commit":{"parents":[{"commit":"461ea1ef7d8ed477471bd664af4e2ab6c4e54f87","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/461ea1ef7d8ed477471bd664af4e2ab6c4e54f87"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-21 04:19:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2020-04-23 21:33:21.000000000","tz":0},"subject":"func tests: Allow test_etag_quoter to be run multiple times","message":"func tests: Allow test_etag_quoter to be run multiple times\n\nChange-Id: Ifb1d091e58b9331b027edf981f6af5c015b56936\n(cherry picked from commit 72a6177520f2f4f1cea6cb4f59b328844b335414)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f4dfe779b9383ffe94f5e812ec55e92ef23ba22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f4dfe779b9383ffe94f5e812ec55e92ef23ba22"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
