)]}'
{"id":"openstack%2Fswift~722570","triplet_id":"openstack%2Fswift~master~I25bc872a1a9c8f9ea11b2ca1c973d131cdae70f5","project":"openstack/swift","branch":"master","topic":"shard_cleanup_tool","hashtags":[],"change_id":"I25bc872a1a9c8f9ea11b2ca1c973d131cdae70f5","subject":"WIP: PLaying around with a basic shard-cleanup tool","status":"NEW","created":"2020-04-24 03:42:06.000000000","updated":"2020-09-09 19:45:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":35,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"805db11cdb01b11c55a7e4b58c8b15a34ddcc32c","_number":722570,"virtual_id_number":722570,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"date":"2020-09-09 19:45:57.000000000","_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-16 16:24:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 15:00:41.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-06-16 16:24:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-09 19:45:57.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"4842c501108ede85a29271786d3b74b2c53d4ea1","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-04-24 03:42:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e3278b0e16fddb9b80297a68404767030826cd0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-04-24 03:42:31.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"2ad50f4742d550827d6d12b97bb05e4cf59766a1","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-04-24 04:00:01.000000000","message":"Patch Set 1:\n\nIt probably would me smarter to attempt to find a complete set of ranges sorted by object count, to minimise data movement. Though it\u0027s only metadata, so maybe it doesn\u0027t matter.","accounts_in_message":[],"_revision_number":1},{"id":"afca9439ff75718a027b35f7c66a24afe738dac4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-24 04:09:43.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nHaven\u0027t looked at the code yet, just the commit message. Thanks for poking at it already!\n\nIf we\u0027re trying to minimize data movement, we\u0027ll want to pay attention to whether the shards are FOUND/CREATED vs CLEAVED/ACTIVE -- I\u0027d sure *hope* they\u0027d replicate fairly quickly and get stuck in the \"audit failed\" state, though, so there should only be a couple that have cleaved from each set. Hopefully.","accounts_in_message":[],"_revision_number":1},{"id":"3e38fa65a00ed47ebe03813a70b9c3aaedd47615","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-04-24 04:39:17.000000000","message":"Patch Set 1:\n\n(2 comments)\n\noh, I wonder if we can make misplaced objects set the src bounds to be (\"\",\"\") like it does with SHARDED if own_shard_range is deleted.\n\nThen the root only really needs to mark bad shards as deleted, and misplaced objects will clean them all up. It would make the code _much_ simpler. I.e just the root container.","accounts_in_message":[],"_revision_number":1},{"id":"7e83ad9313a6617e49e254bf55705a09b15373b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 04:44:11.000000000","message":"Patch Set 1: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ed8368d5dfa4dc6979d3ffe820f14b7 : SUCCESS in 10m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e131777b72554db3ac405261d14dc1a1 : SUCCESS in 1h 00m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9bb92bb87f6c4b2b861b86107c160a99 : SUCCESS in 55m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/603ee914fbe8464bb44dd00e5878dd4a : SUCCESS in 7m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c02a4b62e39047dfb1e2772dbb6a3591 : SUCCESS in 8m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d95d4af1655040ef8f2fc517b5939050 : SUCCESS in 34m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/906ece24af244dd0aa0e5c44cc5500eb : SUCCESS in 34m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6645de6a220d4da39d4d1387ea1552cc : SUCCESS in 14m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/da3c6eaac3ad4224aff2c387422f98d2 : SUCCESS in 30m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dd00a2f5d3704e14b8ac09426fb020c2 : SUCCESS in 28m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c8b898bb7e8649eb9b5aab6737e7fd0e : SUCCESS in 31m 33s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d08ad18aec0e47abbe962540b5276162 : SUCCESS in 28m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/625a6770d7494096969c8527f1033118 : SUCCESS in 28m 58s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a03176de6ffa43af82ec60104ae8b58e : SUCCESS in 20m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9280fe123e8e49bba4742a4568b0038d : SUCCESS in 28m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7a8de7db913547f39203ff530162d99f : SUCCESS in 32m 16s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4c308322ab6f4de2a07b121842d914a7 : SUCCESS in 9m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a72ce82140f54104af2189f18d521556 : SUCCESS in 29m 14s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1e23ad2b408d4315ab10858ed92e3289 : SUCCESS in 20m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e6790d64337340fbbed1b8c3ab186e3d : SUCCESS in 38m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a700e65b5ba049e3a4eb06f6bab91148 : FAILURE in 45m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9363e0f151b4b8aa23091458e5849a6 : SUCCESS in 39m 10s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/76f401a0e2fd43c892ff8f4e8c24d07f : SUCCESS in 39m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/990d8251eef24705b4af6917b57a7599 : SUCCESS in 34m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a151a06d472e4ad882dff24d69e683fc : SUCCESS in 12m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e72b0813488b4358a71b1c29f4572132 : FAILURE in 7m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/29eef6ba97f84a488b9e1b592b015ec3 : FAILURE in 29m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aca791952639493b82d79e7e89def3ce : SUCCESS in 55m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3737b6a23a2b3eb4a0adecd3500874fb0646e87f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-04-24 05:14:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"19648c75be989ec44ce76a39df7f30b4d05467c4","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-04-24 05:18:23.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f95b94a6f125e0a5584bb53d5b69393fd4e90412","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 06:33:21.000000000","message":"Patch Set 2: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cdf8a0671554ddb8d8c58f305654ff0 : SUCCESS in 8m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/702896222c084819bce6906ebb73a2df : SUCCESS in 1h 12m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a0197345b0f143e18e0daa8a47d50dae : SUCCESS in 56m 36s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1fa25fe8af8e4b31976cf68a75255756 : SUCCESS in 9m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5073204a459a4f5c8e84818e06ca31cf : SUCCESS in 8m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8098044b0cec4f0f884514530671c24c : SUCCESS in 12m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f85e3fe7441d46faa92e2b11554613dc : SUCCESS in 12m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f6f9a1fff7cd48cc8afdc61d2d01a867 : SUCCESS in 12m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a171a044b2a546698d7dfa3f0dd90700 : SUCCESS in 12m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d5759de06b4744c39cfa445bfd3c224f : SUCCESS in 11m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/49138ce693c343f587c37bde67a614d4 : SUCCESS in 11m 48s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/3aa335267d514eabbbb31cddcfe542eb : SUCCESS in 15m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/323147a4d9644f77a9f5a076c88a0d30 : SUCCESS in 11m 10s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/53d5337d794341dd94577c7dfc876b58 : SUCCESS in 15m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d27ec456c38a43af96390a4b73a11c05 : SUCCESS in 11m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/77accaaaebe04e1fbf036aec2e0b8bda : SUCCESS in 12m 53s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d7610aa9f80042cb9675c7cda4f4d4d8 : SUCCESS in 12m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7834def318744ee284ca94ceefdb66a7 : SUCCESS in 13m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b42ca0e93c79453983fc5cc5edda5276 : SUCCESS in 15m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eb12c2c2da6146aa89f95611459afdc1 : SUCCESS in 24m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fa9f5a55ffcb4f28af442dee472a5d5a : SUCCESS in 44m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/048ba98fd25d41c0a4af4ce10fc99bb1 : SUCCESS in 36m 36s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/2254a400e31b47798cb1b1e9db51445e : SUCCESS in 28m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fbfa86fb01554220a026dcb086864189 : SUCCESS in 33m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea4c046f2b0646c2bf26ccf9adf9522a : SUCCESS in 21m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f773070084494d97ab9ebbfdea6e6a8a : FAILURE in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b628fb553dbb42a4a2ea7a8593fb9f40 : FAILURE in 21m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a78b80dd940e48a39c61ec8424e3e213 : SUCCESS in 1h 04m 19s","accounts_in_message":[],"_revision_number":2},{"id":"06d5b0c069443fa52723a6b96b1d2b34ba122f5f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-16 15:00:41.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nHrm. For some reason I\u0027m having trouble getting the fixed-up shards to replicate to the other DB in the split-brain. I\u0027m also a little worried about how the own-shard-range epoch can differ from the in-file-name epoch :-/\n\nGot a few minor fix-ups coming, but I think we\u0027re going to have to keep thinking about this.","accounts_in_message":[],"_revision_number":2},{"id":"09a5efbc36c6f7e7410f51257effc0ef11ca1670","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-16 15:01:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"231b2de442d8ebf7502f1e913244c5730ea2739b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 16:24:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12f8830d952949ef920f4e75a7a00b9b : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/30afea9bbb664b139e02ac6146099c61 : FAILURE in 23m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aa7cc8a3405d4c63a82942a68066a1f7 : SUCCESS in 48m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/010ed13eed254848ac30bc51ae326559 : SUCCESS in 11m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d1301f4ccc354da18020d85235e41a7a : SUCCESS in 11m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/94e9ce6b35754ddfbb960ff6d5765d1d : SUCCESS in 20m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e6f11694765f4526bb3e7532113e0a4b : SUCCESS in 17m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/232467bcaca8464abbe1dea2a2e65446 : SUCCESS in 17m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/408deec23cd745a1aeefaf1d17157fea : SUCCESS in 32m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8d9fc1fa34054e22b64b52508b30c95c : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6780898f726142328cd821a85b8ecb67 : SUCCESS in 20m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/201963c541b04cb5939fa12d1db64e53 : SUCCESS in 17m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/49798b6c30dc4f9f81a297cdbd49a416 : SUCCESS in 20m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ac65ea02b3a049d48efeda7befa415ec : SUCCESS in 21m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2d4a28b2ca844384b04aca540dfcefb9 : SUCCESS in 22m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c80d560a25034cf49e4f49c2eb77571e : RETRY_LIMIT in 5m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2ea6cf96d1af458d9651e110caf30892 : RETRY_LIMIT in 4m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cd48ed36c1074d49926b1068ed1e73a5 : SUCCESS in 57m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0c4dedf0d4fd4989baeff6529d72a322 : SUCCESS in 52m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be4e31be27c64b8db50eb1488f8d26fd : SUCCESS in 16m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab2617e98764ee8898a284409a86998 : SUCCESS in 8m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d2bc7a942e5e40e6a3592e0a8ae497dc : RETRY_LIMIT in 12m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/890fc2da054b4b47a969aecaa68cad9c : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":3},{"id":"6e4eb3f877be4340e1eb9859c59ec76297ff8d8a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-09 19:45:57.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nso shrinking is working great - I would love to have some code that helps me find potential donors (instead of picking names out of giant garphviz images)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"69341eb6c5b74a5127e9796924565061623db790","revisions":{"d961a6922d1450485db2b0222839205fe0f63e34":{"kind":"REWORK","_number":1,"created":"2020-04-24 03:42:06.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/70/722570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/722570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/722570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/722570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/722570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/722570/1"}}},"commit":{"parents":[{"commit":"a5afe767581d2cb97cf3690067e6d626c7682c2c","subject":"Revert \"Make rolling-upgrade job non-voting until virtualenv problem is solved\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5afe767581d2cb97cf3690067e6d626c7682c2c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-04-24 03:34:37.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-04-24 03:34:37.000000000","tz":600},"subject":"WIP: PLaying around with a basic shard-cleanup tool","message":"WIP: PLaying around with a basic shard-cleanup tool\n\nI\u0027ve been testing this on Tim\u0027s shard probetest problem, 721376.\n\nThe script takes in a root container and the container ring and:\n\n  - Searches through to the shards for the latest set of ranges that\n    covers the namespace.\n  - Goes inserts those into all the bad shards, those that represent\n    incorrect shards, which the state set to found. Also updates the bad\n    shard\u0027s own shard to trigger sharding.\n  - Updates the root shard table to delete all the bad shards.\n\nYou are then expected to let replication replicate the root container\nshard table and sharder to shard away all the bad shards to their\nrightful location.\n\nNOTE: We added a patch ages ago to skip over empty ranges when sharding\na container, so inserting them all is just much eaiser to code and this\nis a POC/WIP.\n\nLooking at the sharder, when we\u0027ve finished sharding a shard (non root)\nwe mark the shards own_shard as deleted, but the container itself\ndoesn\u0027t seem to be. So that could be a bug.\n\nChange-Id: I25bc872a1a9c8f9ea11b2ca1c973d131cdae70f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d961a6922d1450485db2b0222839205fe0f63e34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d961a6922d1450485db2b0222839205fe0f63e34"}]},"branch":"refs/heads/master"},"5743e64855418cad97c248b111da822314a4aca3":{"kind":"REWORK","_number":2,"created":"2020-04-24 05:14:29.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/70/722570/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/722570/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/722570/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/722570/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/722570/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/722570/2"}}},"commit":{"parents":[{"commit":"a5afe767581d2cb97cf3690067e6d626c7682c2c","subject":"Revert \"Make rolling-upgrade job non-voting until virtualenv problem is solved\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5afe767581d2cb97cf3690067e6d626c7682c2c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-04-24 03:34:37.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-04-24 05:08:04.000000000","tz":600},"subject":"WIP: PLaying around with a basic shard-cleanup tool","message":"WIP: PLaying around with a basic shard-cleanup tool\n\nI\u0027ve been testing this on Tim\u0027s shard probetest problem, 721376.\n\nThis patchset is a more delicate solution. This only works on a root\ncontainer. You pass in the root container and it\u0027ll:\n\n  - search for the latest set of ranges that covers the namespace\n  - then those that are left are marked as deleted and state set to\n    SHARDED.\n  - Then written back to the root container.\n\nMisplaced objects will now go and clean these containers all up.\n\nThe marking as SHARDED seems like a bit of a hack. But the audit cleanup\nof shards expects state \u003d\u003d SHARDED and own_shard_range.deleted \u003d 1.\nWe can of course change that. But for the first version just doing what\nthe sharder is expecting.\n\nChange-Id: I25bc872a1a9c8f9ea11b2ca1c973d131cdae70f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5743e64855418cad97c248b111da822314a4aca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5743e64855418cad97c248b111da822314a4aca3"}]},"branch":"refs/heads/master"},"69341eb6c5b74a5127e9796924565061623db790":{"kind":"REWORK","_number":3,"created":"2020-06-16 15:01:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/722570/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/722570/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/722570/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/722570/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/722570/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/722570/3"}}},"commit":{"parents":[{"commit":"0bed59e3faae94b96f98bbbc61b09abe54453267","subject":"Merge \"Use ini_file to update timeout instead of crudini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bed59e3faae94b96f98bbbc61b09abe54453267"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2020-04-24 03:34:37.000000000","tz":600},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-16 14:52:04.000000000","tz":-420},"subject":"WIP: PLaying around with a basic shard-cleanup tool","message":"WIP: PLaying around with a basic shard-cleanup tool\n\nI\u0027ve been testing this on Tim\u0027s shard probetest problem, 721376.\n\nThis patchset is a more delicate solution. This only works on a root\ncontainer. You pass in the root container and it\u0027ll:\n\n  - search for the latest set of ranges that covers the namespace\n  - then those that are left are marked as deleted and state set to\n    SHARDED.\n  - Then written back to the root container.\n\nMisplaced objects will now go and clean these containers all up.\n\nThe marking as SHARDED seems like a bit of a hack. But the audit cleanup\nof shards expects state \u003d\u003d SHARDED and own_shard_range.deleted \u003d 1.\nWe can of course change that. But for the first version just doing what\nthe sharder is expecting.\n\nChange-Id: I25bc872a1a9c8f9ea11b2ca1c973d131cdae70f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69341eb6c5b74a5127e9796924565061623db790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69341eb6c5b74a5127e9796924565061623db790"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
