)]}'
{"id":"openstack%2Fswift~724141","triplet_id":"openstack%2Fswift~master~I3d10e121b403de7533cc3671604bcbdecb02c795","project":"openstack/swift","branch":"master","topic":"bug/1719860-patch1","hashtags":[],"change_id":"I3d10e121b403de7533cc3671604bcbdecb02c795","subject":"swift-get-nodes: Allow users to specify either quoted or unquoted paths","status":"MERGED","created":"2020-04-28 19:42:22.000000000","updated":"2020-06-09 17:58:55.000000000","submitted":"2020-06-09 17:54:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724141-1591725266139-1e1c92d6","meta_rev_id":"74adc116b6024cb6ccd7b5f1bde8f2f947ef9893","_number":724141,"virtual_id_number":724141,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-09 17:54:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-08 19:05:50.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2020-06-08 23:59:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2020-06-09 15:02:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 06:00:29.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2020-06-09 15:02:36.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-06-09 17:54:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3381aac4728ae8234fc21914a3493edc59c92281","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-28 19:42:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4c3c41ca7125928d0366d2304e37953ae5d0b99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 22:23:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/379d0d5b6048427fa46a308e8fc67291 : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/f4f026162c2d451aaa4d7d6ff0114d4b : SUCCESS in 1h 02m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b71bb476cbb14626a70b9b1670b25cea : SUCCESS in 52m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/67558d4ac952440199d199861dc9590c : SUCCESS in 13m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fe0cc7b5212b474bac7eaf0717038e18 : SUCCESS in 12m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a5b8c3c69469495490543604b6a11ec0 : SUCCESS in 33m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1e9fd1393c7b420d9c3fb67b1536038e : SUCCESS in 24m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cefecd550cf749219f0b2098f3ac22b9 : SUCCESS in 34m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/efa480c5ef204788b94eea6acba76baa : SUCCESS in 14m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b22df002d09d4354a36238de8ec2530f : SUCCESS in 10m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/13f12d72b8294a4190b28c46c7f5c757 : SUCCESS in 18m 07s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f5f251d411ab4870b8ddbb5f686f0a01 : SUCCESS in 10m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8547e682c6e24442a5075e0e7cb3746c : SUCCESS in 11m 28s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/291088833c9a48728f0c298d667fb7c4 : SUCCESS in 17m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/137bb370a14c441ab08a0bf4461c23b4 : SUCCESS in 13m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/83d86492e5a243dd88d3017c0dba5762 : SUCCESS in 41m 30s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0b589553bae94c6ba4acb210ae45097f : SUCCESS in 11m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8423a86c17b6481a903d522efb1e56d5 : SUCCESS in 12m 04s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/af63bd7d137b4bdea30c038576d380e7 : SUCCESS in 17m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e46bfe927be476889f2f1d52a0b5339 : SUCCESS in 25m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ed1b4e9dbde644c18c5b8e39a9feadc5 : TIMED_OUT in 1h 01m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/19de67e7e42747c5bf15082d7f782b78 : SUCCESS in 32m 18s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c3e1645f09ee495a833a63275131b202 : SUCCESS in 29m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/69b65eccd1e247159ecc7045932cadc1 : SUCCESS in 31m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd61f68a6bfb43fbbd3b08179eac56b7 : SUCCESS in 34m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/952635b9af9644f0bf7f3bf7d6255a4e : SUCCESS in 6m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/53d55ada2aa2431b9f8803364944c038 : FAILURE in 55m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55b2e6e62faf4fbb9b122644b8bf2149 : SUCCESS in 1h 05m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/84328aa0fea2497688d17ebd02bb226a : SUCCESS in 1h 12m 26s","accounts_in_message":[],"_revision_number":1},{"id":"87fa7c31e6200d6dc76925164fab7d0cac415d36","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-28 22:35:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2a4fecad46fd5a551e058077724931057a7e603f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 00:30:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/127664770d5f49ca82dccebff4583219 : SUCCESS in 10m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/4baa40f506924e1f89d55b20da80ea6d : SUCCESS in 1h 04m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4470d630f6c246d08582b5ca4c72fcef : SUCCESS in 50m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/13446751d88d406fbf03afc893b11a8e : SUCCESS in 9m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/56c36cc617c74535bc9493ebb1dc77fa : SUCCESS in 11m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/30262ac4c5b143c7874dc24fc6d9e8d5 : SUCCESS in 14m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4bd5bfeab11e4ba3b9622698a7e1f39e : SUCCESS in 14m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/156093cc66e6402c840f24c7dd28b3e2 : SUCCESS in 15m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/961705bfe0f04010abe3b1cea5c302f7 : SUCCESS in 15m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/70ea6636b02642d89658fa5c34e3f215 : SUCCESS in 11m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ba19b74efafe4984b7fff6a44ce39f0b : SUCCESS in 17m 24s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/002f990e46744d4e92292a4b6ab889d1 : SUCCESS in 15m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a36852aac5814bbb8c9b4c12c30bf5cb : SUCCESS in 19m 00s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/41430a3cb44340f996013f30f91cdc23 : SUCCESS in 17m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/535c7b90bb284a2bb8c6a311c971b29d : SUCCESS in 12m 20s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6dc1f067e5ec4cfd92810cd1a0e6405a : SUCCESS in 16m 47s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/34efa24559384e8a96ecef5a972ca42f : SUCCESS in 26m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d620f99eef7a4586b01e68597a29fd46 : SUCCESS in 28m 58s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d40f0bc63bb94ffe8dacd2abff595ffc : SUCCESS in 39m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f059408c47044530bceae91fb8a57484 : SUCCESS in 20m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b0918060fd384bf290d67dced326471f : FAILURE in 54m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d29399e35afc4e58a38ede7db639a6d9 : SUCCESS in 29m 40s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8d231999edf24f109a3dfbb79752cf0d : SUCCESS in 27m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/02df166a9dc44daeae9fe9af81dcf1e2 : SUCCESS in 33m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5535dbe739ef415190d0a2584cced7aa : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3075dc679dd7460889d5ca205499db25 : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c54cf927afb04cd98f4464f37f313438 : FAILURE in 24m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e9c0a945bb044b689d305862d8bebe3 : SUCCESS in 1h 04m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/77b0f784c7db4df081deba018bf8f2c4 : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":1},{"id":"aa02f7b1a549757b9e0b524031fa6436c546f576","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-29 05:47:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"42693dff8929d9d1be86a17c54ddb0fd4ddda38a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 07:03:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ddc1adafb3d411e82e2af871c1f26be : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/010f3544bf174f378b1b0d889412b2ef : SUCCESS in 1h 13m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/25f2534bf02c4a58a7383e08de75cd83 : SUCCESS in 1h 01m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/66062545aab042faae931e4f6605a338 : SUCCESS in 8m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/85790d18ff794e46810b8e4951d3daa1 : SUCCESS in 9m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b5b8a87ffcde40e6a347b9b64d79d95c : SUCCESS in 18m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/333ace94ec044ef5a24a58b80a158a02 : FAILURE in 28m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/17bbf13bd3f4423c8e84eefa6f6a55fe : SUCCESS in 31m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a29a06e887fd4ef8811ead755af738ca : SUCCESS in 12m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/89547528bda94bd79d31c14a3218998b : SUCCESS in 9m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1ba5db026fe8487a9cfd92321dfeb899 : SUCCESS in 12m 48s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/0268a71fca0248498063cfa55a27098f : SUCCESS in 26m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/39a5d037b6b04818830537925bfaf039 : SUCCESS in 31m 39s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/5f9b86e474834f9f9aefacdff33b494f : SUCCESS in 35m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0df9df83712a4c1ab2cf6c1dfef4942c : SUCCESS in 12m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2f3a4bb1a0aa4c17b8cdde7d9d2b3502 : SUCCESS in 31m 16s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5557fe2badf2480d8ee64ced3569eb44 : SUCCESS in 12m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dabf435ff6124167be07ae8e5efaac21 : SUCCESS in 30m 36s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f57479fcd4f84bf3ab3988231bd80c8c : SUCCESS in 25m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/548e2f7f8e00451f8ac94453fc44f13e : SUCCESS in 19m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/987174a7186d4d72ae18420793060164 : TIMED_OUT in 1h 03m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8ef266b2127422d9e827ecc1b6aeaca : SUCCESS in 45m 31s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/790b394ea29148329276a76656513183 : SUCCESS in 30m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ce3172989d7042b882ab0643448a6663 : SUCCESS in 33m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b95f4abfa7e74e358dcb3cc84012a3cc : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd9d39b6284e4014b63fbb57e537034c : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/29661eec04d34cc4804e64e0f5f1142e : FAILURE in 22m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c4ec3a619f744c8ba835b39f39527ca : SUCCESS in 1h 11m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6ec0ebad955d4c4faab3dcbd3ecdeb01 : SUCCESS in 1h 09m 53s","accounts_in_message":[],"_revision_number":1},{"id":"55e351efb5d4810ec759384a9f8c1fceda190da2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-29 18:09:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7d736dbd535cabd53f2e7c4606191d71def84519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 19:31:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edb78b107f0946719ef30e30d3979103 : SUCCESS in 10m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/eec818d1dc56435ab9e7a41b01815b98 : SUCCESS in 59m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ddf0b59c4ce04ad980e7ef8a00fca2d8 : SUCCESS in 48m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/027f0cf8717b4d56ba17c8548da066b0 : SUCCESS in 12m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a0779e2f767748ec8fa6d1dcfb03a41c : SUCCESS in 12m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/077d149469c1466c8d3f16ad36d5ac5c : SUCCESS in 17m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe5acac75dc440e0aab447fb0f4d165e : SUCCESS in 14m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c94028a9d58842929413314024cd36ae : SUCCESS in 14m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/003cee502ab64942ae8a962525b19464 : SUCCESS in 26m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ac69aa7140be447bb681f4c27731bfae : SUCCESS in 20m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/63e4ed844972460296ba7fc0862432cd : SUCCESS in 14m 34s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/edac5306649943e5961742ab9740f5cf : SUCCESS in 11m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/31d66bb968e14b52b67e3d0d9ec553c2 : SUCCESS in 19m 00s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3f0a512888234bbb99eabbcf95ce3929 : SUCCESS in 16m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/17563ec623134e36921b31451a55be35 : SUCCESS in 22m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e57a3902092b4513ad20b6a475e819dd : SUCCESS in 21m 21s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d4f8d97ebfc146548210a4411dc95878 : SUCCESS in 14m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/87ff09b1a9b34eeb93b7eccaf7783550 : SUCCESS in 20m 38s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c9fcc9f71d04450d9a5d480d1138e8da : SUCCESS in 16m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2c49e362ac714c96bb21b012dbdc1943 : SUCCESS in 18m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/67ba699b214c45618d96f13831303b91 : SUCCESS in 45m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f46ef222499141e8b3d65d971bf9cede : SUCCESS in 30m 47s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/fbfdb6524a3c4ed0a282f1c30e470677 : SUCCESS in 46m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4406f17623214d869f9e6ceb1cfa8f0a : SUCCESS in 29m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28df1053d85e4c98bcbf51083e16f38f : SUCCESS in 20m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5591858e81ba4e9eadb68df5ceb365ce : SUCCESS in 9m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3158c813f3ed4733aa505b24c3e6e9d1 : FAILURE in 1h 00m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a445fecc4aa41c4ad3b63ba4a11d886 : SUCCESS in 57m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/04c567362b1d4471ae9e8381e56b5376 : SUCCESS in 1h 00m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f7de8327f9e2d1d262d4d2807b6e1d1a54598429","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-26 21:57:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"056da080e3bbad1f37f880110700a5a5b707982d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 23:01:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b985784cbf640199ecfec54ec9313ac : SUCCESS in 10m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/5cb26a01e9c045f9be2725bf9cb0c31e : SUCCESS in 1h 02m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5cd48b57d8044f018c98d31667ab551b : SUCCESS in 44m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8d17309dbd134643a6cabc0b198b55de : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/efcf0eebbc944885b4f31e638dfa9503 : SUCCESS in 8m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/026ab4b6f0e34f0296c128e900f56505 : SUCCESS in 16m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/311cc41ae78f4aa28938404512c79b1e : SUCCESS in 13m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/071f0f3b38f047689bad6bc8d603aead : SUCCESS in 17m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5d54a527fe5742bf969946e5959a1521 : SUCCESS in 17m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4eec19c4a85c4e1f9f63f8b3a88f1400 : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3afcbb624b894592b06a88a6a88c7239 : SUCCESS in 21m 48s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/67d407c80803403ebf73635ec1108201 : SUCCESS in 11m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e3d61b36e9cf4fb89cf1e1c4c13732ea : SUCCESS in 11m 59s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f483e978080a4136a370074496bfb1cb : SUCCESS in 17m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/52fd22ee70af4211a07f8342079fcf37 : SUCCESS in 11m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/588424bb90a84236b1bac550fd5e6a97 : SUCCESS in 13m 53s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/6e7bbcb9d2a94e6b9a665bcebeb9ea56 : SUCCESS in 11m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/05bc225d2f9b437c979a15b74604daf2 : SUCCESS in 12m 35s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ba79bc22764b4f2780a0e5b0f25bd359 : SUCCESS in 20m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f905639af28c4f48805de9b353263e7b : SUCCESS in 23m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f964baf640b74778a6364af50cf815a8 : SUCCESS in 45m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0134576ae9b34e4db3b7a7d03af76812 : SUCCESS in 34m 02s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1ed8c547824043c889d6183d2b6cccb3 : SUCCESS in 33m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9cc6bb2b9b0c458591f922c8729276c4 : SUCCESS in 39m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9465a8853dfd4e1da2f420bd7d5f9074 : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ece33bbaa2948208ae04bda36163d7e : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fe3d5af1883d42a796e54ac1f89a1d02 : FAILURE in 8m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/767cb7ba8f9446a0a67717f748ea4ff8 : SUCCESS in 54m 10s","accounts_in_message":[],"_revision_number":2},{"id":"70d4f47de88fa9a1a94f38304bee0aa97c632d68","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-05-27 04:36:19.000000000","message":"Patch Set 2: Code-Review-1\n\nDid you think about adding a command like switch to enable this?","accounts_in_message":[],"_revision_number":2},{"id":"b23bacd075a59fd98f01e1578731476e05b9689b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-27 05:24:55.000000000","message":"Patch Set 2:\n\nSo I\u0027d originally gone that route with patchset 1, complete with another flag to explicitly opt *out* of this new behavior and a warning that default behavior would change in a future release if you specified neither flag and the new behavior would have changed the result of the query.\n\nWhen I brought it up at the meeting a few weeks back [0], people seemed to want the new behavior immediately. But if the only flag is to revert to the old behavior... it seems silly to bother: if an operator is going to change their tooling to include this new flag, they may as well change it to use quoted paths instead.\n\nI suppose this could go back to a tri-state but with the unspecified behavior preferring to unquote() and a warning to help publicize the new behavior... but then we\u0027re adding two new flags that mainly add friction to this operator tool yet (knowing us) we\u0027ll still support them indefinitely because if we *remove* them we\u0027ll just break something else!\n\nThe thing is, I *really* want bug 1875734 fixed -- it sucks having containers and objects out in your cluster that you can\u0027t easily find.\n\n[0] http://eavesdrop.openstack.org/meetings/swift/2020/swift.2020-04-29-21.00.log.html#l-13","accounts_in_message":[],"_revision_number":2},{"id":"be6247329a43401b2bd301ee490b6567c7c15a90","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-05-27 06:00:29.000000000","message":"Patch Set 2: -Code-Review\n\n\u003e When I brought it up at the meeting a few weeks back [0], people\n \u003e seemed to want the new behavior immediately. But if the only flag\n \u003e is to revert to the old behavior... it seems silly to bother: if an\n \u003e operator is going to change their tooling to include this new flag,\n \u003e they may as well change it to use quoted paths instead.\n\nWell, duh. Of course it has to be opt-in, default is compatible\nwith the old behavior.\n\n \u003e [0] http://eavesdrop.openstack.org/meetings/swift/2020/swift.2020-04-29-21.00.log.html#l-13\n\nI think Clay was wrong, and your -Q idea was the right choice.\nNo tri-state is necessary then. But this is what I get for\nmissing the meeting (I joined but didn\u0027t look at it, apparently).","accounts_in_message":[],"_revision_number":2},{"id":"b5209cb8d7b43ccf689250c862bf1b7897201590","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-05 23:11:14.000000000","message":"Patch Set 2: Code-Review+1\n\nI 100% agree with Tim and Zaitcev having something merged is the best state - I did not realize I was the hold up on urlquoted paths working from the command line.","accounts_in_message":[],"_revision_number":2},{"id":"d26594192b3b326cbe56e41d5d44c46ed6f1ce5b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-08 19:05:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"79551070cd5edab8685dbb51f5333b581242d4c0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-08 19:05:50.000000000","message":"Patch Set 3:\n\nOK, one new option, no plans to deprecate or change behaviors.","accounts_in_message":[],"_revision_number":3},{"id":"e96e6ba7141791ffa4231fec25a550f26c74bf4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 20:23:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9952af904b34634ad33b8c004e17c21 : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/b985bad224f0452f89544ff4853943f1 : SUCCESS in 1h 09m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e9a965e5b5034bd697195863c8472593 : SUCCESS in 49m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca85d89ce4244d6b8db5e5796d62896b : SUCCESS in 10m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7ed6465b727b4c67be281c24cf8cd7b7 : SUCCESS in 12m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/005c89714d1445ddbe3e0775157a4664 : SUCCESS in 12m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/546e9cfaf32d44dfba67522b5079346d : SUCCESS in 19m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9312ff1e9a0548afa39853c504459dde : SUCCESS in 13m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/85fb2ae5438d46369c8282866ed45f9f : SUCCESS in 24m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/842cd7766b164feda19b872ee63113b4 : SUCCESS in 20m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b73724f46c484d8897b1eec08ac7e5eb : SUCCESS in 17m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c688b6095954431a4f2bc20a2ba263b : SUCCESS in 22m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4dcdb62642114a39b67044bec40da308 : SUCCESS in 19m 38s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cf52046c5f62485d810bbe02bc30ec3d : SUCCESS in 19m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/955298ccbc524c47b9e424d8dc80d309 : SUCCESS in 20m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/46865b80d9d143faac1f295eaea68d05 : SUCCESS in 25m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/db08848361f0420b992fe3b2c1612708 : SUCCESS in 55m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/74f87270d613442ebea5d8e1350a12c4 : SUCCESS in 44m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/528ea9a06383468e9090ac9a5cd4ddd4 : SUCCESS in 41m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3305ec2c9ae14d34871380c8e5f12556 : SUCCESS in 15m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1499447f4134477955880d58b2ac496 : SUCCESS in 7m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9372d2c9622546238101a9976550f6d8 : FAILURE in 11m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/def5cef414dc4d4b89527ae6fef64fb7 : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":3},{"id":"51a389ce242368bea940af78768b7c5ea34f306d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-08 23:59:56.000000000","message":"Patch Set 3: Code-Review+2\n\nI think this is a great direction","accounts_in_message":[],"_revision_number":3},{"id":"0935144c2633a6b402b0194b97f83b964dc5e918","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-09 15:02:36.000000000","message":"Patch Set 3: Workflow+1\n\nthis works 100% just got to remember to always type -Q\n\nvagrant@saio:~$ swift-get-nodes /etc/swift/object.ring.gz /AUTH_test/%00versions%00vtest/%00test%008408285287.20706\n\nAccount  \tAUTH_test\nContainer\t%2500versions%2500vtest\nObject   \t%2500test%25008408285287.20706","accounts_in_message":[],"_revision_number":3},{"id":"1124f77cb23d6169c5d749a9792ff278aefc42bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 15:02:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0c74bc45baf6acf94a90cf475b56e8cefab69ea2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 17:54:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de4a1f5551dd447d816bc2634cb70489 : SUCCESS in 14m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/df02505b075045558aa332523dc3a931 : SUCCESS in 1h 05m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/67ec1514cd9f4cc58bdf27ce67bab3a0 : SUCCESS in 52m 35s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f4f57c0e91b145a6b3dc9d6c4b9597cc : SUCCESS in 15m 16s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/66f14405876a4edb8a53cf4165a622af : SUCCESS in 12m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/08e8a6b44cbe433cb87429eccb0085e7 : SUCCESS in 15m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c7f3cfa6f1774f1499fda76e561efb91 : SUCCESS in 14m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1bbd380d1d084be5b46cb4edbe0a8e19 : SUCCESS in 18m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c54297c4c56641e69b62579e56904002 : SUCCESS in 13m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/44efb0de2d274c27b75e0d6e7d61ade8 : SUCCESS in 18m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ddac6740c5ea4fc79d9c96997324b940 : SUCCESS in 18m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/599d6432f0b049e0abf5c691ae714679 : SUCCESS in 17m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7613a89815144fd2ad4ca776024c24ac : SUCCESS in 17m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d82cb3d311564ccaa3ae317c99870227 : SUCCESS in 19m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8b3b48e2f8e5400ca4e4559435f9dcfd : SUCCESS in 46m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e309669d0d7841699c39d181b7d03ffb : SUCCESS in 44m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/874db577b76d46c5b2802275327bb5d2 : SUCCESS in 45m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35ae600be7b740e08863bcb02d739b7f : SUCCESS in 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/778d518df0964477b8e629a76932175a : SUCCESS in 8m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d6bce6fb6e2453fa914245e6beb9f94 : SUCCESS in 58m 46s","accounts_in_message":[],"_revision_number":3},{"id":"ab6d630a78becf12150276cbdd4219599a58ba41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 17:54:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7a0956f59b615fc68aff95f09751d94145b9d8eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 17:58:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71ec02a4e7df4436a62c2f862a9ba325 : SUCCESS in 3m 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ae3d90a7aa9248fea16be20134f03add : SUCCESS in 3m 13s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d6fee5285a2145119f0d525f7ae858ba : SUCCESS in 3m 08s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1dfa41dada30c139129cb2771b0d68c95fd84e32","revisions":{"a08d8bd2c10c482339a0b48c15b6004d61b4757a":{"kind":"REWORK","_number":1,"created":"2020-04-28 19:42:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/724141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/724141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/724141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/724141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/724141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/724141/1"}}},"commit":{"parents":[{"commit":"69b8165cd87693f077a2385d9b6d53a7f459b098","subject":"obj: _finalize_durable may succeed even when data file is missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69b8165cd87693f077a2385d9b6d53a7f459b098"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-28 17:45:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-28 19:35:14.000000000","tz":-420},"subject":"swift-get-nodes: Allow users to specify either quoted or unquoted paths","message":"swift-get-nodes: Allow users to specify either quoted or unquoted paths\n\nNow that we can have null bytes in Swift paths, we need a way for\noperators to be able to locate such containers and objects. Our usual\ntrick of making sure the name is properly quoted for the shell won\u0027t\nsuffice; running something like\n\n   swift-get-nodes /etc/swift/container.ring.gz $\u0027AUTH_test/\\0versions\\0container\u0027\n\nhas the path get cut off after \"AUTH_test/\" because of how argv works.\n\nSo, add a new option, --quoted, to let operators indicate that they\nalready quoted the path. Since it seems like we should have been working\nlike that from the beginning, immediately deprecate the old behavior and\nindicate that the default may change in a future release. Add a\n--no-quoted option to explicitly opt-in to the old behavior (and\nsuppress the warning).\n\nDrive-bys:\n\n  * If account, container, or object are explicitly blank, treat them\n    as though they were not provided. This provides better errors when\n    account is explicitly blank, for example.\n  * If account, container, or object are not provided or explicitly\n    blank, skip printing them. This resolves abiguities about things\n    like objects whose name is actually \"None\".\n  * When displaying account, container, and object, quote them (since\n    they may contain newlines or other control characters).\n\nChange-Id: I3d10e121b403de7533cc3671604bcbdecb02c795\nRelated-Change: If912f71d8b0d03369680374e8233da85d8d38f85\nCloses-Bug: #1875734\nCloses-Bug: #1875735\nCloses-Bug: #1875736\nRelated-Bug: #1791302\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a08d8bd2c10c482339a0b48c15b6004d61b4757a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a08d8bd2c10c482339a0b48c15b6004d61b4757a"}]},"branch":"refs/heads/master"},"f2a9f0619045a3babf88ecbf400e381c054bf8f7":{"kind":"REWORK","_number":2,"created":"2020-05-26 21:57:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/724141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/724141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/724141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/724141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/724141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/724141/2"}}},"commit":{"parents":[{"commit":"69b8165cd87693f077a2385d9b6d53a7f459b098","subject":"obj: _finalize_durable may succeed even when data file is missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69b8165cd87693f077a2385d9b6d53a7f459b098"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-28 17:45:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-26 21:34:26.000000000","tz":-420},"subject":"swift-get-nodes: Unquote paths before working with them","message":"swift-get-nodes: Unquote paths before working with them\n\nNow that we can have null bytes in Swift paths, we need a way for\noperators to be able to locate such containers and objects. Our old\ntrick of making sure the name is properly quoted for the shell won\u0027t\nsuffice; running something like\n\n   swift-get-nodes /etc/swift/container.ring.gz $\u0027AUTH_test/\\0versions\\0container\u0027\n\nhas the path get cut off after \"AUTH_test/\" because of how argv works.\n\nSo, unquote all paths coming from the commandline. This lets users run\n\n   swift-get-nodes /etc/swift/container.ring.gz AUTH_test/%00versions%00container\n\nand have it actually find the versions container.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nThis subtly changes the behavior of swift-get-nodes in a\nbackwards-incompatible way. *In most cases*, this unquoting\nwill be a no-op, as\n\n   unquote(\u0027AUTH_test/container\u0027) \u003d\u003d \u0027AUTH_test/container\u0027\n   unquote(\u0027string with spaces\u0027) \u003d\u003d \u0027string with spaces\u0027\n   unquote(\u0027100% beef\u0027) \u003d\u003d \u0027100% beef\u0027\n   unquote(\u0027☃\u0027) \u003d\u003d \u0027☃\u0027\n\nBut, some paths (such as \u0027100%beef\u0027) will now need to be quoted.\n\nDrive-bys:\n\n  * If account, container, or object are explicitly blank, treat them\n    as though they were not provided. This provides better errors when\n    account is explicitly blank, for example.\n  * If account, container, or object are not provided or explicitly\n    blank, skip printing them. This resolves abiguities about things\n    like objects whose name is actually \"None\".\n  * When displaying account, container, and object, quote them (since\n    they may contain newlines or other control characters).\n\nChange-Id: I3d10e121b403de7533cc3671604bcbdecb02c795\nRelated-Change: If912f71d8b0d03369680374e8233da85d8d38f85\nCloses-Bug: #1875734\nCloses-Bug: #1875735\nCloses-Bug: #1875736\nRelated-Bug: #1791302\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2a9f0619045a3babf88ecbf400e381c054bf8f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2a9f0619045a3babf88ecbf400e381c054bf8f7"}]},"branch":"refs/heads/master"},"1dfa41dada30c139129cb2771b0d68c95fd84e32":{"kind":"REWORK","_number":3,"created":"2020-06-08 19:05:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/724141/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/724141/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/724141/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/724141/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/724141/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/724141/3"}}},"commit":{"parents":[{"commit":"69b8165cd87693f077a2385d9b6d53a7f459b098","subject":"obj: _finalize_durable may succeed even when data file is missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69b8165cd87693f077a2385d9b6d53a7f459b098"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-28 17:45:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-08 19:03:56.000000000","tz":-420},"subject":"swift-get-nodes: Allow users to specify either quoted or unquoted paths","message":"swift-get-nodes: Allow users to specify either quoted or unquoted paths\n\nNow that we can have null bytes in Swift paths, we need a way for\noperators to be able to locate such containers and objects. Our usual\ntrick of making sure the name is properly quoted for the shell won\u0027t\nsuffice; running something like\n\n   swift-get-nodes /etc/swift/container.ring.gz $\u0027AUTH_test/\\0versions\\0container\u0027\n\nhas the path get cut off after \"AUTH_test/\" because of how argv works.\n\nSo, add a new option, --quoted, to let operators indicate that they\nalready quoted the path.\n\nDrive-bys:\n\n  * If account, container, or object are explicitly blank, treat them\n    as though they were not provided. This provides better errors when\n    account is explicitly blank, for example.\n  * If account, container, or object are not provided or explicitly\n    blank, skip printing them. This resolves abiguities about things\n    like objects whose name is actually \"None\".\n  * When displaying account, container, and object, quote them (since\n    they may contain newlines or other control characters).\n\nChange-Id: I3d10e121b403de7533cc3671604bcbdecb02c795\nRelated-Change: If912f71d8b0d03369680374e8233da85d8d38f85\nCloses-Bug: #1875734\nCloses-Bug: #1875735\nCloses-Bug: #1875736\nRelated-Bug: #1791302\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1dfa41dada30c139129cb2771b0d68c95fd84e32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1dfa41dada30c139129cb2771b0d68c95fd84e32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
