)]}'
{"id":"openstack%2Fswift~724393","triplet_id":"openstack%2Fswift~master~Ifd744044421ef2ca917ce9502b155a6514ce8ecf","project":"openstack/swift","branch":"master","topic":"versioning-acls","hashtags":[],"change_id":"Ifd744044421ef2ca917ce9502b155a6514ce8ecf","subject":"versioning: Have versioning symlinks make pre-auth requests to reserved container","status":"MERGED","created":"2020-04-29 18:07:18.000000000","updated":"2020-05-27 00:17:30.000000000","submitted":"2020-05-27 00:15:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724393-1590538558537-60d2867d","meta_rev_id":"653422637a4eefe0b5a7a77aba3442344f93b834","_number":724393,"virtual_id_number":724393,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 00:15:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-26 18:12:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":1,"date":"2020-05-26 18:20:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 02:32:27.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2020-05-26 18:20:46.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-05-27 00:15:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3cbcdc14c85dd479b7b86c50ad23d756ca625af","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-29 18:07:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db7b45d107c5c7d459f8cfe75757e8f0ab307100","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 19:11:23.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52e7451f35424760b86f93041e0feb2e : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/c32b8a0e834346c2a13f37181ca46c65 : SUCCESS in 58m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b7843144e34d4cdfad99e8d1c9ea0636 : SUCCESS in 57m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/185690afe1f64add858374fb37a5d09d : SUCCESS in 10m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5d95cbeb2bfd49e78e28071b9221eb83 : SUCCESS in 10m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e7e21f95f1bf433c8448fb92e24b7977 : SUCCESS in 19m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e5cfad0b39bc416d9d2abe38d10349d0 : SUCCESS in 16m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/059282248aa94d9a9350f81d2a224f1d : SUCCESS in 17m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/322b31aa0347472894fb03fc2995f4a1 : SUCCESS in 26m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6c3307e80c9242ea8cac1d98534287d7 : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2216beb2a92a453b965cd89e35a0940c : SUCCESS in 14m 33s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/9a0276f4afd146ada7f17f47c0755384 : SUCCESS in 26m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d7540d41ccd34c45b00548e5babc6d54 : SUCCESS in 11m 32s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/41c5c60be26e4cbb8a873500248beba2 : SUCCESS in 16m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0bc88fc1cf864660a0533bdb39345a38 : SUCCESS in 11m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4e8d521a7a3e44ce8b2e3d8ef33d5c54 : SUCCESS in 12m 52s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/27ad43faf5d142f3b420d9411e314d2d : SUCCESS in 26m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d31d6f11d043443e8744ef137a323d5d : SUCCESS in 26m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/272ad52944c14b0a9e0d3ed0056df6f7 : SUCCESS in 17m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/254e16a982de4ef2ac8483d646017935 : SUCCESS in 20m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/77e9dca243854d3f8504e8346cb868d3 : SUCCESS in 51m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9178364742ed46b3873bf8c34d5b60c5 : SUCCESS in 34m 01s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/6fbbd9ebcdbf4b308c9ef38405912008 : SUCCESS in 30m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/873758a7d3bf463e8c47e5817a4662c8 : SUCCESS in 30m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87174d287aeb4e9f86f4dbdecfe0404f : SUCCESS in 36m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bde666a28a64745800f880e929172da : FAILURE in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/457781f3ee454518ba2a5c5c100f9669 : FAILURE in 39m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aea6f5e3c3f944a4b9cccffdeeea8742 : SUCCESS in 48m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a037d2245e0943cfb17a264bc342482f : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a8b8904fd78159ed2ea2f6e64f96e283ca51e0be","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-04-29 19:57:22.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nlgtm, still needs to fixup pep8","accounts_in_message":[],"_revision_number":1},{"id":"7f332b945f8f057e66008b9fd8bee7387b876eba","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-04-30 04:19:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b9133beed7455b0b143f82a58795395dac77e5b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-21 21:09:10.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)\n\nI need to circle back to this :-/\n\nThanks for writing up a bug for it, Clay!","accounts_in_message":[],"_revision_number":1},{"id":"656f654f9496ea50a794642e5d0590286806c411","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-21 22:08:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"44fa4845ad135791e9e5958b6cad0b0797de1674","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-21 22:11:54.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4af083a5ddc1f5a3b9011bd2e18eefaa9adf7b2a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-21 22:13:26.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nI think the making the sub requests based on the orig request and preauth-ing only the subrequests that target reserved symlinks is maybe ok and not obviously unnecessarily coupled with versioning","accounts_in_message":[],"_revision_number":3},{"id":"ea337b3a6652dbb8650ce1add751666b48290c11","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-21 22:27:57.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"518795405185becdbc605ac83c0b95afc27fc479","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 23:26:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5ce51d7bcc461fa3cda24ca13946c7 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/670419a84cf640fab175958815cc5ede : SUCCESS in 1h 07m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f1ee7287ecf34895bf89268c0ec8f49e : FAILURE in 50m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c9050fc46ff446a99f00783a84d1e58c : SUCCESS in 9m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c3ba40fcc77c4393827dbf5c701a8b2e : SUCCESS in 9m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/934a5bb370dd49a391384d39504bf47a : SUCCESS in 13m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/977677921aa04ed0a7cf876caeefcf2e : SUCCESS in 25m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/70de3c15c03e4e9d81c34963c718d8f4 : SUCCESS in 13m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/00c248ec80764aee85dd388e6ab448ef : SUCCESS in 12m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cf18d0e646374227959b2a39b6d4eee2 : SUCCESS in 13m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a8df5db573b64e0f920a78030a6cd29b : SUCCESS in 25m 45s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/04d938d700364498bebe038f18737030 : SUCCESS in 14m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/585dbf025b24407c8249241903527ebc : SUCCESS in 11m 06s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/af8791397c25477aa1c83cc735e7c397 : SUCCESS in 16m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e725ca9852964469b3b0d2e9fd808adc : SUCCESS in 9m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/96e4bdef9ee74b96b584b1a67bccf18c : SUCCESS in 29m 49s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a62c305fb0944b75980e5e87ca680f03 : SUCCESS in 20m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a52cf09508be47daa8c27e3c3397125f : SUCCESS in 14m 22s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f36b79860658451fbcb90c069eb5baa3 : SUCCESS in 30m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4f55e40e75bd41e988773bab82bc151c : SUCCESS in 18m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fc8a6d374c624ee38ea32db85b0ba9e3 : SUCCESS in 46m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d68a257d193a44bdbf1c4ee47dbb54d6 : SUCCESS in 38m 21s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/5e5364d9fd71420193e8c560299b0e5d : SUCCESS in 33m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d33803a0cb104bb38bc8fd802a022184 : SUCCESS in 33m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bacbc9d056304afea089c9ec482d168f : SUCCESS in 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50672fa621e648edb318ff93316d3b93 : SUCCESS in 6m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d8190bf978e435c90ba76b3fe42bc7e : FAILURE in 9m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad9c55e4271b4bb4a2f0eb54cbf3785f : SUCCESS in 1h 03m 37s","accounts_in_message":[],"_revision_number":3},{"id":"6d6b92a749db9ea90496c184ff0b52f0ec4bb21e","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-05-22 02:32:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5f71c525b5cf8072b4de41f66718d23bdb3518ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 03:38:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1aa9106185a47e9b730e51b79238796 : SUCCESS in 9m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/be9f7821b2514d08abb962118d6beff1 : SUCCESS in 1h 00m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/245292774b2a4b75ad20bed44f1481cd : SUCCESS in 40m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3f7062c7bf8042b4863414d5cc5d1fca : SUCCESS in 9m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/574478fb081444bb80d56d8237cf2954 : SUCCESS in 9m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9a7f46c0e9e24de5a6003a3091c7151e : SUCCESS in 13m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1016fa2cfebd4a0db0a6ed1788849120 : SUCCESS in 15m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9ea3945b97da4f9dafca815368eb7cb4 : SUCCESS in 13m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/430efb32edb34f51b7ae3a50d7ab9905 : SUCCESS in 25m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dfa25660e7d347549c59e2b470804826 : SUCCESS in 11m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2ad6ccd3127546f08fa5ec294eb96fc1 : SUCCESS in 14m 46s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8c290d1e4b0949988691b04c2c1c88c4 : SUCCESS in 11m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/43eb87985f514823987f73a0d8471541 : SUCCESS in 10m 55s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/133282c488c44e699c76e04a9782751c : SUCCESS in 21m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0898884bcb7547b78939fff1a9ea35e5 : SUCCESS in 12m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/28d48eefda674288946b46ac1b121093 : SUCCESS in 19m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ab5ea05d737d4f5fac3aa75956beb40a : SUCCESS in 14m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bb5ed403a8a643eeb2309a4e25352567 : SUCCESS in 23m 41s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/da1fb650c1b844259368971599af543f : SUCCESS in 20m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c56143dff2ce476d8b3473688e3a6d3f : SUCCESS in 19m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/976e9c93bd9b4cad990d7f685c55248b : SUCCESS in 45m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/74589f7984ba4effae93599670a18e0b : SUCCESS in 30m 12s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f8cbce6cc9ea4e01bfe73ebeab3074d4 : SUCCESS in 34m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/57aeb154ce61452fbf857b903f09b789 : SUCCESS in 40m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a05e1b7cc5b14b2bbb63ea9a5e61c205 : SUCCESS in 15m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb28a69e8904484e849bb207124d7847 : SUCCESS in 6m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5b76df7df1e4906a151d98af5540767 : FAILURE in 8m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/227d90e6c0ec4f7ab11eb3dd486c1a74 : SUCCESS in 1h 04m 13s","accounts_in_message":[],"_revision_number":3},{"id":"61f4c81aa4a05275f8590efbbf2dd9a9cf2d0dad","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-26 15:10:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b36d4a984355fe7f2e3c7f13bf8e96a0f4f0a392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:15:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/922ee5a6bd3d48bd9df83a8c6f90145d : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/bee324130c0c464a8d7214fd74bff43f : SUCCESS in 1h 03m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f38c0e9007c745c7b711404780345584 : SUCCESS in 49m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bf59e5273bb04fbdaa43c73c72343bba : SUCCESS in 10m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/932d3b66d1c74064948ad6243969ae8a : SUCCESS in 11m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5aab0a1c471b462da373ab86ef43a166 : SUCCESS in 17m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cc4399bb26f941b28f2bc3332aef6fd2 : SUCCESS in 18m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4b8032659fdb4a1b9a957cac5b6bf368 : SUCCESS in 15m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/99147a2cea214d8abc1a1f2b3506079a : SUCCESS in 13m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4462e4a8ecd84aa3b0d0139627dcfefc : SUCCESS in 22m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4b9a85d6d70a4be79d27525787ded844 : SUCCESS in 31m 12s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/2e98e1274f6b446c827e0b74a85bc816 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b8d8d3c4863445cb773cbf5e6c32926 : SUCCESS in 16m 47s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/36b8aa2d8508475fb424c58e58600781 : SUCCESS in 26m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/aa2960f0923844e3aa8b98ebf575f805 : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6504d187dcd3452e8ed90973f2d16cfa : SUCCESS in 19m 38s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/94896d424f5d4df1b41a26198146a544 : SUCCESS in 22m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c5af929307f94ba1b073cebe5f9102cd : SUCCESS in 24m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/cd4b7f25880646c0a0f7701dfeb43c50 : SUCCESS in 15m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/65a699a0701741519c0841a07ba54419 : SUCCESS in 23m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c9d8855bc90d48ea8364746924895c44 : SUCCESS in 49m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e37684f98ffb4676afdd142720ea9829 : SUCCESS in 37m 27s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9e2fe01ecf33473a9e9b5b31731ddf14 : SUCCESS in 36m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d6bea82fc7064900a55f2f73f1b6771e : SUCCESS in 39m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56041f4d490d439c839f411c64545008 : SUCCESS in 20m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0378f78cf854e90a5946bf9da842422 : SUCCESS in 8m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a742aa61c1648d98ac1cd7facc448f0 : FAILURE in 13m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cda7249a653f42848ae85e71dd63d694 : SUCCESS in 57m 02s","accounts_in_message":[],"_revision_number":4},{"id":"6ec07382ab778cea40a0acb5455e8dd072ed7118","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-26 18:12:19.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"c85172a85b12397084cb2243c49b39675bf5db29","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-26 18:20:46.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"459b626c9986d2be1335437b0371b6e6bdc89c4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 18:21:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ab47729f31b25d1b18cdf50e59b02169a75e40d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 20:36:00.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c315bbe5982a4fecb71e3c372c27cf7f : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/12797ae1e91f4d519ba3218d0a97a582 : SUCCESS in 59m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/de3da46174af4a6e993de3b2b108f41c : SUCCESS in 47m 52s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cf6aeebc438f43c4897127c7e226090f : SUCCESS in 11m 10s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/75a30582169b4a2fb07d9c52bcec8e4b : SUCCESS in 15m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f42c38ddbd2a4d9fbe34e237824ab008 : SUCCESS in 22m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c9927156a8a54a2cb05592f06983fe0c : SUCCESS in 16m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f6ab8165c596426baea8dea4ac1268fb : SUCCESS in 15m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/651ab25bfd9c47ddbfa483f480e018ea : SUCCESS in 15m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d858a29b9b24449b8d077d736f916c23 : SUCCESS in 13m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/92f05522459b4c47a99c702c3202179a : SUCCESS in 19m 37s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/72c58a901a8348099d7f17e01d3e8a46 : SUCCESS in 13m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c654628856f74c08869bf4a7a82db549 : SUCCESS in 20m 50s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/8c7f724c51c949f5ad8c376e1d554193 : FAILURE in 33m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/64275e258a2c44e1bef72c34b57e8011 : SUCCESS in 23m 21s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f16d85d9db2141b2935f51f19f86b3de : SUCCESS in 12m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/db875523ba4f425287563f37a86fb3ac : SUCCESS in 16m 44s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a5d349531105457dadf08920750d8cfe : SUCCESS in 21m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5d35c8b066614b199a16a2103c31b09e : SUCCESS in 51m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7d86879078c14372b8cadce323a2302e : SUCCESS in 36m 17s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/20393e3e1ea84b98a85d81fb6fa6baaa : SUCCESS in 48m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4e0b56327724493688cb583def51eef6 : SUCCESS in 43m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8d2ed37afd94d06a77a82dd4031980e : SUCCESS in 14m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45eb9560d9894973b4bc13f7f5829c3b : SUCCESS in 6m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3672ca4c74e44c2ab4235cfdf6bc591 : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":4},{"id":"f35f73367269a5a4f9658395ee34582dec7ce00f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-26 20:39:33.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"53d1e740f87481ee721235d22f55412df28a278c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 22:29:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd7dba5515db4ca7b015a9d347ae29b7 : SUCCESS in 11m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/068162d5d5584805919e099d7b364bf9 : SUCCESS in 1h 01m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3a5185d6595d44d7bcd1f3ccf864d522 : SUCCESS in 48m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f3c77f59e2a54d7f9b03d686d5f63955 : SUCCESS in 10m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4f71d3dd0d7f4184bd35b52e528bb440 : SUCCESS in 10m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0bc172615db74f76b7828d60215cc302 : SUCCESS in 15m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9523760efeb84cb8a94b54cae786676b : SUCCESS in 15m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4f196fbc9e3345f9ba41795f70645552 : SUCCESS in 16m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/aa70bea1cc9f486087a3f2e89629d2aa : SUCCESS in 17m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1f8f4fc11e0b4b19b411294613611996 : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/69ef779937bb4569bede10b713c55f52 : SUCCESS in 16m 40s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/30a7eac0e8714b6897c3e5bbfb6c80d8 : SUCCESS in 12m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/853258e632fa4a9c82bac28336d1d15a : SUCCESS in 13m 05s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f71a3881ccfd4ddbbd67d5345e59d41b : SUCCESS in 17m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d37610dadc14416394ecfdb28e0f3b66 : SUCCESS in 13m 38s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/06e5cc9734054d8dbdc78feb41e53e15 : SUCCESS in 16m 56s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/df3b062c82cb488aae930044587b483d : SUCCESS in 13m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/261da943e7c64eccaeab1310b1317547 : SUCCESS in 11m 51s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f99c7501aebd417fbc2755d597898d2c : SUCCESS in 21m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5fa792729f6840b9aa283cf53b195a16 : SUCCESS in 26m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bbf40e0865da496099ec72c1feafd3bf : SUCCESS in 45m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/427a856e7cc7403b863f4a15976b4ab5 : SUCCESS in 37m 16s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/93d6f29204664b02949bc3f3efa589b2 : SUCCESS in 37m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0b0f23d3c16b4ab190a3f17af1e26cdb : SUCCESS in 36m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69daaa04846643349d158cec80130597 : SUCCESS in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3480235e0b7447aeb0cbc3453659fd5a : SUCCESS in 7m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d256593e58a040b1a4b3220ab04fb4ae : FAILURE in 14m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6084bd31a77649d195ca615bbccfa6e2 : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":4},{"id":"415feeb898261cd1aec2c0f615953af92c83c079","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 22:30:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f1e2ca5731b0db10d71576d558943dfd7e261f48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 00:15:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2fcb82d61a64502a4298bb55b5176cf : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/3551391d7ff3446d9aca339eaf602fd6 : SUCCESS in 1h 09m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9095c249d45045efbfb3f1658d961bc0 : SUCCESS in 43m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bd3d22a469dd44c79453378f89adb4b8 : SUCCESS in 10m 11s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/bb84afdb2e544d139c4d76fcafff2831 : SUCCESS in 10m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/afe7aa2b3edc448589dec3928051f8da : SUCCESS in 15m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/792ae36b6ea74829865035718d1b506a : SUCCESS in 20m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ba3e31d6c5184ff78cd72fc1501b776f : SUCCESS in 16m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/974e64f2d84f4fec813229b4feed023e : SUCCESS in 15m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1c99f47770a44bd3b03a988f14cc3fbe : SUCCESS in 12m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bb1bf243e9914b5fb5b70f9978fdbb0f : SUCCESS in 16m 09s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/804946b0e1ca46408c0602a902e6e5ba : SUCCESS in 13m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/226ef7af098d4664a5713d503556bae4 : SUCCESS in 30m 45s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f31112a0e2604ed89728b14db81d1385 : SUCCESS in 20m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e27b3eb5e37f4f72bad6bf6da6a14563 : SUCCESS in 17m 07s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/383b92193c004370881de8d296d95195 : SUCCESS in 13m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c888208b8a4643de9e56cf7b83a5579d : SUCCESS in 14m 34s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/bc5866e026d2422591d4ee2188c4fa0e : SUCCESS in 18m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cdf64bb2cadd4bd78faeae5d5402bddb : SUCCESS in 59m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e6af497d4f5e4173a5e3fba6dd8adb28 : SUCCESS in 36m 56s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/2a46dd1bb7c4407facdfe37da6735bb2 : SUCCESS in 40m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b170f593f4744cce8c75e7babd50c78a : SUCCESS in 38m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49ea98d9964641e78592773277862d96 : SUCCESS in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c5243a566b8486da9158582937f9986 : SUCCESS in 6m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/593243b894cb41e5b121418a8dee5d19 : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":4},{"id":"4d0b4d48f6b14c71efb1149ec0f36a6b73eb3dcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 00:15:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ffcc6fcb89b8967542618ad7e0b193c5516d3e78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 00:17:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a7e5caaadcc4a969595fda4641b1361 : SUCCESS in 1m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/30c6dfe1ac2f40869160f470cd8e37e6 : FAILURE in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/4a55a916eba647aaa502b1ee9afedad5 : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a8e03f42e09de3f3c660e0d4786bb35fe1317482","revisions":{"4e0aa29776acc7d138635925978bc4d18bcba941":{"kind":"REWORK","_number":1,"created":"2020-04-29 18:07:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/93/724393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/724393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/724393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/724393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/724393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/724393/1"}}},"commit":{"parents":[{"commit":"58ddca8fa5ccb99447f7dcc0745cc619449a5513","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58ddca8fa5ccb99447f7dcc0745cc619449a5513"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-28 23:48:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-29 18:05:43.000000000","tz":-420},"subject":"versioning: Have versioning symlinks make pre-auth requests to reserved container","message":"versioning: Have versioning symlinks make pre-auth requests to reserved container\n\nPreviously, the lack of container ACLs on the reserved container would\nmean that attempting to grant access to the user-visible container would\nnot work; the user could not access the backing object.\n\nNow, have symlinks with the allow-reserved-names sysmeta set be\npre-authed. Note that the user still has to be authorized to read the\nsymlink, and if the backing object was *itself* a symlink, that will be\nauthed separately.\n\nChange-Id: Ifd744044421ef2ca917ce9502b155a6514ce8ecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e0aa29776acc7d138635925978bc4d18bcba941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e0aa29776acc7d138635925978bc4d18bcba941"}]},"branch":"refs/heads/master"},"87caae284da9a2551b05384477cc93405201f697":{"kind":"REWORK","_number":2,"created":"2020-05-21 22:08:49.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/93/724393/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/724393/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/724393/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/724393/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/724393/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/724393/2"}}},"commit":{"parents":[{"commit":"58ddca8fa5ccb99447f7dcc0745cc619449a5513","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58ddca8fa5ccb99447f7dcc0745cc619449a5513"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-28 23:48:52.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-05-21 22:08:27.000000000","tz":-300},"subject":"versioning: Have versioning symlinks make pre-auth requests to reserved container","message":"versioning: Have versioning symlinks make pre-auth requests to reserved container\n\nPreviously, the lack of container ACLs on the reserved container would\nmean that attempting to grant access to the user-visible container would\nnot work; the user could not access the backing object.\n\nNow, have symlinks with the allow-reserved-names sysmeta set be\npre-authed. Note that the user still has to be authorized to read the\nsymlink, and if the backing object was *itself* a symlink, that will be\nauthed separately.\n\nChange-Id: Ifd744044421ef2ca917ce9502b155a6514ce8ecf\nCloses-Bug: #724393\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87caae284da9a2551b05384477cc93405201f697"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87caae284da9a2551b05384477cc93405201f697"}]},"branch":"refs/heads/master"},"c27ce02aaea5e018ca6cfce294d906a14721f25a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-05-21 22:11:54.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/93/724393/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/724393/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/724393/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/724393/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/724393/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/724393/3"}}},"commit":{"parents":[{"commit":"58ddca8fa5ccb99447f7dcc0745cc619449a5513","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58ddca8fa5ccb99447f7dcc0745cc619449a5513"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-28 23:48:52.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-05-21 22:11:51.000000000","tz":0},"subject":"versioning: Have versioning symlinks make pre-auth requests to reserved container","message":"versioning: Have versioning symlinks make pre-auth requests to reserved container\n\nPreviously, the lack of container ACLs on the reserved container would\nmean that attempting to grant access to the user-visible container would\nnot work; the user could not access the backing object.\n\nNow, have symlinks with the allow-reserved-names sysmeta set be\npre-authed. Note that the user still has to be authorized to read the\nsymlink, and if the backing object was *itself* a symlink, that will be\nauthed separately.\n\nChange-Id: Ifd744044421ef2ca917ce9502b155a6514ce8ecf\nCloses-Bug: #1880013\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c27ce02aaea5e018ca6cfce294d906a14721f25a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c27ce02aaea5e018ca6cfce294d906a14721f25a"}]},"branch":"refs/heads/master"},"a8e03f42e09de3f3c660e0d4786bb35fe1317482":{"kind":"REWORK","_number":4,"created":"2020-05-26 15:10:11.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/93/724393/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/724393/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/724393/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/724393/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/724393/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/724393/4"}}},"commit":{"parents":[{"commit":"63e02fa9fa4e43cb20dd96eee81d0749edcca7e6","subject":"Test for Versioned Object If-Match","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63e02fa9fa4e43cb20dd96eee81d0749edcca7e6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-04-28 23:48:52.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-05-26 15:09:56.000000000","tz":-300},"subject":"versioning: Have versioning symlinks make pre-auth requests to reserved container","message":"versioning: Have versioning symlinks make pre-auth requests to reserved container\n\nPreviously, the lack of container ACLs on the reserved container would\nmean that attempting to grant access to the user-visible container would\nnot work; the user could not access the backing object.\n\nNow, have symlinks with the allow-reserved-names sysmeta set be\npre-authed. Note that the user still has to be authorized to read the\nsymlink, and if the backing object was *itself* a symlink, that will be\nauthed separately.\n\nChange-Id: Ifd744044421ef2ca917ce9502b155a6514ce8ecf\nCloses-Bug: #1880013\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8e03f42e09de3f3c660e0d4786bb35fe1317482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8e03f42e09de3f3c660e0d4786bb35fe1317482"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
