)]}'
{"id":"openstack%2Fswift~725190","triplet_id":"openstack%2Fswift~master~I4afc73301965b582fd338012f0ddc60f9b85fbb0","project":"openstack/swift","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I4afc73301965b582fd338012f0ddc60f9b85fbb0","subject":"Switch to new grenade job name","status":"MERGED","created":"2020-05-04 06:41:54.000000000","updated":"2020-05-05 07:03:19.000000000","submitted":"2020-05-05 06:57:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725190-1588661858510-2682b9c6","meta_rev_id":"46fd126c816e49a829a2ff5ae15e2a518c968b4a","_number":725190,"virtual_id_number":725190,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-04 14:56:56.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-05 06:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-05-04 11:54:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-05-04 11:16:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-04 18:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-04 18:34:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-04 11:16:25.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-05-04 11:54:55.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-04 18:34:49.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-05-05 06:57:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff679e11f1e46db0d5ce0aab755c5b1f78138a6e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 06:41:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebe73f65c7594320134d64f52f6808f08d04691e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 08:45:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01592d6531824bbdb8f2072b8ed55235 : RETRY_LIMIT in 2m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/b427be680baf42beb5ac54491a1e48fa : SUCCESS in 1h 01m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/90ef85d178e5441c87836cb02b2f9b64 : SUCCESS in 47m 48s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/602d769706444450a7717650bb3f30cd : SUCCESS in 10m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/53b5b697fd5342e799f4a978ec32f627 : SUCCESS in 12m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e9c857a039984dbcb2d1c91b60577696 : SUCCESS in 13m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3c1f91dfcd23450c8ab554a28e8e539d : SUCCESS in 28m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/12f378c580af4dfab1de08ccd861cbf1 : SUCCESS in 17m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2e5a9291cdb74f0a95dc610f8ae098a0 : SUCCESS in 17m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/539d68e3d82f42279e2f8d2ec60056f5 : SUCCESS in 10m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bf6f9c46ae8d4fc88dc62a957cb7ef40 : SUCCESS in 14m 40s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/cccd71b9b5604021a9d1f5c9e58ae296 : SUCCESS in 12m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d0cab92e16424b1992e573ae6795f0b9 : SUCCESS in 11m 08s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3d3ac913f576421c8f6759bcee7c83d6 : SUCCESS in 15m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/314d41a724ae4554af98a289f5169bae : SUCCESS in 23m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0054599caa454c35a1b36fc07bdda33d : SUCCESS in 13m 59s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c86387325ff041ee91a4798c170ed514 : SUCCESS in 10m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a3114f8bb70744a58671137285a11147 : SUCCESS in 15m 41s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2429e042f37a4569bf1ef1d8a8dcab13 : SUCCESS in 16m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cdcd5c53161f4fab84c4b6b9bee384b9 : FAILURE in 6m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ed4707407ad6451aad8e1ddbc1943b93 : SUCCESS in 56m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3fd5a402814449fba819d2438f89e78c : SUCCESS in 32m 04s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8b47cdd0ffd448ba92c65da7411dd391 : SUCCESS in 30m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f5aae7e309174d18a062d954b972d5c5 : SUCCESS in 31m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96d27839da0e4e83886d404dbfddc89e : SUCCESS in 25m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e05b6e1e5ca403ead2a3e1a5295e55f : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03e5eb3dd0274607823b95427088c91b : FAILURE in 11m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/624c682fdfc44dce847f3db5618d764b : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a96c1ee7821859014e3ea2c7030942dc53f37772","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 08:57:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ec36aa2feea637520508d5c8ad0355d6d72242f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 10:07:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0af53ec274274974925f49a7ac0c492a : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/e9f591f06f5049728d00a4922372ca09 : FAILURE in 7m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a4e90a5e1b334eafa49727d7029e1d6e : SUCCESS in 42m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b456922b228b420b924d71bc25a0f6f2 : SUCCESS in 9m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ec2b89a75e8a4c548e7f4de6bde904fa : SUCCESS in 11m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ffe600ccf1a649369057c5f5bd793a73 : SUCCESS in 24m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/158a6fe6a3c843cda9fd2fb5a8eb13e5 : SUCCESS in 12m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b696ef6dbed44cdeb1ca50fccb8a9918 : SUCCESS in 24m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2066d5375a5b454fa005e2ef7fcdafe3 : SUCCESS in 13m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a8c32f497ea840489a8b437c723d19d2 : SUCCESS in 10m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1698a215b6244e2784de969f6574bf83 : SUCCESS in 12m 07s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/48c84be3ff58411a93efefaa01f8f202 : SUCCESS in 10m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f43aa9a57ca841079930e9df619939f6 : SUCCESS in 11m 55s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0c220a84e07d4a0596d7c4c6173197f9 : SUCCESS in 16m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c3e8db6547e54854b651498381f4e4d8 : SUCCESS in 9m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bbda657dfa6f4da0adc335ecf55c856a : RETRY_LIMIT in 2m 25s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/fe71fab5c2d24dfdbd3ae83b484820d9 : SUCCESS in 10m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/629ea3b4d9354c99b947c8139b221ea1 : SUCCESS in 12m 01s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/89d41d1ce7204e338af763fc1c0d7d77 : SUCCESS in 15m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8d4ac2dbcf08485a8ec92e87d376f898 : FAILURE in 4m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/55136d489d054f0dad0a791658ec6188 : RETRY_LIMIT in 3m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2145bb5ece14439b0f101ac7b794c7a : SUCCESS in 31m 48s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/97461a0855d6458cba5376cd04f8baf7 : SUCCESS in 30m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c59992b7ae048a68b642909d7b730fd : SUCCESS in 28m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe34000128934906b08db214a1dad9ca : SUCCESS in 14m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf43b5dbce7497eb212b8c6615a4bb5 : SUCCESS in 7m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/69a2e7836f49417ab7df6c8f31df46aa : FAILURE in 7m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f52c321065b34d35b9b16a2bddf54b11 : SUCCESS in 1h 03m 17s","accounts_in_message":[],"_revision_number":1},{"id":"732d5d844167fc530118bb49ab1745db5c0860cd","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-05-04 11:16:25.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f17ebf7c13ed67264f62accb443fbff1ad239cbc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-04 11:54:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7424eab725e9ce154ebf4444f480279090e4504f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 12:34:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/138e11872d384a809dd4365991dd4fd9 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/802202e2b1c04374a5c9e671d889b365 : SUCCESS in 1h 01m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cbd8fb2fe5e24f9aad8213479d932840 : SUCCESS in 58m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/695fa4555cf641fb80e62ebf93d18f69 : SUCCESS in 10m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aa80affff03b4c8d88272ffebd675ff6 : SUCCESS in 10m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bbf655eb24f34c9b8c3cf13283a35870 : SUCCESS in 20m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d330726346648a9a74aa9216c62b50c : SUCCESS in 16m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b1d4cd5f49024586bef520cb7e817dbc : SUCCESS in 32m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/64048ac0f0f04a13bfea0204bae06884 : SUCCESS in 24m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/659a2806422a4f31968fc91c6fdc6a28 : SUCCESS in 32m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c9e5a28cca8d42b4ad969dbf5b0e20c2 : SUCCESS in 26m 34s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/6ea13e066b3a4360b8708745121a2404 : FAILURE in 32m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ab27c6f9000749b4bde21612c80a556f : SUCCESS in 32m 28s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/cd945ccb90644b2681c702ef5bef5550 : TIMED_OUT in 41m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dd0dfa4b95c246b7a0ce5a02340a4539 : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5e8a37be87e3405f859bacc4a5feca05 : SUCCESS in 39m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/bf12d3f4df7d4f8194520f684084dda4 : SUCCESS in 32m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bd145190572d4283a6452a5b5306f88a : SUCCESS in 34m 02s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/28706b3e653a42df893571457b527a24 : SUCCESS in 41m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/776490c0407048b8938eb6cb4df0a874 : FAILURE in 5m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/39b1d98e62554d44891c71943b18d10c : TIMED_OUT in 1h 01m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/63181cf3772a4962b2575f8fa6634b4a : TIMED_OUT in 48m 51s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d947a1c581264ed0acebe2211c4da069 : SUCCESS in 47m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6dc1d01b11484ad587ef7518bb0b7350 : TIMED_OUT in 48m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fecacab5d5834eb19f18dbc4e6a96b68 : SUCCESS in 30m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5b158c57b294b969b6cfe6b80ea5655 : SUCCESS in 7m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bfff0a5c9c4a4bf6a076004783961937 : FAILURE in 12m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/672df4dc5ad141a18a0a2a09c5672752 : SUCCESS in 1h 15m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e0cb6746a483e019f4533867ca848a191cb3ea18","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 14:56:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ed81644f456de5680d0575e792cf9cfe41d7c3b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:28:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80f3cfe3dd25496c99a67f5c885ad598 : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/ebcb3315319544c0bf31464de7a6fa86 : SUCCESS in 1h 00m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/40599a2005e14b92b78f684be36aefe5 : SUCCESS in 52m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/739c15bb124c4b0dbef01536a9739ea5 : SUCCESS in 14m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f22875047c0f4c39892bc11886f233be : SUCCESS in 13m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5ebdfe893a7b452aaa1979516046ea46 : SUCCESS in 13m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/debc713345564e3c9413d5d34a88e3f0 : SUCCESS in 14m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/40b3fc3561c746c9a59eca5fb0e4f253 : SUCCESS in 15m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c4fd0e8bfaec4865b32a22a3e8f9bba4 : SUCCESS in 13m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/64250e765bf1443782797e46e4a97528 : SUCCESS in 10m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/83743840b4234c4eb32ef1e3bceff4d6 : SUCCESS in 12m 41s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/2be14f925fa344f98e30ada0f6861cf6 : SUCCESS in 11m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b65d84600a294d57b01a6a3afe64d523 : SUCCESS in 11m 15s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/1ac5a6cee3a14fa68bddfd385e7cb7c5 : SUCCESS in 16m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/659cab75d959423a9c67a384db169e78 : SUCCESS in 11m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3c98c02edc2b4f20a9653c81e0b6ad1a : SUCCESS in 13m 21s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/03cea5a73ebb410892be1bf61da34dae : SUCCESS in 10m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3f2a8b4bd5aa4a5cb06fbc4e5deb2100 : SUCCESS in 11m 41s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/0d6a3df44bb44e00b4ef0f3f9ca335f9 : SUCCESS in 18m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/32634670b6914d9ab0510b4da64f2945 : SUCCESS in 20m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4f50c905c03b40498fba0d7d7b77b7e7 : SUCCESS in 43m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de4f8a2450af435a80dbec5f223116c0 : SUCCESS in 37m 08s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/278bd765f5f6484a9efe1e1812f9ebec : SUCCESS in 33m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/08a2360aaea246219da71e226c869e61 : SUCCESS in 31m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/548e1ab1a5e442ff835d0ba2238fdf4d : SUCCESS in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/614047dbb3f84f36966673379e12a4ab : SUCCESS in 7m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/44fefcf994324a12ab99437f356700ec : FAILURE in 22m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/792aba6245f84dd180fe5d529e3abc57 : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":1},{"id":"4b5370e44e448cd572c2393ba01c776e12042079","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-04 18:34:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"08b2a3db8179b1f34781dc02226f25ccf2e452ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 18:35:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f62503b27b8f3aed2fb96ccebe04244e72f2169e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 21:48:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293abf875ca84f3690729c8b2aca6802 : SUCCESS in 11m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/e135f29851cf40f882f2fbd621b35c69 : SUCCESS in 58m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/224e0e1dea824869bca6dc35e5402535 : SUCCESS in 41m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/242ef535826f44d4a75ef91aca734249 : SUCCESS in 11m 53s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f5e46acac930488f9ac8909586d1cf0e : SUCCESS in 15m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/24ed5c1a858940619485cef939df239e : SUCCESS in 18m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c2a9ad106a0b48a89228a32602e51434 : SUCCESS in 14m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d232424a954e4d3c808c230084247dd0 : SUCCESS in 14m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a5be84d2e434490f883cfc209635426f : SUCCESS in 16m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1d1558e39aab4655bbd7c4ebdc8a0700 : SUCCESS in 13m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7540fb4f99dd4edfbe7c0d9925eeac75 : SUCCESS in 16m 03s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/62a774417b2d4b129446b2603db76522 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/40a7e598029f49cd9a6ebc3f627b8370 : SUCCESS in 13m 29s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/6a66c0130163470bb5144270d8f0688a : FAILURE in 23m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cd97544790c64f0b96fb9d1aa5bd5d58 : SUCCESS in 13m 47s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d89b28f437004b1185844e1bc93ce67e : SUCCESS in 13m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/43e0562374f84e8f9d235d443811709f : SUCCESS in 19m 33s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/0d29b574b35643f2a391915d512f474a : SUCCESS in 26m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4a4e547f1d40444588d6b7a0c950e3c3 : SUCCESS in 1h 00m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b3b8c43ababe4d668d59d6f59d9fb25d : SUCCESS in 28m 41s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9f379cfc5fc64420a4dd6109818e5a20 : SUCCESS in 31m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5dcff935764d49a588730d6bbfc87476 : TIMED_OUT in 50m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3025b56b0efb4c00bd2c7ffad56fadeb : SUCCESS in 16m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9085a73d8e614e84b63005f89bc778a6 : SUCCESS in 9m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/218f91060fe94cff83cf2c03471399b1 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":1},{"id":"50ce126220d69bdf4c0d9b73ae088cea28c634a2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-04 22:35:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a63f6395d31dafcfcdf16d970dc7a58625e4ce96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 23:44:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8297e6aa014f4a8388f5c17fa446513c : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/580cdb2db0584229872210e0b0729e37 : SUCCESS in 1h 06m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9cb864fc1681424ca0ed28ce64c0a41d : SUCCESS in 48m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/91152c95d98d4c17a28f660e48316a84 : SUCCESS in 9m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/25eb70fefb2941b6934ab98300638210 : SUCCESS in 10m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5b3d0a43147e4bcaa14cdf1f3c5612b6 : SUCCESS in 13m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/95dc270d665c4aefa574f6aea5d9f1fb : SUCCESS in 13m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/567fcba517fe4fca940303d8730a77a0 : SUCCESS in 15m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c2bdc55b9e7a4565ac7082b802ae2c1e : SUCCESS in 15m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/44b29586a46f4c209d5abee7ff5da20f : SUCCESS in 11m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5541211a9aaa4e3098612086eaa8f327 : SUCCESS in 15m 48s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c7802de75a7c478a94091d99bf9d64d2 : SUCCESS in 12m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/10c189562d4f4577b2fb85c741e1a655 : SUCCESS in 15m 23s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/72247b99b99d45e7ba74029ff97b608f : SUCCESS in 19m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d16c1f87c0744d92ad24e55a4325cf14 : SUCCESS in 12m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/55c513c77f794b0084d15b3aad52d8d2 : SUCCESS in 16m 34s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/00706771615a499185fcefac5e78a62c : SUCCESS in 12m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/de03beaaa6bc411e8019ac9736f11aa0 : SUCCESS in 11m 23s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/36dcfad207f449b3b14e793d2526fb9f : SUCCESS in 18m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3c13b6ae9b3043fb86cc724b1c971742 : SUCCESS in 21m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e3c4f426f1ae4f2f85c1400f86d5ffa1 : FAILURE in 51m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/941efa8ee0db434fbc0ca4cf533d95d0 : SUCCESS in 30m 49s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/239aaabe555142f4bd6c25e75b8a78b8 : SUCCESS in 36m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d0d706b794294d5298ad0ea2a6ada3be : SUCCESS in 31m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d71fd6c326bd443da98c9b739f6c8443 : SUCCESS in 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a92e589df1c4323957f0de42a00bdd4 : SUCCESS in 11m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/609cb79a7322481f9f7c12400e1d1fa4 : FAILURE in 24m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ba2c95af51e446ab7c494d972b3f80e : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8ee746cd4d6268e1738a448aed40aea3570cf4c3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-05 04:10:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bbb4f7333d2aed1cab1540fbee85f32b18532227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 05:19:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3208e921d3de42f7aa0e570f36076e7a : SUCCESS in 7m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/459e233b9764411182867eecb4d31dd3 : SUCCESS in 57m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/efed82e2a5294d12b27c0d7ad46ab3ae : SUCCESS in 41m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3fb6af5817424e6a8658df4e5dfb7834 : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6a190fd41a914b7596f78ee99b029e82 : SUCCESS in 9m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bc2893b856ca4a87b229a019cb2d6782 : SUCCESS in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0f07824e9ecb433ca1e494910806cbde : SUCCESS in 18m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f5477c93ee6e45128483ac01ef637196 : SUCCESS in 16m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c00a4b6c267243d39864e50415aeeb51 : SUCCESS in 16m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e6323536be394d9aa63b11db924560ea : SUCCESS in 12m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/68dc6271dfca463a898d4bd7c12cafbc : SUCCESS in 15m 53s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/59494fe56eac411794ab8e0bfc058104 : SUCCESS in 14m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2470aff7ffd648f698d734a0e0071461 : SUCCESS in 13m 35s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/62e76f50f4e244ec9be6e71fa82c5362 : SUCCESS in 18m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a2d8dd2efd66414f800152651512a7d7 : SUCCESS in 12m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/596e0378057a414a99539ea692a5a0ca : SUCCESS in 17m 08s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a56a9372f7544321903d267de7c341a6 : SUCCESS in 13m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ea6cd9b9195a4003b797780aa49b71a2 : SUCCESS in 10m 54s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/64b2f63afe044a0198d3b78c3c918eac : SUCCESS in 15m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a95271efb53b4587939208a1f99ce155 : SUCCESS in 18m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/328242177add42b6b845d1e2e8d9a69a : SUCCESS in 44m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d854fe66253446bf82b7185d54ffb299 : SUCCESS in 37m 13s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d8e981bf4a3e40f4b97ef43a5082a6f7 : SUCCESS in 29m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8220f7be6650459eade5ec5f5c691869 : SUCCESS in 30m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0614d52adeca4297bbaaa0510fdd402b : SUCCESS in 13m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe17dd254e0e46e2bb4dacf4638c9c27 : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2ece1cd07350466c8ec7031dff921118 : FAILURE in 20m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82c5148a7c0b4a5688d65cbf67b7b40a : SUCCESS in 1h 02m 17s","accounts_in_message":[],"_revision_number":1},{"id":"35ee4c88c4a26508d9da3a710e93e5b063374258","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 05:19:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d51c2d12b31557411f466af4fd38aaae1107d41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 06:57:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/789d654a9f7d41efa56f482fb9b3c05a : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/edc85a2d9ada4df7ad010b6fbdec36c7 : SUCCESS in 58m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/337d8a19a11542b3a4c2df62f331177c : SUCCESS in 43m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d860023519d944dc96047afa9fa38c40 : SUCCESS in 10m 42s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/63ca47d78b604082bb4f68149bfd8a7d : SUCCESS in 10m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c78c5ed11050411ca0c7f71b9930d463 : SUCCESS in 14m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7cca14766d6b40989f371625592e466e : SUCCESS in 16m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7e1b1458e5c7471e89a3b690d66fadcd : SUCCESS in 16m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8cad0ea488334eb29858d6675c89975d : SUCCESS in 16m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b908c07ed09f4eb7997cc72e21e91b28 : SUCCESS in 12m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7bfba88b93c54c929dc567d3a38f0942 : SUCCESS in 12m 06s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/839df6a948bb4ec3a69a073ea808f50a : SUCCESS in 9m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fa40ec6b073d472cb07c21606db29ff4 : SUCCESS in 12m 33s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/db296b5bb4c349838817aec3ec9c486c : SUCCESS in 18m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e51f560195f5489689e4375c20249375 : SUCCESS in 10m 19s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1315ff3544244d11b32669380ac8ec01 : SUCCESS in 14m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4636008ff09f42ec82ec9f42b1508396 : SUCCESS in 11m 11s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d7d69729e23e4dadab3f5fae041a51f4 : SUCCESS in 17m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3101f4791ff54cdc8a0df09f9301e4fa : SUCCESS in 55m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/910886ba63cf4c5a8f9aa1e2ce7721f3 : SUCCESS in 34m 20s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/202239ebddb947b6acd6ce4c5be39a7f : SUCCESS in 31m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eafb94bbf0a44a49ad60b076c9610abd : SUCCESS in 34m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/401638d27bc6478d80d6312567bfbd2d : SUCCESS in 18m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e979f6a7b8048fb9ee564a10f847868 : SUCCESS in 6m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03ec26e4ce2c4188a6a48c01a603c8b1 : SUCCESS in 1h 10m 17s","accounts_in_message":[],"_revision_number":1},{"id":"aa35bc6aac0d855ba25b0d17fbd78e221d86dc97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 06:57:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d6935181be14dc56dd8e257ef5a98a4b3074c71f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-05 06:58:32.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 16f6af9c97afb12cdf1071bf30b83e33582e2aef","accounts_in_message":[],"_revision_number":1},{"id":"3914392a1d0a26db6ad5a537c9dd059731911047","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 07:03:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50998ce4cb294acc89286d4bbf635cfa : SUCCESS in 3m 29s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0bca381d7a674ac7bfb74dd40b14eb9f : SUCCESS in 3m 03s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/cab1341b37e340bfbf3e3cc6f4657525 : SUCCESS in 2m 16s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69529500700bd3dba575cf0a766737a7184a2121","revisions":{"69529500700bd3dba575cf0a766737a7184a2121":{"kind":"REWORK","_number":1,"created":"2020-05-04 06:41:54.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/90/725190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/725190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/725190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/725190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/725190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/725190/1"}}},"commit":{"parents":[{"commit":"207e08c64bc933afbfd5f5c4ddca9a75a1c7d07f","subject":"Merge \"sharding: Add probe test that exercises swift-manage-shard-ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/207e08c64bc933afbfd5f5c4ddca9a75a1c7d07f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-04 06:41:48.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-04 06:41:48.000000000","tz":120},"subject":"Switch to new grenade job name","message":"Switch to new grenade job name\n\nThe integrated gate templates has switched to the new grenade name\n(grenade-py3 -\u003e grenade). This repo uses an integrated gate template\nbut also listed the job for irrelevant files an extra entry.\n\nRename the job following the template change to avoid duplicate\ngrenade runs.\n\nDetails:\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014602.html\n\nChange-Id: I4afc73301965b582fd338012f0ddc60f9b85fbb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69529500700bd3dba575cf0a766737a7184a2121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69529500700bd3dba575cf0a766737a7184a2121"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
