)]}'
{"id":"openstack%2Fswift~725429","triplet_id":"openstack%2Fswift~master~Ic06d5aa2da9afd44718f1de9dc164b95039ae280","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ic06d5aa2da9afd44718f1de9dc164b95039ae280","subject":"Add a tool to peek inside async_pendings","status":"ABANDONED","created":"2020-05-04 23:16:30.000000000","updated":"2026-04-17 22:03:59.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bc0abee50bc9418eafe8c4c4a717752071efc3c4","_number":725429,"virtual_id_number":725429,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-05 00:29:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-05 00:29:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"524ab92cb7c22f0db78c3f9c6a1d44c53d929e36","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-04 23:16:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe9db29fc79476a045c6381df3ef95026277c826","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 00:29:14.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ecf810b55b5f4e9685ebbed6a341a3e5 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ea55b23945d48afab357129d7516beb : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/b814c66b48fa4c6f80301bfd807b9cc3 : SUCCESS in 55m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bac885f8949646f9875c2721bd9db6ca : SUCCESS in 42m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e2e71303063b44ea9dc6da5bb52cda52 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2ca014d0048744dfaf4e8a21aa3710be : SUCCESS in 11m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/467792a192b64b76a87c6c15a1791211 : FAILURE in 18m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a557bbdb52b54d61a31ddaf31111d309 : FAILURE in 16m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b657668adbff4b899d503c93b25cc61f : FAILURE in 21m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0913560fd52b41979a5bcf6a9e1d24d1 : FAILURE in 18m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/18394821c6304d96bceb2ac6818a5b9b : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d9de6ce1715a43e9b9f529d7d7fd41f9 : SUCCESS in 12m 42s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ca5facffc33f4774aa9913be8102f7af : SUCCESS in 13m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0214bbba7fc749b88bffad3d73701a2e : SUCCESS in 15m 38s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c8047df3b653497eb9e23e629037c655 : SUCCESS in 17m 25s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/69d60f1d532d4a84b2736c6c68cc79df : SUCCESS in 10m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ec19aa35bcc7483ead32b196d5baf754 : SUCCESS in 14m 45s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b130b76c0f044318b0b485b002b11030 : SUCCESS in 12m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d663e6d40a2340ab8c2e2c53764e652c : SUCCESS in 12m 11s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/e3244b7e1ae744e4879760d1036d7a6b : SUCCESS in 16m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3ed2f1eef6414502b766fc8f7b5c0a9c : SUCCESS in 17m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6a14809fe33c4d12ab54129e3736840a : FAILURE in 46m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/70117e93446e469e90eed73862f0ae5a : SUCCESS in 30m 57s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/5319dcca38dd4e5ba63acf09b5d73b78 : SUCCESS in 28m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cdeb258e696f4bf490a6be572d5a37f5 : SUCCESS in 29m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3a538a7a3134853b5a7fcd347ce0bc6 : FAILURE in 15m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca1719d19f1640a9ac6025d8a0d63018 : FAILURE in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a9fb80e3b9e949ef9ddae3e2e91f1d80 : FAILURE in 27m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afed5873d35942eab67b8f0a834471f1 : SUCCESS in 57m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/dc9284f544c14815be98948e563ba063 : SUCCESS in 1h 07m 18s","accounts_in_message":[],"_revision_number":1},{"id":"bc0abee50bc9418eafe8c4c4a717752071efc3c4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-04-17 22:03:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b136797f30239ae39e9540d7766ad842dffaf44e","revisions":{"b136797f30239ae39e9540d7766ad842dffaf44e":{"kind":"REWORK","_number":1,"created":"2020-05-04 23:16:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/29/725429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/725429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/725429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/725429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/725429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/725429/1"}}},"commit":{"parents":[{"commit":"eca39d05661778ce40feafb18eeb790196c0468e","subject":"WIP: Batched updates for object-updater","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eca39d05661778ce40feafb18eeb790196c0468e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-04 20:17:27.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-04 23:16:13.000000000","tz":-420},"subject":"Add a tool to peek inside async_pendings","message":"Add a tool to peek inside async_pendings\n\nChange-Id: Ic06d5aa2da9afd44718f1de9dc164b95039ae280\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b136797f30239ae39e9540d7766ad842dffaf44e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b136797f30239ae39e9540d7766ad842dffaf44e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
