)]}'
{"id":"openstack%2Fswift~728298","triplet_id":"openstack%2Fswift~master~Ic25f32d3c3fdf0e4f338ab8fd2f785e16fffe109","project":"openstack/swift","branch":"master","topic":"s3-logging-and-metrics","hashtags":[],"change_id":"Ic25f32d3c3fdf0e4f338ab8fd2f785e16fffe109","subject":"proxy-logging: Use swift.backend_path if available","status":"ABANDONED","created":"2020-05-14 21:27:02.000000000","updated":"2020-07-01 15:56:26.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"44f5170e0f61500be0bec331739e24a5913e7842","_number":728298,"virtual_id_number":728298,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28499,"name":"Romain de Joux","email":"romain.de-joux@ovhcloud.com","username":"rdejoux"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-30 22:23:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28499,"name":"Romain de Joux","email":"romain.de-joux@ovhcloud.com","username":"rdejoux"},{"value":2,"date":"2020-06-30 21:23:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28499,"name":"Romain de Joux","email":"romain.de-joux@ovhcloud.com","username":"rdejoux"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28499,"name":"Romain de Joux","email":"romain.de-joux@ovhcloud.com","username":"rdejoux"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 12:38:36.000000000","updated_by":{"_account_id":28499,"name":"Romain de Joux","email":"romain.de-joux@ovhcloud.com","username":"rdejoux"},"reviewer":{"_account_id":28499,"name":"Romain de Joux","email":"romain.de-joux@ovhcloud.com","username":"rdejoux"},"state":"REVIEWER"},{"updated":"2020-06-30 21:23:55.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-06-30 22:23:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"248bc313917bbfe287fe5f2f7c114c5b640fb571","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-14 21:27:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81d073487016560343243ad3dab331b19e34628a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 22:32:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45b844f58826480094341f56bbc37dbe : FAILURE in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/e5a8eeac98934d6494d998a6266e28a0 : SUCCESS in 1h 00m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/297f0acbd3034abdab691d8d6508d328 : SUCCESS in 45m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f960cc8c8a4c436db6f2797e2c070867 : SUCCESS in 11m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7b828b340c334f7ea4c85bd6896715c0 : SUCCESS in 11m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a4da8184e81643f38bcd7f0104723825 : SUCCESS in 15m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/61160ab310824c3ba208430fa4d08867 : SUCCESS in 13m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/728f9fb6f9104f49af62118185a5e1e8 : SUCCESS in 21m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/03ff6332f8bf40f098ad63648a0fb04b : SUCCESS in 14m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cc478891b437484e9eee970651a9824a : SUCCESS in 10m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/075631f1ae84424aaf035e95540dad6e : SUCCESS in 12m 17s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/377e13b090fd4c54aca9133ae5df8e6c : SUCCESS in 10m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d8aeb0289c084458add616d9a940e93d : TIMED_OUT in 41m 18s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/115af8044c66496daf1be972e44e319d : SUCCESS in 28m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5f7ad0f1a1e84e9bbbd6ef8997166c52 : TIMED_OUT in 41m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/66aa93914ed5491888bfb1b51766e850 : TIMED_OUT in 41m 12s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/809022c9aedf4f12b10cbc43846afddc : SUCCESS in 23m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9651e68c9c8547a1b1444064873ec06f : TIMED_OUT in 41m 18s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d45fd79a6c55491295786b8e860db5d2 : TIMED_OUT in 41m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7dbece82253047ff9ac3ecdcd3beb7e1 : TIMED_OUT in 41m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c418ca590ebf4aea826bf8eac29955f2 : TIMED_OUT in 1h 01m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee42def22dcf44b49de0af89ed23bb2f : SUCCESS in 39m 21s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8924c6c6585c4df180f86a6466a9f9aa : SUCCESS in 34m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c167bc87a55d462c999348e186da390c : SUCCESS in 34m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abe51e332e88421bb626c59e90562937 : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bed46fb55a343768e65e4ffb95f4c6d : SUCCESS in 6m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e9c8b8077c23433cbf0943a7856510f3 : FAILURE in 21m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75b41c2685d14b78bba7248c084aed5f : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":1},{"id":"78833ce3e36242387e1a06a5bb325c4515c33acb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-15 16:42:52.000000000","message":"Patch Set 1: Workflow-1\n\nrecheck\n\nCould use tests, though -- should be fairly easy, just didn\u0027t have time yesterday.","accounts_in_message":[],"_revision_number":1},{"id":"5f7ad10eb48797fc8889dc53a6ec622ce3287894","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-15 17:39:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b154d9356aff37cc64622591287e53b7e5eb5986","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:34:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e62fe24fe92641da9f239233687a0c10 : SUCCESS in 10m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/6428365803f94e3b85502022dd8db0c3 : SUCCESS in 1h 08m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/199f1da6efa34bd08a5fefe9ef42d955 : SUCCESS in 46m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6e9ae5da98d94984b3f84cebb28ee42e : SUCCESS in 11m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1108b04140e0480eb1668316194490fa : FAILURE in 12m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5f671bfecd824bfc90877dfcd0e89c61 : SUCCESS in 13m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/43d7576aa4074325ac3457bfc8593f89 : SUCCESS in 15m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ee4f983318b8413f9741863fbf0deaa7 : SUCCESS in 14m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e922f96672e64219a7f0c865885649de : SUCCESS in 13m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e443df65ef5f41c799b421f6b9431e02 : SUCCESS in 13m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eed355c8b10a40329a2e5f142f9925de : SUCCESS in 16m 15s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/004a97d836b540759b2ca71115241ee0 : SUCCESS in 10m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e809e58cb0b542898920047ffeb0f57b : SUCCESS in 17m 03s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ac52d897f9f344adb5feec1c3ed532d2 : SUCCESS in 22m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/da43a1999613403bb5af4627c53835b9 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b8589588ae7e4a1dace342999ef7078f : SUCCESS in 16m 52s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/52b8d7defa234bbcb18a68776cb03f73 : SUCCESS in 26m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/62f757443189405d915366a874b079ea : SUCCESS in 22m 19s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/167aaae5d90741d0ac9ab6b07774b5de : TIMED_OUT in 41m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d707fcb5eea9497090644592719d278c : SUCCESS in 34m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/60ea5c4600b748d8bdff2e143281790a : TIMED_OUT in 1h 02m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cdc8ecdd8ee54cfe86fde06e5d40fc07 : SUCCESS in 41m 57s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/97b8d86ac83241e8b5360e5567e1497c : SUCCESS in 43m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/43ede6f52d8e440cb5cf094cc4606c2f : SUCCESS in 37m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23e3c0f695084a7c99d4e4c7aec07ad6 : SUCCESS in 13m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/659cff9b286d42588af8ae6dc6b98180 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d969701e8144cd1a353c48d94980188 : FAILURE in 22m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f55028794854f2eb37012beb6c1f204 : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":2},{"id":"8b9a7eeafabb1bc9709274184586e1807c6a9d3e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-16 00:50:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ad4a4156c20cd13bc953b2fe46d53d372c2ad913","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 02:07:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/301a28cc29334f868d50bfac3620540e : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/ab3f30ef82ff43ad9b1545ec1a2b50f8 : SUCCESS in 1h 00m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aebfca0563564971a42009197a3e5a55 : SUCCESS in 53m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2fd1683347f14f04a2968d0d6986c338 : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c6339c377a3042ddae1852dddb88d970 : SUCCESS in 9m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/156f524d462146efb779554b7d212d37 : SUCCESS in 14m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4e2f0a95c1b84909b4aa612a527bc1ca : SUCCESS in 13m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/513af4f9155b42ffb2e684376a3e00aa : SUCCESS in 27m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/36836f9ebc44436995688c7d62fc0c07 : SUCCESS in 14m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/446096a8606349c88cbb49a38a70b492 : SUCCESS in 10m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b924696306384c6d95a195ab6fdd6d31 : SUCCESS in 12m 31s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/3fa9e182c7804b3dab7f05f804b820a3 : SUCCESS in 25m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/672d8ec35e544d97a088f86f3f6cae3c : SUCCESS in 15m 01s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/688c180448fd469eb196fba5b30762a5 : SUCCESS in 16m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e5ef80bc078b478b8f623c7e6846bfb5 : SUCCESS in 10m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/564637a471cd49d499dc489030a05662 : SUCCESS in 29m 00s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/67813a7ef7654ff7998a03e252cb1252 : SUCCESS in 11m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6f373edc406341a98e40f42d16120a54 : SUCCESS in 26m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1e70b8a17647460e92947ee50b355325 : SUCCESS in 16m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5ed8700db5a54ed581afcb65a1e9213d : SUCCESS in 20m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aff6080e93a94b858ccf542e35bf2136 : SUCCESS in 48m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/775effbfc72b440bbd7aaf46ce16bdf7 : SUCCESS in 45m 12s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/54a30c99dba34ddaaafe73d285b51557 : SUCCESS in 35m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6890ff6c2c3b47a8af4607e7174c19ee : SUCCESS in 35m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d8fd82597ca486d927432af6690a778 : SUCCESS in 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c874ed70fb76448c932769b421941cd2 : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf90126a3d9d4c25813dc5fcba4d64ca : FAILURE in 22m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eaaf797feee0477ea9b4911afce4e54f : SUCCESS in 1h 11m 22s","accounts_in_message":[],"_revision_number":2},{"id":"dd69cf5b923e35aa58bdd47a4ac8f04b434a2921","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-18 16:13:35.000000000","message":"Patch Set 2: Code-Review+1\n\nthis *seems* reasonable to me, but slightly changes the semantic of a, c, o keys (which WE don\u0027t currently use our template)\n\nWe should see what OVH thinks (since they DO use these keys AND s3api?)\n\nThey may either have a solution that\u0027s better, or have some expectation that this breaks... or maybe it seems fine?\n\nThe obvious addative/backwards-compatible change is just log a new key as s3api/backend path.  But having the splitting is a pretty nice improvement, so hopefully this works for everyone?","accounts_in_message":[],"_revision_number":2},{"id":"b5813c8ec69ac21d7a0063f567f60f7e790e7873","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-18 16:15:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a63aeabb0fb109260d8e201f9d9f481a352fb89","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-18 17:38:09.000000000","message":"Patch Set 2:\n\nFWIW, I\u0027m pretty sure OVH\u0027s solution for my problem involves ceilometermiddleware, which takes a similar-ish approach: https://review.opendev.org/#/c/718085/\n\nStill, it\u0027d probably be good to hear Romain\u0027s thoughts.","accounts_in_message":[],"_revision_number":2},{"id":"961b681c24ba834567600acc705b39f92703ccb1","author":{"_account_id":28499,"name":"Romain de Joux","email":"romain.de-joux@ovhcloud.com","username":"rdejoux"},"date":"2020-05-20 12:38:36.000000000","message":"Patch Set 2: Code-Review+1\n\nThis is a nice change !\n\nWhen I prepared my first patch I only had billing in mind (and so ceilometermiddleware).\n\nActually we use `force_swift_request_proxy_log` s3api config to have log with account/container/obj field but with S3 query on multipart object that produce a lot of log entries.","accounts_in_message":[],"_revision_number":2},{"id":"a8e92a9a442a5b1194101d8da4c32eb739705ff7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-08 18:52:02.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b1ff5140f9906aa3f9bbf7c30ea59801e1bf85b6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-08 19:21:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"02a6d9217f293a64878348ed3659fd7f48bd2897","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 20:20:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e272debb7674a51a17952b0f7e2a889 : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/8fedd8e192144285b9e80198c803e5e2 : SUCCESS in 56m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f6421ebc226841ca99fc1edb9f853240 : SUCCESS in 48m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c9027f9a188b4d8689d5107a5a6ca448 : SUCCESS in 13m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cb129b56b4d542e5a694d85c854bc4c2 : SUCCESS in 13m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8d89c103e95e48208e018ffebfd0c0c7 : SUCCESS in 15m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2875c9ef9390448fb2c3499e58dab5a6 : SUCCESS in 17m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/192e2e07eaaf4821aaf311fedab52a3c : SUCCESS in 18m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f4cedba5e9c842bab9ec8cf7b2ceff1f : SUCCESS in 13m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0601e7de7a95443a941f577aab459fb5 : SUCCESS in 25m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1b3c02fe55374bd79706ffbf51f0e096 : SUCCESS in 19m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4dc24e4477e4561b6c5ffc930313e33 : SUCCESS in 24m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/07d2dc3d805c4336b062959d39befbfe : SUCCESS in 18m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5cf963408eac445d8b91e3db2f78a6b1 : SUCCESS in 24m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/77687cb91b644a30ae2f888e93d3bfac : SUCCESS in 22m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c9fa243f22314ed998da19ec0c2b5210 : SUCCESS in 26m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dcd080607d6847a5a282bce700c7a251 : SUCCESS in 52m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2cc7261ad0a241c588551291ce5b1aea : SUCCESS in 49m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1c586ad0c2c44e2eac2b7be2f4a85115 : SUCCESS in 49m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cd4cbf08cb748878dc737f8172358a4 : SUCCESS in 21m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6573c28a87fb4d51a098cdaa8ecdbd44 : SUCCESS in 7m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5ebdcd40a56a45d1a365944314b3d485 : FAILURE in 11m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ea39ec16015466ebb5055d4bc7489f3 : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":3},{"id":"75b6bd85f56e9113b41f3a5d64ec509d2e60a151","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-10 21:06:04.000000000","message":"Patch Set 3: Code-Review-1\n\nthis works ok, but with s3_acl turned on I\u0027m not getting anything parsed out in the object field:\n\nJun 10 21:05:28 saio proxy-server: 127.0.0.1 127.0.0.1 10/Jun/2020/21/05/28 PUT /s3test/test HTTP/1.0 200 - aws-cli/1.18.70%20Python/2.7.17%20Linux/4.15.0-101-generic%20botocore/1.16.20 - 9 - f73bcdd750b2707f9238ecb417b140a2 txe797bec0f6694671bead0-005ee14b18 - 0.0718 - - 1591823128.434513092 1591823128.506356001 - AUTH_test s3test -\n\n\nI think something is wrong with the backend_path getting stuck on the acl request","accounts_in_message":[],"_revision_number":3},{"id":"365fe06f2fca1c8949e4a0539d026b317b28d00a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-23 17:14:50.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nFWIW, I think https://review.opendev.org/#/c/735221/ should fix the s3_acl issue -- maybe I should squash the two? I\u0027d like to get Romain\u0027s input on the s3api change, though.\n\nI also noticed a (potential?) issue with transfer stats, but I think https://review.opendev.org/#/c/735220/ may give us a more-or-less reasonable way out.","accounts_in_message":[],"_revision_number":3},{"id":"1dfd3c8503d5a4e8fb1fbed00a20b7f1c77a205b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-30 19:09:43.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"9480e82b6db767180472d5ddee2ed2f777094ba3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-30 19:13:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e42d2055e1a763476be559c53a4a19f89ffd63c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 20:17:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/661781e706174c75bc6bf4d4e5a9d43e : SUCCESS in 15m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/1b1b5423f8524b8c9c13ce3d7a1e9d1c : SUCCESS in 1h 03m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cc60cd058c7540dca67e1affea256302 : SUCCESS in 43m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ef5fe3f9b0224a32bc0c2bbcd67d5e8a : SUCCESS in 9m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/db1938146b4e424c81798e573ee8b655 : SUCCESS in 11m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/965262532fab412caeea7302ab6731b7 : SUCCESS in 13m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0eacbab919c44e42b205f169c9b0257d : SUCCESS in 16m 59s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6d522a5605974daeab5c280fac885e3a : SUCCESS in 18m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/926a4e57eef44297be0c843f2ee25b48 : SUCCESS in 15m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/383c4271c4834200bbb01b2417346590 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/be10cf40412f47e5b75ae0c3a7aecf91 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e3e282633e7c466fb0d07272b4b8ba97 : SUCCESS in 17m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5b4b961a4fd04aa9b76b8a4b3ffb7fe0 : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b405c7e75bb74e0899b012e3b5e815a4 : SUCCESS in 23m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4444b3268f8241b29a5cf20b28bec741 : SUCCESS in 19m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9f808c46ed8042d299edf4b2cafce69e : SUCCESS in 21m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2940a5b876014b61866ec9c9ad5bc150 : SUCCESS in 46m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7ae81a4149c64cd9a28349102daa385e : SUCCESS in 45m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0398cd760f3f45c581768cfc5aa4abfe : SUCCESS in 48m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b7d8a272dde444ca77095e9a5c98c63 : SUCCESS in 13m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dac982af11a84389a3a0fb12930bcad1 : SUCCESS in 8m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/93d5eb1639034c959b385f6f1c24dd54 : FAILURE in 10m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebadabcda868445c824af6afc43ac49e : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":4},{"id":"81b84b43bfa864eaa77ad9db7763101ae571a2df","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-30 21:20:42.000000000","message":"Patch Set 4: Workflow+1\n\nThese new s3 stats are *amazing* !!!\n\nI think we should do an UpgradeImpact or something - it\u0027s gunna be a whole new world","accounts_in_message":[],"_revision_number":4},{"id":"3ddb5f14217918a4a17cee70a44b198b8968b9bb","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-30 21:23:43.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"61654a58b4484ad3ae7a8ae6b5425fcf25a20c80","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-30 21:23:55.000000000","message":"Patch Set 5: Code-Review+2\n\nlet\u0027s do it!","accounts_in_message":[],"_revision_number":5},{"id":"1029960d7e807e43cb78f2b5968abe5cf8399b9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 22:23:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e18bfd6a729945fb9cd9e3444a9d4a17 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/62527d5a27b9401b9e382095a56f168c : SUCCESS in 42m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a44002b18d8d40d696fe90df24198280 : SUCCESS in 45m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/304ed4988d894237aca224fc84364881 : SUCCESS in 8m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3febeed1075b4deeb035c1c8967e5100 : SUCCESS in 8m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59b97bed437848d28052c9d4e086fb0b : SUCCESS in 12m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a3a65f81718443cae1ec425b19717a8 : SUCCESS in 13m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8ae65c6fbb444f7b9f76c736efa5e0c2 : SUCCESS in 13m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/68a60d040d704eedbdf34ab59c5d75f4 : SUCCESS in 16m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b03b00299807401ebe89c0483b1a3902 : SUCCESS in 19m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2ccf7a902f7f4aa68624ae915ca50691 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/038f778f086c48aca1aa5f1d747fcf0d : SUCCESS in 19m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/39e95061b33a44cdbe5ce2160d92b24c : SUCCESS in 21m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e2da7cb900c648d2a2b5f97a5047617e : SUCCESS in 18m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b2bd119916a9472687feb728e5909d87 : SUCCESS in 18m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49228bc488464431ad3cd6bf84a9c5d9 : SUCCESS in 26m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/334a0ec8933c416aa3cfe04181766c66 : SUCCESS in 38m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/709399ac443e4ee1bac1ea1b8b8b89bb : SUCCESS in 45m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ec1dcba0b0424e47b60a1318613e17ef : SUCCESS in 44m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff685baae66940e3800b8ad386c54b6d : SUCCESS in 21m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1cc95e2b5174d84b2d0f62ba64883df : SUCCESS in 7m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/766cc471418a4ead97639b12eeeed978 : FAILURE in 9m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d48f4e15c9e0486c83d9a386ffb123d5 : SUCCESS in 54m 08s","accounts_in_message":[],"_revision_number":5},{"id":"3949eed263092340ae6e9650b9da403c1328c5b2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-01 14:44:30.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nAFAICT this change is really entirely about logging s3api requests","accounts_in_message":[],"_revision_number":5},{"id":"110e4be320088f02476a2aa6da81894285a93563","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-01 15:56:26.000000000","message":"Abandoned\n\nsquashed to https://review.opendev.org/#/c/735221/","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e9b3adbedf5c40dfb6a218bb76f13978ce194e23","revisions":{"5ec7631b7a9e8d7ba85e4cfda98ed5336fb03622":{"kind":"REWORK","_number":1,"created":"2020-05-14 21:27:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/728298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/728298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/728298/1"}}},"commit":{"parents":[{"commit":"08db36a295377fc437c56f6fc470be035fd74c4f","subject":"Merge \"py27: Suppress UnicodeWarnings in ShardRange setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08db36a295377fc437c56f6fc470be035fd74c4f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-14 21:25:06.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-14 21:26:45.000000000","tz":-420},"subject":"proxy-logging: Use swift.backend_path if available","message":"proxy-logging: Use swift.backend_path if available\n\n... to determine {account}, {container}, and {object} template values.\n\nChange-Id: Ic25f32d3c3fdf0e4f338ab8fd2f785e16fffe109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ec7631b7a9e8d7ba85e4cfda98ed5336fb03622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ec7631b7a9e8d7ba85e4cfda98ed5336fb03622"}]},"branch":"refs/heads/master"},"a8c0845739c894ef5c72f941272ad8ffee984638":{"kind":"REWORK","_number":2,"created":"2020-05-15 17:39:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/728298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/728298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/728298/2"}}},"commit":{"parents":[{"commit":"08db36a295377fc437c56f6fc470be035fd74c4f","subject":"Merge \"py27: Suppress UnicodeWarnings in ShardRange setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08db36a295377fc437c56f6fc470be035fd74c4f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-14 21:25:06.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-15 17:38:22.000000000","tz":-420},"subject":"proxy-logging: Use swift.backend_path if available","message":"proxy-logging: Use swift.backend_path if available\n\n... to determine {account}, {container}, and {object} template values.\n\nChange-Id: Ic25f32d3c3fdf0e4f338ab8fd2f785e16fffe109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8c0845739c894ef5c72f941272ad8ffee984638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8c0845739c894ef5c72f941272ad8ffee984638"}]},"branch":"refs/heads/master"},"dbf73990207c655d782b6db224e5b711ba4b6403":{"kind":"REWORK","_number":3,"created":"2020-06-08 19:21:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/728298/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/728298/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/728298/3"}}},"commit":{"parents":[{"commit":"08db36a295377fc437c56f6fc470be035fd74c4f","subject":"Merge \"py27: Suppress UnicodeWarnings in ShardRange setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08db36a295377fc437c56f6fc470be035fd74c4f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-14 21:25:06.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-08 19:21:28.000000000","tz":-420},"subject":"proxy-logging: Use swift.backend_path if available","message":"proxy-logging: Use swift.backend_path if available\n\n... to determine {account}, {container}, and {object} template values,\nas well as statsd metric names.\n\nChange-Id: Ic25f32d3c3fdf0e4f338ab8fd2f785e16fffe109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbf73990207c655d782b6db224e5b711ba4b6403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbf73990207c655d782b6db224e5b711ba4b6403"}]},"branch":"refs/heads/master"},"18b6d893e47a921148f4f374741c7799b59b3d17":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-06-30 19:09:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/728298/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/728298/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/728298/4"}}},"commit":{"parents":[{"commit":"16e85f6d7d11252fc2f3af97ef7c65bdeecb0656","subject":"s3api: Set swift.backend_path to the last-used path instead of the first-used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16e85f6d7d11252fc2f3af97ef7c65bdeecb0656"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-14 21:25:06.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-30 19:08:55.000000000","tz":-420},"subject":"proxy-logging: Use swift.backend_path if available","message":"proxy-logging: Use swift.backend_path if available\n\n... to determine {account}, {container}, and {object} template values,\nas well as statsd metric names.\n\nChange-Id: Ic25f32d3c3fdf0e4f338ab8fd2f785e16fffe109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18b6d893e47a921148f4f374741c7799b59b3d17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18b6d893e47a921148f4f374741c7799b59b3d17"}]},"branch":"refs/heads/master"},"e9b3adbedf5c40dfb6a218bb76f13978ce194e23":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-06-30 21:23:43.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/98/728298/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/728298/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728298/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/728298/5"}}},"commit":{"parents":[{"commit":"16e85f6d7d11252fc2f3af97ef7c65bdeecb0656","subject":"s3api: Set swift.backend_path to the last-used path instead of the first-used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16e85f6d7d11252fc2f3af97ef7c65bdeecb0656"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-14 21:25:06.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-30 21:23:39.000000000","tz":0},"subject":"proxy-logging: Use swift.backend_path if available","message":"proxy-logging: Use swift.backend_path if available\n\n... to determine {account}, {container}, and {object} template values,\nas well as statsd metric names.\n\nUpgradeImpact:\n--------------\nBe aware that this will cause an increase in the proxy-logging statsd\nmetrics emited for s3api responses.  However, this will more accurately\nreflect the state of the system.\n\nChange-Id: Ic25f32d3c3fdf0e4f338ab8fd2f785e16fffe109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9b3adbedf5c40dfb6a218bb76f13978ce194e23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9b3adbedf5c40dfb6a218bb76f13978ce194e23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
