)]}'
{"id":"openstack%2Fswift~728491","triplet_id":"openstack%2Fswift~feature%2Flosf~Iad53f5a521df514917a2b04b466399e1b29808b4","project":"openstack/swift","branch":"feature/losf","topic":"merge-master","hashtags":[],"change_id":"Iad53f5a521df514917a2b04b466399e1b29808b4","subject":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master","status":"MERGED","created":"2020-05-15 16:09:51.000000000","updated":"2020-05-18 23:24:48.000000000","submitted":"2020-05-18 23:22:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728491-1589844124794-df490097","meta_rev_id":"760b3dd36cf4c13b195b647bc903b1651d4d8e18","_number":728491,"virtual_id_number":728491,"owner":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 23:22:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-18 17:40:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-18 17:40:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-18 17:40:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-05-18 23:22:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0cabdc256ba246abcba112a06c2d0b5f8f7e7bce","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2020-05-15 16:09:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80c5063a4949be54b9f5b26f5d6621571a22d418","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 21:38:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8f77f418e08473bb06ae9a6b5fcf88d : SUCCESS in 14m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2524e7a31bd46519659eb3421baa988 : SUCCESS in 8m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e6db25fbfc42444798137e22fe2b071d : SUCCESS in 10m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6c58be409952407e835515b0ed21d5ff : SUCCESS in 10m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7b2f2bfd49c84b3eb5352f26c045d583 : SUCCESS in 15m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/977ebf3339aa48e9a2084f6d1aee1cad : SUCCESS in 13m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/75f36bbd9a47448ea2c3834f4f3a75b1 : SUCCESS in 12m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fa34627f42ff459ab58c301dcdd6574b : SUCCESS in 18m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eed237e8449447d78bf13e386ca3217d : SUCCESS in 12m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/30e79351695848c98ae7aeb3cd5bff06 : SUCCESS in 14m 19s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a78b85c3d8ea474ba96b19457db4ef7e : SUCCESS in 12m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a14d33f69592411083f9717796073f69 : SUCCESS in 10m 56s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/74ccbcdda8594a19b61b09994a83fb6d : SUCCESS in 19m 04s\n- swift-tox-func-losf-py27 https://zuul.opendev.org/t/openstack/build/768106b9cd71487487d450f27bea3b98 : RETRY_LIMIT in 7m 44s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b8d477ae3f344bc9841cd6c13ebcef26 : SUCCESS in 23m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bba0650a07c14ca1bf953e126dfad849 : SUCCESS in 33m 19s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d741ebf070e642e68bde79da202bc0c2 : SUCCESS in 22m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d1dd8bde70ed43f4b448da2c8ea306fe : SUCCESS in 12m 45s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1b069bc61b8948089043a94ad05623df : SUCCESS in 35m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ba44c61cc9634997be9b449d62860b57 : SUCCESS in 20m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9205e1dcffb648018816a83ae3209f42 : SUCCESS in 47m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b668826c1214a8d8c5129298b569b5c : SUCCESS in 41m 16s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9353b6d7b297424e80bc7d506cd83932 : SUCCESS in 41m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ef54b4e6fd0b4e82be1ccb545b6b42b8 : SUCCESS in 41m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e10aa5ab90c4f14ae4f167aaa683943 : SUCCESS in 21m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72fa7910bb4046c783d10dcda8f7c647 : SUCCESS in 9m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05be2576145649f6bb062d2fa8a1a24c : FAILURE in 23m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b1fe0afe7aa6327f48fadf09351212f08664f607","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-18 05:35:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d8b8d970c39a18002789f7cc056b52c3b7904916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 06:35:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccc737dcdda549f5ab526d428e5a7f27 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85e180bc20ff4ff593d91d14c90e0883 : SUCCESS in 4m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e00468a27cc6436997ac4758ecf5af57 : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7cad42ca0fd0464581bdbb2cb4c54a1d : SUCCESS in 9m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4840dfe3bd3346cdb85707bed81b17c6 : SUCCESS in 14m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4e0fb1db9925468e8dda81ff8338ab4d : SUCCESS in 15m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0f795ed6266147f0a8748261efb1b3d7 : SUCCESS in 16m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ae53317e0eb34ed7a8333c625db1a2d9 : SUCCESS in 15m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5dbb96c809d645b080187f97d59096f7 : SUCCESS in 11m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9cf180abc216473fb2a7b70c1d27f789 : SUCCESS in 15m 20s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b3a4575a7ef946b4a7f5fcca6d013273 : SUCCESS in 11m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/84f1c8dae0584754ac09cec35d009a9d : SUCCESS in 13m 19s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/05c219d2f35b494e83e1d7d5d8c2a46e : SUCCESS in 17m 10s\n- swift-tox-func-losf-py27 https://zuul.opendev.org/t/openstack/build/9d196767027b412a8b0ebb6e8e97c8c4 : RETRY_LIMIT in 7m 21s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/70ace6ef2c7a4c67af656fcbe144804b : SUCCESS in 11m 49s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7ca4b657f0bf430f94d78c21ff73a94d : SUCCESS in 15m 54s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/133002b7edc440e19f70aecd781d64ad : SUCCESS in 17m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e76f3782ac23462ca35ca5908a730e82 : SUCCESS in 13m 30s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/519fcf0f61884b14bfe73a2f7d2fb8cb : SUCCESS in 17m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5d74b8de643b46e381006640c3687297 : SUCCESS in 23m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/372ac24dee8040439b2b0fcfe5a09d26 : SUCCESS in 57m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a93ed6180794e778156388fd04e7743 : SUCCESS in 47m 07s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d527789be24a4a87882637d25ed6326e : SUCCESS in 35m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dae32757b06e4823a94dd091b17339f6 : SUCCESS in 37m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad3e24e1328a4449ac648cb00bc17be1 : SUCCESS in 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e5c24058eaa4f45bd9508322ae1bc7f : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/251d8cf6d7f94253a0ab5d2e6a4f650e : FAILURE in 7m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34703dd811ad7f373470a715e3d8dfed075dd622","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-18 17:40:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"928b05315372b12a9161bfdd3e04645b45928230","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 17:40:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ebb76e139e6c93b80d719b204ee44ac672afeddd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 19:46:09.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85375d7fe47948369db9abde91a6bc05 : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b70a0f505e884c048e633371c2baadc2 : SUCCESS in 6m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b7bf7ae69d4f4900876f80645091535a : SUCCESS in 15m 28s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/dba1b6da71d24345b5db27e6f583aa8f : SUCCESS in 12m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1005d76028764ea48623192818faef51 : SUCCESS in 15m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/272ff364289b49639ff95af1ea1db65a : SUCCESS in 18m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/54bfa4542e664a8fa02a611c6bc8312d : SUCCESS in 22m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9a139df809684ff997b8b0f84d466bd4 : SUCCESS in 15m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3856f09253304d2bb8ae5384c3f35054 : SUCCESS in 12m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bf13c24eaa3f4edaaa1c81df5181dbf9 : SUCCESS in 13m 55s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/81b7a053abcc49a4a67b302f24fa0da5 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c0c687ded89d4279b7d9f2c15aa0e158 : SUCCESS in 11m 37s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f23a34f08e2e413ba380507cfdbca059 : SUCCESS in 25m 34s\n- swift-tox-func-losf-py27 https://zuul.opendev.org/t/openstack/build/59c384024e1b4887bf8c741313a280dc : RETRY_LIMIT in 5m 44s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b807cd3d89ea4b019abbf19c49c30483 : SUCCESS in 16m 18s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1e927675f3ef4bd992d034b5cb85a07b : SUCCESS in 13m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c13730bdf2994cbcbca95b80271f01a5 : SUCCESS in 14m 47s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/3a9789c07b994dbb9b974ef6e94b80b7 : SUCCESS in 16m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/06b19af38d3b4e05a7dd75f8fd9db045 : SUCCESS in 52m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a45b14b4b0d45ce82440ebf293935eb : SUCCESS in 50m 14s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9942f38a8bd04ac189754dec49aeb559 : TIMED_OUT in 50m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7185080c630044afa45f2425533fd5a6 : TIMED_OUT in 49m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58ff214dc43f46aaa9195015d5637985 : SUCCESS in 16m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe396e88a09144b89a4438e8d35bd018 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":2},{"id":"db23b8bea76593e098a69e9f7ac3bb6d2896bef5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-18 20:06:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cf9b49a2d7c605e0a00ba655538985e5ccd0f5c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 21:12:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/684640aa838145bdb3726e564a6fa499 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e727c82abad4cd3a6ffaf91dee684f9 : SUCCESS in 5m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6d1cef960f554926b1da8d42d76a4a63 : SUCCESS in 9m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/95805928a3844064a6e6c010398aa7d4 : SUCCESS in 10m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/70f12014afa642ab9a1b7d79b9758b37 : SUCCESS in 14m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7c8820eb2b5740f49ce922c0a90d2e9c : SUCCESS in 14m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/090acf7934e14392a30af35d3f12802e : SUCCESS in 13m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6d8bddcebd6d454cb2756a94cdbb5554 : SUCCESS in 13m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/08325a6a35c640238fae83b49617db83 : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/074873087de74a08aee03a16e80cb355 : SUCCESS in 24m 33s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/5f783b6ad2d74d0dbf005c534acf7931 : SUCCESS in 11m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/62fa2bfdd74e414f87fcc19b9ba262ce : SUCCESS in 18m 47s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a3fdd7dde12f4aef810aa3059fec267d : SUCCESS in 15m 48s\n- swift-tox-func-losf-py27 https://zuul.opendev.org/t/openstack/build/1f393dad1f8c4bab83c6f027fdcd9fc3 : RETRY_LIMIT in 7m 43s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cb25b3227c894d7c819eaab2e0d29943 : SUCCESS in 19m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5f789e922c3d449cace56136ceb80fa3 : SUCCESS in 24m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1fbadaed287e401e80769b59536f0bef : SUCCESS in 12m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8f2e4f2186ed4378b2afb221ca8d3b05 : SUCCESS in 13m 17s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2e53d81d323a4caaa2191fc77c146ec3 : SUCCESS in 19m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/76c9f920d03041b089f113fbb65a96c2 : SUCCESS in 19m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a6bb5c0deba649c2a1ea1e4c710e01bc : TIMED_OUT in 1h 01m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d6097b25c5b0478db449cf8e7badf010 : SUCCESS in 35m 21s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/450a4663e5f94e9e9a4e438b38a645ae : SUCCESS in 35m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/baabf046a68c477ba8fd466a4faace81 : SUCCESS in 47m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4108f5b95e094a4690eccd98117791c2 : SUCCESS in 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3f2ef6b4a08450886351a83b046927c : SUCCESS in 8m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/41d45baefc0145f4b986930ccba05bb3 : FAILURE in 8m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24f4254486f2e221cf6057b75d8484d3f0059d37","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-18 21:41:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9f942e4861b425dc3831966af6a10c02e7578964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 22:31:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/940752b920fb45baaa170c3dae6d8916 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b230f2fad5ba4313a6505081e207ce14 : SUCCESS in 5m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/521b870ffd7a45bbbc6a53dbe4c5956f : SUCCESS in 9m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e72846ef95d4406db819d353f5aa3396 : SUCCESS in 9m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/454cebfb08b5442f8a7e05ea5a559e97 : SUCCESS in 13m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/57c78ddaa2304b5ba449947ae01f6025 : SUCCESS in 13m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2ff6f8835934490aa4be24521303f0bd : SUCCESS in 13m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8ad4dca546df41689c52d99f878eca89 : SUCCESS in 12m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4a4b76d9bf2542dbb583f3c009ce64aa : SUCCESS in 10m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/526b3285a7d7449fa4fddfdab807e432 : SUCCESS in 13m 08s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/6aa7fb592de34e009a298e775b3fc48f : SUCCESS in 10m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/be61d9fc64eb4452b07b520adf48edb9 : SUCCESS in 11m 04s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/799ae4f89c0d4482b6cdfd33f133e7e2 : SUCCESS in 16m 52s\n- swift-tox-func-losf-py27 https://zuul.opendev.org/t/openstack/build/5d7a82f3032d45bcae0e99fc1bce34b4 : RETRY_LIMIT in 5m 41s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d7c5992e27c54d79a8f3f4819f230d6a : SUCCESS in 10m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/620e1609c44f471e9471287e05e0bb38 : SUCCESS in 13m 50s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/45779cf50e004858b60036b5be44061b : SUCCESS in 11m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e36d6fcf4b1d4d62ae692cc2271f38cc : SUCCESS in 12m 01s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/e45c45bef13646ac806fc4244eaebd8a : SUCCESS in 19m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b28d9a5770224c038726359563e6652f : SUCCESS in 22m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f375d82797e1413cbed08d87c07c67ea : SUCCESS in 46m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/124badf821ae435ca4e1de49c2e278f0 : SUCCESS in 38m 42s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/2019608a11e44b68a81333d74514dda8 : SUCCESS in 35m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a7571a6cbf4b44058d80c929316f7f96 : SUCCESS in 38m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13f3b0c5660f4e34a11315dae5b58bc9 : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d99608934f8b4c44abb9f0a5998f30c9 : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/844fc9352b43405cba3bade15227ccc8 : FAILURE in 9m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c4aba15661847cff85bde0282367222c36a33031","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 22:31:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8b122e318e9b704ed6dd93f74da554f3f2e83062","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 23:22:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0383d82b30742538d7ac20d5991945d : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab00caf30c6e485099bc42026de48a91 : SUCCESS in 4m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/719aea836f5e49559396d6a3640376b6 : SUCCESS in 9m 55s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/83ffc364880f4cb989de5f85a4a82a47 : SUCCESS in 9m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/70c0351a287f4003af8ba42a680d9898 : SUCCESS in 14m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/134948a99e4145a884c0810329d213e4 : SUCCESS in 19m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a001e8e350dc4f1d9e4852c015c5f2e9 : SUCCESS in 11m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a9d86f46d3d847f8acf23d76a8a96775 : SUCCESS in 13m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e6531deb2ba140c686cfa221fbf56063 : SUCCESS in 10m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5cc9b357f5fb4798b3156f6824d5c8de : SUCCESS in 14m 34s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b5df6998125a49dea5255f6f199ec227 : SUCCESS in 9m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b2344549cce24a8e96bd005c7ff5f224 : SUCCESS in 15m 07s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/3b579ca3907f469aac1448aa8e1fa88f : SUCCESS in 14m 25s\n- swift-tox-func-losf-py27 https://zuul.opendev.org/t/openstack/build/611c9db93ea84dc7ab1ffedeea931bdc : RETRY_LIMIT in 6m 06s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3f01b18ee2d74281bd4113885e4f5d21 : SUCCESS in 12m 18s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/adaa8c73f8d84baaa26ecd7e88e854e1 : SUCCESS in 10m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c11ebdc917b04b62b14c170bb34c7240 : SUCCESS in 22m 05s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c0f1ff232d73477a97f7ade9f6cf51d0 : SUCCESS in 15m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9498667c7bec45cfafc2154b8e2e5246 : SUCCESS in 45m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3bd0d4f7f2e43bc9c1e9544d90a959f : SUCCESS in 35m 53s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7b06429a06174ad8987ee77b4a95232d : SUCCESS in 42m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4647727bf1ed4c90a368c250768d64b9 : SUCCESS in 31m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8162451737424e58ab665b17b9c1a813 : SUCCESS in 12m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/774f96439ba34fbd8d3c877455bb337b : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":2},{"id":"32c4c577d7048d5d66742139409fe0c29053294f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 23:22:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"62f0638c2d95ef1cb8193eb4be3a072ee29bff33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 23:24:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a23734c8fa9147dfb0a141b799a0a67c : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/965e6c4514794a0681f129dc3c7dcaed : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/be226f3c8d3e4e869a1e09430318897b : SUCCESS in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/40fe42d1ee814ae095e96c3fa5fae98d : FAILURE in 48s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"414ca6bb6c7ffe12a1f7c4f65526bb0bf98b4b49","revisions":{"c6a00c9882897e131e311b8b9f90f8d12193fd95":{"kind":"REWORK","_number":1,"created":"2020-05-15 16:09:51.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/91/728491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/728491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/728491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/728491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/728491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/728491/1"}}},"commit":{"parents":[{"commit":"c8dac46c4614f1afffdfee495455c2dd9d758027","subject":"Merge \"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master\" into feature/losf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8dac46c4614f1afffdfee495455c2dd9d758027"}]},{"commit":"08db36a295377fc437c56f6fc470be035fd74c4f","subject":"Merge \"py27: Suppress UnicodeWarnings in ShardRange setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08db36a295377fc437c56f6fc470be035fd74c4f"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-05-15 16:09:19.000000000","tz":-240},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-05-15 16:09:19.000000000","tz":-240},"subject":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master","message":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master\n\nChange-Id: Iad53f5a521df514917a2b04b466399e1b29808b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6a00c9882897e131e311b8b9f90f8d12193fd95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6a00c9882897e131e311b8b9f90f8d12193fd95"}]},"branch":"refs/heads/feature/losf"},"414ca6bb6c7ffe12a1f7c4f65526bb0bf98b4b49":{"kind":"REWORK","_number":2,"created":"2020-05-18 05:35:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/728491/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/728491/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/728491/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/728491/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/728491/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/728491/2"}}},"commit":{"parents":[{"commit":"c8dac46c4614f1afffdfee495455c2dd9d758027","subject":"Merge \"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master\" into feature/losf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8dac46c4614f1afffdfee495455c2dd9d758027"}]},{"commit":"08db36a295377fc437c56f6fc470be035fd74c4f","subject":"Merge \"py27: Suppress UnicodeWarnings in ShardRange setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08db36a295377fc437c56f6fc470be035fd74c4f"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2020-05-15 16:09:19.000000000","tz":-240},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-18 05:34:50.000000000","tz":-420},"subject":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master","message":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master\n\nChange-Id: Iad53f5a521df514917a2b04b466399e1b29808b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/414ca6bb6c7ffe12a1f7c4f65526bb0bf98b4b49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/414ca6bb6c7ffe12a1f7c4f65526bb0bf98b4b49"}]},"branch":"refs/heads/feature/losf"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
