)]}'
{"id":"openstack%2Fswift~728495","triplet_id":"openstack%2Fswift~master~Idf8db3b299ecc5289bbee53549292a376ddd8c9d","project":"openstack/swift","branch":"master","topic":"batched-reclaim2","hashtags":[],"change_id":"Idf8db3b299ecc5289bbee53549292a376ddd8c9d","subject":"this works too","status":"ABANDONED","created":"2020-05-15 16:28:28.000000000","updated":"2020-05-22 14:08:50.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0253f4de93f5e484e33e026c8c4c284f954a8cdc","_number":728495,"virtual_id_number":728495,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-18 23:11:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-2,"date":"2020-05-18 22:05:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2020-05-18 22:05:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-18 20:06:25.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-05-18 23:11:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96edee1a940f0b4c8d11b37f023cb258fa0ec17c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-15 16:28:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2812f5b9d1062495fdb48ea2e41f5a0ed0c8f82e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-15 16:29:22.000000000","message":"Patch Set 1: Workflow-1\n\nI\u0027ll test this one next and make my recommendation - both seem more or less maintainable","accounts_in_message":[],"_revision_number":1},{"id":"6e8315f76d82c753e8eda3db50c36678b55496fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 22:23:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9696ef27ea57438f8561630dbc02fd1e : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/2ed82fb0cc4f493784afa7f736121848 : SUCCESS in 1h 05m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0a3f372e8ad94b3e89dfd1ec45f6a19b : SUCCESS in 54m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7ee90c9709304d5086017d7e95a67657 : SUCCESS in 10m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c21df0bdf6a142aeb94ce3dcecbba5a1 : SUCCESS in 11m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6c927f12296a41dbbb7876c3ee2c3d45 : SUCCESS in 17m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03ee707d19cb42bdb44988cea3829297 : SUCCESS in 19m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/20592358ac1e498baf11aaeea4364edf : SUCCESS in 16m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c8123bd8097442b483f749ce5810eaf5 : SUCCESS in 17m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/623db23408974aeaa6d5eb55ee99789d : SUCCESS in 25m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/22055e72a90046fb818c2da39ba37ebe : SUCCESS in 15m 35s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/471634f2ae0e4b01a5e02ac68f571d0f : SUCCESS in 13m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf1b07ad100249d4921fca32ec4fabf5 : SUCCESS in 11m 13s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ae27c8b23a3d4b278f6788eadfd23122 : SUCCESS in 15m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5c0d6d57eb23451c8807688ba565a21e : SUCCESS in 12m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/973f14b569fd415abe0b1395a1ed48db : SUCCESS in 14m 21s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ee7a2ecb69e04443965149cdb8226790 : SUCCESS in 10m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/28514155dbf240c7bdc98ca189a2c9a7 : SUCCESS in 11m 48s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/317403b95c624034a9609b7a0d7971c5 : SUCCESS in 17m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ae9f2b0c48734fbb8b47af2e02f10d2b : SUCCESS in 26m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aba341e89c4e4368b233c1df18dc2363 : SUCCESS in 49m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15f77e36bb094206bdfeff53a0a4d156 : SUCCESS in 42m 27s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/11074b43fc54414ba977fd7ba72b37b9 : SUCCESS in 35m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/76164855f91c402a97bacfa506a1e48f : SUCCESS in 39m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31530a5f59c84268bbf094b2a0082310 : SUCCESS in 16m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a923a71d030f4c41914581bddd69c6ff : SUCCESS in 7m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2a647f7604ef40619cc7e3a2170a7988 : FAILURE in 22m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b957c2e586d1488da7f7e6914789b2d8 : SUCCESS in 1h 02m 17s","accounts_in_message":[],"_revision_number":1},{"id":"971db87e9b1510baea2278eef57f16dfea51124f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-18 14:58:03.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027m testing this one, but I don\u0027t really expect it to do any better than https://review.opendev.org/#/c/727876/\n\nI\u0027m going to try to I guess reword it so I can go into _reclaim for ever batch and bounce the connection each time until it\u0027s finished.","accounts_in_message":[],"_revision_number":1},{"id":"b509f98fe19b3c78a8c1c98805fe2b8c071bdd11","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-18 18:21:00.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nSo this was surprising:\n\n[root@stgss0235 ~]# ls -alhF /srv/node/d48/tmp/test-*.db \n-rw------- 1 root root 258M May 18 15:55 /srv/node/d48/tmp/test-orig.db\n-rw------- 1 root root 1.7G May 18 16:56 /srv/node/d48/tmp/test-patched.db\n\ni.e. the patched version allowed writes to get into the container WAY better than master (more than 6x)\n\nPart of the reason was that this reclaim method was REALLY slow:\n\n(\u0027delete\u0027, 1589820918, 3586)\n(\u0027delete\u0027, 1589820919, 3840)\n(\u0027delete\u0027, 1589820920, 3744)\n(\u0027delete\u0027, 1589820921, 3620)\n(\u0027delete\u0027, 1589820922, 4111)\n(\u0027delete\u0027, 1589820923, 3522)\n(\u0027delete\u0027, 1589820924, 3227)\n(\u0027delete\u0027, 1589820925, 3620)\n(\u0027delete\u0027, 1589820926, 3678)\n(\u0027delete\u0027, 1589820927, 3562)\n(\u0027delete\u0027, 1589820928, 3634)\n(\u0027delete\u0027, 1589820929, 3612)\n(\u0027delete\u0027, 1589820930, 4144)\n(\u0027reclaim\u0027, 1589820879.5930278, 52.1286985874176)\n(\u0027delete\u0027, 1589820931, 4128)\n(\u0027delete\u0027, 1589820932, 3639)\n(\u0027delete\u0027, 1589820933, 3509)\n(\u0027delete\u0027, 1589820934, 4130)\n(\u0027delete\u0027, 1589820935, 4136)\n(\u0027delete\u0027, 1589820936, 4349)\n(\u0027delete\u0027, 1589820937, 3923)\n(\u0027delete\u0027, 1589820938, 3744)\n(\u0027delete\u0027, 1589820939, 3566)\n(\u0027delete\u0027, 1589820940, 3864)\n(\u0027delete\u0027, 1589820941, 4068)\n(\u0027delete\u0027, 1589820942, 4365)\n(\u0027delete\u0027, 1589820943, 4088)\n(\u0027delete\u0027, 1589820944, 3487)\n(\u0027delete\u0027, 1589820945, 4024)\n(\u0027delete\u0027, 1589820946, 3786)\n(\u0027delete\u0027, 1589820947, 3620)\n(\u0027delete\u0027, 1589820948, 3585)\n(\u0027delete\u0027, 1589820949, 3536)\n(\u0027delete\u0027, 1589820950, 3739)\n(\u0027delete\u0027, 1589820951, 3959)\n(\u0027delete\u0027, 1589820952, 4186)\n(\u0027delete\u0027, 1589820953, 3874)\n(\u0027delete\u0027, 1589820954, 4080)\n(\u0027delete\u0027, 1589820955, 3600)\n(\u0027delete\u0027, 1589820956, 3815)\n(\u0027delete\u0027, 1589820957, 3636)\n(\u0027delete\u0027, 1589820958, 3954)\n(\u0027delete\u0027, 1589820959, 4101)\n(\u0027delete\u0027, 1589820960, 3670)\n(\u0027delete\u0027, 1589820961, 3660)\n(\u0027delete\u0027, 1589820962, 3741)\n\nShutting down...\n\n... but OTOH at the end there it WAS operating on 2M deleted rows - so I don\u0027t have a direct comparison to how long that would have taken on master (but probably more than 10s or we would have filed https://bugs.launchpad.net/swift/+bug/1877651","accounts_in_message":[],"_revision_number":1},{"id":"a41e254f3825de65329dd0530cb373709c1169bf","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-18 18:24:18.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nMAYBE it\u0027s worth the overhead of looking at each \"deleted but unreclaimed row\" by hand and if we\u0027re going to do this every cycle since we wouldn\u0027t have to aquire a write lock unless we want to delete.\n\nMaybe we should do the lock free SELECT and then grab the write lock to remove reclaimed rows by ROWID ...","accounts_in_message":[],"_revision_number":1},{"id":"9dc55ad97a13d6ccb7ee2782cb395be49eb6343a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-18 20:06:25.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ecb2c20299b45870f07ae153259bbbbfb965226a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-18 22:04:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ce0944a7c8626ccbcc2f82e3b2281351be03bd35","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-18 22:05:23.000000000","message":"Patch Set 2: Code-Review-2 Workflow-1\n\nif this is better than the parent we\u0027ll just squash it down","accounts_in_message":[],"_revision_number":2},{"id":"8b3912f41a6e8ffa0bfe2cb3618bade6516c439f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 23:11:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e315817f54ce4775bc9dd00dbeb5f0dd : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/2ae67c41a0d3491885549d976e133dfa : SUCCESS in 1h 05m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/87cdb21512d149c587c74207c43cb847 : SUCCESS in 51m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2467c9a2c9694517bc35dbbed9d03e73 : SUCCESS in 11m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5c89bf4e4f124aeb9fd9032ac5b24630 : SUCCESS in 13m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c559a52dbb594e4e8d2c8920e9ccfc46 : FAILURE in 11m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/72a770a6e88041f0bcb6b32cd8edd0d4 : FAILURE in 14m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9a0e1b3caf144da3addf85deb2df9d3b : FAILURE in 29m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d913d9dc227943028f94ce6aa5514ef8 : FAILURE in 40m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7af20e0acf3549558bd4ddc2481d1447 : SUCCESS in 19m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/13596eb7d4af4ee596b33eb20f28f6f4 : TIMED_OUT in 41m 20s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/84fed70783d345fbb490f87185af08ba : SUCCESS in 13m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/42a59af413f34f5faa103781ac018189 : SUCCESS in 38m 06s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/5a7f19894e5f43a8b1c8f0cc636fbd85 : SUCCESS in 18m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dfd6d9a9c21e44a7a77b65baebb4142a : FAILURE in 38m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/34e85b66aa7a4438aadbba41ae0528f4 : SUCCESS in 22m 07s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7f918aa691b5455fa0d3a5c2e42af202 : SUCCESS in 28m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6c05f65667e24d56a00c6d931a579004 : SUCCESS in 19m 51s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d78dba91dfac4891b1d3f60731b749b8 : FAILURE in 38m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/085b3ce7fe3946609eea487499b9edc7 : SUCCESS in 32m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/53f033a38dcb4f40af9b250ada24ba1e : SUCCESS in 1h 00m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/17d6c1870a68420b86fc0631985a76e2 : SUCCESS in 36m 19s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d9c3d5ccbc104bfcad04669927609cbb : SUCCESS in 35m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e23fbbcb1ca422b90c8c7f993bfe2f9 : SUCCESS in 41m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77fc123487fe4521a67fcafeca2494a3 : FAILURE in 16m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d57c211b58ac4640a190810468c32275 : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ca834fcb8584cda8c3179c89299bd03 : FAILURE in 9m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac110c0bca594e8cb7031660d213c813 : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":2},{"id":"0f2ad429708929fba4489bf91073fa76c776c4b2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-20 18:40:52.000000000","message":"Patch Set 2:\n\nI tried this with a smaller flush size too, but I guess it just sucks (or I\u0027m testing it wrong)\n\n-rw------- 1 root root 256M May 20 17:11 /srv/node/d48/tmp/test-orig.db\n-rw------- 1 root root 255M May 20 18:11 /srv/node/d48/tmp/test-patched.db\n\nit seems like the reclaim is plenty fast but also still locking:\n\n\u003d\u003d\u003e patched.out \u003c\u003d\u003d\n(\u0027reclaim\u0027, 1589998252.5878477, 3.235980987548828)\n(\u0027delete\u0027, 1589998255, 821)\n(\u0027delete\u0027, 1589998256, 183)\n(\u0027reclaim\u0027, 1589998255.8239462, 3.2637202739715576)\n(\u0027delete\u0027, 1589998259, 985)\n(\u0027reclaim\u0027, 1589998259.0877938, 3.348170518875122)\n(\u0027delete\u0027, 1589998262, 998)\n(\u0027reclaim\u0027, 1589998262.4360816, 3.3184611797332764)\n(\u0027delete\u0027, 1589998265, 1030)\n(\u0027reclaim\u0027, 1589998265.754679, 3.444554567337036)\n(\u0027delete\u0027, 1589998269, 994)\n(\u0027reclaim\u0027, 1589998269.1993747, 3.035109281539917)\n(\u0027delete\u0027, 1589998272, 975)\n(\u0027reclaim\u0027, 1589998272.2347033, 3.0975534915924072)\n(\u0027delete\u0027, 1589998275, 982)\n(\u0027reclaim\u0027, 1589998275.3323865, 3.351872205734253)\n(\u0027reclaim\u0027, 1589998278.6843767, 3.3519420623779297)\n(\u0027delete\u0027, 1589998278, 929)\n\nI\u0027m skeptical of these results, but will probably focus on the validating and polishing the other (working) change before I drop more time here","accounts_in_message":[],"_revision_number":2},{"id":"c6c172df8490c0929249aee45a40f63864513a26","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-22 14:08:22.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nSo i re-read https://www.sqlite.org/lockingv3.html and I guess as long as the reader is reading the writer can\u0027t write.","accounts_in_message":[],"_revision_number":2},{"id":"533f98138f420f7102314258253bdad8cdfc452d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-22 14:08:50.000000000","message":"Abandoned\n\nBook-ending the reclaim DELETEs with the name index and bouncing the connection is the way to go for sure.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8b841aa5f7b42f93fa86360cfa6317d40ebd81b2","revisions":{"f86f3df42360bcca8c735d923fb6035481e03e60":{"kind":"REWORK","_number":1,"created":"2020-05-15 16:28:28.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/95/728495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/728495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/728495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/728495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/728495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/728495/1"}}},"commit":{"parents":[{"commit":"55a9028046399b4a92ca2a7c27879343442abfcf","subject":"Breakup reclaim into batches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55a9028046399b4a92ca2a7c27879343442abfcf"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-05-15 16:28:11.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-05-15 16:28:11.000000000","tz":-300},"subject":"this works too","message":"this works too\n\nChange-Id: Idf8db3b299ecc5289bbee53549292a376ddd8c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f86f3df42360bcca8c735d923fb6035481e03e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f86f3df42360bcca8c735d923fb6035481e03e60"}]},"branch":"refs/heads/master"},"8b841aa5f7b42f93fa86360cfa6317d40ebd81b2":{"kind":"REWORK","_number":2,"created":"2020-05-18 22:04:06.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/95/728495/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/728495/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/728495/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/728495/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/728495/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/728495/2"}}},"commit":{"parents":[{"commit":"86eab2f45f1be65c7538fb1ffd5374cf33a9d097","subject":"Breakup reclaim into batches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86eab2f45f1be65c7538fb1ffd5374cf33a9d097"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-05-18 22:03:44.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-05-18 22:03:46.000000000","tz":-300},"subject":"this works too","message":"this works too\n\nChange-Id: Idf8db3b299ecc5289bbee53549292a376ddd8c9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b841aa5f7b42f93fa86360cfa6317d40ebd81b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b841aa5f7b42f93fa86360cfa6317d40ebd81b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
