)]}'
{"id":"openstack%2Fswift~728588","triplet_id":"openstack%2Fswift~master~If274a01a6660776cf01b71efbdd102159c520794","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"If274a01a6660776cf01b71efbdd102159c520794","subject":"versioning: Use CONTAINER_LISTING_LIMIT when doing container listings","status":"MERGED","created":"2020-05-15 23:03:23.000000000","updated":"2020-05-17 08:23:41.000000000","submitted":"2020-05-17 08:21:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728588-1589703699466-f8705a57","meta_rev_id":"debafad2f9b07c5f395ee2c606eaf74c54458779","_number":728588,"virtual_id_number":728588,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-17 08:21:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-17 05:57:11.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":2,"date":"2020-05-16 13:30:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":1,"date":"2020-05-16 13:30:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-16 13:30:03.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2020-05-17 08:21:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a814376d14381fb936bb6f723f01bc596631679c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-15 23:03:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eca48ac23a38f002bef70248f6e52147b7180d0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 00:07:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7dc4e1b0696401c85e278049c788429 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/63fddba6da49424286533634f217ac6a : SUCCESS in 52m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8facdf0066c64d97a3bbd8fe26aaf6d6 : SUCCESS in 41m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8ea304856892472db53d574791c31938 : SUCCESS in 14m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/eea8db6f85c746c0b45c02b1332c5945 : SUCCESS in 14m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5e805bd1630f4425be315f4a0d6278de : SUCCESS in 22m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/665091c7ccca45e7b266feac407f198b : TIMED_OUT in 41m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0676e80790214018b2b91936e2600b26 : SUCCESS in 15m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/af774c604b3047f587f6b22db471e218 : SUCCESS in 14m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1333d5ff48e743b7b28b6654594853e0 : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6df433d2251b4a518dd959502165881e : SUCCESS in 15m 26s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/04f386c150e147a9961d0723e1e07c0d : SUCCESS in 12m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5c641a82f115479e907ef6252ea9e80a : SUCCESS in 12m 45s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0d5b39fbd470416091d05323b1f26be6 : TIMED_OUT in 41m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9589947de62e448d9b470b7c2113a11b : TIMED_OUT in 41m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4c2215a5d35d4e8da9dba34428f52036 : SUCCESS in 15m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/92cb2ea33a69448e85ab0b7989f6f393 : TIMED_OUT in 41m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/48c3114cec0f467ca66ccb86392faab4 : TIMED_OUT in 41m 10s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/63a8b6e95ba440938491860fe8cf8535 : SUCCESS in 20m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fffabc4cd7e5485c933a5b610a7e3103 : TIMED_OUT in 41m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fa0e314e212b46c888fb55b27f9b0e02 : TIMED_OUT in 1h 01m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/36a56ef0eaec45f0abaf004439211cb6 : SUCCESS in 36m 21s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9b905b5f72944732b16ee1b4f99662b3 : TIMED_OUT in 49m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f638301cebd64d01abab577fc03164a4 : SUCCESS in 36m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3883196abd834105b1e5e226c6164bee : SUCCESS in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fc902ca1e6a40cdbf33be9b3d56e1a7 : SUCCESS in 6m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/04382dd2fc294ddc9f2db08111ff98f7 : FAILURE in 10m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0a04b1a0de6402a9f11924f31c12585 : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":1},{"id":"63c20909865d665b13ec9c8b7ee82f8d7118c543","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-16 00:50:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f3f38b2c8cfbf5dd35e66af12e00d412a8483a3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 01:52:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3951a29dec274c5e9cd3ff752127af35 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/c700ff690a5d47edae9b699113c8d5db : SUCCESS in 54m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/143922ce4ca541a0ba7c186f12448eac : SUCCESS in 46m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bbd7df59a8a1497ca93f75cc5778ee77 : SUCCESS in 8m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1dd85b14fba8455786a7e4ff14092e2d : SUCCESS in 11m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d05e222ca07b4121b56bbc46e7a276ad : SUCCESS in 12m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/450dd4e97c3a4578928156206c64eb8e : SUCCESS in 13m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7671de82195c4e54ac3d96596d583264 : SUCCESS in 14m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2e5f2189f64541c2b8b10ec4ad5ec4de : SUCCESS in 12m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b96ed6230faa4467aad7330ea3926b38 : SUCCESS in 9m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/db102653206a42408575f7f0dc7473f4 : SUCCESS in 12m 18s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/34537531ed2a44de863eef6e1a313400 : SUCCESS in 9m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5ff65511119449fb9156e2f1e4d6e337 : SUCCESS in 28m 26s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f81d2ba1a69841b0a20b242a4156e52c : SUCCESS in 14m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7c144933b3284682b13e1e30b0e160fd : SUCCESS in 10m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1d5145d4b3b44a6c98c04d771e5c4135 : SUCCESS in 32m 29s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/bf918547a9da4c5180e1196f88fb54f8 : SUCCESS in 9m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/65b12d8833024940ac180414666be8bb : SUCCESS in 29m 18s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2aab92f4fd5649e3bc7b696ac8e0a0a3 : SUCCESS in 15m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fe0d0abcc5ee4d53bad79bda22ea7cdf : SUCCESS in 40m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/61c7b68769974dedb2966241f8fb72a3 : FAILURE in 40m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0ffd0b02a32a4747b0c24036b3c637a3 : SUCCESS in 36m 32s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/3e9ec60ec74c4ae0a4dceeaafc2dd773 : SUCCESS in 34m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0977951272c74be383be46ae29fa30b5 : SUCCESS in 28m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd0dbfe223ef44458fd5963fc94cd3de : SUCCESS in 14m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd26ea91e1f14c58823ee349056cb99f : SUCCESS in 7m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e3086c861946461a99ec9063a73d5663 : FAILURE in 25m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b21a1039dd0440bbdbdd925f7ab7b45 : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":1},{"id":"51ebfc8e5f3acb5150f44a47f9debf197355fd5c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-16 04:30:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5d4068c590d153066062580f30e48ef534bec996","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 05:36:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf5bd71ed1fb42bba448cfc0a095e164 : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/2c4c3f5fee0b46a2a7bde9a929f74852 : SUCCESS in 1h 05m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d516208024934c609630e71ee79c16c4 : SUCCESS in 43m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/64e8a6869a2c434eab262906c5153ae0 : SUCCESS in 9m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6de9263deca34ad69680ba14f1e977f1 : SUCCESS in 10m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4e5f8c8cd93a4b1d9eb257cfb1d15521 : SUCCESS in 31m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2701e8a2914c4e14bec7fb02b186f17b : SUCCESS in 12m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2c0e02e251f04828a25117039ccc091c : SUCCESS in 12m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6a09c5868ee7483d8f7a6a884d3cb8b1 : SUCCESS in 26m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/da0c9c639a8a4f1ca47160f68edd059a : SUCCESS in 29m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0ece8f2aa5184f4593a61c0faac850f0 : SUCCESS in 12m 59s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c3d388e268584829a076bec6b26ee55a : SUCCESS in 24m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bfe0dc86493440519b7d25a1db5a4b35 : SUCCESS in 10m 45s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/075c728fc7ab48da90aab0a802c937fe : SUCCESS in 15m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/84d472a094744c9ea91585620ada8466 : SUCCESS in 29m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/77b83e241b104e9b8669e3e1aaa99c6e : SUCCESS in 34m 35s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/88e1a961896a4c7ea892bad4d28981a1 : SUCCESS in 10m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d252aa8f527a441fa4c7035c4b7fbf40 : SUCCESS in 16m 34s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/05ed8a1494674d2283af4576a6b22068 : FAILURE in 33m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/45d80241e0fc4fa3935a32d8e14f71f5 : SUCCESS in 26m 04s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7ec0180f4d7949dd9f21813bed4281c0 : SUCCESS in 47m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e14ef5b2734d4866a984eae5245e254d : SUCCESS in 41m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ae124aef67e44a8d982746510ef442ee : SUCCESS in 47m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ab0ab6f4c4464306a4439babf4be9a83 : TIMED_OUT in 49m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/000ad1aefda0468e827782c2819d3bb6 : SUCCESS in 14m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00f2975c749f4aed8c8061fe36f17174 : SUCCESS in 7m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/388a7a501b194fd1948cfa824f112794 : FAILURE in 24m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/911dfc8ce91b4d8ebf9c42ee0a1073f3 : SUCCESS in 1h 01m 22s","accounts_in_message":[],"_revision_number":1},{"id":"6a4482f19b293e822363cbf18039483622e762ed","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-16 05:39:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6625d0932504961e69cf046eae5031670539897a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 06:44:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7021cb20d9c744378d0e2ceb87f92285 : SUCCESS in 7m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/22bfdd6e6de04265b1a8f8bff2ab687e : SUCCESS in 1h 03m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5172b50ccb224cbca8aeb81d5f494aed : SUCCESS in 45m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1fbdd23059c547f897d643f0fc52c450 : FAILURE in 10m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f952cc876a304ce69ce0eeb77bb3eec0 : SUCCESS in 11m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/462877002c0d4b4fb199121c508bdd79 : SUCCESS in 12m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c1035d78afbd4789a5d9827f653a8859 : SUCCESS in 12m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/653f61baaf0943bf9bdd9b53a9bd3a99 : SUCCESS in 14m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9351b220e8b6469290d2b8893b7ec65d : SUCCESS in 12m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/103e9070f1ca400a9d696814c4ba763e : SUCCESS in 9m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cf76b4e4ce49430681e566dc0268b53d : SUCCESS in 14m 34s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d00bf3a9a1cb41048beb97d016061384 : SUCCESS in 9m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ef46a4ef86bb4d40859db55aff40cb39 : SUCCESS in 14m 03s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/fe0c65b2e2ab419487a185952bdc834c : SUCCESS in 15m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9f724451e27247a78b28525262ce5a24 : SUCCESS in 9m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cdf55e875e6e46918f273c129c36691c : SUCCESS in 32m 41s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/20d9f359dc5c477a870c3a2bbd01ad70 : SUCCESS in 18m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f35f303ffbfb430f9364882fe41dcab7 : SUCCESS in 10m 37s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ffa635f48bab4f0f8fbbbdae93573ac6 : SUCCESS in 16m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a924c02188d845e79e148e0c97a1d22e : SUCCESS in 17m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/29a6b7ed3fea481cb56404f964d6ba78 : TIMED_OUT in 1h 02m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15adf51465f14ac298cab86ccea8ab24 : TIMED_OUT in 50m 45s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9adc4e4cced84769a7e6a82c703732fc : SUCCESS in 38m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0835f1700260408199192e3385c0e8f4 : SUCCESS in 39m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3db94626e15043bfbfed60afe680c3c5 : SUCCESS in 29m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a953bc28a544e729752c77933b2b769 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4d6c57e42d945be88d323c5d004c8fd : FAILURE in 32m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5447bd74d4b45a0a689951a8f8f828a : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":1},{"id":"114f40b2d18fd16d99d0ef80e9a3040d1e9afa0b","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2020-05-16 13:30:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eaa8a72ca2103d5b982fe926b07f990cd7bea9a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 14:41:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9472c635fe1f41278c6407457d7493be : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/7e30bbc3770b4947a733d18c1405c1a0 : SUCCESS in 1h 10m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/962620ea204a4b72872a4163078d30b3 : SUCCESS in 45m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ccd95eaa61bb449a8ed6aff1beeae49c : SUCCESS in 10m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2543cc01688e4eec8ac570144a3bec75 : SUCCESS in 10m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fccb57b3eb214d51ba1af0f4c0ccedcc : SUCCESS in 30m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/325984f93e5f4ce0893a55134d346e88 : SUCCESS in 27m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ddb6add474a34112a34c1646f4d3ffcc : SUCCESS in 13m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8b53e8e994f947f2824a34f9ca609cd6 : SUCCESS in 31m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/777b314c83f14315946cc9c8917647ae : SUCCESS in 26m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5dc25eaabbee4430a473b3914c33320a : SUCCESS in 13m 39s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b371957def424461bffb0ac65122d40f : SUCCESS in 29m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a3f19f791f4143cb96d44c3d5f107343 : SUCCESS in 13m 07s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/04d3bb01155b4268b531ba19cb296038 : SUCCESS in 18m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e2d966796287467ebec12a832fd3fe62 : SUCCESS in 26m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/efc699758a0845cf959d8826ef3b93d5 : SUCCESS in 15m 57s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3d993fc1aba14e29a606262cd720961f : SUCCESS in 29m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6a3058433bf5498d82599c4902fdf52d : SUCCESS in 14m 24s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ca06aeb14dd541759c05e4785f8e9f0b : SUCCESS in 20m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a5c609dd57ee40a1bfbb3d66867690e7 : TIMED_OUT in 41m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4915d6c1cb084853a16ea707a5dcf742 : FAILURE in 56m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ee93b8342c34c60909526416da6d610 : SUCCESS in 34m 26s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/77e3522852bc4ae99e7047126869ff01 : SUCCESS in 38m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb134090e0c24476960a43b0038137e2 : SUCCESS in 44m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80f95a1bab1e4b8fb474675f7cec30fb : SUCCESS in 31m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f11edcacb66b4e129b5847a593a6cb52 : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/80c8e67e26904b288dde1b000f4fdf57 : FAILURE in 11m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b98dad0909d4087aed0b21154ec98a0 : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":1},{"id":"2d3cfc272f2ce649d0b1258dfb3873defd4cb95c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-16 15:21:43.000000000","message":"Patch Set 1:\n\nrecheck\n\nI maybe should\u0027ve written up a bug for this. w/e","accounts_in_message":[],"_revision_number":1},{"id":"f7ab569b98b992d35326684f819c5441ef8241d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 16:36:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97669a3f277247b4ad62da5665c13676 : SUCCESS in 7m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/68645f2b1d084a458511260a8882c369 : SUCCESS in 1h 07m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e408b1d20117451aa54740efde22e5c6 : SUCCESS in 41m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47203cf32495413986e0a956aad1d1ab : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9bfc8e1b4b1f4cf28c2edf4ca99b3a00 : SUCCESS in 9m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/64c583228e7b4d4693c0c9aa02a3a041 : SUCCESS in 15m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ca6a0b8e4b184de49ea30c7d37367520 : SUCCESS in 10m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ffa77a4767b84f55b5e764a1b967fec5 : SUCCESS in 15m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7cf0b80e5a704ea19d5d9580362ae2bc : SUCCESS in 16m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8908f16481904b5aa48ebb79c2179c1c : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/600b12301aa84255a2a8e1af7e26ff3f : SUCCESS in 12m 49s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/2bdad3ece4d341a0ada1752367562361 : SUCCESS in 14m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3f879ca5191f42b694c061740d50bd21 : SUCCESS in 10m 18s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7f868f907f234bf5baa3ce654be29936 : SUCCESS in 19m 13s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/95ee1ecd052446089dd930923fe3ff3b : SUCCESS in 10m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8a0bd49cd7ba438d94e1fc99e795e965 : SUCCESS in 16m 44s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4fc8cbe3bcd843248f9aff1276f443af : SUCCESS in 10m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9c6464c3a4bc4f33826ea78ce6868730 : SUCCESS in 10m 38s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/434d744c80274c91a233adb32d57ff0d : SUCCESS in 20m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/196ee824f2b447c39b66c2575a1d808e : SUCCESS in 18m 41s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/97162a519cf043d2a2422bc185689f12 : SUCCESS in 47m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a9ef53a21eb04856a063d5dd6e942259 : SUCCESS in 32m 53s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9957e2bdd78a4d0b8dd54e044d9f99ec : SUCCESS in 31m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2cd8e4b7b8b349738f8898c18a36097a : SUCCESS in 34m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e567fdb6b61c45df9e254c85f78c496d : SUCCESS in 16m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7df40d3dd3d4ffe95daf5b5dab7ac22 : SUCCESS in 5m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/414fc06b284e45f69bdad8babcf02984 : FAILURE in 8m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d57ab30a1f964f04b0ca7a12199b8dc6 : SUCCESS in 1h 08m 46s","accounts_in_message":[],"_revision_number":1},{"id":"37cf14310305a4b8ad739841b16a7a5c97234f06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 16:36:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41dd44c5a15a0c2847ec4be9dc630326dd0c80f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 17:48:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e381692d09a2447891d1d3c8e7e7ff8c : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/ffa9eb7e14fc4216852f89c059dce23f : SUCCESS in 54m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f436b61e2042469f856ac2fd5e8d3366 : SUCCESS in 49m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d175eb7ea0e642c5bdbad8dd8006d40e : SUCCESS in 9m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3d6003fc9b0f4d66b871cbfd957617c1 : SUCCESS in 9m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/401cc9f1acfd4b2d925d9d5da596abee : SUCCESS in 14m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ea586432ea0f4d979d8be07092109359 : SUCCESS in 23m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dac591cf3f5748bd9721284e15ba242c : SUCCESS in 24m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d018bb41c9fc41c1a8b63dacff443f37 : SUCCESS in 32m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/af09877d248c45b69167b2f9f160b2df : SUCCESS in 9m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/67447823ce86480080e01426d78f31d6 : SUCCESS in 23m 34s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b95b14322fec4639af1c75f4cf645335 : SUCCESS in 29m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cb9fb7277efb4a09b0dc37ea2cae4533 : SUCCESS in 21m 01s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7b91c35d92f5424eae122b63cda874b1 : SUCCESS in 14m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0ba955fb2ce24fa5b5728d2cd0830dd4 : SUCCESS in 29m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/56dc9e664bce466fb276d810815f4b0e : SUCCESS in 9m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/77dcc2c281cd4a09b182d7e23f3a09ad : SUCCESS in 21m 23s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/9e360e7d542a4627803d69fab4b7d2f3 : TIMED_OUT in 41m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/985130888709473ba6527fda1920c841 : SUCCESS in 40m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/117a0e779d054631b0376266aa8417e4 : SUCCESS in 44m 45s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/21a13369856749918c85cf1cb2aa8bd9 : TIMED_OUT in 47m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a1d0ec86b5f3408e90442ec8db935119 : SUCCESS in 30m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51e58134d5084a8c81e2044ff05157e1 : SUCCESS in 21m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d6c22f911214d65807bb17f82cc7f19 : SUCCESS in 5m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b63faf893a534554a4dfd1b5d2b92cab : SUCCESS in 1h 10m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ea43baa40f4c15c4fa9f8cb56bdac1773a38e030","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-16 22:05:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1f18fbc25e4425c8fdae31975ba814b62bba37d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 23:06:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d04ce25267b24e149785a3811582e5f8 : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/60f4fbdf741d4db8912b4493c826ee91 : SUCCESS in 59m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e31a7c5ac0e94e259ab543b9d5967b10 : SUCCESS in 45m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/051379092b4340e6a4ff08d73b199e0f : SUCCESS in 8m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cc606c44cb4f4d6eaec8fd37d3cdb3cc : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/15ff7479dbb746a0a0ecf36cdece78d3 : SUCCESS in 16m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0c1b3fb4b0844f5e9de6e7f7d9f7bf7c : SUCCESS in 16m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ca833f61c9d147bfbc43dad32f589613 : SUCCESS in 15m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6b24f33eeb2f44a591aa9f56a3ac7ec3 : SUCCESS in 15m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a46a3ae592bd4b948cfda1a2ae96b831 : SUCCESS in 13m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/423bf269a9e941e8bfa1cc6a52dddb03 : SUCCESS in 15m 57s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/baa043b4a72c4d7d97daa04621ce6cd0 : SUCCESS in 13m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/128f9e95593d45c09fcb48db175c3238 : SUCCESS in 14m 28s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/4f3f7f7dd72246b1bfe6bfdf8e60d3e3 : SUCCESS in 35m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8afa582af8424407918ca7dc9c1d97cb : SUCCESS in 13m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/09769d7103244a9f8549448bcaa06b95 : SUCCESS in 17m 26s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/45eccb1184274c11b56eef46cd59e0e5 : SUCCESS in 10m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/407e38b15e9c4442a6ae369bbad0329a : SUCCESS in 15m 04s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/da9d64126a85492cb4d7156a4eefec27 : SUCCESS in 38m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/73665e1ed2de4a6495d638f8d6b4bd73 : SUCCESS in 22m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e1f2eb816ffb4d9694a7a095079858c1 : TIMED_OUT in 1h 00m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0600802f6edc4e999786179ba24f8cab : SUCCESS in 42m 49s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/54a2ac18449e40e29e1a593e43ea75ae : SUCCESS in 33m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba371fc765b5417886ea6b4afbea1c63 : SUCCESS in 33m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab5b2778470f467695d730ed2caf37d9 : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f536432d7434c35a1415e348b3dd7e7 : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13a6ed87527847d9853f3b212a33fed9 : FAILURE in 7m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db54ad5c9f804b87b7f14e56da78ce2c : SUCCESS in 59m 08s","accounts_in_message":[],"_revision_number":1},{"id":"416bc94092a0f529630b5c3ff6e22f0f98f83f48","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-17 03:10:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9794151c1f2d47aa47b30df578f7dd01acac8630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 04:11:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f9ef3b6a62466aa6ba8fccaaf97064 : SUCCESS in 8m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/380abf5e61ad472cbf212ebb0909581a : SUCCESS in 1h 00m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e3099758dcbd430e9defea7e62d46d2e : SUCCESS in 47m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b94cdce6f1c24a50892538933f7aa610 : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a062dc3269a24613bd699eccfbc404fa : SUCCESS in 10m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9e7dd030ec6d42d7a0510a3e43d89185 : SUCCESS in 12m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/70acfd4fd51e40ba852151d6b0dbdfa3 : SUCCESS in 16m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6a3f439145ec419db6932619c03dc603 : SUCCESS in 13m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5da060037ebd484a9160687593314bdb : SUCCESS in 11m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/99ca8d78957e48afb7ebe1d29fa5800f : SUCCESS in 10m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/87a15d959a3c4809bb05343fcaf9f21b : SUCCESS in 12m 01s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f5ca68c33528414099011c313daca866 : SUCCESS in 10m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2ae434e7ce1043159149c4c9238b88df : SUCCESS in 9m 43s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/25167db46eed4a90bf27d77161c4b36a : SUCCESS in 15m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/35280368967848ac9eaad0cef5e11a46 : SUCCESS in 9m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/131b904b046046c5a69cb621e472f660 : SUCCESS in 13m 22s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/304978eaece9458e937ad939ea3857eb : SUCCESS in 10m 10s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/15ae61d49ce44f669b69601614df876e : SUCCESS in 10m 39s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6cd744f6a13a4b49b553ea3bab2911d2 : SUCCESS in 15m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5b3026a3be344b41b167e4393b0d7b9a : SUCCESS in 18m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c9a7e8e2cd7e4af1a0b931d4c1679d19 : SUCCESS in 41m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86bda4baa7774b5d8df918fbad1127e7 : SUCCESS in 35m 02s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/912888fcd85b4d818de026a8ba96ca98 : SUCCESS in 34m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0a6cdf62d8b248a58d1784bf924a19ab : SUCCESS in 34m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83355c1b55364674ad602f8d6f8497f4 : SUCCESS in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e08d6bbdf6cd4f24b3a6a2ee0fb4bb4f : SUCCESS in 6m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/64bd506a5bea47fd9b67e7c3c3447723 : FAILURE in 7m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8cd54b2733d14b3ca0bf3d9ecc19f8d0 : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":1},{"id":"fa36da0be8c46c65bb853efc63d0f060957de600","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 04:11:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"803dae9400fd1b4d6416b29efd06d63805700573","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 05:24:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c0a02ebb5f4840abc0a418b3d0116a : SUCCESS in 14m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/dcc3b2e979124598bf4c7b2ac8d50f14 : SUCCESS in 1h 03m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b4670302f6874bd488568a7494e70df0 : SUCCESS in 49m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7e8f29750a334ba9a68d63f5a14f595a : SUCCESS in 11m 45s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/209e6aa79cf84fb38f5e23ea1c91aa82 : SUCCESS in 12m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/89ce042269524056b6f2fe60f36bfc7e : SUCCESS in 21m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5ceb09ff3dd64b8394dd8fdc62ec2d5e : SUCCESS in 21m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2e815c090a0c4dbba74bb8d39efc1667 : SUCCESS in 21m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0ea5f15648ce413d9053f136359dae20 : SUCCESS in 20m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1a4a0c1ddeac4232abfee1446ec926c1 : SUCCESS in 20m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6b283af84adb49c988a4d6ef1db676f7 : SUCCESS in 21m 54s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/add101ec7a4c42eebaadd70533f3e930 : SUCCESS in 19m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5b8aff5f4db04315ad44ed165f47be98 : SUCCESS in 20m 39s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/936aa90df7bb4715a28225d4cd6c1f19 : SUCCESS in 23m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b10208408b0a4a2b9ecf149b61f55550 : SUCCESS in 35m 52s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/68690285b1094452918aaac5884a7aba : SUCCESS in 36m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f3847911820a42738c40eb08aa7f7640 : SUCCESS in 38m 04s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6bac893cdd634170a3abdf9fbe0f6118 : TIMED_OUT in 41m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aeaa01d93c07424ba78746416a37b41a : TIMED_OUT in 1h 01m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4df77d64a71a4863b440ecb31663c47b : TIMED_OUT in 48m 01s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4140ca9e5a3a4faf88d50c58013a2187 : TIMED_OUT in 48m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f20c04ad7aec42ebb679de1ac925b8e1 : TIMED_OUT in 47m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b692fd8177f240668ef823688217ddbf : SUCCESS in 36m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b6a2ead69834e3b91b4b9fd5bc5ca9d : SUCCESS in 7m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c27f77d532f41d8be4420b61eede4d4 : SUCCESS in 1h 11m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ab54848614fb5657403c4b8364b2cb9e3ebd8f0a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-17 05:57:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"00aab564d13a70d34b855533d6071c86ca37aaf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 07:14:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/026872463e234a07ad0c3e0732fb88c1 : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/73c3f3b2f21e4c0eadc5552860ebc9a6 : SUCCESS in 1h 11m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d3262423727d480b859a1ca2b746d6fa : SUCCESS in 42m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/049bc45c49ee46f1b24fafba652bcb09 : SUCCESS in 8m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f467f3ecbc3c457bb5b6e6e4db9323ba : SUCCESS in 10m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e36830e287534111b384b7514b1e2260 : SUCCESS in 19m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ce6fbb8f387d4eaeb89762a3a3e0668a : SUCCESS in 15m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b770ecc051b047d98ed4b127b8904e10 : SUCCESS in 20m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f9615762824b4f3dbf4f639bfe93b860 : SUCCESS in 17m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cccc334dcd264b9f8a53484870acc394 : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b7d0c1961a354af185d569e642a38c7f : SUCCESS in 13m 51s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/54298d99e10c454b820331999b6eefe1 : SUCCESS in 17m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8eeb9d31b1574af289614211388bbba2 : SUCCESS in 13m 04s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/8b92665c66204fdb80edd0cc70221bf1 : SUCCESS in 22m 03s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/96038df995b64d669e6a3730e313d6ee : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c558e683b5ea4f828413130a09e40fa5 : SUCCESS in 19m 49s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2da69138f62b420ba24fbfd16afee83b : SUCCESS in 13m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1b162472f5d7416d8069a971baaf69cc : SUCCESS in 23m 10s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/3aa28652655a474d94a47871dcccc2ac : SUCCESS in 21m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dfd9ba5997a54d8b9d30cc33030d17bd : SUCCESS in 25m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/58808c5cbf1f4553b6bf5826d0e6432f : SUCCESS in 54m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/657e1c482a8f40e2b0e8eb8ad6f421cd : SUCCESS in 35m 53s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/441ee0be75ac46939e5bf66e8a45bbe7 : SUCCESS in 39m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97cf6123875f44c18baf5afc8f7c19ba : SUCCESS in 44m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3781ee6263fa471ea47ec5a559794ca7 : SUCCESS in 14m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e3c7cbe00b3466c828ea9356fe96293 : SUCCESS in 6m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4bed82c947c04018b530692b6c12bec4 : FAILURE in 10m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79e8a435bc864774875517292833939a : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":1},{"id":"600d2e496801f9a56a6e476f8a14f0f29390c8fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 07:15:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39f7bb2e2a484ab7326ff9cdad0b9fa032edb261","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 08:21:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/885a98fbcdf645df934810893fca2e14 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/9b68919d69e6483d80743c1492a23386 : SUCCESS in 1h 01m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c2a758c064d54835a3c39d644b7e87b5 : SUCCESS in 53m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d1416d70e40a4e2cb910864c53b269b2 : SUCCESS in 10m 06s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ecd08e31d67341308921f8994f9a4dd2 : SUCCESS in 10m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eab9f70cbd9e47a1a39425b31198343e : SUCCESS in 19m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/77f08d55367a4dd29b287e289c6736ff : SUCCESS in 20m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d2a1f6c2dc5847b4860c34b8cf41fb2a : SUCCESS in 19m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b659090278964396923e07699acace4b : SUCCESS in 14m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2bbd33e282f34f4280b808d4d0e314d4 : SUCCESS in 9m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f92eb63143bf4ae7a2815a8eb97ef512 : SUCCESS in 16m 23s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8dcd064fcca24526ad00995f341f057c : SUCCESS in 13m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/75fe4b2f3e72423387757fe7015eb08e : SUCCESS in 13m 33s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/2fbb174f75b64a5fb6156d412d483a12 : SUCCESS in 18m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/91894163d1344914b7044b0dca1f04b8 : SUCCESS in 13m 52s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/fff8dfa97545408887f1bf7e051c939f : SUCCESS in 11m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/17c495eec7b142bfa3e3ce17dfb538bf : SUCCESS in 22m 11s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6eb66eb3407b41c8b9f64bf150d7cc3a : SUCCESS in 15m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0145e953295d4a3ba6a7fa0d96233753 : SUCCESS in 50m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/829a3010d57a4b44963d2c2733374206 : SUCCESS in 34m 39s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/abc3fe06c8fc47dca7f643ebb813c7a7 : SUCCESS in 37m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/383b027eaa10460286b1f466e618cec5 : SUCCESS in 37m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f538f564f041441e8dc67cc497813fde : SUCCESS in 12m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc2792a121b74ad089605e5939f0d847 : SUCCESS in 5m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56a9979196494157953613953449b6a3 : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b396edcbdf9750ffc3938defd8347c6f87c960d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 08:21:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"debafad2f9b07c5f395ee2c606eaf74c54458779","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 08:23:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc81bc8716d6401e897cb1dcbeedcf2a : SUCCESS in 1m 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/89791e72858e4c33ba470f1034968668 : SUCCESS in 1m 28s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/a9be4b80fdd6445eba912103c0f18711 : SUCCESS in 1m 15s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1358d37328e1ee3245a61777c2b6aea269af8832","revisions":{"1358d37328e1ee3245a61777c2b6aea269af8832":{"kind":"REWORK","_number":1,"created":"2020-05-15 23:03:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/88/728588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/728588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/728588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/728588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/728588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/728588/1"}}},"commit":{"parents":[{"commit":"08db36a295377fc437c56f6fc470be035fd74c4f","subject":"Merge \"py27: Suppress UnicodeWarnings in ShardRange setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08db36a295377fc437c56f6fc470be035fd74c4f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-15 23:03:02.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-15 23:03:02.000000000","tz":-420},"subject":"versioning: Use CONTAINER_LISTING_LIMIT when doing container listings","message":"versioning: Use CONTAINER_LISTING_LIMIT when doing container listings\n\nChange-Id: If274a01a6660776cf01b71efbdd102159c520794\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1358d37328e1ee3245a61777c2b6aea269af8832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1358d37328e1ee3245a61777c2b6aea269af8832"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
