)]}'
{"id":"openstack%2Fswift~728589","triplet_id":"openstack%2Fswift~master~I0dd3b72e374b9eb93a13ed7ef5fd24cd28e66396","project":"openstack/swift","branch":"master","topic":"upload-id","hashtags":[],"change_id":"I0dd3b72e374b9eb93a13ed7ef5fd24cd28e66396","subject":"s3api: Add config option to include uploadId in GET/HEAD responses","status":"ABANDONED","created":"2020-05-15 23:13:04.000000000","updated":"2020-06-10 21:42:47.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f650d590cb58af3103bad90e76f232aa9bdff183","_number":728589,"virtual_id_number":728589,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-05 20:18:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2020-06-05 19:14:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-06-05 19:08:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 19:14:48.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-06-05 20:18:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2721541907e1057ac7d14ff87f1b24ee7f87ddbd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-15 23:13:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccb32ef2d41cbec034cf45069b713eefbc28866c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 00:25:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d2245e3f86d4d27ad315f4c193b2e49 : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/daa7e473ab5f4d569704524e4e0beac3 : SUCCESS in 1h 06m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9188f76f0acd402a996babdf45b7446d : FAILURE in 1h 04m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/da5989a60af04d558411389d131dc7c7 : SUCCESS in 9m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f53fc55dc220449fb2236549dad66a3e : SUCCESS in 12m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bf1c8efcf5524dc981be531568fef949 : FAILURE in 15m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e23daf276ea24c888d340a103b25118f : FAILURE in 19m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/635fcc8615a84c2788327d6c96e3497a : FAILURE in 12m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/558016146c78493eb6240bf7529f405a : FAILURE in 14m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/54375bd2ec824cbc91782f24d873de6e : SUCCESS in 11m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/55adb2a928474d3486366aaebf4391d7 : SUCCESS in 15m 07s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/6b07760bf91b4c58bd8c802e48d19eb8 : SUCCESS in 20m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ef1b9c8e642740c4b3ad944e1c250356 : SUCCESS in 13m 21s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c8dda70ff504448599b901de793a115c : FAILURE in 33m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fb3a18779ab5492caa9cc488e5e66948 : SUCCESS in 21m 04s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/44ce9198adc94f678e462d1fe3c306f2 : SUCCESS in 13m 39s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f1381de88e29465ebef55a137610374b : SUCCESS in 10m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/041dc106002d4e76aa18887f368a1ae5 : SUCCESS in 11m 50s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1ab4730de14941e3b9881daa359ed948 : FAILURE in 25m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/139439133ab448f1a50295f2e6022616 : FAILURE in 33m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b3ea76c115e84fa4a22dc7bf24de3618 : SUCCESS in 44m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/30ce5662a9f4450898192026407a897d : SUCCESS in 35m 10s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b0b9e442c7cb4dacbaeb655f1552cbfa : SUCCESS in 36m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/48ff02785c134ea79c731cdfdb4ea985 : TIMED_OUT in 48m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6859ba72c2b34776993beba2d401c6c6 : FAILURE in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5209a972b59a4d40bb309ccfc8cb7995 : SUCCESS in 7m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4e70542e547545ac9ea65d56ddda754f : FAILURE in 36m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6870d28822be42bf9bd60a50910da56b : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6cde8c9aecb380ce6fd6f15813d2a087b683d5fb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-16 04:33:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4813d0d5d8bc3fc111150706eeaf5b178de61eea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 05:42:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd220717dbf44026a4c75c276c858939 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/f487f6395ec64f419f09351eb87078cc : SUCCESS in 1h 01m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/56e3020e5c534e7cb86bc070d0c0a654 : SUCCESS in 46m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8965c4aa21db46659931014563d9ec90 : SUCCESS in 10m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5a046ced75c54bd1a6b94f84e45a86e4 : SUCCESS in 9m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d4ebd4d91f784e86938c372a152248b0 : SUCCESS in 14m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b4b71fde26224d35bf5e18340ef612fc : SUCCESS in 15m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bdcab4961f094b07b90c6d80fbdb74d0 : SUCCESS in 15m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f6d4b22e0a4845729058906b8ad2aec1 : SUCCESS in 15m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5b7f3fc8c9ff40f8a33ff216c37148c1 : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fa5a07ff72a6473fb31670e6de525d52 : SUCCESS in 17m 37s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/002f135b6aae4301a2f9abcd09fbbfe6 : SUCCESS in 12m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c25fa2e645fa4d73bf9ebf37a7a20e3b : SUCCESS in 14m 42s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/94f18f9051724966a408feee193d2d9c : SUCCESS in 20m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c6713552d24d4f9faba19f4844cade8c : SUCCESS in 27m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/61f4490738b64e7b91b519027d5def24 : SUCCESS in 16m 19s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/930e5e06e50342b1818907dd83f028dc : SUCCESS in 13m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4170a6822fa2496b835bb4d8206819c6 : SUCCESS in 14m 40s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a7163c5560d040daa3ed6f172018f5f9 : SUCCESS in 20m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a1a9dec9ae0e4d859981bbc7b1125141 : SUCCESS in 22m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7884a9ba5fd14b899b02b2db98ac5661 : SUCCESS in 52m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e3712ec590a644e0918b9996cf113730 : SUCCESS in 35m 37s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/57571857aa7648dcaa58aa91965481c6 : SUCCESS in 36m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/192fce4e39024398bbe86e52c5596751 : SUCCESS in 43m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e02604698c144a51b227b850887dade5 : SUCCESS in 30m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c45112eab924efe977b6907036e0a00 : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9f74f7e4c9bf43ecaab9181e917c4934 : FAILURE in 27m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b3c361ce5ec4f94b3c3efddfcd86666 : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":2},{"id":"9c33701b98f8de63407755ae3db35af42677bd0b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-05 17:45:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(6 comments)\n\nimplementation seems reasonable and well done\n\nit\u0027s not obvious to me when operators would know they need this or how s3 clients could come to depend on it in a useful way","accounts_in_message":[],"_revision_number":2},{"id":"619ac4d6bb36baa8db000d95264728c1c9750d51","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-05 19:06:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"04b773b3eb3024bc192ed4a8b434dbbcfaaf77ba","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-05 19:08:34.000000000","message":"Patch Set 3: Workflow-1\n\nOK, I at least got it out of the x-amz-* namespace, but I want to try getting a server-side fix so clients don\u0027t actually need to see the thing.","accounts_in_message":[],"_revision_number":3},{"id":"295f5e6fe42d7f83225e108ba2bb721be42fdd6e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-05 19:14:48.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nyeah i\u0027m -1 on this too; but I fully support doing the work to help clients :+1:","accounts_in_message":[],"_revision_number":3},{"id":"4b12dbf0580a9b50cd3a3c5ecf517f33af1a66b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 20:18:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b22b00715a544351ab4a0635260d0cfb : SUCCESS in 12m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/3db7adc8a766467cb6468d1cd6d08b67 : SUCCESS in 1h 10m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a58d5ac554d84c638f4d5b269dd3fce7 : SUCCESS in 47m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8169dda80c06425bbb2505c171031d7b : SUCCESS in 8m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a80bfb93fb4e4317ac89002b6869c047 : SUCCESS in 7m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3a1a4817df254f76988a04bb4d865638 : SUCCESS in 23m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f4273244e6864158a2c49c9e577a8edc : FAILURE in 20m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2573dfe76eb54d168ad18103c56c485d : SUCCESS in 21m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/91ae4203490f4e6c8b7eb04f6bb72751 : SUCCESS in 19m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d09639445acc423db04f3acf9dd2f05b : FAILURE in 17m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/57c35525f7c04a2e985e6b22f1abba76 : FAILURE in 19m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a5337de45fb44d0a10c29422d2e6409 : FAILURE in 38m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fe4204e6fd2b40db8465785b8ee4bf8a : FAILURE in 19m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fe85f6e74fd94e768292f8d5cabf5d6a : FAILURE in 20m 21s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e0d402f468f046e697a31bdf12f96bff : FAILURE in 31m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8167274ddcc84a98a9fcbc81afe1e5ee : SUCCESS in 32m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a3f8fd819cb54a55a537f110a9e59d49 : SUCCESS in 45m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8ed98c30b9ba4c47a30b186167490663 : FAILURE in 36m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f89e86270f454b32abc0694bcc7e7628 : FAILURE in 34m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcf5ed09fc934f34ac03fc9e244fb3c3 : SUCCESS in 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bd4f2c5beca4833ac7b5193ea0f29e8 : SUCCESS in 8m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/82e8738aeeb543d49458eb7625a6dea1 : FAILURE in 7m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84660ef8d72842379f5043a9cd4aba8b : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":3},{"id":"1b4a9da6610e770508301bafa57b9e0723fc53b9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-08 19:39:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"af968d4b9393451762e3d314c0d2ffd686f4a269","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-10 21:42:47.000000000","message":"Abandoned\n\nnope, just retry https://review.opendev.org/#/c/733919/","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6bc8b7ebcb842623e5c6ef668c441e9a26a1fbd9","revisions":{"8b5a513fb0bfe600cb0b3b4c3a43abcb80fb0be3":{"kind":"REWORK","_number":1,"created":"2020-05-15 23:13:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/89/728589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/728589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/728589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/728589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/728589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/728589/1"}}},"commit":{"parents":[{"commit":"08db36a295377fc437c56f6fc470be035fd74c4f","subject":"Merge \"py27: Suppress UnicodeWarnings in ShardRange setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08db36a295377fc437c56f6fc470be035fd74c4f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-15 23:12:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-15 23:12:58.000000000","tz":-420},"subject":"s3api: Add config option to include uploadId in GET/HEAD responses","message":"s3api: Add config option to include uploadId in GET/HEAD responses\n\nThis may be useful for clients trying to decide whether to retry a failed upload.\n\nChange-Id: I0dd3b72e374b9eb93a13ed7ef5fd24cd28e66396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b5a513fb0bfe600cb0b3b4c3a43abcb80fb0be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b5a513fb0bfe600cb0b3b4c3a43abcb80fb0be3"}]},"branch":"refs/heads/master"},"affac56609732876c78524f5326145827e264ad2":{"kind":"REWORK","_number":2,"created":"2020-05-16 04:33:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/89/728589/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/728589/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/728589/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/728589/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/728589/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/728589/2"}}},"commit":{"parents":[{"commit":"08db36a295377fc437c56f6fc470be035fd74c4f","subject":"Merge \"py27: Suppress UnicodeWarnings in ShardRange setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08db36a295377fc437c56f6fc470be035fd74c4f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-15 23:12:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-16 04:33:14.000000000","tz":-420},"subject":"s3api: Add config option to include uploadId in GET/HEAD responses","message":"s3api: Add config option to include uploadId in GET/HEAD responses\n\nThis may be useful for clients trying to decide whether to retry a failed upload.\n\nChange-Id: I0dd3b72e374b9eb93a13ed7ef5fd24cd28e66396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/affac56609732876c78524f5326145827e264ad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/affac56609732876c78524f5326145827e264ad2"}]},"branch":"refs/heads/master"},"6bc8b7ebcb842623e5c6ef668c441e9a26a1fbd9":{"kind":"REWORK","_number":3,"created":"2020-06-05 19:06:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/89/728589/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/728589/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/728589/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/728589/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/728589/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/728589/3"}}},"commit":{"parents":[{"commit":"08db36a295377fc437c56f6fc470be035fd74c4f","subject":"Merge \"py27: Suppress UnicodeWarnings in ShardRange setters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08db36a295377fc437c56f6fc470be035fd74c4f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-15 23:12:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-05 19:06:16.000000000","tz":-420},"subject":"s3api: Add config option to include uploadId in GET/HEAD responses","message":"s3api: Add config option to include uploadId in GET/HEAD responses\n\nThis may be useful for clients trying to decide whether to retry a failed upload.\n\nChange-Id: I0dd3b72e374b9eb93a13ed7ef5fd24cd28e66396\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6bc8b7ebcb842623e5c6ef668c441e9a26a1fbd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6bc8b7ebcb842623e5c6ef668c441e9a26a1fbd9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
