)]}'
{"id":"openstack%2Fswift~728998","triplet_id":"openstack%2Fswift~master~Ifa298ac668c22541b6db5fd021236a757b985d40","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ifa298ac668c22541b6db5fd021236a757b985d40","subject":"ratelimit: Log proper container path when returning 498 for listings","status":"MERGED","created":"2020-05-18 19:04:27.000000000","updated":"2020-12-09 08:50:11.000000000","submitted":"2020-12-09 08:48:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728998","meta_rev_id":"d7ffef9b8ec1dadbde62d12bad0fc8544485c104","_number":728998,"virtual_id_number":728998,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-09 08:48:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-12-09 05:02:42.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-09 00:45:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-09 00:45:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-18 22:44:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d3e52e9d3d27e54269a9b13d385d494a663d2dad","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-18 19:04:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5cfb0b48187664ff09350422753cede5a8a3076","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-18 19:06:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ccb1f9c347b14ea581297b0cbae0d28a7381def5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 20:10:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1380ce103bfc4deaa3fd090179a89be6 : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/332a16e47a9c4a6fb0b595b7828d01df : SUCCESS in 1h 01m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/09838fc31c1e4971946936ef29edaa32 : FAILURE in 48m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cd5beb66b002464baef9ae38bd388036 : SUCCESS in 11m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dfee1ae697794bff9be76753eda5ec17 : SUCCESS in 11m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8119a163ef7c4fb280b64773d369c144 : SUCCESS in 13m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9952f4af327d4848a563a80a4392d332 : SUCCESS in 13m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/260f8b27761844f9bee5b21fad8abcba : SUCCESS in 12m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cfab6081a5414d2fa2650918c2d18e20 : SUCCESS in 14m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4438c6a1c6fa4f42b34a0a32d519a9ab : SUCCESS in 10m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f7cd0946f5bd4fd386f0f85328984e93 : SUCCESS in 15m 56s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/06c346db65fe47ceb3265079e13c9923 : SUCCESS in 10m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/409db0570fd14ff7972ad5736bdb79a5 : SUCCESS in 11m 21s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d9d61d9911f8413abf38024981a281b7 : SUCCESS in 17m 24s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2ab1c2b78fb44c3690b0fda13fc9aa67 : SUCCESS in 13m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/579b4f5c2faa4152b4905d0153c18d7e : SUCCESS in 26m 01s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f54942a44131421986e14f54b2b12620 : SUCCESS in 20m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/96e2eae564d34216b51ebadd316599d6 : SUCCESS in 21m 52s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b8c1e1460abb4ccfb927480da370ade6 : SUCCESS in 17m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4a5b5d87654b4eacbd1183d8973b3903 : SUCCESS in 20m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/49aabbb95724415a916698d137b005c9 : SUCCESS in 47m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e67dddc5aa294c4a99a04dbf6ae9b97c : TIMED_OUT in 48m 29s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c2b5a6fb2be24f12a8abac83a04b8a5d : SUCCESS in 32m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/602c1bc331014c8b814210d61d0cce1e : SUCCESS in 38m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fb02324b8a54c6d8dc2a16ecbd56add : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1be8b0ba3bca4c3c9a30b3d078f9c7e9 : SUCCESS in 7m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/634767e8aa1041569d38d6d3bdd75ff2 : FAILURE in 9m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb7390fdb47e4a7f9971028057c335cb : SUCCESS in 58m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4c1a40fb8b9c9abf2c1a3601229bf0786d3e2ddf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-18 21:41:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"38339f30a9795b490f3c971a4c68a2d7ffd54b7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 22:44:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b15c0762bb49c88a55470537866fb6 : SUCCESS in 10m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/91efdb75d75c40a883998b5103b4fe12 : SUCCESS in 1h 00m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fc78f7bb16b24c70a5eb2474c1a21e99 : SUCCESS in 47m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/079a580f2dfd4151aae93dff3b61edc0 : SUCCESS in 8m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9b65125a3c3b444b8fb391b58dceab15 : SUCCESS in 11m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b5454eb1f86f4bcdbff82eebbe8649d6 : SUCCESS in 11m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f1f1f1589d704225a14ac5d5c4cc253e : SUCCESS in 14m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/195d4b091d5145e5838ce490a5e7d33d : SUCCESS in 12m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d73354afef4347fd9e34dce22e257c8b : SUCCESS in 13m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c53fe2b05c2e4beeb94f3210a2c185df : SUCCESS in 10m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e8fb1d3391fc46bcab25fa7e72168206 : SUCCESS in 11m 59s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/cea9afbf2f764e5399a4f1614950624f : SUCCESS in 9m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a5d915b68c7c4b77959d854afd4a17db : SUCCESS in 10m 17s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/37a87757776043089b36b13a135d37d1 : SUCCESS in 14m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8e93e463c34b47adb2438fd8b11b8704 : SUCCESS in 10m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e8adcfc4199347269c24e982e3e4ef4e : SUCCESS in 13m 49s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0c12df6002784de7ac0038d311375deb : SUCCESS in 10m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2771a0d8b2d448b791347fe31a090e07 : SUCCESS in 10m 56s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4b16a3b6b318494f9575e755b80a2aac : SUCCESS in 15m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8383126ce26e479c982bb700783e9b2b : SUCCESS in 17m 04s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7a5e8876be2b4bf4926a1acf4fe6cad0 : SUCCESS in 44m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/06482bcbbceb4c0a8861bc667450508a : SUCCESS in 32m 54s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/fef92867ee584022bf228646658e2cef : SUCCESS in 32m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e28c6e30aad74846920df27900d87c75 : SUCCESS in 39m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa0c28094b294c3ea76b8e23ad7160ee : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/113d03ef3fea4a08bf1c44cc7388fff3 : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a82e2f581764ee8bdbf2a1c2853be0f : FAILURE in 8m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2832e3dac6c4e64a7759ca80bcd6ca9 : SUCCESS in 57m 43s","accounts_in_message":[],"_revision_number":1},{"id":"048ffbea2adfac9365501d9cddbac9ecc210dcfc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-09 00:45:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nYeah, this is still stupid.","accounts_in_message":[],"_revision_number":1},{"id":"6a5e4ad38297579926a9d6e3d3f6de7611911d0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 00:46:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35fe86365c7308b94ff3d9ee48786a3a1f40ff70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 01:56:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014de0da04624d3cad64243c4d422943 : SUCCESS in 7m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/0bb51fe32ad84e74b643ea0b972b9c34 : SUCCESS in 1h 04m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2d2b69d7c0234aaba6be93eb3fdf4a40 : FAILURE in 47m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ab7c68167db142b5b005b396c2b02ae9 : SUCCESS in 8m 38s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/aa83ace463544b34bbae9d0549a34a8c : SUCCESS in 9m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/06713e0837234daba41477eacb2af2de : SUCCESS in 12m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e92a3b3e23584692902c46c5d09258bd : SUCCESS in 13m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c16b449bc45549e9bee5f4c8341edcca : SUCCESS in 12m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fdb88b9bd19442dcad81e1bf048310a2 : SUCCESS in 12m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e85565180cbe4f4da9c1597340ab3f3c : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c0a1d191da3c4c5b954ae48efc01739e : SUCCESS in 17m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e3479f131bbe4526bad814d936ca1a32 : SUCCESS in 25m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/990604e9054e40868ad32f838dfc9f24 : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e023a39cda1d4b30a86e8ffdc9f5c760 : SUCCESS in 19m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1608f3ca0dc04c75a3d1201c52310741 : SUCCESS in 18m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1e92ab6def3d42ddbf6af85d25f16ed2 : FAILURE in 50m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1dd5483c47254605b9ab9540423f3af3 : SUCCESS in 43m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/312090788445474a95fc744f95b62f6c : SUCCESS in 40m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c10f5393e394c4aaf0299778a95e13c : SUCCESS in 13m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13c0bc7c19bd47f28daafa3a2b2709da : SUCCESS in 6m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cbe7da1176247fc9ef4a9c2efdcb3d8 : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7e1108a4e1fc6210b5dbd3f49554e7b2770cb8d7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-12-09 05:02:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"305cf23f83551d6fbcb962a7757be020dda13b12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 06:33:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/663b2d7dabf8407b986d427c8de636b8 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/866e64be12ea4713aee8cbed9b247278 : SUCCESS in 1h 10m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c57862a755b24ca3872735fc165e17b3 : SUCCESS in 48m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/46d12f92019644128c81369844b1216d : SUCCESS in 10m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ff2dc05f03be4101b16bf6b2d29b5723 : SUCCESS in 10m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/082510e88b964e9ebf40f8b3f8b0b8ef : SUCCESS in 13m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b535c2831c214a108c3d53a84ed9d35a : SUCCESS in 13m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/da59daf6d2a94f659753bbda64011df4 : SUCCESS in 14m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/662a0fbf1ab04f929f22d5094617f302 : SUCCESS in 11m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fb85d308bde64323ac48ba2e836eecfa : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ab224f10c0cf4575a3a00adad4d1ca57 : SUCCESS in 18m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d60b29f5589c4bedaf4de8c0bc64ffe4 : SUCCESS in 18m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/67a5be17fb7a40dda25a208feafa58de : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c6e6c497779a4552be3bcf80b5d207e0 : SUCCESS in 18m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/396de6cd31ce40fe929fff1e46a8002b : SUCCESS in 18m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d56f650c8054437ca8532ad8e9bec9b6 : POST_FAILURE in 5m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/43e315042a4946c7b400ece8664d8e4c : SUCCESS in 48m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/281c0a6389e04b908f4db8fed92d248a : SUCCESS in 37m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fb47fe304d784225adb39d05590deb9a : SUCCESS in 42m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43561e3df2564e848deaf52a17fbcbef : SUCCESS in 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c83f694d30643e09b676e7b0cc4c651 : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d17c5b3ef171493ea9beb366a46c101f : FAILURE in 8m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21218c81189141de92910585b9477c74 : SUCCESS in 1h 03m 35s","accounts_in_message":[],"_revision_number":1},{"id":"a3fec626c3a320ca0c7dadf64eaa811b45b83131","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 06:34:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a17ec090d557d6724d36302d894d4763ea2a8776","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 08:48:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3203911e878045599a346bc7e4864119 : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/193e8b2243c14962bcbdb622684d736f : SUCCESS in 1h 16m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/17f6109b05bf431b9bd7566a9dc56e98 : SUCCESS in 53m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1409049962494dcf9ea40ff0f91b6dbb : SUCCESS in 8m 20s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/0d5cd20824e441a5800f96c3a77208e0 : SUCCESS in 9m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ff6ccf5490c54423885027979ebe5719 : SUCCESS in 12m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef2c30bcb2e64249871e435f56cedaaa : SUCCESS in 17m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b51e7a95147e456ba5913342174d88ac : SUCCESS in 15m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/534ad302ed8b4a19b912775956405943 : SUCCESS in 15m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6f80c9e8988d4db0993d0018f4c74881 : SUCCESS in 20m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/81fad8bfe7a047f89b750e4c4f40a228 : SUCCESS in 15m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2c9ec6467b974850a2387fbba4cf10ad : SUCCESS in 21m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/afdbaa1e3a36420a8227a9f8ac9471ce : SUCCESS in 58m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/33a31d53d5d245a5a725e151d6ead409 : SUCCESS in 17m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/61626a8c3ab641cdb5bf6e4a4cc75234 : SUCCESS in 18m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a21b2394a0044993855a2865958b6bfc : SUCCESS in 56m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d99ad69134d14f28b167e90d2d172e5b : SUCCESS in 40m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0826ff83370046d1b46198af997ff80b : SUCCESS in 48m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce81e10231cb4d9f90ca29b0779ca219 : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6b240d3d003403799231b7220b429e5 : SUCCESS in 6m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95edd630f4c848bda38454b1dd119789 : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0bfc6ba0688a229856b73a10e15128606e54384b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 08:48:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d7ffef9b8ec1dadbde62d12bad0fc8544485c104","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 08:50:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92930e910edf4ffa96d66f534427035a : SUCCESS in 1m 14s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8abc3c80e4064228a9a950077e3aaef2 : FAILURE in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/62ca91689375472f863c758990cde94c : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"802d80f316020c5e8de2d71814eef738ef62f363","revisions":{"802d80f316020c5e8de2d71814eef738ef62f363":{"kind":"REWORK","_number":1,"created":"2020-05-18 19:04:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/728998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/728998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/728998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/728998/1"}}},"commit":{"parents":[{"commit":"70dede1ae9b714f0eed3804022b49517a562bdd6","subject":"Merge \"versioning: Use CONTAINER_LISTING_LIMIT when doing container listings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70dede1ae9b714f0eed3804022b49517a562bdd6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-18 19:04:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-18 19:04:07.000000000","tz":-420},"subject":"ratelimit: Log proper container path when returning 498 for listings","message":"ratelimit: Log proper container path when returning 498 for listings\n\nPreviously, a ratelimited listing could log lines like\n\n    Returning 498 for GET to AUTH_test/bucket/None . Ratelimit ...\n\nChange-Id: Ifa298ac668c22541b6db5fd021236a757b985d40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/802d80f316020c5e8de2d71814eef738ef62f363"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/802d80f316020c5e8de2d71814eef738ef62f363"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
