)]}'
{"id":"openstack%2Fswift~729091","triplet_id":"openstack%2Fswift~master~I4a65f50828b4e90ff6be2c3b343b295e442cc59e","project":"openstack/swift","branch":"master","topic":"729093","hashtags":[],"change_id":"I4a65f50828b4e90ff6be2c3b343b295e442cc59e","subject":"prefactor: Pass s3api Config to S3Requests","status":"MERGED","created":"2020-05-19 02:36:36.000000000","updated":"2021-01-15 02:41:06.000000000","submitted":"2021-01-15 02:39:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":9,"has_review_started":true,"submission_id":"729091-729093","meta_rev_id":"dc9f7947ae0b40e6d3f4f022a52f1c067ab53aaf","_number":729091,"virtual_id_number":729091,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 02:39:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-14 02:10:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-14 02:10:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 23:46:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-12 14:41:00.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"0297dc56c1da4726dd966b8ac6160761c00ecf73","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-19 02:36:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d4d2bbc356b6015b25cf0feb2a1576c4ccef915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 03:39:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c2c0dbc9e0747cdb8e2becc206cc00d : SUCCESS in 7m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/1cd93db8a2ec43efbf51b1d5a588eb36 : SUCCESS in 57m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a5743b2a759341939978aac5dd81c904 : SUCCESS in 39m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/17421bc3fcc448e4b22acb2b817f12a4 : SUCCESS in 13m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9c3b2e46dace48cca5d76a87d62a2c43 : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/20151d2843ba48cebe2aa8476a6419b8 : SUCCESS in 12m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c93f50729265445db11474f4fc472620 : SUCCESS in 12m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f9c5054791134fa68910590453c8092e : SUCCESS in 15m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c4467cd9c0f0486187a33af09b2ba0c9 : SUCCESS in 13m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/20240707c3fe4673835ef614a9e184c4 : SUCCESS in 10m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ac0c75622d9044fd91c3002aa21747b2 : SUCCESS in 12m 43s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/548c01d4e0f84b59ae953eb404393f49 : SUCCESS in 9m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e3c9fd917dc84a98b5485850751d12ca : SUCCESS in 26m 54s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/060e9294ed6e46009a34e422f71a88b3 : SUCCESS in 18m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/57ae00ea417f404788dc819be0704641 : SUCCESS in 9m 49s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/16f06004a9484784bb95d4e996afd683 : SUCCESS in 16m 49s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/cf2a0bac413b4e01bdd1f06ff2a6559b : SUCCESS in 9m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/995f7a78b73545fa85d160a74013ca60 : SUCCESS in 11m 12s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a40c405e1cf440f4ba338890777463e6 : SUCCESS in 15m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/831203d38be04305a492bb86f8508c34 : SUCCESS in 18m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/38e1b881aec64d9893d4f93b0b096e4a : SUCCESS in 47m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0ab371cfaf294bafa415639a95791cb7 : SUCCESS in 40m 28s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/01c403e754544ce8963d880b4eda7840 : SUCCESS in 34m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e42d506ed1cf468ba38c3db80c9e17f1 : SUCCESS in 36m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f5d889c1b6842549fef27da37958b67 : SUCCESS in 15m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/150fde04df9946e897baf85cddd90a55 : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/924e6bd5bcff49599a233bcb7f1d3a17 : FAILURE in 10m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2af1b08ac05d42ebb9d5c4de7fbdf72d : SUCCESS in 57m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ed98abbd720a0fc0ac456e63b742a5484a436412","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-26 22:39:10.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"6d45a2d75c76d4d3f5d50160b74bf17d092e7d56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 23:46:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ed8dbb365a341aeb410a60d353bf9db : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/33ea7130b1e54242b824f8528bfa1980 : SUCCESS in 1h 02m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ff73e6e3b73a4176b499e24c67bd70e5 : SUCCESS in 49m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f930abad057640cd87c0c104f3ecb931 : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/da40429edb074643b1b0a57da0f1ef78 : SUCCESS in 9m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/72de8349cfb84489bf275588c84c7a99 : SUCCESS in 19m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3212d559e1994669bf2457e7f2a514ac : SUCCESS in 14m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4ae61898b7244de99df540d57bae9393 : SUCCESS in 13m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d88436b5c3b34abf94dcfd06f65d9ac7 : SUCCESS in 17m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a375c29aeb754001a5f1c16dcdd02750 : SUCCESS in 23m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/082ce922edcc4dffb495cbb58f5d6994 : SUCCESS in 12m 17s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/55b2211fed8145688c4b30f7633e821d : SUCCESS in 9m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/41e48afd22c6420794500c1c245e6ae7 : SUCCESS in 30m 44s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/cf33ee4e1f844733a2720b7aba446632 : SUCCESS in 15m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b9974856e59c4b039901c1a874ab8926 : SUCCESS in 13m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3c66b16558d94d77a30b0f0698687533 : SUCCESS in 27m 48s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d07f0b54b0f64022b3ace456ea88292e : SUCCESS in 11m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/29077c28f61e41d4a2b781842ad3ecd8 : SUCCESS in 17m 40s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/69d0c0a183a04be4b35cd4e6fedf1e56 : SUCCESS in 37m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d6d7596ae52d4884ae2d13efe7514dc6 : SUCCESS in 20m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d03b7f3015794873bc97534403445cc5 : SUCCESS in 53m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5459cda8f4341cb82739c0690251c09 : SUCCESS in 45m 34s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/31d9b1987690498b929ccb71cb0c3967 : SUCCESS in 35m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77302589e2a2490d89a435b94b8819d0 : SUCCESS in 57m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42d7338e216848b3aa5526529699634f : SUCCESS in 24m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a3f12ac07ed44fb9a7af1809ce5d81b : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fe5228435b574f3198b74fb0e355c9a1 : FAILURE in 8m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/783cf49ef9da4a4889b143a4886e7e74 : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e4562b9166a8c576d47de12409f08e1647b02436","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-08 22:25:45.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"92246a098a43a22ddd518276af9a1acb69ff13be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-08 22:54:41.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8adb27497cf3d8954c7003db70ff7ae779da583c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-08 22:59:29.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"2b4ec718791322414223e11ce241d18d648e4c15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-09 00:39:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b819f6e9f247bfa1f2229c5722c9fb : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/7c114145fad04faea66c147f35ab3628 : SUCCESS in 57m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/40d5c46113694b63b90fe176134406c8 : SUCCESS in 41m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b93b7e1eea494e51ae83b951487ce87c : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d29be3b709de42698f80948fa28621f7 : SUCCESS in 10m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b1d037b6d54049df932e1db2e7d43e24 : SUCCESS in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a0697b5a06ae45619f6183dfa4167945 : SUCCESS in 13m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a8a621be57e64e22a0d4c327aa39273f : SUCCESS in 14m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/11fbea48d6a84fcc991f3b00f9e20b6c : SUCCESS in 18m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5edbae08f3304606a756cbf49966c609 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/20e00885f4db410b8339e806066ca676 : SUCCESS in 20m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bf5c56fcc7854220bcccd3e59cf47cd4 : SUCCESS in 17m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/296b0f674ff5419fac097db80767b19a : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/744804ec8104467c90b1b252c63804b8 : SUCCESS in 20m 10s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1fd50cfc265948b982ef7d9b4e77386d : SUCCESS in 17m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ebd6da1d79784140a725238306dfce12 : FAILURE in 22m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/37ee7c8f84c84c01a3d77df91ca99ab6 : SUCCESS in 51m 14s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6eda36426d2341768e17eaaa0fda8bb4 : SUCCESS in 1h 05m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d83bce9a3926453d9a27eaca52d97593 : SUCCESS in 42m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/292b821659a84c289f5854782ce85b27 : SUCCESS in 51m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb173449184c487d91b5971e812f4d78 : SUCCESS in 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01e04454fdfb4602b8a6cf83032b5f95 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca7a997f516e41e4984709b5dddecb10 : FAILURE in 8m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fd3a7ecae8a4a79890bdadfa26d2fd2 : SUCCESS in 1h 08m 16s","accounts_in_message":[],"_revision_number":5},{"id":"312b0c3cb5f170cfe2139e98f14ac82575d4f2f2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-12 14:41:00.000000000","message":"Patch Set 5: Code-Review-1\n\n(12 comments)\n\nthe intent is laudable, but there\u0027s a couple of issues in handling slo_enabled and location","accounts_in_message":[],"_revision_number":5},{"id":"2649aa30ea0a8934215a34aec064d0fd9872046e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-13 13:47:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"890e522590572e45e987e7ce484e02a75fcde54d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-01-13 15:35:46.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"18188b5a4710eeccfdace298871b114000ffbe07","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-13 19:14:16.000000000","message":"Patch Set 6: Code-Review+2\n\n(5 comments)\n\nThanks for the cleanup!","accounts_in_message":[],"_revision_number":6},{"id":"7d63dcdef91afce63676803a4eda374dd6210bad","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-14 01:36:27.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)\n\nSubmitted https://review.opendev.org/c/openstack/swift/+/770685 and https://review.opendev.org/c/openstack/swift/+/770690 to continue cleanups, but I think this is all reasonable. Thanks again!","accounts_in_message":[],"_revision_number":6},{"id":"f88f206ab1dc234b6611616de3c63f834044421c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-14 02:08:27.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d8bc2005ca6ecf225181826b230dd24523a43df9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-14 02:10:15.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nRebased to accomodate the new tests from https://review.opendev.org/c/openstack/swift/+/770109 -- still looks good.","accounts_in_message":[],"_revision_number":7},{"id":"c00a6407b1e3b73018d1a4ed8d59608d8cf876fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 00:23:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31179249a834d378f1b8b830271783f : SUCCESS in 7m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/01d6c280f89d4059bdd7388f1031aac0 : SUCCESS in 1h 02m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2fee3522f522482c9586f4191b4dc35d : SUCCESS in 58m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0b2d6c4302e04248b0db7c63676f2758 : SUCCESS in 24m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7bdd16b10f684298b3aa5aa14cbcd170 : SUCCESS in 25m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e9b4e77859224edf82a4d5d349566cb8 : SUCCESS in 22m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cac81586eeb44ef6a70bc0799a82c8f9 : SUCCESS in 20m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6ae0f1c4fa2449339a1abe918e77b6f3 : SUCCESS in 13m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9e99ec5cce62454f9739e094c73f8f01 : SUCCESS in 13m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/827a95aaa8804a71a498262e010402e1 : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/75ed1969653343e9bb003296b5bcce80 : SUCCESS in 19m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/728c6bab2bce497f96367ae55962851f : SUCCESS in 17m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/75b0626cf1814db5b4011db96b2d16a6 : SUCCESS in 29m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9d0b3a82ddc74bb28bee6ed8900b5455 : SUCCESS in 16m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d673113248cb4a9ea77dff1ac118cfbb : SUCCESS in 18m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2f2ef221ec4f47039b2f3b9b034445e5 : SUCCESS in 20m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c6246959bf714c5f96cc674bccfe0891 : SUCCESS in 48m 47s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/258718acd1504367bae9564b1183eec3 : SUCCESS in 46m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e40d63d2305049219853e6c2e71eb7ce : SUCCESS in 1h 00m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2b7517f9b1a848e483f0063c5ba54fba : SUCCESS in 45m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc25a91cb65842f5a1402f24558b52d2 : SUCCESS in 12m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40457028645d447886d0d501a74198da : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e3b07e61fb474d738dfc8640ff8386f8 : FAILURE in 8m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/040d8da89fab4ef189fab9f9890501e4 : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":7},{"id":"aa73bb42c156ded218ec838fe06d006307635900","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 00:24:10.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ce841bc032e4a291a3c57f680d5196d926332449","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 02:39:12.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5037186e3abd4965a02f78e0ed6bfd25 : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/5f3c6ae5e72f44eb9388835f19b49889 : SUCCESS in 1h 07m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3987f88b36fd4cfcb7c5f06300cf1700 : SUCCESS in 49m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7900eb841115480a89863a2ceda61152 : SUCCESS in 11m 01s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a3312118846543d681b23b18525ce184 : SUCCESS in 10m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d661039cda68492da2880764fd517a16 : SUCCESS in 13m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cb335862aa55424eb83108a25b5eed93 : SUCCESS in 15m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8105ffbc8c38411f84291a94db7de5e0 : SUCCESS in 14m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/70227c556efc40ce851e357b28dc127f : SUCCESS in 14m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fe19ffaad6894b6992e9ef5b24a2d4c9 : SUCCESS in 19m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a02dc2870bdf45d18d868ee0980154da : SUCCESS in 17m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d596ed64f70046009ce7dc1dedf55b4a : SUCCESS in 19m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/394bc8b54fb94ed1857be31389cf5ebe : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fdaaee5a352645388516b0eeedb6afda : SUCCESS in 19m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5c332db9e26b43409e511d0d59d76353 : SUCCESS in 18m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/193bf2d6244145bbae536e23e4ba32ff : SUCCESS in 50m 12s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5955f40684aa40c1b36817e489e82a2f : SUCCESS in 42m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee96484f1e934a7fb160896fdda1bf9a : SUCCESS in 44m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f455c115450e44cd81d87cfcc59e523c : SUCCESS in 41m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/703bb9f280ae476f80a8461732544139 : SUCCESS in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8771b79635b4e74907c237c227a9cbc : SUCCESS in 6m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77f9dc00802a428eb7b9637818c2cda5 : SUCCESS in 1h 02m 41s","accounts_in_message":[],"_revision_number":7},{"id":"c1a89e51c48fb1eb340fa48db20fd8f622efca3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 02:39:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"dc9f7947ae0b40e6d3f4f022a52f1c067ab53aaf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 02:41:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f3cf25e1cbf498fb7d206820033a0fb : SUCCESS in 1m 14s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/82637013e5bc46adb3f82c4a8fa7c54d : SUCCESS in 45s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ca1ca9852bdb43e7b0bc65904b3b21ab : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"83233e7b36a451330c18eeefa19cf170eaa49abd","revisions":{"51bb90bc7dc59d253448d2ec4e1be56ac8c4e2db":{"kind":"REWORK","_number":1,"created":"2020-05-19 02:36:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/729091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/729091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/729091/1"}}},"commit":{"parents":[{"commit":"70dede1ae9b714f0eed3804022b49517a562bdd6","subject":"Merge \"versioning: Use CONTAINER_LISTING_LIMIT when doing container listings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70dede1ae9b714f0eed3804022b49517a562bdd6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-19 02:19:51.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-19 02:19:55.000000000","tz":-420},"subject":"prefactor: Pass s3api Config to S3Requests","message":"prefactor: Pass s3api Config to S3Requests\n\nThat way we don\u0027t have to plumb in some half-dozen options one-by-one.\n\nChange-Id: I4a65f50828b4e90ff6be2c3b343b295e442cc59e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51bb90bc7dc59d253448d2ec4e1be56ac8c4e2db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51bb90bc7dc59d253448d2ec4e1be56ac8c4e2db"}]},"branch":"refs/heads/master"},"9ff51f59141b41d28a97b81729239c8d580819e0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-26 22:39:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/729091/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/729091/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/729091/2"}}},"commit":{"parents":[{"commit":"7b6546b0762ab6950d2ba0d6c8f46508637f132e","subject":"Merge \"fix s3api functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b6546b0762ab6950d2ba0d6c8f46508637f132e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-19 02:19:51.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-26 22:13:07.000000000","tz":-420},"subject":"prefactor: Pass s3api Config to S3Requests","message":"prefactor: Pass s3api Config to S3Requests\n\nThat way we don\u0027t have to plumb in some half-dozen options one-by-one.\n\nChange-Id: I4a65f50828b4e90ff6be2c3b343b295e442cc59e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ff51f59141b41d28a97b81729239c8d580819e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ff51f59141b41d28a97b81729239c8d580819e0"}]},"branch":"refs/heads/master"},"b7673f5b3b297576c41d58d34284da7bb1eae59f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-08 22:25:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/729091/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/729091/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/729091/3"}}},"commit":{"parents":[{"commit":"1cab97ab7a09c6a68c11663c2147f66e92f201b7","subject":"s3api: Make UNSIGNED-PAYLOAD requests work again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cab97ab7a09c6a68c11663c2147f66e92f201b7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-19 02:19:51.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-08 22:22:50.000000000","tz":-480},"subject":"prefactor: Pass s3api Config to S3Requests","message":"prefactor: Pass s3api Config to S3Requests\n\nThat way we don\u0027t have to plumb in some half-dozen options one-by-one.\n\nChange-Id: I4a65f50828b4e90ff6be2c3b343b295e442cc59e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7673f5b3b297576c41d58d34284da7bb1eae59f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7673f5b3b297576c41d58d34284da7bb1eae59f"}]},"branch":"refs/heads/master"},"a0de3529397426b8b6b721c779f5f3fe96b31b91":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-01-08 22:54:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/729091/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/729091/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/729091/4"}}},"commit":{"parents":[{"commit":"873d0ae2185c90db11344161c7768895fa6dd2be","subject":"s3api: Make UNSIGNED-PAYLOAD requests work again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/873d0ae2185c90db11344161c7768895fa6dd2be"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-19 02:19:51.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-08 22:54:35.000000000","tz":-480},"subject":"prefactor: Pass s3api Config to S3Requests","message":"prefactor: Pass s3api Config to S3Requests\n\nThat way we don\u0027t have to plumb in some half-dozen options one-by-one.\n\nChange-Id: I4a65f50828b4e90ff6be2c3b343b295e442cc59e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0de3529397426b8b6b721c779f5f3fe96b31b91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0de3529397426b8b6b721c779f5f3fe96b31b91"}]},"branch":"refs/heads/master"},"899d24e130b1db22582967671964ad91bfb6e422":{"kind":"NO_CHANGE","_number":5,"created":"2021-01-08 22:59:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/729091/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/729091/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/729091/5"}}},"commit":{"parents":[{"commit":"09a2f93933d42342be3ee33096c4dea28b2a29f7","subject":"s3api: Make UNSIGNED-PAYLOAD requests work again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09a2f93933d42342be3ee33096c4dea28b2a29f7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-19 02:19:51.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-08 22:59:11.000000000","tz":-480},"subject":"prefactor: Pass s3api Config to S3Requests","message":"prefactor: Pass s3api Config to S3Requests\n\nThat way we don\u0027t have to plumb in some half-dozen options one-by-one.\n\nChange-Id: I4a65f50828b4e90ff6be2c3b343b295e442cc59e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/899d24e130b1db22582967671964ad91bfb6e422"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/899d24e130b1db22582967671964ad91bfb6e422"}]},"branch":"refs/heads/master"},"149c0a78038a4336487f065504c7d36a2288e498":{"kind":"REWORK","_number":6,"created":"2021-01-13 13:47:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/91/729091/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/729091/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/729091/6"}}},"commit":{"parents":[{"commit":"09a2f93933d42342be3ee33096c4dea28b2a29f7","subject":"s3api: Make UNSIGNED-PAYLOAD requests work again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09a2f93933d42342be3ee33096c4dea28b2a29f7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-19 02:19:51.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-01-13 12:31:18.000000000","tz":0},"subject":"prefactor: Pass s3api Config to S3Requests","message":"prefactor: Pass s3api Config to S3Requests\n\nThat way we don\u0027t have to plumb in some half-dozen options one-by-one.\n\nAlso, increase test coverage for s3request.py\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I4a65f50828b4e90ff6be2c3b343b295e442cc59e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149c0a78038a4336487f065504c7d36a2288e498"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149c0a78038a4336487f065504c7d36a2288e498"}]},"branch":"refs/heads/master"},"83233e7b36a451330c18eeefa19cf170eaa49abd":{"kind":"REWORK","_number":7,"created":"2021-01-14 02:08:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/729091/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/729091/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/729091/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/729091/7"}}},"commit":{"parents":[{"commit":"3cb58f8efc76f5d49c3836c4288d5a05a3678ef4","subject":"Merge \"unit test S3 SigV4Request X-Amz-Content-SHA256 handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cb58f8efc76f5d49c3836c4288d5a05a3678ef4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-19 02:19:51.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-01-14 02:08:08.000000000","tz":-480},"subject":"prefactor: Pass s3api Config to S3Requests","message":"prefactor: Pass s3api Config to S3Requests\n\nThat way we don\u0027t have to plumb in some half-dozen options one-by-one.\n\nAlso, increase test coverage for s3request.py\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I4a65f50828b4e90ff6be2c3b343b295e442cc59e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83233e7b36a451330c18eeefa19cf170eaa49abd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83233e7b36a451330c18eeefa19cf170eaa49abd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
