)]}'
{"id":"openstack%2Fswift~729778","triplet_id":"openstack%2Fswift~master~Iffbb567124096df02b04981550faec8204d5d1ec","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Iffbb567124096df02b04981550faec8204d5d1ec","subject":"Bump up timeouts for DSVM and probe test jobs","status":"MERGED","created":"2020-05-20 22:24:39.000000000","updated":"2020-05-22 17:01:10.000000000","submitted":"2020-05-22 16:57:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729778-1590166664938-2e69b6b6","meta_rev_id":"dfbc4359692c6208b6f5789816aea8224b02543e","_number":729778,"virtual_id_number":729778,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-22 16:57:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-22 14:35:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-22 14:35:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 16:57:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f254dd7f9ff7f1842366a300eb8d370660a17b5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-20 22:24:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"170334deeba0a578e41aa1607a32dc30bf431309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 23:39:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26c07c33bd884b658c6aaff7c130421e : SUCCESS in 9m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/2ec8ab8a05dd437098f00aa53cf73f88 : SUCCESS in 1h 01m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a26fa85e317347d4be0491f7f9e368e9 : SUCCESS in 50m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e778375218034ae38c47b87624cb3518 : SUCCESS in 9m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b225e31462b340ccbb77e35670bd40c1 : SUCCESS in 8m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b8b521bd0c3c42f1b230f9c2de4bbdd4 : SUCCESS in 15m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/909176dc656044cab8798cd163216c50 : SUCCESS in 14m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/afe8df7f613c44b1a100005c85a6d5a0 : SUCCESS in 14m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e585cfc6541845cd8456408f21f5049f : SUCCESS in 23m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/afc29563daaa4fcdb662d87e22dc5547 : SUCCESS in 12m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5445e906ac5145f5a622c28536790356 : SUCCESS in 19m 31s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/19eb1c982175445698da08ce5db63e09 : SUCCESS in 15m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dced850a2b764a0baa57a4aa43df1750 : SUCCESS in 20m 15s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/48618853481f4592913aa809d572ae62 : SUCCESS in 30m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/036158efd58b46108e585b70b45a6e93 : SUCCESS in 11m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fc8beac3e80243ec9d7ecfac3e2d9bae : SUCCESS in 13m 15s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/dc87345fcf144b5c97852da0282501cb : SUCCESS in 16m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/de409b28878d4cbeb22ac4d3cc791613 : SUCCESS in 17m 34s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f64cf677ac994f52b330169e2d8c98fd : SUCCESS in 18m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e785508e7b7a4fdcb5a56ef840689326 : SUCCESS in 47m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e58056d6db874c249bc0e2e0a7cda764 : TIMED_OUT in 1h 01m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/65b0af28ec184831be6429b9e5cdd204 : SUCCESS in 55m 26s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/510ae5e005ca4b2ea0fabf4eb02f991a : SUCCESS in 38m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/798eaebcacc441dda64a45399e26df80 : SUCCESS in 39m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b629fecfb37a4d539c69ca3fd1656e49 : SUCCESS in 17m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4fe757769c44f508e90622b99ceb493 : SUCCESS in 7m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6751de4b92a84fe495df1cee21368a4d : FAILURE in 8m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/568ec8e0dcb24be6abcc4554ea688650 : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":1},{"id":"821d3d0c0227ecf5242e3ed06dce34a098d35770","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-20 23:59:50.000000000","message":"Patch Set 1:\n\nrecheck\n\nMaybe I\u0027m not being aggressive enough with the probe tests :-/","accounts_in_message":[],"_revision_number":1},{"id":"a163fd6456e2335d53fb4fec7b3df2a268c997a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 01:06:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34bf1407e51b4fff8de88a38c1851159 : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/8a474165f4334b438bc6971fc1b30bce : SUCCESS in 55m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/529c5fb367674045a3587d6115e9b5aa : SUCCESS in 41m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/81cad3f2daba4c50ab513b988c7127f8 : SUCCESS in 8m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/39ab26bf573047eaac197eef34646a03 : SUCCESS in 8m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/45b971aa4ff64ab1925cd052177aa49c : SUCCESS in 16m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4c6eb1025d304961aab862ac5846fd1a : SUCCESS in 15m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2d778603066646c898d9b126ca54c504 : SUCCESS in 25m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/181429dc8ed842e28979dbbd0e25b4e6 : SUCCESS in 17m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0087f43e5e024c108cf38a0abf1285fe : SUCCESS in 23m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7c0e526bd7f04007937fd9158b042e7a : SUCCESS in 18m 18s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/036680f8f2fd4474aef2ef2dca9b5c53 : SUCCESS in 16m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b3411c03ff59445bbe92b8a21eff2b23 : SUCCESS in 13m 18s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0d090ae842234a17bda12fb0fc23ec64 : SUCCESS in 25m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/00cb5e7a74264fca89c058491d74dbf9 : SUCCESS in 32m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bbb5a297e33847ab93b500cbe39741ec : SUCCESS in 38m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e569a38cac2d415a9b69b9616585259b : SUCCESS in 34m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ac9a7edc10194e3b88f7ab53d143fa6e : SUCCESS in 34m 33s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/73509dd47132443fbc6c4d0e39b8bc7a : TIMED_OUT in 41m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bfed1754750d4ac4bddfb026a218ce0a : SUCCESS in 24m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3e087bed1e8b43ca9ccadef3d5150300 : FAILURE in 45m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/51ed51f627eb457c82bd7387b15cccf8 : SUCCESS in 37m 19s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7bdfce1fcee14a0783811d3704602260 : SUCCESS in 36m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a74a8067437b4e6d9b7c1a9ea19c5ee0 : SUCCESS in 35m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d1cbfc81c9c40dbba63666ae3a5388a : SUCCESS in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e814a31721e4c83a2277fa9353b70b9 : SUCCESS in 7m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f2ba14a3e16e481ea4a765184d6feaf9 : FAILURE in 7m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54fc14b0a2d84c0882d09ba21159e4c4 : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":1},{"id":"12ba218591ca7f3be1e82814bacc0ff665eb4964","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-21 02:47:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b9fc7408bf8d00e8d064dc4e30acf4b880bb8bd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 03:55:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd3726ac15241beb6325e0cf21b2934 : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/f7591f8718ad4659b88af9d86e913366 : SUCCESS in 1h 02m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0a0af7f2a21c4ef885424d9bfad8e230 : SUCCESS in 50m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bcf37750a6194003ae31262c21e1a8cd : SUCCESS in 8m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/87c0ef6ef4344e9e8f844b01163ff2d3 : SUCCESS in 10m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c531ee6fa07043488833338b6ac6f867 : SUCCESS in 12m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/41004e94ad16416c9291ffb08bbfccc2 : SUCCESS in 13m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1cc81cee5e9e4d22aab853114c902fdd : SUCCESS in 12m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f90f5aab42e74001a1765e0ddc5e8dc2 : SUCCESS in 16m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ec92a4462e164721bcddd302fad4255e : SUCCESS in 10m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/92591382505b49f0b3a7c56003a112a8 : SUCCESS in 16m 41s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4b83fe9f790a42d29f0b81e98548f64d : SUCCESS in 10m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9d770f8f4cb34f778c7f0e19ccd5c7cf : SUCCESS in 10m 59s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/1f4ce1e78d6e4823b58dc2b0aacf9f2b : SUCCESS in 19m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6e594509006f41e49863bdb677f1c20f : SUCCESS in 20m 20s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c136d4cd42a040ba9b98f5b43a40a418 : SUCCESS in 16m 23s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/590114525cd74aa1b4d77593f45dfe69 : SUCCESS in 11m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5197e94e8fa3491cbe79d4a984ee55ce : SUCCESS in 12m 05s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a1830f68282a4bb0bb10502bb497cd2d : SUCCESS in 20m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0688350b4d0141f8bf0df9bad3c5ba06 : SUCCESS in 21m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5b4cac43e8b3486faa05d46eb8ca9ed4 : SUCCESS in 49m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/74206168a40f496aaa3758e61898f2ba : SUCCESS in 35m 27s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/55edb11056d141219539bc3196563fd8 : SUCCESS in 35m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9d75bd846dc432cadaeeded7f981d99 : SUCCESS in 33m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58b88881c4914be1933b3e21909d472f : SUCCESS in 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e77e1aaecdb2447db18fe33ae8049332 : SUCCESS in 6m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b16367734c1547c1894c0155a5ca82c1 : FAILURE in 9m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3768116ea844baba3d4c8b8c508fead : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ce2529a10e0d6a08a3377f70e5653282e334266f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-22 14:35:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a5c2e1d6178f07ed2a59c1a30e7da169d0b5c0ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 14:36:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0283b39be28e6eb6a5a09dd62318a3a8ec43e338","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 16:57:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13f22b5245e04f4b9424ff914a8d5748 : SUCCESS in 11m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/e25c0fcfc6c94a498c9aa5dbf6dd3153 : SUCCESS in 1h 14m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9335d02855744700b3b6227593460309 : SUCCESS in 47m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7a9e3a89d3ef4a5bb65e6c8e4e62a79e : SUCCESS in 13m 36s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ab353ae8be894743a031c89244e41dcf : SUCCESS in 12m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3eb94f2811614837a8dd929635fa7df5 : SUCCESS in 26m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e57dee39494746199e151e1ec7cecbb9 : SUCCESS in 15m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1049bd9c679449729d3fadccea54f0ad : SUCCESS in 17m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c0872f6ffab04a80a39eeb3ca6f6d3b7 : SUCCESS in 21m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/21ecc00836fe4f90b2945c6c7a86c7b9 : SUCCESS in 12m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/30ed27a30d1c485a9cb533761754cb5b : SUCCESS in 17m 36s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/82202fba21d44615afb0e6512608e8ea : SUCCESS in 15m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/63c5441c7e5548099481d8f96f233d3a : SUCCESS in 18m 37s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/40a334a3845a43b6a614656e699df683 : SUCCESS in 41m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/08b89d8c2d6d4479872a9f5b78bca024 : SUCCESS in 14m 20s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5c8435d223604ac284571bcd58e41ea6 : SUCCESS in 13m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/889fb0dfc022451aa4501a7c31e06a76 : SUCCESS in 13m 18s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1e5f133f8881483199db16282127e374 : SUCCESS in 20m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3f9162728d68475e9d9343a54b547019 : SUCCESS in 53m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a890fff132ce4da7ba67f142a4b7a20a : SUCCESS in 42m 40s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8a3e84940cbe4a819d5ee56dcf04bd6e : SUCCESS in 46m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cffa148e8a214277aae06fac09c7c549 : SUCCESS in 42m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4804417b5fa342bb9433a033918d8acd : SUCCESS in 27m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/468c1dca91b74a8baa2294cf2e8f9369 : SUCCESS in 6m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32608e378fa240a5950d8e3087419403 : SUCCESS in 1h 05m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a8a9c33f088fc4989366245526d872cd1035a90d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 16:57:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dfbc4359692c6208b6f5789816aea8224b02543e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 17:01:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd78048c3544480ab9118c86e72e88f5 : SUCCESS in 1m 23s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/31664885c7534850bad9826935a2f5ce : SUCCESS in 1m 24s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/68c2a874ce93460b80dbac57f978fc0b : SUCCESS in 1m 09s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"67598b3a4a6f841726b7358245ae18edbfc58250","revisions":{"67598b3a4a6f841726b7358245ae18edbfc58250":{"kind":"REWORK","_number":1,"created":"2020-05-20 22:24:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/729778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/729778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/729778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/729778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/729778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/729778/1"}}},"commit":{"parents":[{"commit":"70dede1ae9b714f0eed3804022b49517a562bdd6","subject":"Merge \"versioning: Use CONTAINER_LISTING_LIMIT when doing container listings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70dede1ae9b714f0eed3804022b49517a562bdd6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-20 22:22:10.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-20 22:22:12.000000000","tz":-420},"subject":"Bump up timeouts for DSVM and probe test jobs","message":"Bump up timeouts for DSVM and probe test jobs\n\nDSVM recently got a bunch more middlewares enabled, so it\u0027s running more\ntests than it used to.\n\nI can\u0027t think of much that\u0027s changed for probe tests, but I feel like\nI\u0027ve seen it popping timeouts more often lately.\n\nNote that the new timeouts are still lower than the typical run-time of\nour longest-running jobs (currently grenade / tempest-ipv6-only).\n\nChange-Id: Iffbb567124096df02b04981550faec8204d5d1ec\nRelated-Change: I3cbbcd2ea9ced0923bee4a6b0783e4cf5e82e95b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67598b3a4a6f841726b7358245ae18edbfc58250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67598b3a4a6f841726b7358245ae18edbfc58250"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
