)]}'
{"id":"openstack%2Fswift~730414","triplet_id":"openstack%2Fswift~master~If69dbf2a4c876fa4b830f6d8652f22f39ceaf549","project":"openstack/swift","branch":"master","topic":"versoin-if-match","hashtags":[],"change_id":"If69dbf2a4c876fa4b830f6d8652f22f39ceaf549","subject":"Test for Versioned Object If-Match","status":"MERGED","created":"2020-05-22 20:10:38.000000000","updated":"2020-05-26 18:06:40.000000000","submitted":"2020-05-26 18:04:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730414-1590516269545-69ea3769","meta_rev_id":"f73b0a3ea2771dde29e8b3218f8241985c567194","_number":730414,"virtual_id_number":730414,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 18:04:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-26 16:06:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-26 16:06:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 16:06:02.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-05-26 18:04:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ad257a1a8adbeff776298a551f4f749c67b4691","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-22 20:10:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e674c6687e083fb334afa2ef87bcb5e683d1fe3e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-22 20:12:33.000000000","message":"Patch Set 1:\n\nI came into this trying to think of reasons to pass through headers to subrequests in https://review.opendev.org/#/c/724393/2 - but it looked like this was already broken on master?\n\nwhat\u0027s weird is that both static and dynamic symlinks both already seem to support if-match !?\n\nI\u0027m going work on writing up the bug and see if maybe I\u0027m just doing something wrong","accounts_in_message":[],"_revision_number":1},{"id":"84575339fb5f6b0e1dcb4c41ded49600fa88c50c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-22 20:16:24.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI have this passing on my vsaio...","accounts_in_message":[],"_revision_number":1},{"id":"0fa0627d724d6dd82a5e338c9844e603dba76206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 20:55:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14f0a496640741f180ceab41cb0954ef : SUCCESS in 12m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/06f8935a68d142da96969664e4609814 : SUCCESS in 12m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/148fb7baff8b48959201527104d02825 : SUCCESS in 13m 43s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/262c34ee50fd46fbac3617963cd4a63a : SUCCESS in 11m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6adf1e5fb7c645a3a63c025736e7ab0b : SUCCESS in 14m 15s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/2f7582e1ac044dceba23e5e9fbbdfaa1 : SUCCESS in 27m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c1ba5aec7812474885ade8a130e75252 : SUCCESS in 20m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/69655dc72ab448d2acf7fd15c597c013 : SUCCESS in 21m 11s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9fbcdfdeab864b19a22cd8ed407a4349 : SUCCESS in 11m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f3ccb2b57ef64d1d887e74ab2b84a70b : SUCCESS in 11m 09s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4d4460ab7d7a410f9193b0ecbe6f4a35 : SUCCESS in 18m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/88940c4bae4e4b4481c1f1d0dbfc680d : SUCCESS in 22m 36s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/586b58517c8e4f5798dcd2934abb2857 : SUCCESS in 35m 47s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c235913583194397bfe5bb442777900a : SUCCESS in 42m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/51934ae5ccd347e6b0b9bb8924e6b28e : SUCCESS in 37m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0242fcc07dd24e12b8e32495955ad07f : SUCCESS in 6m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4337329f1d5049588cf8f550b213ef56 : FAILURE in 8m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"094d807557089afd239927b0ae56573efb8151e4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-05-26 13:49:24.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e48c5c35a09cbef82ecded693e1eb64461f5916f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-26 16:06:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nWe should probably add a similar test with Range...","accounts_in_message":[],"_revision_number":2},{"id":"5ac75c525f8c33b47f635b18da5dfb21436f820f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 16:18:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d55f2a63db4194aec67c77fe158b46 : SUCCESS in 11m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/19682c5ca3d947efb684a05a3689938c : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9fb073538c2e4655b1b8a37b3cdc63d4 : SUCCESS in 17m 54s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/7e6d1207d281437a991eda53a4dd918f : SUCCESS in 14m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a5d511309c14b7083a54588e6b70058 : SUCCESS in 10m 51s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/a7dc98e19ba94fdcb2cc64008f56d657 : SUCCESS in 18m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/505de6d009634d0098c7db8592af8221 : SUCCESS in 12m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/76b9778e1b354e428ac666d600b44f64 : SUCCESS in 20m 23s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3e3022e387ad4a20b7d25a1ba6f559be : SUCCESS in 16m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b4e1443abdd34ceba4092a9874e50d51 : SUCCESS in 22m 20s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b183f5f680e5404d99c0c071effef409 : SUCCESS in 22m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ed3f8b3b2c254580bc4c33f9d0055c1a : SUCCESS in 21m 31s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b104739783c149d0bd1124e2655d3b2c : SUCCESS in 1h 00m 10s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e7f24fa692db49fc8120de18b7d1e970 : SUCCESS in 1h 00m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b7585493f2934b0b83f456212158b028 : SUCCESS in 42m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/541dd48022b041538d9f48c637eb6ec7 : SUCCESS in 7m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ef0ee98ad38b4aa6824f14d5e24dbfdc : FAILURE in 14m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1305cb3142c246985b2056a470d6dfb29c0f5632","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 16:18:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76508a7f0a710f47ed36ba549bd7e482ae007dd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 18:04:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c18bb16d34e14a0aaf47ae7e614d5ff4 : SUCCESS in 11m 41s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e4fdada2160c45d3977460158e134f9a : SUCCESS in 12m 35s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/5db912610c294f29948893d5aa4e45c8 : SUCCESS in 16m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/370ef402621f44eb9d64919ff7a15411 : SUCCESS in 18m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b6ce867899b840e1aec85b399d1792ab : SUCCESS in 22m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/40e084faba56437a967c27db192bfebd : SUCCESS in 19m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9ae5578610ee4b1ead1bf43b125ea7f6 : SUCCESS in 29m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/697f54e155a94f389c10e882c73180c8 : SUCCESS in 18m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cd3d9c42caf6436eaa5857f909a7f109 : SUCCESS in 19m 27s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/7760c8ef451e42d081ccd82e864ae661 : SUCCESS in 14m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/971f81238a5d4749bd954ba81ebc21a4 : SUCCESS in 13m 42s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c0969e7ebeca4f25a69ba9e29c57653b : SUCCESS in 18m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6492abb403814a3b9167b5c7b00b48a0 : SUCCESS in 14m 38s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/82c0efac4a0745f3a2d0a918423a301b : SUCCESS in 11m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ced00be7c14f4d4d8d34be8c93b757c6 : SUCCESS in 11m 37s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/374fa526bb024ac8b1801ef032af867f : SUCCESS in 17m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f99a48e32e9a4f2695115ce7df2bd6c1 : SUCCESS in 43m 37s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b46bffa372184d52b6728ed5c40c06be : SUCCESS in 37m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6b23ac32a4394f5480400a3da0af25c2 : SUCCESS in 37m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee7ef0aad3f5439da6d422e811a2ad3e : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":2},{"id":"80a1ff35964e19f7589fb715aad026c99c8bd840","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 18:04:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f0c9a65b331a949a502d184b59daaf29ea45fd53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 18:06:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5dcf4e9ea344aaa5e80627849a3f44 : SUCCESS in 1m 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2eff9358186c4b69a8ff1c71eba96907 : SUCCESS in 1m 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1cf092ec6ba049e39ba92cb81f158249 : SUCCESS in 1m 38s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"63e02fa9fa4e43cb20dd96eee81d0749edcca7e6","revisions":{"00740d465d5b018cb6ebd318be4d92a216a3730a":{"kind":"REWORK","_number":1,"created":"2020-05-22 20:10:38.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/14/730414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/730414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/730414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/730414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/730414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/730414/1"}}},"commit":{"parents":[{"commit":"67598b3a4a6f841726b7358245ae18edbfc58250","subject":"Bump up timeouts for DSVM and probe test jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67598b3a4a6f841726b7358245ae18edbfc58250"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-05-22 20:00:59.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-05-22 20:05:24.000000000","tz":-300},"subject":"WIP: Fix Versioned Object If-Match","message":"WIP: Fix Versioned Object If-Match\n\nRegular objects can do conditional requests, versioned objects should be\nable to as well.\n\nChange-Id: If69dbf2a4c876fa4b830f6d8652f22f39ceaf549\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00740d465d5b018cb6ebd318be4d92a216a3730a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00740d465d5b018cb6ebd318be4d92a216a3730a"}]},"branch":"refs/heads/master"},"63e02fa9fa4e43cb20dd96eee81d0749edcca7e6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-26 13:49:24.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/14/730414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/730414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/730414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/730414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/730414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/730414/2"}}},"commit":{"parents":[{"commit":"67598b3a4a6f841726b7358245ae18edbfc58250","subject":"Bump up timeouts for DSVM and probe test jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67598b3a4a6f841726b7358245ae18edbfc58250"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-05-22 20:00:59.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-05-26 13:49:15.000000000","tz":0},"subject":"Test for Versioned Object If-Match","message":"Test for Versioned Object If-Match\n\nRegular objects can do conditional requests, versioned objects should be\nable to as well.\n\nChange-Id: If69dbf2a4c876fa4b830f6d8652f22f39ceaf549\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63e02fa9fa4e43cb20dd96eee81d0749edcca7e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63e02fa9fa4e43cb20dd96eee81d0749edcca7e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
