)]}'
{"id":"openstack%2Fswift~730957","triplet_id":"openstack%2Fswift~master~Idcc33c71869ccf7ad5a3134767d36ae94aab4598","project":"openstack/swift","branch":"master","topic":"kitchen-sink","hashtags":[],"change_id":"Idcc33c71869ccf7ad5a3134767d36ae94aab4598","subject":"dsvm: Run domain_remap tests with tempauth","status":"MERGED","created":"2020-05-26 19:55:45.000000000","updated":"2020-05-28 00:27:38.000000000","submitted":"2020-05-28 00:27:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730957-1590625658595-41f9ad85","meta_rev_id":"e428f436bd72cfb8668091f8f0afc8d1818e3940","_number":730957,"virtual_id_number":730957,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 00:27:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-27 21:38:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-27 21:38:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 00:27:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"503b2a3a57911fd14ff76ee9306f929be581d207","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-26 19:55:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba3b36224722f73f02fc88b866569bdbc96b6277","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 22:31:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ae78083dcf429f8b73e414e3a6e89b : SUCCESS in 11m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/81d07f73559c4038a682e6cc0c53803a : SUCCESS in 1h 00m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/836383b0b50344bc9b93153344120a0f : SUCCESS in 1h 21m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5d8a80b9ef17461f91509603c607e4b5 : SUCCESS in 14m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7ae143cb08ba4f53bd4fa3282ec5d920 : SUCCESS in 14m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a8b6d1f3ea674c31ab809cf0ef4dc5ab : SUCCESS in 27m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4547fc7780d24045aeea1c595aa20a21 : SUCCESS in 18m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fa6adfbf577c4ac48bec81a12642f87b : SUCCESS in 28m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/079ff2b3a0d7453c98270c6cee237d80 : SUCCESS in 25m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8fee4cb1aa914a2a9b01af7125937477 : SUCCESS in 19m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/60775e42dd7443c0bed1272fb4b285d2 : SUCCESS in 32m 59s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b65718c4617c461b9ae559fef85afcf4 : SUCCESS in 22m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3b263d58c4d24afea2ab9ad7055b1895 : SUCCESS in 30m 17s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/433b87e2355f432c9775c3bc8a1e36b2 : SUCCESS in 20m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cd64f9c1102a46568de3e74633d39efc : SUCCESS in 11m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6d47cf50876c4e6cb42893b9b9c20a5b : SUCCESS in 15m 08s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5edaa6531a394fe0b1cde07b0e77622e : SUCCESS in 12m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6cc91324cf8445bda54748e7b60f910e : SUCCESS in 13m 15s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4a5f2d44998a45668157c2c4da733d6d : SUCCESS in 19m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd996af0e680476891444d35b1eb912a : SUCCESS in 21m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/257f30b999c34ad49a55c1db70a7c4cb : SUCCESS in 48m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/54bf80c7f79843f1906409badd56309e : SUCCESS in 42m 32s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f159ed9d542b4e5dae55e9e0a105bf74 : SUCCESS in 45m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/357206c10d4748b9af14e780aa7d6d24 : SUCCESS in 46m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97db7bd3568e4c5ba87be992194efb42 : SUCCESS in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a41a0d27b2449aca345b7ce878160dd : SUCCESS in 8m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9da89bb2657348b6ab54cd6a796d8a61 : FAILURE in 18m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f98321f9553740e2b45059169191096f : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7d20d8d652d65b95b56cd592fba94585edb43a8d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-27 21:38:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"11af56acf673cf695bb4157eaf7711c35b79ce55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 21:39:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64e7bbfe65413c7792828b72068c8e70ff235644","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 00:27:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"58f30e5f65ba73bd89654ad2baf9ae8ec5215d56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 00:27:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98336c950ee741758bd4308e45549757 : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/e639a25dfcc04e21ae81ac0b867236c6 : SUCCESS in 59m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4a91fd33321c4550ac412aeccb6a3605 : SUCCESS in 45m 01s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5a542831611846be8cf6f69c959a0dd6 : SUCCESS in 14m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/9db3f84cce6f4a97af6359da2201c2af : SUCCESS in 12m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/612535ee52f94ce39c173e23ad9f36e3 : SUCCESS in 15m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2b7081b5b6914bf68410871ef251bcf8 : SUCCESS in 15m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/12f763d3bf9a4507b15775580cdaa498 : SUCCESS in 20m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8c48a42c22da4bdc81e07d9614da8f7c : SUCCESS in 17m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6c22979674be4219b03bb1165fb76689 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e9df914a0e7e4b2b8616242a85106337 : SUCCESS in 14m 10s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/95c796644738410d9e2a7d6b53c470db : SUCCESS in 13m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a5abeaab95264653a9ff562d065650ac : SUCCESS in 21m 42s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/feac43e0f90948689c702eaca9310be1 : SUCCESS in 27m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/72c6c062f2204062b38c704c55224f66 : SUCCESS in 23m 45s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4e2a295fe22b42619f9d4542af36b6da : SUCCESS in 16m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9fceaf0306ea441e887a3ce45af2f953 : SUCCESS in 17m 35s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/beb7ad7ac7544ca48aac1746fc1e40b2 : SUCCESS in 21m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/50dffd23f21149179f51fee0989d97c4 : SUCCESS in 47m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/63ef64e8e66247c895617de5e3f79b7a : SUCCESS in 48m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/71bd6353fcde4806b8c6ee06d278de2c : SUCCESS in 47m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b1d3e5e48e840fb9e0eea75ec53b87e : SUCCESS in 13m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/463540fd7bc64437b2cdf3ce6d9ec7ef : SUCCESS in 10m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36352c090b8e44c598966275238de8ca : SUCCESS in 52m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63b16a368366d7c1fb6976841ce28a914128ce6a","revisions":{"63b16a368366d7c1fb6976841ce28a914128ce6a":{"kind":"REWORK","_number":1,"created":"2020-05-26 19:55:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/730957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/730957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/730957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/730957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/730957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/730957/1"}}},"commit":{"parents":[{"commit":"bb9b0326fde08768e6d609a210a1d1a5ec1c32ff","subject":"swift-dsvm: Enable s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb9b0326fde08768e6d609a210a1d1a5ec1c32ff"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-26 19:41:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-26 19:41:03.000000000","tz":-420},"subject":"dsvm: Run domain_remap tests with tempauth","message":"dsvm: Run domain_remap tests with tempauth\n\nChange-Id: Idcc33c71869ccf7ad5a3134767d36ae94aab4598\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63b16a368366d7c1fb6976841ce28a914128ce6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63b16a368366d7c1fb6976841ce28a914128ce6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
