)]}'
{"id":"openstack%2Fswift~731318","triplet_id":"openstack%2Fswift~master~Ie285f513e1ed71dbaf4e12fe747ea6087664f843","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ie285f513e1ed71dbaf4e12fe747ea6087664f843","subject":"Remove swift-dsvm-functional-py3 job","status":"MERGED","created":"2020-05-27 20:17:09.000000000","updated":"2020-05-28 00:30:19.000000000","submitted":"2020-05-28 00:27:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731318-1590625652681-65991141","meta_rev_id":"39177a8ca1877b9372fb2870e0d053e1d444e478","_number":731318,"virtual_id_number":731318,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 00:27:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-27 21:38:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-27 21:38:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 00:27:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df04b13c77d677ba75ac7f8c8c0f6b556e76b4f4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-27 20:17:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1e7f32ab50bfa1311e2b93a5d81f3655d2faabd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 21:24:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c5dc76903ee4f77b2fb438f547eaecb : SUCCESS in 11m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/65916ac22afa4db7a9c1162c2df6cfd2 : SUCCESS in 59m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8d6d6bd9594f4fa9af33f752330ecfdb : SUCCESS in 51m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1029f703eb6141efbad11c4c57529927 : SUCCESS in 15m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/21fa64bc62064b60867ab1c75f17a227 : SUCCESS in 11m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c2a172e3381e4a7683204d013ee7ac1e : SUCCESS in 14m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/21b32a69d6d347a9850b5d5e94afe8b9 : SUCCESS in 14m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/58c255630caf46828bce87dcaa026cd4 : SUCCESS in 14m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c69f11fd5af844ada2ad2691044c6726 : SUCCESS in 17m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/abdb49bc049f4c6fbf03f6d0f362ef17 : SUCCESS in 12m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3b5994a7c599433d8b5f7a607c3c6fc9 : SUCCESS in 15m 29s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/9e5a1faae3b84c288f3cc5ac22898f6d : SUCCESS in 19m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ccb2662be96b4e659bbf1dc56ae67e41 : SUCCESS in 12m 58s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c600fe93a44046a89b5e7bff161c5513 : SUCCESS in 17m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9a67f5f68fd14fc382ebd826a68611c1 : SUCCESS in 12m 34s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/406c243ab4e94ef48f2ee2821286a972 : SUCCESS in 20m 11s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/21f12a3b0b9741b69617ae4f84523775 : SUCCESS in 12m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/644848325dc44744a44cde0ee6b9666d : SUCCESS in 13m 26s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/cf71a4b97bc545bdbf03febd47052fb8 : SUCCESS in 17m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d2be2ac32cba461482bc84dcac41c916 : SUCCESS in 19m 41s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/145d490bb6774b27a8e55a20bbfb35c9 : SUCCESS in 42m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aca74020264e41f6a9e0b356b53436d3 : SUCCESS in 44m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/781b16830de648e290ddf6154cb654d8 : SUCCESS in 36m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04672dd4c418459aab19ab6b0e4dae15 : SUCCESS in 16m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93f68e29cd714b399a41aa996d67676e : SUCCESS in 7m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/191f0bbc6f4642de84f444ccfb88be36 : FAILURE in 13m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9d1f5dee2254e098142c24488eaf309 : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":1},{"id":"74591f2ff5a4d1224cfa16accb4c298230894095","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-27 21:38:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7346a58c035018928e3467f72cd0b73797bc8f98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 21:39:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7e764f525c87e516f34e84bb36b0b7e0757bd8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 00:27:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65ba70f7c6445cea681a5926e12397f : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/08fbe79510cf49099b2251b189ae05f9 : SUCCESS in 41m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0d18c0842794453581854fd7f3724caa : SUCCESS in 48m 11s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2457772e0d0d46e1bd9893e0f70949be : SUCCESS in 12m 16s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/42453b90f1f14db88fe03b5772683943 : POST_FAILURE in 16m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d4db9482a4f1431c83f6b44803fe191f : SUCCESS in 19m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fc9cd57e939c474fa4f50640b99c194e : SUCCESS in 21m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/45a3d46c8a814797b8c3c2f55c265ce5 : SUCCESS in 21m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/260b3250768240ee94903f3952e30e60 : SUCCESS in 21m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8f7eb530926847a2bdb5517708cfe425 : SUCCESS in 18m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3cb13c5953364e1dbeb7579f236819d8 : SUCCESS in 15m 17s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f769095fb1894f6d8a8d7c4b4ce44222 : SUCCESS in 12m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ead303c8cff644ca83c0fb7c09b6db21 : SUCCESS in 12m 23s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e79201f7b118447e9c0abafbea42d687 : SUCCESS in 20m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/42415e07297047ef8660828c12d4c017 : SUCCESS in 12m 31s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/044da5e47c48430d8207984e25db858d : SUCCESS in 12m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/714c408e302a405ebff552e28bd555dd : SUCCESS in 11m 48s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1ba2badc32fe4898b2f35f357f683b2e : SUCCESS in 23m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d54493ff55a54670b2d5e6d4314e9c21 : SUCCESS in 48m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3af7730d2c8a4765a4c07804bc185374 : SUCCESS in 39m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f843a55dd5b4c6b9473702b97d65432 : SUCCESS in 54m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/157fa017cc4d49f1b496c5b01dc59614 : SUCCESS in 24m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3b7bec13cb64d80aca802a666fe0e97 : SUCCESS in 7m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c3436b82d2c4fc8b968da98873b9cbf : SUCCESS in 1h 03m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4f1eddb2c403b00544d997418bdc6cef6dc3ee41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 00:27:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"39177a8ca1877b9372fb2870e0d053e1d444e478","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 00:30:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f43a0e1e99e461d9b2d022dd0072b86 : SUCCESS in 1m 17s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c58b9046c7ea4f969b8e27c3d8b41637 : SUCCESS in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/59a226ef1e514b1f93554a1c8fd05f0a : SUCCESS in 51s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3aa4e3ec85fb6d8686bdadcb455817bb3b80543e","revisions":{"3aa4e3ec85fb6d8686bdadcb455817bb3b80543e":{"kind":"REWORK","_number":1,"created":"2020-05-27 20:17:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/731318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/731318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/731318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/731318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/731318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/731318/1"}}},"commit":{"parents":[{"commit":"650272eacb6a71d24ba690599c48e710bf8993f9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/650272eacb6a71d24ba690599c48e710bf8993f9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-27 20:16:55.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-27 20:16:55.000000000","tz":-420},"subject":"Remove swift-dsvm-functional-py3 job","message":"Remove swift-dsvm-functional-py3 job\n\nFollowing https://github.com/openstack/devstack/commit/6b6bdc711,\nthe swift-dsvm-functional job has been running everything under\nPython 3 anyway.\n\nChange-Id: Ie285f513e1ed71dbaf4e12fe747ea6087664f843\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aa4e3ec85fb6d8686bdadcb455817bb3b80543e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aa4e3ec85fb6d8686bdadcb455817bb3b80543e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
