)]}'
{"id":"openstack%2Fswift~731358","triplet_id":"openstack%2Fswift~feature%2Flosf~I91c72a7a98468497ab323a920e9573fcec765b66","project":"openstack/swift","branch":"feature/losf","topic":"finalize_durable_losf","hashtags":[],"change_id":"I91c72a7a98468497ab323a920e9573fcec765b66","subject":"obj: _finalize_durable may succeed even when data file is missing (LOSF)","status":"MERGED","created":"2020-05-27 21:53:46.000000000","updated":"2020-05-28 00:54:09.000000000","submitted":"2020-05-28 00:51:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731358-1590627103145-3a602513","meta_rev_id":"bef021fea74a3b249cf16e0235ee96ba50d1d08a","_number":731358,"virtual_id_number":731358,"owner":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 00:51:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-27 21:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-27 21:59:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 21:59:39.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-05-28 00:51:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a20bc9e7aa95089edb6e916378032b44fc4f4068","author":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"date":"2020-05-27 21:53:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b488cb973d00566bc0605a5675fc148963f90e04","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-27 21:59:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nYeha, that looks right","accounts_in_message":[],"_revision_number":1},{"id":"879b661bc280bbf28a1413e4ca446524c2353031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 23:00:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56b83ddbe52e4d88ab985617da667af2 : SUCCESS in 9m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec3fb4597cb4423887836914269ca83c : SUCCESS in 9m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/32aa1323ed634032aca152a66dc80d87 : SUCCESS in 11m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f52ac3cc0bea43dab02a452a207066ac : SUCCESS in 13m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b46a0567926e4afcb47cbe3da07d37fd : SUCCESS in 15m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3b5121e911544c5aa76bad0c67438e66 : SUCCESS in 15m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8bd296fce7414739abb1edf4cf8449a0 : SUCCESS in 14m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9ef5054c3f834e3c93ba4ab56ba75ff1 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3602cc29c4554a52a97b8d8b02318375 : SUCCESS in 13m 32s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c5977c7d78fd415f99c05d5b846f2d69 : SUCCESS in 11m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/547b02b22a0045ad99b6c0951ec19984 : SUCCESS in 11m 32s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/aace40e7e714431a8c448a92bc5d1152 : SUCCESS in 16m 53s\n- swift-tox-func-losf-py27 https://zuul.opendev.org/t/openstack/build/97f6a699034e4240b45d408cc385b3bc : RETRY_LIMIT in 8m 16s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/57035c566da44ecabb17008652cf23b4 : SUCCESS in 10m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d1ed66901ac2454ca697b968053757ce : SUCCESS in 14m 25s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/62282787354c46b9a86faf2e46807763 : SUCCESS in 11m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/78eca573d086443597bcbfddd434a55d : SUCCESS in 13m 29s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2e50e5759f184ae59cd201d334e6784e : SUCCESS in 20m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/264e9fc805a14d58b01a5d26bff54020 : SUCCESS in 22m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/65d75b5e91c5461c8523b62fb6b4f6ae : SUCCESS in 52m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/48cb7e92f25a499880967c4b766a3016 : SUCCESS in 39m 52s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e56a96b08bd6467fbf1a0d762827faa5 : SUCCESS in 41m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ef0cac43c18f44029067238ec272a786 : SUCCESS in 39m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18e5d53532034d2c92bb5cfef19a0b16 : SUCCESS in 18m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83f2bc3dfb5a4adba0a8a521a94dd210 : SUCCESS in 6m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/27bc36f867d9485b90ac1f8948bb2494 : FAILURE in 9m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8f21a6d2900e2f27acd1663c01b6971e034fc09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 23:01:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"679cbc9507803f466a847d202e75bce3bc6e36fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 00:51:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/435edc89cd1e41b586f7d9f20dfcded1 : SUCCESS in 8m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ad5c92b6360043a09d606612d52179b0 : SUCCESS in 9m 32s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2cf82ba9cb224bec9c2ff148220bd1c3 : SUCCESS in 10m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/12b6bfca4db04660bda26bd88fac0dbd : SUCCESS in 12m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8142ce12cd864c01aeaf28e0a7d8e359 : SUCCESS in 13m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2f3f17172897429bb8d01a6a85a91b6c : SUCCESS in 13m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2f067c233c424b7e9fdc6b085d5f17ce : SUCCESS in 13m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cb8fc7930100458eb760753e9a78c80b : SUCCESS in 11m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2345b39cc21643b696f5991010d36820 : SUCCESS in 12m 46s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/22d7f12be13b4826a55c3b19559b2fb4 : SUCCESS in 10m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/489481d8be39479397e74d4d95ce58b9 : SUCCESS in 12m 16s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ac9be28acede47c0a56751c4cee67efa : SUCCESS in 15m 57s\n- swift-tox-func-losf-py27 https://zuul.opendev.org/t/openstack/build/7885c2acf93e4a238382ff900d57d6d8 : RETRY_LIMIT in 6m 05s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0207de468ac74927b00c030c4edf3014 : SUCCESS in 10m 46s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f9176508da8446259cc6a242339e7f3d : SUCCESS in 11m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e936f441e7564236b89bdf712ef575b4 : SUCCESS in 12m 29s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/0c7ceb59d11a428b9c885f64d264971d : SUCCESS in 14m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d8efebb453c342b19dcee0d10fec333d : SUCCESS in 49m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/baf6ea78dc5c4f78a11194307af45e6f : SUCCESS in 35m 08s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ebf8081bf24249e687dde40aaf16a736 : SUCCESS in 36m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4a190fc7d5e34bd2875cd0fe1ddd6e21 : SUCCESS in 44m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04cb1d8e8f1444a5aefa92bfca900b64 : SUCCESS in 12m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/951a09337427473cbf4e2c1a7aec994e : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a8540d9c1adede1a1aa59e296b179c3c45f1ee87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 00:51:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a6870a2cda04f2e4e02c026940f1486357c30cc9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 00:54:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/019c15953f894f29862f1d769dbf9af1 : SUCCESS in 1m 38s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/36cef6819220445b9fd6f136ff5e9bf4 : SUCCESS in 55s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/aaefb87fbb174fc5a1b6c3b85f1d4e29 : SUCCESS in 1m 06s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3fd0bd9d82160305a821e742b2cd968036911b2","revisions":{"b3fd0bd9d82160305a821e742b2cd968036911b2":{"kind":"REWORK","_number":1,"created":"2020-05-27 21:53:46.000000000","uploader":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"ref":"refs/changes/58/731358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/731358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/731358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/731358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/731358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/731358/1"}}},"commit":{"parents":[{"commit":"414ca6bb6c7ffe12a1f7c4f65526bb0bf98b4b49","subject":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into merge-master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/414ca6bb6c7ffe12a1f7c4f65526bb0bf98b4b49"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2020-05-25 13:36:15.000000000","tz":120},"committer":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2020-05-27 21:52:09.000000000","tz":120},"subject":"obj: _finalize_durable may succeed even when data file is missing (LOSF)","message":"obj: _finalize_durable may succeed even when data file is missing (LOSF)\n\nBring 69b8165cd87693f077a2385d9b6d53a7f459b098 to LOSF\n\nChange-Id: I91c72a7a98468497ab323a920e9573fcec765b66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3fd0bd9d82160305a821e742b2cd968036911b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3fd0bd9d82160305a821e742b2cd968036911b2"}]},"branch":"refs/heads/feature/losf"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
