)]}'
{"id":"openstack%2Fswift~731362","triplet_id":"openstack%2Fswift~master~I5b0ae0b78d37a31928e0b34a1b6118802a3a8236","project":"openstack/swift","branch":"master","topic":"kitchen-sink","hashtags":[],"change_id":"I5b0ae0b78d37a31928e0b34a1b6118802a3a8236","subject":"dsvm: Run service-user tests under keystoneauth","status":"MERGED","created":"2020-05-27 22:08:07.000000000","updated":"2020-05-30 01:08:51.000000000","submitted":"2020-05-30 01:07:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731362-1590800854883-60d2867d","meta_rev_id":"03c8135def38c96f960139b5efba017f40307fcb","_number":731362,"virtual_id_number":731362,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-30 01:07:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-27 23:26:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-27 23:26:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-30 01:07:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1440975e3dccfb2fe93eceaae46f9a4a8ad4b521","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-27 22:08:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5326cedee64096a5db1962a10f293b231398b533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 23:19:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f2a21a91213460085c3e47501aa821e : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/c4b145eadd1c4c8ca3c763f200817828 : SUCCESS in 56m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6477b5d69b774b449d5f991ce80c0980 : SUCCESS in 49m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fc5d9dd61f334be1903804bc95292fb9 : SUCCESS in 9m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a30466578649424fa6ef9849fd014886 : SUCCESS in 12m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3022c62bc51a40a8939b6c69d57808dd : SUCCESS in 14m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e8c2ec036c4649a695de61b097bd4290 : SUCCESS in 18m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cc630fca258a42cba9f329ef2fa1eacf : SUCCESS in 32m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1d57e19bbe3140f7b402e6321da93e61 : SUCCESS in 35m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5ec5c2da862143b5a5aafa9e553b114a : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d610e25d8310424b90e8c7d6bb725bf0 : SUCCESS in 21m 21s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/c90f3af2a8c34daba2d22b3e816a648a : SUCCESS in 13m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1ee51a0cf2594b98b00f1db9289bda13 : SUCCESS in 33m 20s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/0ffb4f5e22a14950be2b9221e1052bf0 : SUCCESS in 19m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/99fc67eccd0047cc8b74b29f1676c2cf : SUCCESS in 12m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2411a56793244d2a80fe392c590cb5b4 : SUCCESS in 15m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f775a1650d544c0fb28f070807a9848d : SUCCESS in 14m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/35fa21001dde4553973b5030cc46746a : SUCCESS in 13m 25s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8471ff63694946459526f76596ed2f5c : SUCCESS in 16m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/06d0e05a6a5c465d8c6406fc3c8b281d : SUCCESS in 35m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2c3ad83aed5b4f2280a436ef02154f9e : TIMED_OUT in 1h 01m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e6233f2e475549dda675f77cbab5ccbf : TIMED_OUT in 1h 03m 35s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/bace242c73584bbf98ad33b08b969707 : SUCCESS in 53m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/542c35c3ecf949c99c9b3b98a1771b1c : SUCCESS in 50m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ceaebce191354077867668f3697995b2 : SUCCESS in 34m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a067918ce134364bbc983f2ae766344 : SUCCESS in 9m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/474c578cb6c8435fb1c46b7921bb53ee : FAILURE in 12m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c0fc29f7bdd46dbb0b3c33ed3d8b1f8 : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":1},{"id":"76daf3c88f3952e7dbc5d7a2f22222a59b1dbf06","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-27 23:26:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc10a258eea18bc486cd3a4ec505268c0303f82c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 00:35:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa63416815947439a2300ab1fcec535 : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/98801dd8c95845edb9ead2deec4158a9 : SUCCESS in 56m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5cb68e5edb6b46de9ad0a4a8d61a4a01 : SUCCESS in 49m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/30decbf15b1e4145a3b1f7512b440c8a : SUCCESS in 10m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/74dfcd764e004bb9812b8a2606d31b28 : SUCCESS in 11m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f7303202d5b247abab90614f4dc894e2 : SUCCESS in 27m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/00d772433b0d4c7bb7a5fbe248a7f912 : SUCCESS in 35m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/973e645ff3044bd08c1011ab7ba22525 : SUCCESS in 27m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/da333bba23b641029db667a14a29872d : SUCCESS in 35m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/938196f80f6744f7ad19941188fd7c6d : SUCCESS in 26m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1f30a43feeb041e6bdb89bc417c4bb1c : SUCCESS in 17m 40s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b9ef48d6ff5d4de78131526f99f82261 : SUCCESS in 26m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9eedac4486134e4486802684dda18ee4 : SUCCESS in 15m 06s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/cbb5e124a2e942d4a4abbfe33db4cdec : SUCCESS in 31m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7d22db773c304afba95a5e8a24a9fbad : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1a5c958f224a4fbe858d5f790967b1a6 : SUCCESS in 41m 02s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0fb7b9fee51e40e284276487d188488b : SUCCESS in 33m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ca241eae571b49abaed3008c42553ab1 : SUCCESS in 34m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/50bcd043f3564b6f8784c03346d3cfff : SUCCESS in 22m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ac4a326f45b541d6809f041f9ae067b1 : SUCCESS in 52m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/13279c459ff44cb5b1ddbd71027511ea : TIMED_OUT in 1h 01m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f70a2d26dea432997755e17aeef66d6 : TIMED_OUT in 1h 03m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/135990f9d8ef458498db77e529edf82d : TIMED_OUT in 1h 03m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16ae8f547d914c42a0c79e09d10d1ca0 : SUCCESS in 34m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/381634972df54e4ba6c0734ef42781fb : SUCCESS in 7m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/99ae2c1714c74f2f811f673d72433c86 : FAILURE in 10m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54d39fc4a21743369f76673f8e440c6f : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b8ae19e91888c01ae6cdc9ff269ee3d9358f41e6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-28 01:01:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b040d95c9882a9d2d6eb1cfaad097a7426dad886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 02:36:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78cbd6418aab48c2b84e2e7533507e80 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/4f0b1f696b0e4267a8f6853198cc08f0 : SUCCESS in 1h 07m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2aa3326936984ece8743f50bd5cdb96e : SUCCESS in 37m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ca2768a5db34fee9e7b1daf451f977b : SUCCESS in 9m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fd904f847e97499c8ddeb9779b34286e : FAILURE in 11m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/285d5bdad88342f4b4da71581e4c5a07 : SUCCESS in 12m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/83d0c279a4b64dd6a5bcc3569b1d3122 : SUCCESS in 14m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2df8d71d7a0e41de99e4b88e28998a7f : SUCCESS in 12m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/688afd08af394d57b78650132019700f : SUCCESS in 14m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/581ad02719db4c87abb7fc6fbb535fe0 : SUCCESS in 9m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0266a125d4fc458a8e5c3c644b58559f : SUCCESS in 14m 18s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/2a4dc81b45cb4195bf46a2e5fe7cd44c : SUCCESS in 11m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/07435db7e8b6421fab81e2ee62f0653c : SUCCESS in 11m 57s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ea19c2a47f1941f19086cd410517917a : SUCCESS in 29m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/378ff7662d274e9f86b6c0b4d9317230 : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/481d9c0f02324bd0820195c21dbdfd54 : TIMED_OUT in 41m 21s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4d0986bec34a444d949c1025bd839aba : TIMED_OUT in 41m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ac4d0d7c8a5b488fb3dda8186ce99a2e : TIMED_OUT in 41m 15s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a6db4b9012dc40f68cbc6be14a88f16b : TIMED_OUT in 41m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49549c93f0ed40f8a023de7ff7af74d8 : TIMED_OUT in 1h 01m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e1cb9be854424bb19052f6b23b98d569 : TIMED_OUT in 1h 01m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b86dcea180147e4961364b77764e17e : SUCCESS in 56m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9c15a5b4abcf4ba295494f86ef1dcf1d : TIMED_OUT in 1h 04m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c713f499bb35437db9047084bdd1f79d : TIMED_OUT in 41m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49a790d25f6b441d8b6c8de984879c49 : SUCCESS in 9m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/509b1d45e09e4426ae322ab664ff7ed5 : FAILURE in 13m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c5329eb5c8342fc81ddc14187441af8 : SUCCESS in 1h 32m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c1877e9b022b69add62c4453a7b2017b1a8a4afa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-29 20:28:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d3c40039f2d2e6f1dc474aa11f4fabf6c1f45679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 21:37:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37a71075262418fa9f21b6f2840b9db : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/ba23634275d840418f780ee748cc65d7 : SUCCESS in 1h 03m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d5272e4023d746fcb25775630bfa62ac : SUCCESS in 43m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b442e8d60f644c4da6c1a490341da6af : SUCCESS in 8m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/343b48d1ea79489abb34f1bc80f17cb5 : SUCCESS in 8m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/23208af9205546968edb0be08f122c31 : SUCCESS in 16m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/22316907e254448892ae89570b5885c7 : SUCCESS in 16m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/04d8ee39236347e3b30d030cf0e6b68b : SUCCESS in 12m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/78adcdb39d6b4d0cbe17a3b2c019d356 : SUCCESS in 11m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/85a062ae369047ef884f229244adbde1 : SUCCESS in 10m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fe0124dce21149a59687565fd33c7294 : SUCCESS in 11m 39s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d41ec3456f7b4de999dc8034dba0e5de : SUCCESS in 9m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bc033d54be2c42689ceae39cebf9d097 : SUCCESS in 10m 59s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e34cc0ffd8094f36b98e9a4fa6dc4a3d : SUCCESS in 15m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5198581f728d448494c19717b90d6062 : SUCCESS in 14m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/939e95b723b74417a53353eb151395f8 : SUCCESS in 13m 54s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/cf9742b28e3d4f5b9ab8a0ac24b54423 : SUCCESS in 9m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ddee70b735c4496e949994d074e093f0 : SUCCESS in 15m 28s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5b4c4a661c1a4915be94dab0deaffbd3 : SUCCESS in 15m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/85f875b317ea4c8fae649fa9a294bb63 : SUCCESS in 19m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/76529378e555450f808ff33fb11ddc61 : FAILURE in 49m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e0c4855a9fa84d5d84f5131b0dd53d95 : SUCCESS in 45m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8542b8e0190748148da7062aa304f687 : SUCCESS in 32m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef027e8a196945dcafef07777f1b4ffd : SUCCESS in 15m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a72fdd5c87ab424bac8d5bcf9232dc69 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f1f9d25c0984faebd18a58543d49429 : FAILURE in 7m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/546e23488b784a0eb86f5be49ea76b75 : SUCCESS in 55m 26s","accounts_in_message":[],"_revision_number":1},{"id":"02292b253728ce256bd28cbbbed036f299b76dbf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-05-29 22:50:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bd4a04ac05cee5347a7e4a40db2a7b1a8f49f751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 23:50:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7643908539374e1bbe005b448ce0a45d : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/c2db23f05a63406daa41c462181a17a4 : SUCCESS in 39m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/73a30d6824d14d459cbcb5dceb7a25ad : SUCCESS in 32m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/662c812401024a5f8eb7f4e005063f4b : SUCCESS in 7m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9eb764790c6e4b9eb7e731c7118d1fcd : SUCCESS in 7m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0b58d1df31e845b88ba8ab95536aa2d6 : SUCCESS in 10m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e90115cf88454ddb8914eb9afe30b900 : SUCCESS in 10m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a036d8a274fa4582bc4413608b0821e9 : SUCCESS in 10m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f74dc367a8974bcf925bbd5907a31164 : SUCCESS in 10m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f6f2806e7f7d4dc1b47dabdaab5214b0 : SUCCESS in 9m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fffce30737754238bd0b470a07367870 : SUCCESS in 10m 15s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/fd8a6208300d4b5d8bf3da5dccba41d6 : SUCCESS in 9m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9ee9f4210ce444d38378d0582965cca6 : SUCCESS in 9m 20s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/eaf77629360e448ead0d8b3a59c78019 : SUCCESS in 13m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ce7cc387306c4a6a8ed1d356114b1a27 : SUCCESS in 9m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/59f48351d04648ab99e32e0ca09e962f : SUCCESS in 15m 21s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d798c356bf294309b7b4a76ea2c93849 : SUCCESS in 9m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3b1111e70a534137b81d2af0ecf0bd58 : SUCCESS in 13m 24s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d709fcd9fe8b47daa055b758291f9666 : SUCCESS in 17m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b4397103e42142c1aa1c04607209ddb5 : SUCCESS in 18m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e7690505c184498aa3d191127a5a27f7 : SUCCESS in 52m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/97604f94c7d044499e155dc5b952bc69 : SUCCESS in 36m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/097cb1d9137c4cacbc422687bdfd75f5 : SUCCESS in 32m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bd8505fee7146a3aa9121a34c4685f8 : SUCCESS in 12m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09e3a305d71f4f169f9a84df29c399f9 : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f1d7bd211c3741a4aeea3b674014322e : FAILURE in 8m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/465fe9011a6a45588958dafaf3ea4fc3 : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":1},{"id":"eff4c5c197514bdf99d6fa069d906ef36523063a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 23:50:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3bc2640d18ea32b83fa15f528a436414a471728","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 01:07:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/743b48012bed47639fe6a52d23abb277 : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/cd7ca2d99a3f4f5b968ed81b5de301a0 : SUCCESS in 1h 10m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ccc3c0101ce842e691b8ee21d64d8fb2 : SUCCESS in 48m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/974bdeb07c994cd6926c4ee1350c7161 : POST_FAILURE in 12m 42s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8f4c666b5a7d460998f01178d323ba17 : SUCCESS in 10m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bba2b4cc944a43b5af962df5ceab6a4b : SUCCESS in 14m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/78d6c3fdd3134e47b99d3c7e11f96fae : SUCCESS in 26m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ab2a2d6dea4742d1960926ee8e093c26 : SUCCESS in 12m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8b71863ee35c41948c922a897f6a2def : SUCCESS in 25m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e1a533b867f248c28a01709e89da45dc : SUCCESS in 9m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/abae0e3a9a9d41348fa1ee9cfd6e5b5c : SUCCESS in 29m 38s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/91e0d116b7234355be08d13039b3708d : SUCCESS in 10m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/467bd0bb3ecb49a899b45d50792496a1 : SUCCESS in 23m 54s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ba4afa2e196f48de95bc5c03d9cd5354 : SUCCESS in 16m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6a6c8229fb20408a9283a21ce071c9a2 : SUCCESS in 20m 12s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0dbcfca85ce34e839e00f202c09b4823 : SUCCESS in 10m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d443ec5caa8e491fa0ae9aa15cf64c12 : SUCCESS in 23m 19s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/e45d26b30169420ab251e3c32631eba3 : SUCCESS in 15m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/06dfeed4ef424c4d95c807c9d9f4d941 : SUCCESS in 45m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/000de7984da84f1ab95fc3dc4fad4f53 : SUCCESS in 48m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d1202b3c67044e5e856c131cedc53f02 : SUCCESS in 47m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/576e21d8bbec4e359635882e2e6034d7 : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2839834dc9284c8f8840eee098b0288f : SUCCESS in 5m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a01807d99844f9fa843512a15144573 : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":1},{"id":"15b62f1fa508228589e2684b6bba4189bfbc0e73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 01:07:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8087e72aff13b71577b547d0624e25601d531043","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 01:08:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/134449cc65db4dcf87683c3e7e4b3b26 : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e8071ceeac094f9b86ba26a6e053b7e1 : FAILURE in 46s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6a71c45abf704d87b8034b6ac5f36b09 : SUCCESS in 56s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60f052f69aee346cf6f53426327bc81e2580fb6b","revisions":{"60f052f69aee346cf6f53426327bc81e2580fb6b":{"kind":"REWORK","_number":1,"created":"2020-05-27 22:08:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/731362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/731362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/731362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/731362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/731362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/731362/1"}}},"commit":{"parents":[{"commit":"73f0b143d18e4920ec4d8cc35833573491e853d6","subject":"dsvm: Run service-user tests under tempauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73f0b143d18e4920ec4d8cc35833573491e853d6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-27 22:07:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-05-27 22:07:52.000000000","tz":-420},"subject":"dsvm: Run service-user tests under keystoneauth","message":"dsvm: Run service-user tests under keystoneauth\n\nChange-Id: I5b0ae0b78d37a31928e0b34a1b6118802a3a8236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60f052f69aee346cf6f53426327bc81e2580fb6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60f052f69aee346cf6f53426327bc81e2580fb6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
