)]}'
{"id":"openstack%2Fswift~732431","triplet_id":"openstack%2Fswift~master~I4e6889b323598c39b71c3db04e669b7033fcfbf6","project":"openstack/swift","branch":"master","topic":"https-dsvm","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-02-10 23:42:42.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4e6889b323598c39b71c3db04e669b7033fcfbf6","subject":"Enable TLS for DSVM job","status":"NEW","created":"2020-06-01 18:04:43.000000000","updated":"2026-02-11 00:01:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"6f5e2a056ea597cb593a9132d7439eca20565f97","_number":732431,"virtual_id_number":732431,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-10 23:42:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 23:41:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87a409292de30e85e613830d466b28800136dd63","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-01 18:04:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07e72604914bae38566e45e1b8ae089e8790bf71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 19:11:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1079ce30497a43ac987f91ac578a4362 : SUCCESS in 11m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/636cb9a9f25641bcaa2af59c265f064c : SUCCESS in 1h 06m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/36149743f42f4b3aad7986f672fafa10 : SUCCESS in 32m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/28f7b03dfdb64976a2aefbd5bd294865 : SUCCESS in 8m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cba7076ff55a4f92b77e1c24d44b6fb9 : SUCCESS in 9m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4511648dfa19423aa3da84d0e3440545 : SUCCESS in 26m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9ca788d17ebb44eab72e8f960f0a1f15 : SUCCESS in 13m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/31442e4817b14442a29a3297576f670f : SUCCESS in 25m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6e3123b6ab274f5fb42d5e2d8a44e5de : SUCCESS in 14m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/849b8efd260f417e9cebb80ca4daf20c : SUCCESS in 10m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4b67961ffea844d8bc9bfaaa84a422e8 : SUCCESS in 14m 46s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f85e892639984c88ab17ac2f4ea32221 : SUCCESS in 11m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cc33b29eec3945a3a57c9feca79d50d8 : SUCCESS in 13m 20s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d87e7402af79460692305f78c8a080d4 : SUCCESS in 16m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/265ac354b8214fb8abbc1c07e93ba804 : SUCCESS in 11m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7986ff254470457c8cd392e8d396e3f1 : SUCCESS in 14m 05s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2edc1b6c309a432f989ab890f4354642 : SUCCESS in 9m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4f8a25de16624e3fbdca824856a4329a : SUCCESS in 10m 53s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/46876b86bebe4ed2ac403d4f7905e322 : SUCCESS in 15m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/84cffb62b5cf43f9be3960ba2c733d30 : SUCCESS in 24m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1633743004c1414c92d2cf703583760b : SUCCESS in 55m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f39af332c46040acac89a79400baad94 : SUCCESS in 34m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/616a954a42ed4e0ab773326b5a93bf34 : SUCCESS in 41m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47dda4d0ace74927ad659b132c738880 : SUCCESS in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f4812c9c07446dfae3b95bc12d3af3f : SUCCESS in 7m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c2e604157d214bccb8fb2173eb9810dd : FAILURE in 8m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e01304f8fd5b419a9c42b98614e9ee14 : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":1},{"id":"68a99956882e0d93fec783a7f771cff13c6cea07","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-01 20:17:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"86977f6bde814257acc9c8359880560f52860e95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 21:17:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/638fdba5895147a8b0439975db8a76ff : SUCCESS in 10m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/185e49cbeb3a47fa9eac89394b6ea54f : SUCCESS in 57m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6bd0b9b40f0c43ac966f14e342693203 : SUCCESS in 42m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a4304e24047c4ce59c8083ff81d28b42 : SUCCESS in 8m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0f39fc38f8134691a41dee90e3da4bda : SUCCESS in 9m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/31690a8774cb4ae9a7f5dc9a9eac8f3d : SUCCESS in 14m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ebadfe95320b4839ba411ef2e93b386a : SUCCESS in 14m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/601e68d7543944d893b37e1acf953ea6 : SUCCESS in 16m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6cd5d9bcd1dc42f7a701edd17e662ffc : SUCCESS in 12m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e3cd94c963824f80a4b97136eefc20b0 : SUCCESS in 12m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bfc8fe61c528439badc598285fa51fe2 : SUCCESS in 16m 50s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/868aefc7bb03482a864a68c8578f74e4 : SUCCESS in 13m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/46e2a02e085a42b9aa926d7f9fbba2d2 : SUCCESS in 12m 59s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/7da79455d56b45628760ecbf96f6811e : SUCCESS in 19m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/35869fca09f049caba2df930b96b4666 : SUCCESS in 12m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/76c6afbce11f44b3a2fc2faba26fd2d6 : SUCCESS in 13m 03s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2d4dcd7f8c064468850d2465c99d8bfd : SUCCESS in 9m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/348dc009a4884929aee539ae8bd6ac7c : SUCCESS in 11m 53s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d9d2462406124e4b9c1006a86195b53c : SUCCESS in 17m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6529e04c19dd47bda6dcee8725378c4b : SUCCESS in 22m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4ec17d35fb5348eb96f980a8745f0444 : SUCCESS in 46m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8815d281e86643f2890a810f628bd53e : FAILURE in 18m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b23a1d98a5d4397a1f1d85daed738b6 : FAILURE in 24m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a9c06ce085e4ad3a80c5efc02daf676 : SUCCESS in 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4b51fdd55fd4cb4994acae2bdb5cb38 : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/02e7403c045a4edba751310fde4a3c3a : FAILURE in 7m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1575611a9cfc48e1b8c8ad6ef2f04fc6 : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":2},{"id":"eb65309f91c65f7281b052ff2eebb86ea194f616","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-01 21:47:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9c43163384e798ca2539a55db1e6544a3a65b015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 22:42:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b81eb82dc4d4176b4d5fff5fe4d472b : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/da4a6ab7748e445db39cb654cb5e95ce : SUCCESS in 53m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f0feab0c8aa449ca9e2bf6e5124a1e43 : SUCCESS in 46m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/78603804766142d1a1c7f8eca01c585e : SUCCESS in 6m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8c5b13a630914535bf50cf692cf6b7e2 : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cbe646a31d104380a6505fa4a6553695 : SUCCESS in 10m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/06ce7110597c48f6afe16db563941715 : SUCCESS in 12m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/afc3bf120c7e4bd682c9442187f913c0 : SUCCESS in 12m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b0931e352fc54fab8e0bc63d4baa7184 : SUCCESS in 11m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/180f8d67661c4efd84eb975fa35baa79 : SUCCESS in 9m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7fd4697659ae4e4d89afacd73102b9b4 : SUCCESS in 11m 43s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/fa1af53a02264d9896325ad89feea82d : SUCCESS in 10m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cfb090c0408a4e538ed27e60b947b4b2 : SUCCESS in 9m 46s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/e679a4b5ddf94cb7aa9b0db885ddbc61 : SUCCESS in 15m 25s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/45db1b60100f4cd48e91938e90db461a : SUCCESS in 9m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b62188e77e7d40a89d063297e15faee6 : SUCCESS in 10m 45s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3a5b36533e0349d38ba753faeb8a73e0 : SUCCESS in 8m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a6f2594405594f07a3fefc281eb66df4 : SUCCESS in 11m 35s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5af30b9edb744ed79ffba2f5a4fc6e90 : SUCCESS in 16m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/900eadda2c8a48f5b54eac3b92076d13 : SUCCESS in 20m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0af8c204cb5145919a3508a68829e49c : SUCCESS in 43m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e65943dffa2f4887a80f31f7c38b80f9 : FAILURE in 29m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/af5d7dc639704ca2b81fd40f4e5bd407 : FAILURE in 26m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2fe0c3dc05d4aa7b4359d5203cc34fa : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1722735d25d24be9b85450a4323c0011 : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9f2b392969cf4273b5e50dbc6289b1c4 : FAILURE in 7m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/538ba98133c343c6a4bc30e5b705fa21 : SUCCESS in 36m 13s","accounts_in_message":[],"_revision_number":3},{"id":"a5a655e03cbb7ba3ed81683ab4119fc6dd53b704","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-21 23:23:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a01f6beee8e334d9190dc8e02da6bb4454a3fa55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 01:55:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40e9f29245a84df5a6236b4f7bd36d8b : SUCCESS in 18m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/99b8a53efd45457b9dfc264650d01970 : SUCCESS in 45m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0150ebed35be4e46a5d26d5fb87b064d : SUCCESS in 42m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fc6da40116b4453fa3b8176aa7819f6b : SUCCESS in 10m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5d5be37bc7954abb90a7099e771a56ec : SUCCESS in 9m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/35856eac819a45bfb6bd0775434ed101 : SUCCESS in 12m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/13fe7d4ad20c4eb595e55b879fd4fa17 : SUCCESS in 13m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6ce815b03056462ea327e2a1b56257f0 : SUCCESS in 13m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a8d04dab94e74f68aa50f8a3b7238453 : SUCCESS in 12m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0473645c59e64553b19a874f502c30b9 : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/302e8255c26b4edabdf8396d611cc3ea : SUCCESS in 18m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/42f228f626434f008fb66ce2aa094e9f : SUCCESS in 17m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d92e5ee4b44c44e89a082aebe2fa6108 : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/81672f0b66754a8c90148777ac637540 : SUCCESS in 21m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c0aa145d6cfa4a529bdce305dc8117d7 : SUCCESS in 21m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1efea9d073b4258bcf85cc6b970dd68 : TIMED_OUT in 1h 01m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e73d9df72c8e493799e44ff91587d78d : SUCCESS in 45m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8fe175331dbc4c76b3c17154553471b6 : FAILURE in 28m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b2e0546404804b08b9e7c2f85a7ec713 : FAILURE in 28m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cab57c1d20ed4b859ac047849c55be49 : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00bdafa20caf4acfb6a5b97cf904d33c : SUCCESS in 8m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/433ef3b64ec04e3a9defe61b691391d7 : SUCCESS in 25m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05106e79062345edb9906c7f82ed013c : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":4},{"id":"c90ab702c3b8433d401c4f29de5425c7107e0210","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-22 16:29:01.000000000","message":"Patch Set 4: Workflow-1\n\nRight -- current test failures are all exercising s3api; Apache complains like\n\n 2020-09-22 01:05:46.340118 [ssl:info] [pid 15348:tid 140342376507136] [client ::1:36948] [frontend ::1:8080] AH01996: SSL handshake failed: HTTP spoken on HTTPS port; trying to send HTML error page\n\nI think I just need to make sure s3_storage_url gets set appropriately, now that https://review.opendev.org/#/c/571021/ has merged -- though there\u0027s a chance I need to root out some is_secure\u003dFalse hardcodes.","accounts_in_message":[],"_revision_number":4},{"id":"9e3a71cb8d86f8d8211b0c20f31e7bfa23caf3b9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-22 17:09:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"095a48a492cfcb6e3435d8856b8d758086edf566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 23:41:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68fa427ffec84548bac526a5d45489aa : SUCCESS in 17m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/5713aac1c53e475bbeffacffe49effda : SUCCESS in 1h 15m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4aa1118a8ea245cfb8448dd5cd2c41b6 : SUCCESS in 35m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d405326f08414bb88c01f7c734a77f9d : SUCCESS in 12m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a7d77783a8f6420d8151d3b7402f4d9b : SUCCESS in 9m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6676b1830ef147af88764ee8b414f324 : SUCCESS in 12m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2e7a9fc2539c4a29b5f90163a59aa8aa : SUCCESS in 14m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6223769b7d2e48d5a21b0378d11a19ba : SUCCESS in 13m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5c16c8d44f504176a14ab6530c44f0f7 : SUCCESS in 12m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71a4bbf070304b80ad5aff1df8bad269 : SUCCESS in 18m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e85d69c8ef3a43f288b0742b8b0717a2 : SUCCESS in 18m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d9f0f50365e74c5593450c26b85e0aae : SUCCESS in 17m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c16086565d8d4fcebad01d626e5ee35a : SUCCESS in 19m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a15592f49d4947a78a913f7d79f7a19b : SUCCESS in 18m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/102cc06bba814ffa9595e5680c4f43a1 : SUCCESS in 19m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f199426bfe884178b1dcf7f037c8ae4c : SUCCESS in 21m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6edf3ff967bd49f4b26a273102baa109 : SUCCESS in 43m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d62dcb2ee96f4054996f68df8f966756 : FAILURE in 41m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/45e8084288c6453589f39db909277316 : FAILURE in 43m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cefddb258198450497849506c9caf303 : SUCCESS in 14m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ab8bce843044cdabbdf6f6dd9aef01f : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b057badef274c328c2acd49890c46f4 : SUCCESS in 23m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba0f4a0b23bc4c5d8d1b6b1f5467b59c : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":5},{"id":"9574127dd05a1018c2175edeb6febd368eb70cd2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-01-28 06:33:25.000000000","message":"Patch Set 5:\n\nrecheck\n\nI don\u0027t remember how these fail, now.","accounts_in_message":[],"_revision_number":5},{"id":"a834f55c0587522f74a261af78704ae886c3ffad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 09:25:11.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/789accde40524b938681cb653e1d0bf8 : SUCCESS in 11m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/bfc598dcbd344183bbef4f3a62834b36 : SUCCESS in 1h 06m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0b18578c6fca4cb9ad1e01b9203b317b : FAILURE in 50m 48s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/14f4c18028b243b3b5afa4478c543cb3 : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d7dc4181757a4935955495e134b4bf4f : SUCCESS in 9m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f7e62137e4824534b06e664b666321ce : SUCCESS in 12m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/abae982bf4ba475fa8f62ea6e871ffeb : SUCCESS in 14m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/30064c7ca8794bfdbe6321968397a4ef : SUCCESS in 13m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8f083e2635b84537aa19bc3a8e767ed2 : SUCCESS in 16m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/13eeac61d9644da0b6476e49058ffc19 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/31af01c738284971910fc4b3accb4a90 : SUCCESS in 17m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fdbde3d48bb84b9bb827c54c386369d2 : SUCCESS in 25m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/866d92f4c32a40aaa77a19c53c53521b : SUCCESS in 22m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5ab0d2aaf1d346bfb17a741e18f86d46 : SUCCESS in 30m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/470d83dfef4a4d64ba3651c509513a2c : SUCCESS in 23m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bdb6700ae5fe4a089d2a4aa75414b3c4 : SUCCESS in 25m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/30794d90b9e041149d4c2d08a58770cb : SUCCESS in 44m 53s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/20f650ec76fb4c6d9fc61ecdd714644e : SUCCESS in 44m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c06b6ff2b0e94e869f34d6e414de1b03 : TIMED_OUT in 1h 32m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a7e26a8f43a0446c8c785c1690a49a0e : TIMED_OUT in 1h 32m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/849592b1aa43497caf1a73d7c04a9858 : SUCCESS in 34m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecab47347d474764918d7415e5228159 : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f311ca59ffe34063a177a2a172a96e52 : FAILURE in 9m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbffc113ff5a41359adbc549a3ddeebd : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":5},{"id":"2b6348bb437ae0c28f220086f3864413aba7136a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-07 20:55:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8fc3059525653752e01c33c48056bcbe05ab3fc4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-07 21:45:02.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e935ffb0c08b2b6a8297c341303cf3c25118ad95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 21:59:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb0d4d171f134b25bd8f7737a3529941\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/86e981c3848a47a18aa9457278fb3129 : SUCCESS in 1h 00m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/788880f413004c66843e52106daa57ed : SUCCESS in 38m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/920c754234ae4190b84ff3cc5794eabb : SUCCESS in 54m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6270520ad4d54f75a25d4b8b68912b86 : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":6},{"id":"eb3e8f42edae3a12d3afa422e46654da6a881bfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 22:32:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ad5c4a5449a490c932fc201c8206f4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5798b8981af24b87a79a794c648c5ef2 : SUCCESS in 11m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/c669ed37c85a41f595ff25269803dd2a : SUCCESS in 1h 02m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fbdffbd3a86c45738f0c36910a927da2 : SUCCESS in 58m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec0b2cfb3aee41d6a35c555c4f87c99d : SUCCESS in 1h 07m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1e1973b83c04585b4b50fbc7ff95f4b : SUCCESS in 23m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1a791169b5048dab3292dc345b0ded1 : SUCCESS in 16m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f7c07d6bcf44955b408c924b1e1b86e : SUCCESS in 21m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0eaceb6d2b474638b7a9562b412da742 : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/73fad32854df4ef0b585117f7418f5da : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5de6c70c4039424cb96c405281723563 : SUCCESS in 18m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b1bd66e487054e6db72f9baa87c52891 : SUCCESS in 19m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3f0a9574782746369c4876f9f23e2c37 : SUCCESS in 21m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1f2258287a5648e386afd19b303a971a : SUCCESS in 20m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b67c11ae6b4742cfa16cf12a21a4e237 : SUCCESS in 15m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6681369383664609a66a04e7da857510 : SUCCESS in 22m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/12c0a21c9d3e4206b49dad841d768bda : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3d61b035b1dc42ea8d082d5332fea835 : SUCCESS in 17m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6b0c534965564b98bf5b089501c8af7f : SUCCESS in 20m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f2c27e6fe1c74dc2b5b7bebfd7266289 : SUCCESS in 7m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/408adc1b3a5147e8b5b58e6d5dcd5332 : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7199843db3144941917adc3381cbff2d : SUCCESS in 9m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1ca8d217c31642278cb070bfb4b90c56 : FAILURE in 53m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/716c10fec9584f788ad7959a7020f63d : SUCCESS in 1h 05m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ba4ed0dd4cf14434ac63fd77b4e8e9c9 : TIMED_OUT in 1h 33m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/779ce69c0a034777a4928bdc30a91071 : TIMED_OUT in 1h 33m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/685b8a17eb19405da6f63e0bd66313a5 : SUCCESS in 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7985d5e4ff746e792eac4d515d6e547 : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/618a93a261cb477fa50956e8d482d515 : SUCCESS in 21m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2786e7d4e034a1eaad1418831213910 : SUCCESS in 52m 18s","accounts_in_message":[],"_revision_number":6},{"id":"ff6323d1436f33adafe45ae481640fd0778d3b29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-18 18:33:15.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"8152f26f1b282660224010d72d7f801334b4ed89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 19:48:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6363d1b5d8db43b3a4797817474532e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35d985a863e406d8b927d96eaf6fd0e : SUCCESS in 6m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/2e7635890f714189b906e1031e5a09dc : SUCCESS in 41m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/66bc6c267c2947da96368306cae6c1c5 : SUCCESS in 52m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1ec552428e1e4f1ba757927ae81389c1 : SUCCESS in 34m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d35d4789fc248d9a43b7cedeb952825 : SUCCESS in 43m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8721695248da4eb6822d9c0c522c7195 : SUCCESS in 11m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/265d57c9a9554eb4a4f7298e1a9bf0ac : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ddfa22eea6a647d1a6120075e967695a : SUCCESS in 19m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c9159742f32845f095e6fb6530556298 : FAILURE in 3m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/48c24928f42c4ccd94cb0464aa574ca4 : SUCCESS in 20m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2b3a441254a84a769f971c8ad2efdf7e : SUCCESS in 20m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b557e49c19c24ca0ae13f27e82fe3466 : SUCCESS in 20m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0fc27c84a94d4b0e88a453675cf5394a : SUCCESS in 17m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/280ac23821254295a6fa2d7869a19a4c : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/af2fd632ed164e7f945ea59440da1ade : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5d1166bba0964f22ad918eda0d680b58 : SUCCESS in 13m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ebb87c2565ce4dfbb831ec9c48ab0370 : SUCCESS in 13m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/41a0783181da423d882be0fd051758b6 : SUCCESS in 11m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3f95d13e02e44780a5bdf6fa01b30ca3 : SUCCESS in 14m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dcbde7e1e232470b9712478470ee4fe1 : SUCCESS in 5m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f5691c71e1094fa6a098a7d5c430861b : SUCCESS in 14m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7e2e879080864bfbb5cef21b55eec8b6 : SUCCESS in 6m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/25feb93f536f4735b6631c5f1f346031 : SUCCESS in 40m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef66d19b6fdc4d93993e84fccdaf8a8d : FAILURE in 41m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2ed1478a77b947e7b9cb2794471653a3 : FAILURE in 58m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9374858f6f7b459aa447c074af5de18c : SUCCESS in 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ee546591e034fe5939b301d39bc2a02 : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/707fcea1dc35489ca4494d3deccba6b3 : SUCCESS in 17m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7044303a5c6b4fbaae80731bf57a369f : SUCCESS in 47m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41c0b7df4678492eb730f0b3f8c0a94f : SUCCESS in 1h 04m 49s","accounts_in_message":[],"_revision_number":7},{"id":"fe21b8077f116dee3622856efd0b20bdfdbbdff2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 22:54:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f48d22d9da04c0182e1aca82be512ac\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f7a6acdea61b4ceeb81384c9a52624db : SUCCESS in 2h 36m 08s","accounts_in_message":[],"_revision_number":7},{"id":"ef0de5cdbdbfe5bde3036e358e94b76ac58cfc80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-18 23:33:56.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"4d6929bbdb14f2ce68d2ffa2cce18723a888198b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-18 23:39:42.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f9cfa9b419dfcc3f357b66a55acd85cded47aba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 00:55:09.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5cf81b2300b4378b346e28be4a61d25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e78af9e10be144b7bff9951d7d7127af : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/540fb77382ad4d6a853bc64a99019dd2 : SUCCESS in 1h 01m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c04f1bf760ab4a0eb58385ecf75a73ee : SUCCESS in 1h 00m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0b6c0e2b03f0422a9d5b9c9b59881698 : SUCCESS in 59m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79a63b5c1cd147f2930f41b721fc2a09 : SUCCESS in 51m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b39fb75213424530984517d417a0b6f6 : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f8eb66dcd6244778a3a0f154124adee : SUCCESS in 21m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5687a507b53e4579a8dad10e60c9a80a : SUCCESS in 21m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/053ee7cd1e3048e1a273dd3a8228d530 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aa42288a331c4523aa1264cc00de932a : SUCCESS in 20m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fede62e10ae14e66963f7496fb3d31e6 : SUCCESS in 19m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4069e1d9c62340e993f7d0e03114e3aa : SUCCESS in 20m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9a6e3a3c308e4a3a8a076e22482f945a : SUCCESS in 26m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e439199a22a54c7ebc2120a7561f1ab1 : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b94006c6c89f4b869e7ae90cc6ee1f09 : SUCCESS in 17m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dcbf0fd0f421484a809708a680b27259 : SUCCESS in 19m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f112bbf4f5ad4201aa8b8e2717e79fe4 : SUCCESS in 11m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a7099a9928cd41ce95ec6d62e88207a8 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f6e44ad0289b4544afefb3f47d369cff : SUCCESS in 13m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/45d0591c184946ab9770abfb7a11111f : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cf842985da4f4d68bb305a9a83287548 : SUCCESS in 9m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eb08aba86d9a4d4db9893e32d1f61d4a : SUCCESS in 6m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/36e853c5b40f4a71aa8ff3067f049aec : SUCCESS in 50m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d3293800b5d84acbab93989427fae276 : RETRY_LIMIT in 13m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/68b16064aed54d98b8b51bc12167444b : FAILURE in 1h 15m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f2c2dfda067409f85b0db0ab7e0832f : SUCCESS in 20m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5285719a573c45b1992e9a7435b2937a : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b469e5b5c8e24033b7b14f0a4362412a : SUCCESS in 18m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca1ce946c221469da062ab145ecdafca : SUCCESS in 55m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6f58b83ef574a3692884fcb723917de : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":8},{"id":"731113f3aa6280c15a702b6c540fc3843f2759fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 01:17:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/639628c04bb54c0e839df6b75bf96c8c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/de22fb7c82754478a09fb0f36b8ed211 : SUCCESS in 1h 41m 13s","accounts_in_message":[],"_revision_number":8},{"id":"6ab2b6bf8a55b214313c53cc3b3ac1640b1c0dcd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 01:29:35.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"edbeff6046ba8b062d8f2042371e6c3f7e1200d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 01:51:45.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"6d0d8fbcd5a393ef041919a8284e60a6c7a3de7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 02:08:15.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1ddc8e667634370e9635cb29d3810839bd4f570f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 03:54:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d63f61ac5824458b5bf83e902b03c53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eb190efc6204bab856e3bef304dd7d5 : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/5ab20165ef544d0f91dfe25434c127b5 : SUCCESS in 52m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e75fd3b5452a433f83933a2e244f3bd9 : SUCCESS in 1h 02m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/18d8976f20944c9e89849f76030b4aab : SUCCESS in 55m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f44fef5ff02b4386abfa9963b851ac53 : SUCCESS in 51m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f3ff530ef894a9da56f33c20f3494f1 : SUCCESS in 21m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75b5116e4ab6443385ad8174c1586733 : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d59df74b73d34f4b946ca6ff13323cac : SUCCESS in 21m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/af221e2843944a2c9ba892e0c3f35cc0 : SUCCESS in 13m 19s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/df2fbea34b0d41889c4fcede649383be : SUCCESS in 21m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8ceb8fe8118b4a85862dd570aaadd298 : SUCCESS in 20m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/98f3fb64a1d04c669573b93d54d6da6d : SUCCESS in 15m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/32c8c615444f4c94a893de56203bb022 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/44a73aedcc664974b3d2f4e2bdd9cde7 : SUCCESS in 18m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/78462ee10ee54611ba3508727964f7fe : SUCCESS in 16m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/278c9844329042d492bb7c52b88e8725 : SUCCESS in 7m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2a1aff0a162340fabdc3522063853625 : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7c947deba43e4d29abf8cefc7a92240b : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4302f3e10f43466ba3abb611d2c9ef02 : SUCCESS in 54m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a2fecbf7b05b4c0fabfbc96cc41987c4 : RETRY_LIMIT in 13m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d767ae25eecb4e7ebf3f5fd4f72a4e9b : FAILURE in 1h 20m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a2c923732344461a160552e124dd45f : SUCCESS in 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb062a9f5bdd4be4b46b4f7ad4b4288f : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/07680bd9ffdf4e4689fe14147de8681a : SUCCESS in 21m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4625937df144711bdb3207ba4a2370d : SUCCESS in 35m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38ca84cbc6564a729e8aefb50b553449 : SUCCESS in 41m 41s","accounts_in_message":[],"_revision_number":10},{"id":"0ff942f83859c7f981fd6e6750356b63d7434fba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 04:35:40.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4667c415c6a849988eed031edd11c6ba\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2e12074e9dd94b0ea893969f892015d5 : SUCCESS in 1h 57m 44s","accounts_in_message":[],"_revision_number":10},{"id":"3320ae1337b7464d21a447ab57bedc85bc77acf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-27 01:20:17.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"581245bfc6ce377eee802b5e2d81a8786a1a8854","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-27 01:42:05.000000000","message":"Uploaded patch set 12: New patch set was added with same tree, parent tree, and commit message as Patch Set 11.","accounts_in_message":[],"_revision_number":12},{"id":"0339681016d5df571ff9af0cc700db6e68fb64c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-27 03:11:14.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"292d09c4c3635d0a9d14f7bcff17b7bafbda0173","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-27 04:19:32.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"21cb5ab945870ef4b3fd13992dc1a1ea6428cc71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 06:24:20.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8496dc7efc8c4abab09002f50373dfaa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d5ce023e8a47f2ad76c847e176d065 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/713ff89c49884d579befc186f5609470 : SUCCESS in 1h 58m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d875f3521aea444db9fcf39e9a2824bd : SUCCESS in 39m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/391d24590c1c4aa0be043ed8e8323788 : SUCCESS in 51m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ab3b6a326c74ebca917478df475fd06 : SUCCESS in 1h 50m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51fb7be4e30844b7831b8b1f4defaa48 : SUCCESS in 11m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc0997563fcf4b34a3006e409802b9cf : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/044dab7d307b4f14b941412542ef2859 : SUCCESS in 21m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ac49f890a94f40bbaeeb349981e5d68c : FAILURE in 3m 43s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/61aad835e27e4180aaedfcf9a7d18330 : SUCCESS in 26m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/10d5b1deae314a67ab6d33bfec6e3e34 : SUCCESS in 9m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/26a1457d8eb94d168dfdec72fc929af9 : SUCCESS in 31m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/85435128583b4175b1bef8d759c5f0d5 : SUCCESS in 12m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/daf3d04778be46f794e269bc7ac69319 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/95abab1fddba490f8d51926ed9ad15e4 : SUCCESS in 22m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/71e01c1a54274128a9394bcf9c4ab7f9 : SUCCESS in 4m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b2574a8ead4541d393f941f9d9fb6086 : SUCCESS in 13m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bc832754b614438f8933e7c152ee447a : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/884e21afcd99426c84e7cbd98987efce : SUCCESS in 58m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1377194108914b1588b85f6d9a4a528f : FAILURE in 21m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b7393f41e7814f9d8b90618ad0c9b7d3 : FAILURE in 31m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/899fb882568849c98ec4e1034633e207 : SUCCESS in 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54bbb7888c3448c0b3537c8bb74ba9e5 : SUCCESS in 14m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/631b2b1a369b4b379ce93f8f820e1497 : SUCCESS in 18m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bdae521745b43099d2ee3c8c8ffbcaf : SUCCESS in 55m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e9024229547442485fb2f3b94f2cccf : SUCCESS in 42m 20s","accounts_in_message":[],"_revision_number":14},{"id":"24fa75082317818ebf45cb5dcc529ce53313a06c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 07:52:36.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e05f6d8a2842456fbe6c8e9956909484\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/690d72369d6f4c249e004e4daf559b75 : SUCCESS in 2h 27m 49s","accounts_in_message":[],"_revision_number":14},{"id":"46f4583ed2271f9f39ccf6a46edfb4615b6451ef","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-27 21:42:32.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"117b540ed2a771ef6f6640dfb8ef81b4cdf8a1d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-03 22:27:04.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"feaf437bcae1453f44166edb5726112ecc8f00fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 00:01:11.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eead75675e34e908bfbb74d47d190b8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cb9bafb9dc8443d6b8ddfa045e26d15e : SUCCESS in 40m 48s","accounts_in_message":[],"_revision_number":15},{"id":"45a653a92e7ce867adc66939dad68b8a27aa2174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 00:02:13.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b4fc04ee07f46cebc356b155fbfa689\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af8d2abbd4e4e66898934804b8f465b : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/8cf223b714d1489ca92291da88c49520 : SUCCESS in 58m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dc5d3b9614914129bef8dbe76922e599 : SUCCESS in 45m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ecec992f68fc4f819f165dd4e139d5ab : SUCCESS in 45m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a72f79a9b3f42d8bbd4e8d5a0ec373f : SUCCESS in 53m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/22e5c8a5d83044efb29687ef0184a039 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ab71a66dea24af2a02951a3ca053d08 : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b298ecc5aeef4825b0b87552efedd32f : SUCCESS in 12m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e6cc305b4bf4410ea58b1da0cbe16f88 : SUCCESS in 4m 35s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a27dd02f718140269b40407f985ca2b4 : SUCCESS in 9m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a91694d4a28043b1a4578e9beebcc7c8 : SUCCESS in 19m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7a0db95e023d48ea8085b30cf90f7dfb : SUCCESS in 26m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/23805d8f05a64eccbc412d1ddeb7797d : SUCCESS in 13m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5ee72f06b08f447fb75e17d32c035523 : SUCCESS in 14m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e41280bd1ff14eb5a81543e7ea52926e : SUCCESS in 16m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/63f3dabb6072438983af3185d02f39e9 : SUCCESS in 7m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/10a14a0738034589a7ea82c789da7119 : SUCCESS in 16m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bf09129d2a0f4c4ab29cf29e3dd1021c : SUCCESS in 9m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ab5f9ba4cb9424f97635f893b72b029 : SUCCESS in 58m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d777d7dc2f9d4deeb9177781cf08404b : FAILURE in 33m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/31d85768d34242cd8f7508070b916ab6 : FAILURE in 31m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bf2d9edbe9e4e2ba96368a8ffb94b7c : SUCCESS in 15m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d3318fd19643e2914243158e5ebf82 : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8593cb03bf7c40e8a3e93be073e9a857 : SUCCESS in 19m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/382c43745f8941e08e4af97c7ed350b5 : SUCCESS in 53m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c984268bd82c45bda0d2f14f4ed0b6b9 : SUCCESS in 49m 02s","accounts_in_message":[],"_revision_number":15},{"id":"5f6cb3ea1d1bb9e351fa3862111e6f9f1b42b4cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-04 19:58:45.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"55952efab7f263d042c19e4d919679c6a42abd28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 21:13:16.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b4c1759e2a240f391221812c66170e2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32427d2be2e44c32b3f65a959cb248ca : SUCCESS in 1h 12m 19s","accounts_in_message":[],"_revision_number":16},{"id":"c42f55025c3227c65774ac7b7229e5b94d6dd29f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 21:36:56.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08f52d85066548fb8b0d18326d151241\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e76a04fb2c6a413daff89cef4e04beda : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/afd08ba30d5544afadfdd4296ad968e5 : SUCCESS in 50m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/252e8d1ef7ea43129fa64b80dc0f76e1 : SUCCESS in 1h 09m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/423a581a64a34887832e2e3607dfdd03 : SUCCESS in 55m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cc4a65d81c54a8c86c497a0c3f4e0f4 : SUCCESS in 1h 05m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7053c2c7376843bf821b9da8a3a8f396 : SUCCESS in 16m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/774cb81d94e3446d949469ea03602b05 : SUCCESS in 20m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/457f58909a174ff09ac79887c1493a8d : SUCCESS in 21m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/960a3f9210114cdea8e251434bf37d4a : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a546f6bb415f4994a815a78f6374df16 : SUCCESS in 22m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9b9711a7a4e64369a021b8302c37ab01 : SUCCESS in 21m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a5953ed8a63a472fb6da6f64e9b28a0a : SUCCESS in 36m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cf7f98f86efc44439c88b9dacfa6d48f : SUCCESS in 13m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dc85b2fa18a741f1895ade501b2a48f7 : SUCCESS in 15m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f87315eeb28646609160d75687a1b12c : SUCCESS in 17m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c69ca79f2bf74bbeb4d693e82778318d : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/13cfa4a75aa4442989b3914530105d4e : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b296155395884a73a5aeeea1867756cf : SUCCESS in 7m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5305122292e04141b5a9d128e7c8d897 : SUCCESS in 55m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eb8115463bd74effa0770ba460ff3872 : FAILURE in 32m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/877e869b2fa740eea6b0a48aab311798 : FAILURE in 22m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/228cf976f3da473186d38f45e6630078 : SUCCESS in 19m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d12ff9b2ace47b2aab0f10ef33214ef : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/044143e4c7404b679f05016562011764 : SUCCESS in 17m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85381355f4d749018c2d405809a9e0a2 : SUCCESS in 42m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cb5fc088f1042049db171da135a296b : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":16},{"id":"cf01ba9c9fbd11c70d83e786b2ddb696570b69fd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-04 22:56:09.000000000","message":"Patch Set 16:\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"4eb8236c92ef28febaa3e94a19ce7ea81d271ef1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-05 00:40:23.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"f30dab6186da4da0397298584d211dd189a47f49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 01:58:08.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48208de98747493e9da490c6d5f5c24f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a09fd22a7a4db496c162db01fc0f47 : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/4f546f4ebfd447f88fb0eff7bfd77aae : SUCCESS in 50m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8514cb2ce4324986b427a304b81d8e11 : SUCCESS in 1h 00m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fef57bd800614e0b85dce15ec44bed6e : SUCCESS in 1h 04m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02fb339175654939a20ad5b6f684577e : SUCCESS in 1h 01m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d27eaa5675b34679aa87c870016f8fb5 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fcd136b84abc4cca853b40e1d2e9e121 : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d332a95a7ccc4f4d9903e0d0940e66bf : SUCCESS in 20m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/79caa783bce24686bd0a19da81b69dbc : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b105376622224a3c90bc5343404f4836 : SUCCESS in 20m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a746bdcc200046289433ecccf9154b9a : SUCCESS in 16m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b8c22bdad7044fc3a983a11f9a07c956 : SUCCESS in 31m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/adea533f3f384283b188db35eb5d44bf : SUCCESS in 12m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/119486eccbac4e7eade9404a97743b66 : SUCCESS in 14m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/07aa46076b474ca9b8a9f417f3eb7c91 : SUCCESS in 15m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6da31442b4fc4b7bab71646010819833 : SUCCESS in 5m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cd97d2dd41a84990b58663ab6ca41578 : SUCCESS in 15m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e56df1a7175342c09def7c424c324719 : SUCCESS in 6m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9a504ef0b0d4760ae884ca6aefe4be4 : SUCCESS in 46m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc1878ba1a2943dcb742b34f4d193a8e : FAILURE in 30m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/702570cffc244bf88be3a11b598f1c4a : FAILURE in 32m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f70fdfaa9d954d43a87698fdb4a9904e : SUCCESS in 22m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dad75ab46af419e9c80b26c46df604a : FAILURE in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c1c68d49c57d44d3a0b7832ff357abfe : SUCCESS in 17m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/745d1102ea9c41c59e9b2a8e84a8a3fd : SUCCESS in 49m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38ce551c16944112b643323a0ace5812 : SUCCESS in 1h 04m 39s\n\nWarning:\n  Comments left for invalid file test/functional/test_dlo.py","accounts_in_message":[],"_revision_number":17},{"id":"5105eae9ba1eb83cb25d8ba1aa2a8111911499cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 02:13:34.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33fb8bb98f5f4fd8abca4442035bece8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/77fc7cd1a1f34b40be7b415a94a37514 : SUCCESS in 1h 19m 35s","accounts_in_message":[],"_revision_number":17},{"id":"ecaf36f5bc4ba409da7b3b1d0d0334e17a962dc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-05 21:50:48.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"528ae9df3c63db49c84d5f363f0130892c219866","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 23:12:11.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8432551b9b1649919e3fb1ec68b0844d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43928d0b2410437384c3d6d23dbf1e9b : SUCCESS in 12m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/9ee494fbc66440da8064e600bb9ba94a : SUCCESS in 1h 03m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b237fd88b08948c3bd1fe5f5dc2b79cf : SUCCESS in 58m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e94dbe8dc0141d89af3a34fe93882b4 : SUCCESS in 45m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97e8bd58911544dd90560822a2cccf2a : SUCCESS in 1h 02m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2c4b3836be94408964af2a626eb3631 : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3acc7526bed476aaecebec7beaf3d98 : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26a1accdd47c4be2809f6f778d38da13 : SUCCESS in 15m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/29a5d40277984948a4a9306a74f756af : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6af226a8fe754e9ebe764a3755e02b89 : SUCCESS in 21m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b70b46d4686e4231882ab6c3af45df33 : SUCCESS in 18m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eeb86c67025d4666b1324f873e17be71 : SUCCESS in 27m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/831d4f6eded34f5ca5f9d0e1d88ac321 : SUCCESS in 13m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fb95512599114516b3caf932a6549581 : SUCCESS in 13m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7fbb384f19954c3abe04fb0165b95fa6 : SUCCESS in 14m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/01a13484c78a4238b181e4d156e73bf7 : SUCCESS in 8m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3bf75d9db2ad4c58a139efc7f4d854f0 : SUCCESS in 12m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/10ab87df70c44f3a8e23308f3405f8af : SUCCESS in 9m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/af9958b5b32541f5b7591968e60ff801 : SUCCESS in 53m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0719a6ce085542a19463c11dce15be25 : FAILURE in 32m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/382ac9c420b24dad94b4e0386c9cd464 : FAILURE in 35m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d665a8abbfb9442fb3db738a5bb702e4 : SUCCESS in 23m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7a453e553f24ee9a64ea752ddebd750 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5661225852fa45999ad14967cec8bb5e : SUCCESS in 19m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fa710bc845148a3a50ee66a233182f9 : SUCCESS in 1h 04m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae68a9bc5aeb4bba995836734f0646da : SUCCESS in 1h 04m 37s","accounts_in_message":[],"_revision_number":18},{"id":"933388c9d70530a46ad8c3fabb0fb11d358bfce0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 23:23:00.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26185e51afe443908d9417bcf92fd752\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32aaf36f945740dab63a67a6451a315c : SUCCESS in 1h 29m 20s","accounts_in_message":[],"_revision_number":18},{"id":"094def23ccdcf3265640a3d8f5a01288794299a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-05 23:39:16.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"3002fa4a57055654eb283b398e0e15669e806a3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 00:32:10.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"8892adf37e96a5a220220c53d89a09a780ed1835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 01:17:32.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"b7ad6c4a1dfbf1361d0d5e26808146736d4daaf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 02:22:53.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"a2913be537b910f8dcee41cd7a9566ca9db9649a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 04:02:20.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46389000572e4e2fa63b7051c16a5455\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9cc428a287f49f08e52fdc55a1e77c1 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/f3a4f06899ae4fceb019af330b5e1e85 : SUCCESS in 46m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dc423996af3644a38bad9a5b606cd787 : SUCCESS in 1h 14m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aa67f2370b0e4fab9e97f19ceb1180ba : SUCCESS in 1h 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccd7f0f9d4a94318974da9aca5e03b82 : SUCCESS in 1h 02m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fdac0542013c4d8aa824470124b9755e : SUCCESS in 25m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aeee0d985dfb475ba37c4d357b028eff : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58a3cd0494cd480caf612234c5f25af5 : SUCCESS in 25m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/46fcd9415d964aa3ac4e688ef7b70e09 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe680a8645e4483ab513667094430466 : SUCCESS in 19m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ac0413e7ee474c60ad5894b310a73899 : SUCCESS in 20m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dd0813ce391744b5a9a027b66e0c5ea2 : SUCCESS in 24m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/62cc4fa9557d45cdb5f071da27755d6e : SUCCESS in 19m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b17a5bd1dc1b4799bdebd61337e3f0fd : SUCCESS in 17m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/896ab033efe345e1be6cebc87510596b : SUCCESS in 18m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e883c95f7b654ec68ed3ef8c18596901 : SUCCESS in 8m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/905b6273bb604370a0cb842fb5d0f624 : SUCCESS in 13m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b17ac1d059e2473aa910f06a0d33088e : SUCCESS in 7m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6b5cb3a826a4f939a6156ae5a91f426 : SUCCESS in 1h 02m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/769667abc2f4451ebfb80dfb2b65ec4a : FAILURE in 1h 14m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b7e2f3c985944828b989aabf65966da2 : FAILURE in 1h 06m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa335b88da274226994cc9e8c2db9700 : SUCCESS in 24m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80405ead34c445a0a464e276d4b07699 : SUCCESS in 8m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c263ad945b1d4359afd3dd28b3e52087 : SUCCESS in 24m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b53349cdc8f40cebe5357fffc4acefc : SUCCESS in 1h 01m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f67fe795a354967b659c0ed86653a27 : SUCCESS in 55m 02s","accounts_in_message":[],"_revision_number":22},{"id":"16475ec1c27e7234a682ea4ac1eaca1917e60da1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 04:42:10.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1848e442e53415aa16eaf3b9a683aa8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/72c8a0bfb4eb4f47b6a28b25397a703a : SUCCESS in 2h 15m 51s","accounts_in_message":[],"_revision_number":22},{"id":"4eb18dbd68e8bf20377b9c4b200e2466ef78d6c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 04:58:07.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"65402ae83b821709728d49ef8060d5d261ce06d9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 05:01:14.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"a631e1c621e09a29dd46d3c6845b3c84817bf7c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 06:02:45.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b172916b7054bbe9036a8b9cc42db9f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/13e3cc2c2b4f49a48133ac9dccb626b4 : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":23},{"id":"2f14d4e667df75e0fa1d25acc6952313b0a5e1c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 06:08:01.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0dd94625a224763ab2965026f067ce3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bc669eb44d342b79fce22f4e179842f : SUCCESS in 7m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/a17fab0eafd740de80d1a7d9997bb0b4 : SUCCESS in 43m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a334303bc5049f9a8f7c2ebcfd824bb : SUCCESS in 50m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/147d5ca20237498090e608a5ef157bab : SUCCESS in 40m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78c13bf1b5404f1a9e6be82ce6df3134 : SUCCESS in 56m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d0d394000e24569bee833885c9eb129 : SUCCESS in 23m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b37bc1859a04a46ae8f75194fb2b2a5 : SUCCESS in 24m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e67b190a09b24d16a6b992f5367b21cc : SUCCESS in 19m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1f051be6bd7343aaadf2f8c42ac9816a : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a2317df0ac774edea5adf29c5b709b74 : SUCCESS in 19m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/35cbf17d2bcb46ec8c28dd59bc0a8e5f : SUCCESS in 20m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e6a353c585dc491785af7a1c5c893133 : SUCCESS in 22m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/16d6e9abbf554396b985919e2f36e5fc : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/65cefd0aa859484eb6c5414bee9d29f8 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9ff386be96fb46b79c97a80cafefdae5 : SUCCESS in 20m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f959a0582ad4428ba2988d40f899f4a7 : SUCCESS in 5m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/02a34a0885994feb9e5ccb105f547f05 : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/efee5f5f0e9f448dbd25708b49cbf52d : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/da8121cb4d2047b99fd8019bcac822ec : SUCCESS in 55m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/878e3cd8a6a14ab39df28640a1c533c0 : SUCCESS in 47m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1d3869f9ac4e4f288ccc3df0d5974d21 : SUCCESS in 35m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aeb6e8de4947459aa3635e7610ee1e1b : SUCCESS in 18m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f13028df77034b4aa781f7a71aba44c0 : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b519bcd6f0d64b84aa88d3185ed85c41 : SUCCESS in 19m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccf3af06ad004ca7b9853566236e072f : SUCCESS in 56m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a5f00539f27476b9f86117e34372d7b : SUCCESS in 45m 29s","accounts_in_message":[],"_revision_number":23},{"id":"2aae57269f39e48f1161282d5943889353cfc604","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-06 22:28:34.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"3179518a75b300d1e7ed76dc60f649070a822f06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 00:04:29.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f70aadd6a3414a30a78e983ff2181de5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9834bb5a51824937a2a4a9ef94621d47 : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/ce48733aea294e6ca742ecf74c63de24 : SUCCESS in 1h 05m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5e8ea9cafd934b208c9209503a07de0b : SUCCESS in 52m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4430eefe8d9a476da432a30eb65d8ce6 : SUCCESS in 52m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea383988e1974a76aef1c336886c9d70 : SUCCESS in 1h 01m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bae1cc39c4694c94a9a7c76f5fa31ff0 : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/210b6a10654f43a0ae59e9e5723e8172 : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/906e47eea44a4c0ab5fa41eecb3ac266 : SUCCESS in 17m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fd4e824c186a4c4ebb4b735df1801a87 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/28cfb60ca0d14bf9bc8802cbb30fa679 : SUCCESS in 22m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/64ce244359b74ee890f965b8d76010c5 : SUCCESS in 16m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/819ef88606af4e3e8fec86d5a2f4da41 : SUCCESS in 28m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e07779f7c6b44070b7c0c5c7bd7dcde1 : SUCCESS in 7m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c4e0a59ec5bb4dd6aabfbcb036fbce6f : SUCCESS in 13m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d2c52246dec8464385e72fd94b57f0fe : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05cb2eebf0aa42c5b3d8e857cd40cc9e : SUCCESS in 5m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5db8520e4beb4f7d8c1051aaebecbb9d : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/02752c9473804ea794e3dee9f98c22d5 : SUCCESS in 6m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf2a2754e73a49d284b5a5571df271f4 : SUCCESS in 46m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/277d3680406b4c07b7d92c6d24be6e09 : SUCCESS in 38m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c57509497b574514bdff0dc116a6029c : SUCCESS in 58m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a700baab5c844996a7b044b2e4df3597 : SUCCESS in 19m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9446680376864f2b96fba7c9df634e07 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d564cc8f49154f4abc8cd122ebd0534e : SUCCESS in 18m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b2f5067524d4574927ad7be8d5b3570 : SUCCESS in 37m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfcdab5a616f43179b05627f60f92409 : SUCCESS in 1h 05m 45s","accounts_in_message":[],"_revision_number":24},{"id":"a253d486050151a76cddc1e5efd451e63bc716cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 01:14:32.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f54d5df711741439acf0b6bee423707\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e6b774ad7b1b401f921b2c82bf575828 : SUCCESS in 2h 30m 39s","accounts_in_message":[],"_revision_number":24},{"id":"5a5b9f04326e60d86add087d6b7d8afa35362833","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-09 23:43:52.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"44c0d3912ec3baba2af03a1a6951556f6ea461a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-10 22:40:45.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"5459ca3a418b196a95fae049ed4ba00739cc1d89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 23:42:42.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfa2b44240004ffbb98a3ebacc4d63a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae48e25eddbc4797a95f8ef33416cdb9 : SUCCESS in 5m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/9cc9f090961944d2acbf4955bba27de2 : FAILURE in 6m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6090e46cd1994860ae23c06c61765a6b : SUCCESS in 34m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fa4c321a8b6246f5ac197affb7eb7b2a : SUCCESS in 37m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/502ed8c4f7274fd6b0f2b8ecce6e02f7 : FAILURE in 33m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14decc29557643d29f503f1316282847 : SUCCESS in 10m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f8f507c706640f58a06ddfe3970f963 : SUCCESS in 10m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a60a63deade4ea38ea6d2c04399b3f7 : SUCCESS in 20m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f81a52723ff146b2876028397e00cf18 : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/765d3417f6844777a14b40d4a2f5163d : SUCCESS in 20m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b52c43f74ac444d695a0097a08c1619b : SUCCESS in 15m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ba049e6c42aa4cf4b81b55a845fdc37c : SUCCESS in 25m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/af7f820ea3ab4e61a8d79215628d1251 : SUCCESS in 19m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a91ccc0fa4a0498fb38aee42ab96c363 : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4639ebef00b5424c9298ab43217651d9 : SUCCESS in 14m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/29aaace81fad40d7bb2092d4acf5f188 : SUCCESS in 14m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6cb581d1f5624bfc854028c479fe51f7 : SUCCESS in 9m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ffed68dcbc7044e3b52820ed864ac9d5 : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a036ea1b962a41daa7d476d3a9adf719 : SUCCESS in 9m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc595445f0eb44ccb5bb15ba711a931e : SUCCESS in 59m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/14590c26364641439676efbde61651df : FAILURE in 20m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3e7f1d79e4fa49dba0825ae83067c69c : FAILURE in 32m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a656fe1f7534bcca8e99c06898b1458 : SUCCESS in 28m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a569be6fa0ce43899df5c0273310cdfa : SUCCESS in 3m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5f300d12a0674a3f8c28983390f37bdf : SUCCESS in 17m 15s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aaa046a21dd14339b77236705747b2a0 : FAILURE in 43m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/771e079c2d654d65a9d06693b24038d8 : SUCCESS in 56m 07s","accounts_in_message":[],"_revision_number":25},{"id":"6f5e2a056ea597cb593a9132d7439eca20565f97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 00:01:01.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/539678d7de3e4bbeb540929f8133a1f5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/604493030ed246f2985251ec0bd4eda8 : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"c0c7a9bc4d9e88fa072816b2a20f5f8197744b03","revisions":{"e6b96bce2fc4fc4a4e16e9eea0a74067befbbec1":{"kind":"REWORK","_number":1,"created":"2020-06-01 18:04:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/1"}}},"commit":{"parents":[{"commit":"fbbfc8393c4e469d703f5cab96df453da458b105","subject":"Add support for cacert plumbing via cafile in the test.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbbfc8393c4e469d703f5cab96df453da458b105"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-01 18:04:31.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-01 18:04:31.000000000","tz":-420},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6b96bce2fc4fc4a4e16e9eea0a74067befbbec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6b96bce2fc4fc4a4e16e9eea0a74067befbbec1"}]},"branch":"refs/heads/master"},"7a44cdbd4799c637b7fd5d318e9e5de9e0aef145":{"kind":"REWORK","_number":2,"created":"2020-06-01 20:17:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/2"}}},"commit":{"parents":[{"commit":"fbbfc8393c4e469d703f5cab96df453da458b105","subject":"Add support for cacert plumbing via cafile in the test.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbbfc8393c4e469d703f5cab96df453da458b105"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-01 18:04:31.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-01 20:17:28.000000000","tz":-420},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a44cdbd4799c637b7fd5d318e9e5de9e0aef145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a44cdbd4799c637b7fd5d318e9e5de9e0aef145"}]},"branch":"refs/heads/master"},"42bb3a2f336cb9210082b7d23796a3071d923285":{"kind":"REWORK","_number":3,"created":"2020-06-01 21:47:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/3"}}},"commit":{"parents":[{"commit":"fbbfc8393c4e469d703f5cab96df453da458b105","subject":"Add support for cacert plumbing via cafile in the test.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbbfc8393c4e469d703f5cab96df453da458b105"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-01 18:04:31.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-01 21:46:25.000000000","tz":-420},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nAlso, run DSVM tests under py3 (since support for IP address subject alt\nnames was only added in 3.5).\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42bb3a2f336cb9210082b7d23796a3071d923285"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42bb3a2f336cb9210082b7d23796a3071d923285"}]},"branch":"refs/heads/master"},"68d9b6a57b40554d35d70c3e2a9726ebf14c1913":{"kind":"REWORK","_number":4,"created":"2020-09-21 23:23:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/4"}}},"commit":{"parents":[{"commit":"de0780cf3f2c00f31ea26a7558bb95720e640d34","subject":"Add support for cacert plumbing via cafile in the test.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de0780cf3f2c00f31ea26a7558bb95720e640d34"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nAlso, run DSVM tests under py3 (since support for IP address subject alt\nnames was only added in 3.5).\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68d9b6a57b40554d35d70c3e2a9726ebf14c1913"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68d9b6a57b40554d35d70c3e2a9726ebf14c1913"}]},"branch":"refs/heads/master"},"3aace871b5d4838313c4a05747471255ad4b3514":{"kind":"REWORK","_number":5,"created":"2020-09-22 17:09:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/5"}}},"commit":{"parents":[{"commit":"de0780cf3f2c00f31ea26a7558bb95720e640d34","subject":"Add support for cacert plumbing via cafile in the test.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de0780cf3f2c00f31ea26a7558bb95720e640d34"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-22 17:09:13.000000000","tz":-420},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nAlso, run DSVM tests under py3 (since support for IP address subject alt\nnames was only added in 3.5).\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aace871b5d4838313c4a05747471255ad4b3514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3aace871b5d4838313c4a05747471255ad4b3514"}]},"branch":"refs/heads/master"},"a6daf4d8fb4d4d0a57b36ac74452753a99b119d9":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2022-09-07 20:55:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/6"}}},"commit":{"parents":[{"commit":"6ea2d413b0df27eef23d317d10c974c7a9e86da8","subject":"Add support for cacert plumbing via cafile in the test.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ea2d413b0df27eef23d317d10c974c7a9e86da8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-09-07 20:50:27.000000000","tz":-420},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6daf4d8fb4d4d0a57b36ac74452753a99b119d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6daf4d8fb4d4d0a57b36ac74452753a99b119d9"}]},"branch":"refs/heads/master"},"4c4b0225dadd4646a62ad14834dc6863036cdbad":{"kind":"REWORK","_number":7,"created":"2024-12-18 18:33:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/7"}}},"commit":{"parents":[{"commit":"17d399d32746580b4d0dd3be943a44542390342a","subject":"Add support for cacert plumbing via cafile in the test.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17d399d32746580b4d0dd3be943a44542390342a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-18 18:31:49.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c4b0225dadd4646a62ad14834dc6863036cdbad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c4b0225dadd4646a62ad14834dc6863036cdbad"}]},"branch":"refs/heads/master"},"be98087c739d490cbc7f9da7bc1924a8bac252a1":{"kind":"REWORK","_number":8,"created":"2024-12-18 23:33:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/8"}}},"commit":{"parents":[{"commit":"17d399d32746580b4d0dd3be943a44542390342a","subject":"Add support for cacert plumbing via cafile in the test.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17d399d32746580b4d0dd3be943a44542390342a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-18 23:33:45.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be98087c739d490cbc7f9da7bc1924a8bac252a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be98087c739d490cbc7f9da7bc1924a8bac252a1"}]},"branch":"refs/heads/master"},"468f8e1125d2dab6321377dc1fb91c3b72e7b8c3":{"kind":"REWORK","_number":9,"created":"2025-02-20 01:51:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/9"}}},"commit":{"parents":[{"commit":"e3efdf9de9efe28407cea4c24dcb4ab914225a2d","subject":"Add support for cacert plumbing via cafile in the test.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3efdf9de9efe28407cea4c24dcb4ab914225a2d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-20 01:51:36.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/468f8e1125d2dab6321377dc1fb91c3b72e7b8c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/468f8e1125d2dab6321377dc1fb91c3b72e7b8c3"}]},"branch":"refs/heads/master"},"b0ebe6272a4fe15d5e310fd1187071cf4997fe84":{"kind":"REWORK","_number":10,"created":"2025-02-20 02:08:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/10"}}},"commit":{"parents":[{"commit":"e3efdf9de9efe28407cea4c24dcb4ab914225a2d","subject":"Add support for cacert plumbing via cafile in the test.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3efdf9de9efe28407cea4c24dcb4ab914225a2d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-20 02:08:09.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0ebe6272a4fe15d5e310fd1187071cf4997fe84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0ebe6272a4fe15d5e310fd1187071cf4997fe84"}]},"branch":"refs/heads/master"},"65e38a3408a24757230739e115798f2932f598f2":{"kind":"REWORK","_number":11,"created":"2025-02-27 01:20:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/11"}}},"commit":{"parents":[{"commit":"325cbe051e57e5ee2ca825dd746c840bcb9e53ff","subject":"Add support for cacert plumbing via cafile in the test.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/325cbe051e57e5ee2ca825dd746c840bcb9e53ff"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-27 01:19:42.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65e38a3408a24757230739e115798f2932f598f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/65e38a3408a24757230739e115798f2932f598f2"}]},"branch":"refs/heads/master"},"31fecbaf6f6b7ea32bcc4c64c5ec888e8ba68909":{"kind":"NO_CHANGE","_number":12,"created":"2025-02-27 01:42:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/12"}}},"commit":{"parents":[{"commit":"96a56a5aaf9fe8972fd8eeaef5af94fa348be31a","subject":"Add support for cacert plumbing via cafile in the test.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96a56a5aaf9fe8972fd8eeaef5af94fa348be31a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-27 01:41:57.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31fecbaf6f6b7ea32bcc4c64c5ec888e8ba68909"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31fecbaf6f6b7ea32bcc4c64c5ec888e8ba68909"}]},"branch":"refs/heads/master"},"2503fc513b219aa3cd541ebc831f70fb884de1c2":{"kind":"REWORK","_number":13,"created":"2025-02-27 03:11:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/13"}}},"commit":{"parents":[{"commit":"96a56a5aaf9fe8972fd8eeaef5af94fa348be31a","subject":"Add support for cacert plumbing via cafile in the test.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96a56a5aaf9fe8972fd8eeaef5af94fa348be31a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-27 03:11:08.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2503fc513b219aa3cd541ebc831f70fb884de1c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2503fc513b219aa3cd541ebc831f70fb884de1c2"}]},"branch":"refs/heads/master"},"ec017143dedc804e6a315d09b446e28fe83d5e25":{"kind":"REWORK","_number":14,"created":"2025-02-27 04:19:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/14"}}},"commit":{"parents":[{"commit":"96a56a5aaf9fe8972fd8eeaef5af94fa348be31a","subject":"Add support for cacert plumbing via cafile in the test.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96a56a5aaf9fe8972fd8eeaef5af94fa348be31a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-27 04:19:26.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec017143dedc804e6a315d09b446e28fe83d5e25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec017143dedc804e6a315d09b446e28fe83d5e25"}]},"branch":"refs/heads/master"},"ff86f6f72c9b67f93ca767dc54abfb6916626e6f":{"kind":"REWORK","_number":15,"created":"2025-03-03 22:27:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/15"}}},"commit":{"parents":[{"commit":"b9db592c67f59b9bf83acd50fe6979372064747f","subject":"Add proxy_forwarding middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9db592c67f59b9bf83acd50fe6979372064747f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-03 22:26:51.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff86f6f72c9b67f93ca767dc54abfb6916626e6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff86f6f72c9b67f93ca767dc54abfb6916626e6f"}]},"branch":"refs/heads/master"},"4ddfc0751570c21bd9b64b6b9d421f11854d7a98":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2025-03-04 19:58:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/16"}}},"commit":{"parents":[{"commit":"655c86cc539c78e6336023de352a2f8aa7d7a22d","subject":"Add proxy_forwarding middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/655c86cc539c78e6336023de352a2f8aa7d7a22d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-04 19:33:52.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ddfc0751570c21bd9b64b6b9d421f11854d7a98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ddfc0751570c21bd9b64b6b9d421f11854d7a98"}]},"branch":"refs/heads/master"},"67a991000be502b90d7b2e28306d803fb93a339c":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2025-03-05 00:40:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/17"}}},"commit":{"parents":[{"commit":"c83fd1afd4e5f26686f41c21df9c2928804669e4","subject":"tests: Accomodate some apache2 quirks [4/4]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c83fd1afd4e5f26686f41c21df9c2928804669e4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 00:39:58.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67a991000be502b90d7b2e28306d803fb93a339c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67a991000be502b90d7b2e28306d803fb93a339c"}]},"branch":"refs/heads/master"},"6a043e7661e3ea85215236852c42ce9ed021a8a9":{"kind":"REWORK","_number":18,"created":"2025-03-05 21:50:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/18"}}},"commit":{"parents":[{"commit":"e1328a42d27de15dbdfe09fb85123bc9b5c7f2d9","subject":"tests: Disable additional checksums for boto3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1328a42d27de15dbdfe09fb85123bc9b5c7f2d9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 21:49:17.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a043e7661e3ea85215236852c42ce9ed021a8a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a043e7661e3ea85215236852c42ce9ed021a8a9"}]},"branch":"refs/heads/master"},"d8da695c5416d7015335ffede0824e6690b42d5d":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2025-03-05 23:39:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/19"}}},"commit":{"parents":[{"commit":"e6f101f7c1bd7011812bb2e8e8fe3f6a5001ea17","subject":"tests: Accomodate devstack-deployed apache2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6f101f7c1bd7011812bb2e8e8fe3f6a5001ea17"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 23:39:07.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8da695c5416d7015335ffede0824e6690b42d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8da695c5416d7015335ffede0824e6690b42d5d"}]},"branch":"refs/heads/master"},"3b934ee48326b0e8a67f9e83680ca56ae8a7c365":{"kind":"REWORK","_number":20,"created":"2025-03-06 00:32:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/20"}}},"commit":{"parents":[{"commit":"e6f101f7c1bd7011812bb2e8e8fe3f6a5001ea17","subject":"tests: Accomodate devstack-deployed apache2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6f101f7c1bd7011812bb2e8e8fe3f6a5001ea17"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-06 00:31:59.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b934ee48326b0e8a67f9e83680ca56ae8a7c365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b934ee48326b0e8a67f9e83680ca56ae8a7c365"}]},"branch":"refs/heads/master"},"8011836c885f3157459bddeb33a79afdf54470a8":{"kind":"REWORK","_number":21,"created":"2025-03-06 01:17:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/21"}}},"commit":{"parents":[{"commit":"e6f101f7c1bd7011812bb2e8e8fe3f6a5001ea17","subject":"tests: Accomodate devstack-deployed apache2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6f101f7c1bd7011812bb2e8e8fe3f6a5001ea17"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-06 01:17:26.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8011836c885f3157459bddeb33a79afdf54470a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8011836c885f3157459bddeb33a79afdf54470a8"}]},"branch":"refs/heads/master"},"c8f0419239001327a3a9243ed9b4de9e6fc493e9":{"kind":"REWORK","_number":22,"created":"2025-03-06 02:22:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/22"}}},"commit":{"parents":[{"commit":"e6f101f7c1bd7011812bb2e8e8fe3f6a5001ea17","subject":"tests: Accomodate devstack-deployed apache2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6f101f7c1bd7011812bb2e8e8fe3f6a5001ea17"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-06 02:22:47.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8f0419239001327a3a9243ed9b4de9e6fc493e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8f0419239001327a3a9243ed9b4de9e6fc493e9"}]},"branch":"refs/heads/master"},"f61ebc99df039c3fd6cae2fa47364cc33b3c2e48":{"kind":"REWORK","_number":23,"created":"2025-03-06 04:58:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/23"}}},"commit":{"parents":[{"commit":"e6f101f7c1bd7011812bb2e8e8fe3f6a5001ea17","subject":"tests: Accomodate devstack-deployed apache2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6f101f7c1bd7011812bb2e8e8fe3f6a5001ea17"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-06 04:57:56.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f61ebc99df039c3fd6cae2fa47364cc33b3c2e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f61ebc99df039c3fd6cae2fa47364cc33b3c2e48"}]},"branch":"refs/heads/master"},"d5c27e658f85aaf8fb7612f78e0a5544d90dd12a":{"kind":"REWORK","_number":24,"created":"2025-03-06 22:28:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/24"}}},"commit":{"parents":[{"commit":"3a044452a1336733e0988975063abbd51064741a","subject":"tests: Accomodate devstack-deployed apache2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a044452a1336733e0988975063abbd51064741a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-21 23:22:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-06 22:28:13.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5c27e658f85aaf8fb7612f78e0a5544d90dd12a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5c27e658f85aaf8fb7612f78e0a5544d90dd12a"}]},"branch":"refs/heads/master"},"c0c7a9bc4d9e88fa072816b2a20f5f8197744b03":{"kind":"REWORK","_number":25,"created":"2026-02-10 22:40:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/732431/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/732431/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/732431/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/732431/25"}}},"commit":{"parents":[{"commit":"ecbdc48359f2e41d96afd3ff37d5ac5e31752464","subject":"tests: Accomodate devstack-deployed apache2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecbdc48359f2e41d96afd3ff37d5ac5e31752464"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-05 23:20:19.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-10 22:40:05.000000000","tz":-480},"subject":"Enable TLS for DSVM job","message":"Enable TLS for DSVM job\n\nChange-Id: I4e6889b323598c39b71c3db04e669b7033fcfbf6\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0c7a9bc4d9e88fa072816b2a20f5f8197744b03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0c7a9bc4d9e88fa072816b2a20f5f8197744b03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
