)]}'
{"id":"openstack%2Fswift~732441","triplet_id":"openstack%2Fswift~master~Idf7f4ddc947abbc581688e47ded0b68ff0876552","project":"openstack/swift","branch":"master","topic":"auditor_212824_alt","hashtags":[],"change_id":"Idf7f4ddc947abbc581688e47ded0b68ff0876552","subject":"wip: watcher ideas","status":"ABANDONED","created":"2020-06-01 19:40:48.000000000","updated":"2020-06-10 04:11:06.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5c7a5b66725bdb6d1b36489db8c3525a1f0e1857","_number":732441,"virtual_id_number":732441,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-01 21:46:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2020-06-04 15:59:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 21:46:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 15:59:48.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"6370a7453f6f8687fa5098344527d4c9e0757acb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-01 19:40:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be20493687c94368bf210321be2c836339b42aa6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-01 20:32:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7aa23577adceb9c0b67b71edfc01e2ea2cdf16d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 21:46:07.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4da5b00ed3f476184a0bb44abbed94f : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/75fd60e21fb14041ac3d0712747a0dfe : SUCCESS in 1h 07m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d200e67f900d4c3992456b957a9ce829 : SUCCESS in 41m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/59d9da295c67480b848c7db0d82d1943 : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1e48406fb10e4d56958da0e76acf3284 : SUCCESS in 9m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/667e90560c1a41fda86066400f63917b : SUCCESS in 12m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/191a435f01824c92bc4da683eb1b2108 : SUCCESS in 12m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7b0ef76c8e6d4d52adb18c11dadbf228 : SUCCESS in 14m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9ee75b4bc523439aacd0da9000d7f342 : SUCCESS in 13m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9fa21f8a25894e3388c149afad48d055 : SUCCESS in 10m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5b5dce550b764c369da580b52d99b510 : SUCCESS in 13m 14s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8d54bfa6d7c24c5abc8c7c9744ff498d : SUCCESS in 10m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9ade37eeb9dd472c88fe0cde6602f144 : SUCCESS in 11m 10s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/32b3068cc31044a999fd3ff53e80d8cd : SUCCESS in 17m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0570a279562e4ffcb35cb1cdf3653d69 : SUCCESS in 10m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/62a998bb37cb4ce7b2516e99e52434dd : SUCCESS in 13m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/330468f95f1e44d5932ebac1011daf08 : SUCCESS in 10m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dfa8c9935002493985582a79022e0130 : SUCCESS in 11m 09s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f4e29ef550694188a43d9639e5254f33 : SUCCESS in 15m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b7ac1768e3ae427fadec93ae2e0ebcc9 : SUCCESS in 19m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c15cffd6ca1744eabf0da38fc4b9ad3a : SUCCESS in 45m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b154cb014d734f29bdb157b871eab983 : SUCCESS in 38m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0f975e16daf8428e985249eaa0ce4e68 : SUCCESS in 39m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba684acbdb5d4475b4e7ff01ad88f0f2 : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fef5b3505f1f41d38be83aeb8f13f7cd : FAILURE in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f3bbe48b596147778f68874d391e6e14 : FAILURE in 8m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fcd5c00cf66426db87e35bf496f627c : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":2},{"id":"bacf4f40cb8481d325d69e595fb27204e04eb4bd","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-06-04 15:59:48.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"80f7e19e5bab6ddfa0d5a4ef7d0a26fd0c7b5e6b","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-06-10 04:11:06.000000000","message":"Abandoned\n\nLooks good, squashed into https://review.opendev.org/#/c/706653/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e2d99f88d1129d228324de282974f099d9205c59","revisions":{"cbda376158282593037a104cee6102ce41c04be3":{"kind":"REWORK","_number":1,"created":"2020-06-01 19:40:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/732441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/732441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/732441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/732441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/732441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/732441/1"}}},"commit":{"parents":[{"commit":"4a72c1b19ee6eb2c754e8c02bb987c103b200aa6","subject":"Let developers/operators add watchers to object audit (simplified)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a72c1b19ee6eb2c754e8c02bb987c103b200aa6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-01 19:37:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-01 19:37:37.000000000","tz":-420},"subject":"wip: watcher ideas","message":"wip: watcher ideas\n\nChange-Id: Idf7f4ddc947abbc581688e47ded0b68ff0876552\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbda376158282593037a104cee6102ce41c04be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbda376158282593037a104cee6102ce41c04be3"}]},"branch":"refs/heads/master"},"e2d99f88d1129d228324de282974f099d9205c59":{"kind":"REWORK","_number":2,"created":"2020-06-01 20:32:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/732441/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/732441/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/732441/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/732441/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/732441/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/732441/2"}}},"commit":{"parents":[{"commit":"4a72c1b19ee6eb2c754e8c02bb987c103b200aa6","subject":"Let developers/operators add watchers to object audit (simplified)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a72c1b19ee6eb2c754e8c02bb987c103b200aa6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-01 19:37:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-01 20:32:41.000000000","tz":-420},"subject":"wip: watcher ideas","message":"wip: watcher ideas\n\nChange-Id: Idf7f4ddc947abbc581688e47ded0b68ff0876552\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2d99f88d1129d228324de282974f099d9205c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2d99f88d1129d228324de282974f099d9205c59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
