)]}'
{"id":"openstack%2Fswift~733047","triplet_id":"openstack%2Fswift~master~I227e8c2defe96fb5f33054167677ae551830d9e8","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I227e8c2defe96fb5f33054167677ae551830d9e8","subject":"docs: s3api doesn\u0027t support tagging","status":"MERGED","created":"2020-06-03 04:51:09.000000000","updated":"2020-06-03 17:45:55.000000000","submitted":"2020-06-03 17:43:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733047-1591206215201-292dea62","meta_rev_id":"3ba9448e4fa20de08650caad5e856fc5cb522e4c","_number":733047,"virtual_id_number":733047,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 17:43:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-03 17:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-03 17:22:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 17:43:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6c20b1e6cf455797fb13f18ebe85fce9ac5000a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-03 04:51:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50aab8ed1906d3c6242f7808d087c19e22f0b6e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 04:59:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f52f8f459944d86b95904d2b224aeb7 : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b0d836b9c002eebc5a4ab1dc28b796ea499e1e48","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-03 17:22:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e15296d68ca7ed700788c5e4ba574a72bec64958","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 17:22:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"947acfcd9af218c5bc52e0e0b0b93d27d26099ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 17:43:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/385a3641a49a40d1a0dc6836bdb22173 : SUCCESS in 8m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3018dc620e674d6886698e0e58dcbc59 : SUCCESS in 9m 40s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f1f6547ca22c422bb0347056c572deef : SUCCESS in 8m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"426ba1bfb67fceb02f9ef15a4a0b96f9006aca11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 17:43:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3ba9448e4fa20de08650caad5e856fc5cb522e4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 17:45:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/745170ee34834caa9d4a645d95b76401 : SUCCESS in 1m 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/af443b9c86d140c29379478eaa315db8 : SUCCESS in 51s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0c034f877ae24dea86579ad1b544e1ae : SUCCESS in 1m 01s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"738514c16422cb7215fbf3996282395edc09c290","revisions":{"738514c16422cb7215fbf3996282395edc09c290":{"kind":"REWORK","_number":1,"created":"2020-06-03 04:51:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/47/733047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/733047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/733047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/733047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/733047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/733047/1"}}},"commit":{"parents":[{"commit":"45a9ca39a494ca89f4c7366437cf3c419171a62f","subject":"Merge \"Enable s3api and staticweb tests across all func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45a9ca39a494ca89f4c7366437cf3c419171a62f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-03 04:51:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-03 04:51:00.000000000","tz":-420},"subject":"docs: s3api doesn\u0027t support tagging","message":"docs: s3api doesn\u0027t support tagging\n\nChange-Id: I227e8c2defe96fb5f33054167677ae551830d9e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/738514c16422cb7215fbf3996282395edc09c290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/738514c16422cb7215fbf3996282395edc09c290"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
