)]}'
{"id":"openstack%2Fswift~733863","triplet_id":"openstack%2Fswift~master~I9e6e7679f674bddddcc4440b38d5741aaece3393","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I9e6e7679f674bddddcc4440b38d5741aaece3393","subject":"Quiet eventlet exceptions in test","status":"MERGED","created":"2020-06-05 14:50:29.000000000","updated":"2020-06-06 03:33:24.000000000","submitted":"2020-06-06 03:31:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733863-1591414295801-0bd3b327","meta_rev_id":"8b3d2a334e059adca0eab2758cb7c68bdae3e529","_number":733863,"virtual_id_number":733863,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-06 03:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-05 21:16:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-05 21:16:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 21:16:39.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-06-06 03:31:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ae9c2b7757bee5d8ec28d3c5a61403bfdd7a354","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-05 14:50:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1273f4c523f5b941cd3a4672d1804d5a4f9ebfc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:47:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d67a9bd9ed459280e973d369dd3c66 : SUCCESS in 12m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bd68e408bd8f431d99116d14cfffa9d2 : SUCCESS in 11m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a7d31cd1cf494ef5a0c33896a6433fe6 : SUCCESS in 12m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/153e987746c64d7383ac3201ad397964 : SUCCESS in 22m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/358cb92f0e3244939c7a97a7023aed02 : SUCCESS in 17m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a742f21e544841eba0af3865b18d689d : SUCCESS in 17m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f841901c08eb4235b716acf44b88ebb8 : SUCCESS in 14m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3fd909cf273446ad87bedec5921c7afd : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d9d5c2c3c833427094840c764a077a2b : SUCCESS in 18m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/72c83e91b9074d989c73e097288e309a : SUCCESS in 25m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/083d60cf7fcf464fb5b2ec3698f76eac : SUCCESS in 39m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9a26ec65bfc14b2e9d36906ee6122813 : SUCCESS in 40m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/56e3c3a6c2474e5cab2bb99f9be6626c : SUCCESS in 17m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/db11cc66b3d845459f6a51d1cd2b5dee : SUCCESS in 47m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/68716f30e35a4476a346425feb63f2b2 : SUCCESS in 43m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d42d8d57c75b4cecb5a25943cc6d14a2 : SUCCESS in 18m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccfd136d701f4d21aa51e5e4f86b5517 : SUCCESS in 8m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0752d78aeeba48d09f4caa6410013e4c : FAILURE in 15m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ac9c6530939425c1317fa756548c69034535022","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-05 21:16:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nYup, that was lame:\n\n $ nosetests test/unit/proxy/controllers/test_obj.py:TestReplicatedObjController.test_primary_returns_some_nonsense_timestamp\n Unable to read test config /etc/swift/test.conf - file not found\n test_primary_returns_some_nonsense_timestamp (test.unit.proxy.controllers.test_obj.TestReplicatedObjController) ... Traceback (most recent call last):\n   File \".../swift/common/utils.py\", line 3363, in _run_func\n     self._responses.put(func(*args, **kwargs))\n   File \".../swift/proxy/controllers/base.py\", line 1317, in _make_node_request\n     src_headers.get(\u0027x-timestamp\u0027) or 0)\n   File \".../swift/common/utils.py\", line 1257, in __init__\n     self.timestamp \u003d float(base)\n ValueError: could not convert string to float: \u0027not-a-timestamp\u0027\n ok\n \n ----------------------------------------------------------------------\n Ran 1 test in 0.007s\n \n OK","accounts_in_message":[],"_revision_number":1},{"id":"e06c539e5e09f869317f4839b85fecb420742abe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 21:16:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f12971795d986cbf839bb0b2ec48d580f272e387","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 22:22:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d2fda5719a74bb8b307854b2f9c474b : SUCCESS in 9m 52s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e59ec6a484204d1785701503728f5699 : SUCCESS in 9m 23s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fc3b73fc52264d9d84587607fcb94b71 : SUCCESS in 9m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/027847f1dd2346a9a08da19fb2546157 : SUCCESS in 23m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/81cb8be1633e474dbd1a54187f27d69a : SUCCESS in 16m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1293ae93fc734b75892accd7ec77fece : SUCCESS in 23m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/44d52c686ac448bdb1a4b84592a71295 : SUCCESS in 15m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/994c8437028d4cca85a8af899425d478 : TIMED_OUT in 41m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/180be95c2f164b3ba61125caf935466c : SUCCESS in 20m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6307c986651145b7abc020843629e096 : SUCCESS in 18m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/60256ae3ec2a41299ebdc3c19c9f6847 : SUCCESS in 20m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f99167b41fb74b3798ce7b382aae015b : TIMED_OUT in 41m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/51441bd8ae9f4ae5913380a517c7774e : SUCCESS in 41m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/73e393bdaa174b46964fec4d7e43feb5 : TIMED_OUT in 1h 04m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb314fe3132e4ad3b7e329d4026e134d : SUCCESS in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b89589fdecff46f1ad71a7b980e05429 : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":1},{"id":"8bef837fdca2ec5899e8c90e33b236265665e6b6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-05 22:33:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1a8207295e6d69cdd1acfcfe27812900b35c9afe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 23:19:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98579105b6ac40e5bf172e009f9bbacd : SUCCESS in 8m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c4799cab4fd54ef8b457bb6f1a7d2a77 : FAILURE in 11m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ac2dac59ce9544e5b98e4aca3a072317 : FAILURE in 11m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/71738a706bc245ec85b8e767eb7653cf : SUCCESS in 14m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6be97fc28c3645029f8f2942d8e30415 : SUCCESS in 13m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3033c99a852446ad8da0c634eed962c5 : SUCCESS in 14m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b2d7424198634602a8e195b0c5d0592d : SUCCESS in 13m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b92a2f1f80844453b2a563c90e0e9654 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2cdb126f41b249bca780755e098f7d63 : SUCCESS in 17m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ef27b53e560c41259c1797936813fdec : SUCCESS in 16m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e31469b28b9440f3bef6f0ced9c14dcc : SUCCESS in 23m 32s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e17a6042ab1a47b68a4b378e3d6c332e : SUCCESS in 17m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/73743572f4764efca15181609de3ec6f : SUCCESS in 20m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dd23cd57eaa8493bbdb03789e7cd8091 : SUCCESS in 41m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/364f9c89ea3a41d39ccfdc15ba926849 : SUCCESS in 40m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8017a71a6409478ca2a5cba1b5525235 : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a4a06f830a043aeb2fdb0cfb14929b0 : SUCCESS in 7m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a90e4ac8ccde436ca0ea5becd11ad8e6 : FAILURE in 7m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f63752c207d1132232b651e0058afc0d2de2a2fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 23:20:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a0fa57154c042ce99c93db7c01eece6f9381416","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 01:27:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/775738c3330c4e4b948f8d796f1bc94c : SUCCESS in 7m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/067e941d5fd74bceb6d3cb1f61ce1b3a : POST_FAILURE in 11m 54s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c6c9aa2b2ce8413788839eb01f57b33b : SUCCESS in 10m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6144313872394078bb54b98740200325 : SUCCESS in 13m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c689ef2efce1474fb0ce1463064cb2da : SUCCESS in 11m 59s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f313512171594350afb715d996457d50 : SUCCESS in 25m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e836b747b0404ba7846b5b7025ef7064 : SUCCESS in 12m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bf25f64d45f24b47847aaa20fd339e93 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/00dd3f8d9a9c45d1acf76890baa5961a : SUCCESS in 34m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/88d9970bf6c24f998e541b1254ed1ecf : SUCCESS in 16m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4b51ee484cb34fab941fed73761197a2 : SUCCESS in 16m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4a21271a70724f7888eb43c5924f2d32 : SUCCESS in 16m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/defb35f48cc64f45b2db0042322cd288 : SUCCESS in 38m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/37107fe3124240dea9af148f3b77e8f2 : TIMED_OUT in 1h 08m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23113c62c3f34339b76700df62a40616 : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffd5167b05d44011a1d37e6379b73156 : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f7f3ac27a028fdcb07a5a98798104dbf75680ae9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-06 01:57:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eb265939d002a0e8929260981c651918c48cf3d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 02:43:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbfc0ec260a24878ae1b641b49ed6372 : SUCCESS in 8m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca62f62ed2844cae8fbea306a9fe3732 : SUCCESS in 8m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cada4727a2e84804b676b17af5c9ea20 : SUCCESS in 8m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dfc5a4c5de4843e493d330aa4e6d1722 : SUCCESS in 15m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/31256a8716744b249ca12a080d44f74b : SUCCESS in 15m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/73309d0110ad47938e6fd1152ddb059c : SUCCESS in 15m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/84e187bf4f59422c97f08de116d7d2d7 : SUCCESS in 15m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/612df00d73824e96b2303232eeef290b : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/44adb6b4426d46cfbeb1c92b0d6a4e30 : SUCCESS in 16m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f3aa05c03e3949308b00fe7ae04485aa : SUCCESS in 19m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b15ccfd150c44038895679f8f29ebcfc : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/761d69ee86e4421a8bcd65c3f3554100 : SUCCESS in 17m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d8bc3aa8ce7945568a490face5fd58e0 : SUCCESS in 19m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bc6b35fa3c7e4f18963cc50057a2c858 : SUCCESS in 40m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b8cffd9ac30f49d68f9659d6ceddfde4 : SUCCESS in 38m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ec3dd7d52414a02b226c3cf03a96385 : SUCCESS in 15m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dada53a18484ec6853fc856be70d09b : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6bc695140a3d4be998e896e78f792c34 : FAILURE in 7m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4845930f0650da2a182c3f04cba43882c83b0fe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 02:43:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9dbdbf1a92eafdd10cfb641fe4a59a18594fd278","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 03:31:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c6502151d2846e6862cb6e74928d6ab : SUCCESS in 7m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c9a21d200a084713a5c0cb9cf1a01f48 : SUCCESS in 9m 30s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fa6d95c7531546d3bbb2fa9016ef7482 : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d399af6fd0db4e59b7a991557281ecfe : SUCCESS in 11m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9e97d5a32983456f84ce0917f465e5ff : SUCCESS in 12m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/afb327fe829a4987a79dfbd81ae4520b : SUCCESS in 18m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/37f961fded994d8fab176973fd914dd4 : SUCCESS in 12m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ccde03591a7f4b9c91f97118d2766e59 : SUCCESS in 21m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/54364792c27245c6aad2738352b12144 : SUCCESS in 15m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0f23baf13fe645c3912b526976a1cb46 : SUCCESS in 22m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/43cb23e4f9164197bc9e7098ba33bb68 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f1e8ed9c916a4b5a9d9491798f2c31b2 : SUCCESS in 18m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a4b03aa540514b9cb0d99435d5c9e31b : SUCCESS in 40m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/765d57b73fdc4f11b275608321054cb8 : SUCCESS in 46m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cab31bd2411b4767accbca8eb7e5f3b2 : SUCCESS in 12m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d28618ac40494feab814afb2bae11aef : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"00080cbc394c3e059d51e1f467ee8e59be302f10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 03:31:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8b3d2a334e059adca0eab2758cb7c68bdae3e529","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 03:33:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f653afcdc52042b0b04c20770604365f : SUCCESS in 1m 30s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/777962205a4c48869554370849b95571 : SUCCESS in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/5c2bb222f21f486aba561bd5bd2974d9 : SUCCESS in 45s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc731198acec365cd46d2c5e0c7c297e948ebe66","revisions":{"fc731198acec365cd46d2c5e0c7c297e948ebe66":{"kind":"REWORK","_number":1,"created":"2020-06-05 14:50:29.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/63/733863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/733863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/733863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/733863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/733863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/733863/1"}}},"commit":{"parents":[{"commit":"56278100a1f1201a3adf318f083f17f083db35ec","subject":"Merge \"relinker: Improve performance by limiting I/O\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56278100a1f1201a3adf318f083f17f083db35ec"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-05 14:49:01.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-05 14:49:01.000000000","tz":-300},"subject":"Quiet eventlet exceptions in test","message":"Quiet eventlet exceptions in test\n\nThis traceback would probably show up in the wild, but it\u0027s not relevant\nto this test.\n\nChange-Id: I9e6e7679f674bddddcc4440b38d5741aaece3393\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc731198acec365cd46d2c5e0c7c297e948ebe66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc731198acec365cd46d2c5e0c7c297e948ebe66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
