)]}'
{"id":"openstack%2Fswift~733911","triplet_id":"openstack%2Fswift~master~I0dc5644a84ededee753e449e053e6b1786fdcf32","project":"openstack/swift","branch":"master","topic":"waterfall-ec-refactor","hashtags":[],"change_id":"I0dc5644a84ededee753e449e053e6b1786fdcf32","subject":"Start to decouple Object GET path","status":"MERGED","created":"2020-06-05 18:56:09.000000000","updated":"2023-07-24 14:11:41.000000000","submitted":"2020-09-04 14:22:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"733911-1599229337596-9d651cf1","meta_rev_id":"25eca6b2c52d6dd415e3bf67d4473660be8c9cf5","_number":733911,"virtual_id_number":733911,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 14:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2020-09-03 00:55:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2020-09-02 15:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-07 17:37:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2020-09-04 12:18:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-07 17:37:38.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-09-04 12:18:34.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2020-09-04 14:22:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ff6438544e2a7f698c0364c85f4a9cf0359345d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-05 18:56:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"636859bc7b0a3a41cea0e3a874226bc7b61a42d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 20:10:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab26807639642b88fba0f3c347b55ce : SUCCESS in 7m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/f2a5b7e7e2f646c7958493b706ec65ad : SUCCESS in 1h 06m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e216d22ab9374a77b1e8fe7661c5b0b3 : SUCCESS in 43m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/58d71a1b1c2048fa9f0ab8c519d069ff : SUCCESS in 9m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/89e76778c60d4e2ea918ededfc75cc7e : SUCCESS in 10m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a5474645e088430c9f75eff587cfc0a4 : SUCCESS in 10m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/520ac7e4975a4e7c9b03829b69c67264 : SUCCESS in 13m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bc51847075ad439890f97cb6da448596 : SUCCESS in 14m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/54c07a3d51c94c1c85080b4fd1f08bc7 : SUCCESS in 14m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/54335888b56b4f86b700e29e4574b75e : SUCCESS in 20m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d1e1d3bfc9db48ca9fa5dfaf0ae946bd : SUCCESS in 17m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aedefaa5e14b4cd7b2efa2dab59a34c3 : SUCCESS in 19m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4e8362c3a5864ed59163e4d209be4164 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5d6a3a781d6c420480a24d9ec31bae3c : SUCCESS in 16m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f478843ec3714169aea09a53e9d41f28 : SUCCESS in 16m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/532127c34ed647069995a92fd303846f : SUCCESS in 21m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6ccbf18cd47c491796c39577d30f0a19 : SUCCESS in 46m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae29c576081c481cbb6825976074ef65 : SUCCESS in 46m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a4b6bc515a5c4ab8abcd9ceb7f0ea49d : SUCCESS in 45m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/815bb7cce8a042f5a46c37d364b8fadb : SUCCESS in 13m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b174f557165403f9b3efb8ec340b554 : SUCCESS in 7m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de8a2d27f6ba4b57afb28b0730089ade : FAILURE in 10m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/837634f17b1746448313cb41825af71f : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e7c99e970ef80dffb7206e6d6c3c1db1969cac5c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-19 04:30:34.000000000","message":"Patch Set 1: Code-Review+2\n\n(5 comments)\n\nI think we\u0027ve made a pretty decent case that EC and replicated have sufficient differences that make it reasonable to think they may need to diverge. Would like to see how the it shapes up in the next patch before approving, but this is a straight-forward, mechanical change.\n\nStill have a couple references two the old ResumingGetter, but we could fix that in the follow-up, too. I wouldn\u0027t respin just to fix them, anyway.","accounts_in_message":[],"_revision_number":1},{"id":"14fbd26cfbcc332b83456eeaeb3e9a577aca297d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-19 15:18:19.000000000","message":"Patch Set 1: Code-Review-2\n\nno reason to land this until we\u0027ve got a +A on the next couple\n\nI think it\u0027d be nice to clean up some of those docstring references here; although I probably do catch them in the next one.  I\u0027ll squash things around next time I\u0027m updating.\n\nThanks for the vote of confidence!","accounts_in_message":[],"_revision_number":1},{"id":"59dfeff8b5a695ff9c33772aa645f0d4b0f610cb","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-19 22:13:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f1457bf44c4b78f22758b1fd1c1f547db7d4aa69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 23:35:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbf2bd3e65f9454181e1c6db00c9e8ef : SUCCESS in 12m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/578b3f4ed462429dab60c81ac879e66c : SUCCESS in 58m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4e61f0dcc13b4177ba7ce4aa90982823 : SUCCESS in 43m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c27fba08faf34cd8ad8124bea4f19ce3 : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e770141d8b654626959c1303ea8bdbc4 : SUCCESS in 10m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/627684b636334ad4b2e4c7c3e249777e : SUCCESS in 19m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/00c17ba3a09c41239718d23f4f2bcad8 : SUCCESS in 19m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/13720713df5f48d8aa9cf020f360a216 : SUCCESS in 15m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3164cf2b67c0431eabf15d7f365f5759 : SUCCESS in 32m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/916a9e2bff6b4b67aa884e0d1a511fce : SUCCESS in 40m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a531859dbbc04dfd9b7146eda82e5c7d : SUCCESS in 18m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/37c3686b974c47a7a6e6fe23aa143ba6 : SUCCESS in 31m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e5c416100a2b492295381445ef63f271 : SUCCESS in 32m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8678debe940849a680d18396d5b0160f : SUCCESS in 34m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e07d81f5406d4acf8054d818c0d58473 : SUCCESS in 20m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/409b7d211b104fcabfdb10fbfb0fcf76 : SUCCESS in 54m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bd5cb82d4c67408aae2e93ef991282a6 : SUCCESS in 41m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d248337ad08d4f19b1f69e651af7880d : TIMED_OUT in 1h 07m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30502a82ffab47f4966df37d9a84d685 : TIMED_OUT in 1h 05m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/697004e5819e4b7fbab2d7e4732eb0ef : TIMED_OUT in 42m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a65ff266aaef48cca98f83f53709ff1c : SUCCESS in 12m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e8b00d23975b4606b11ab3b53338d869 : FAILURE in 16m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5f293510370445089666ef76060bf6d : SUCCESS in 1h 19m 53s","accounts_in_message":[],"_revision_number":2},{"id":"6b0ea80f4ea023796d823e45b3571f0eaa0860eb","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-07-22 10:14:38.000000000","message":"Patch Set 2:\n\nThis is looking really neat, love all the comments, esp in the `__init__` of the ECFragGetter. Already makes it a lot easier to grok and therefore maintain.\n\nLooking good so far, onto the next in the chain :)","accounts_in_message":[],"_revision_number":2},{"id":"d85a2c26a2636afd90789fff76f3a3efbddbe45e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-22 17:20:17.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"355e406faca0b76886a9be37e489cc31e9cf2adc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:29:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f2bfaa49394a30a0f4d419e94e401d : SUCCESS in 10m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/b7978046895d4080894461128a511d2e : SUCCESS in 1h 05m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/319bbafbe8ad4948afb0ee09ff8d48d1 : SUCCESS in 48m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ecaa9223ec0d4f379b2a6faacf13f3ce : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e293e70fe56c40d7b680ec89efa43a1e : SUCCESS in 9m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f432bfe3e6414e31a9d53d7b525c7404 : SUCCESS in 14m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aba3558a69274aac8aeb037dd2b66969 : SUCCESS in 19m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b4f160fb7f8b44b4915278bb32770195 : SUCCESS in 19m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6bcc5d012b3e4f289ade0e3aca722f46 : SUCCESS in 17m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/008de3668de14e8e810f00ff478ace1e : SUCCESS in 23m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cdc0d9c3913c4933a92f2949773ccb8d : SUCCESS in 22m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a066d859c8a947baa3e7cbeccde96473 : SUCCESS in 23m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/14de3e4819294510850537cd466bb227 : SUCCESS in 21m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/da24abe7fdf24a46b653a5062cf5c530 : SUCCESS in 24m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cabc5d9c9bc6465db9d8ea91b4899c1f : SUCCESS in 23m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b053ac97ee644c9cb99385e68a0e3997 : SUCCESS in 22m 27s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/132a2f391a344a5794ff0f7cce27b8a4 : SUCCESS in 47m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/accad0bedbba4453b7d4f045cd0a0eff : SUCCESS in 50m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7af727d5f5194599a65dd41f7aaff30a : SUCCESS in 1h 00m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d41d0a328d24d8fa4f6371b92290fb2 : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38de5f04cadb404ca3357b13c90bc3bf : SUCCESS in 11m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/832858f8b0e14799a7c0bc737b474aea : FAILURE in 10m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af204f13afbd476ca5624647b48f91c6 : SUCCESS in 1h 12m 04s","accounts_in_message":[],"_revision_number":3},{"id":"19ebbdda2f051690e82c8ee72ad778e558ddf239","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-23 23:11:55.000000000","message":"Patch Set 3: Code-Review+2\n\nStill looks good.","accounts_in_message":[],"_revision_number":3},{"id":"5116c44efc34e9606474bc0ba0725bc85b3456ef","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-27 17:00:31.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"58c1774365e9ef3e7779c192443051c48577bacd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 19:33:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e8dd2ce5bf84428ba8ccf8144665d7e : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/47d612a4ef9b4e2898ada7d2659eda48 : SUCCESS in 1h 02m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/94133ccba0384b2195c5bba17958096f : SUCCESS in 52m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0dce51cc138c4b059489b2f51c81e306 : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4f440d4f1e674c67b11bc6428636bd81 : SUCCESS in 9m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/756ca90b4c624b248f3388629262bfe8 : FAILURE in 3m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/430815d6cc3d4edfad9eb692613f60dc : SUCCESS in 13m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9cfdb5aa052f4cfcbf2b6dfd315c5a70 : SUCCESS in 12m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a2d68795bc66455bbed0997c6f422bc9 : SUCCESS in 16m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/32839b729f5047058598f290b613d55c : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fd5258d2f4054c6da90c487cb2cd7c49 : FAILURE in 3m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/afc894869a6b4c0db445cef16221ec18 : SUCCESS in 20m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d97d81599df84ecca7de53bec9d98c08 : SUCCESS in 24m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/55404f4551334c02b6142afbe9244b0b : SUCCESS in 23m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e3e52ceaa4d4483fa049c8d5ed55836b : SUCCESS in 19m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/622d2a9eddbf4a5395345205f31c4ad7 : SUCCESS in 21m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/191fcab79e2246e4bc90ee0524333e2b : SUCCESS in 48m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/acd8756add854753941fbc665b7acf89 : SUCCESS in 41m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b77ccb47c0ea4042861af59406741a40 : SUCCESS in 41m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b25d458726a44bdcb23ed273a799f06b : SUCCESS in 18m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/208899ab198e484db12fd5940b7b2cf2 : SUCCESS in 5m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/24d1dde2a68f49d484ef5ee2a07df1ad : FAILURE in 11m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d939b45347ec4593a717d6f770695996 : SUCCESS in 48m 04s","accounts_in_message":[],"_revision_number":4},{"id":"25cd7591a91e92366fdd9fc2e21eb50ff17aafcf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-28 19:09:11.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f5c60a23fa87ee0042b68c86c82fd58ccca89a68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 23:02:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad8c8e70089c42409ec59ac570968f8f : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/dcc3644a9fc0441ca4371465bf4af3d5 : SUCCESS in 57m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a7924a65de4a416fade42ed647b0c5bb : SUCCESS in 44m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/183085b19ccd4012966a39cd5ee566a2 : SUCCESS in 10m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/27c75ee68eb4438aadf606fbbe4016e2 : SUCCESS in 10m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/27b471ae94a1410caf789bf011241377 : SUCCESS in 12m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f376208183c46ac9e382c221dd8bada : SUCCESS in 14m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ff5311db2cda4b95820752b167767361 : SUCCESS in 12m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8b94c38616944a84abbb3981b4c8548b : SUCCESS in 24m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8f3e87c88eac470186ba5a1a104e525d : SUCCESS in 18m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/79d51a0547674531ac3cabf4a76e68bc : SUCCESS in 17m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a85d9ba95f04f6d831358ca6213a368 : SUCCESS in 34m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f017c908b44e46349174cb3c6e4d93fb : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a8bdd70cdf554ffda93312adc9f2ed05 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8e3bb8616a754715951680279b472af7 : SUCCESS in 21m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/401fbf817f45418484bf4886f145255d : SUCCESS in 23m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1af41345eecb47c090de7f0725420d69 : SUCCESS in 1h 14m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/736443188d584c43a45f0f618a9dafc5 : SUCCESS in 34m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4a8a1856be9b49a1af2c42dcb9477cf0 : SUCCESS in 48m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c9e0dcea58d4a89a5c5d0e191fdc925 : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e0c7603d6284b9981744313bb7dbbac : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb0f5f1f10544a3e969e45169ce1b255 : FAILURE in 9m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0af0951e8319478a82f731301524ca42 : SUCCESS in 1h 01m 42s","accounts_in_message":[],"_revision_number":4},{"id":"2b2367d3a3a08502d11589e7022d7d6644cac127","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-08-04 18:46:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e2d856465a7682dc202003c63bffa5bb8f43be86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 19:48:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29b5712a692f40ed88e72081769e74ef : SUCCESS in 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/7091cb72d3b34a4abddd6570eba7cf35 : FAILURE in 47m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c5aea06f3e2a4b41b4386bdaf3df366a : SUCCESS in 53m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b298be6b6c94cc0b78362a906493581 : SUCCESS in 9m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d9069477fac74a54a77baa95c3295091 : SUCCESS in 10m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c5b37d4a29034c02953c442f636574f7 : SUCCESS in 13m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f1c75afc94048b6a5a81f93375e9d3a : SUCCESS in 23m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c90bdc03fc6b47e3af2a5f185cee01b6 : SUCCESS in 21m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dd6853654ebf427a82869dc72fff7c55 : SUCCESS in 13m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bbcd1cc9ca2d46ad89f5686ac576c6f9 : SUCCESS in 29m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/485e332eba1a435595a3837733b4c8ee : SUCCESS in 40m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/257c3c2c2bb449ed96cec097f1906c5b : SUCCESS in 41m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d975fa1c581648ae9c74031dd5d68d71 : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3a5a45b082884ca3a9be8abdb333924e : SUCCESS in 41m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a56c957c15684058b383b34be7963602 : SUCCESS in 17m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dce8656041e949ce9ec483e4daa37df4 : SUCCESS in 45m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8ed51f9f9d9d461b96c729b7df64f32f : SUCCESS in 49m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f2a367e408854d29ab671f8a86b05089 : SUCCESS in 45m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1c4e4a0de0794f4e94704e1ef2c132ce : SUCCESS in 50m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/217d8eebbeec40109425ec2d7d051aee : SUCCESS in 21m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04557613cc3344649995c62a8c8f0462 : SUCCESS in 7m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b5b46481fe184ad2bd57a001debcfe22 : FAILURE in 8m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/475b8ae08ec9441199a9379f58c31f77 : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":5},{"id":"b4ef5465768ccc877024e2eb5b6dfa80864b5029","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-08-04 21:57:01.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"04478acfd7469f123a2306a42a708583548f4089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 23:04:25.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaa57a74eb2d4f63a416f09160168a69 : SUCCESS in 7m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/e75890390b344c6aa634374f65546d50 : SUCCESS in 54m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e5c2e727e44f4acd92f325e7246c96a4 : SUCCESS in 48m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/03a4cdaf6bb145c19f21ca387d718bc0 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d4f89dfd1fee428c92c08808383f4f65 : SUCCESS in 14m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ad0c8cc9f8f04e58a9f8c072f193a936 : SUCCESS in 16m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a8252e580ed45ae9e69236481ca70f9 : SUCCESS in 20m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9c71b9ee33d84f99b71282a2c6a7c24c : SUCCESS in 43m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a0c6b75d73134305823f0bd61dce6dcb : SUCCESS in 42m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/32949ba133b8460ca812a6c0090f589b : TIMED_OUT in 1h 01m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1649f48d84bc4585b92e6143be1f371d : TIMED_OUT in 1h 00m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b04def18a92b44b2a3a59a0bf9de4b7d : TIMED_OUT in 1h 00m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/50b6d8737a344999a6168796b9f2756d : TIMED_OUT in 1h 00m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/15c6f853927d43e9ba11192f866a861d : TIMED_OUT in 1h 00m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3b1c5a8710744e48a32f5f1b08cf7a69 : SUCCESS in 45m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/662cd42ece02413894b7d78fd1b8ceec : SUCCESS in 22m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/37afe5507064492b80e5dafabf3a7827 : SUCCESS in 48m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8c4f9bbc61554714a9f611546ed9f2a9 : TIMED_OUT in 1h 03m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d6aeb2a980ab46f89a5b789629231859 : TIMED_OUT in 1h 03m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ecb710918b249399b8ddcb0da083332 : SUCCESS in 25m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e464173058154673985fba8dc321207b : SUCCESS in 7m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7683d17bf04841579e517b5116db3a21 : FAILURE in 8m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4a6ce66b2a64a6c8513dddde11854d8 : SUCCESS in 59m 06s","accounts_in_message":[],"_revision_number":6},{"id":"5f7ab38ff808308dac3c570c22bbf55551e271e6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-08-05 15:26:13.000000000","message":"Patch Set 6:\n\n(3 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"cc12d9512c20609ec91308233b84a5ec8388589a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-08-07 17:37:38.000000000","message":"Patch Set 6: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"9c41704b6bd0fc6b57fc4e1450f938a9303d333e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 18:57:49.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39c2f16676e747538454bb272574f314 : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/26ef912775c048348a041c65a0372386 : SUCCESS in 1h 17m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0a957f044eaf4cb486898bf6be9f2b84 : SUCCESS in 1h 00m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/237ff37822694eab961967b198fb9fc1 : SUCCESS in 14m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3ffa355ca3b6470cb719e23e2704d76b : SUCCESS in 14m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6c96ee7cea164a26b70244e713e7d807 : SUCCESS in 38m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7a28a468a9f146d382e3826c5c80f59d : SUCCESS in 38m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/afb134f2954f4486b5058c4121ee50eb : SUCCESS in 39m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2badffabdd0242998bd29ccd48b36c13 : SUCCESS in 40m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/451c6c4177274caca126acf8eef050fb : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fd473f5729f0480f892cca033be36f5d : SUCCESS in 19m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/99ed53fd6b8b40be86e61ca0531bf5e4 : SUCCESS in 18m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3af489c5e60f413ba2973a5c3c0ef6ee : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/acea426f4ec0438781caa90acf6fac3f : SUCCESS in 49m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/45fcc3b1224647e2a16f04185144820f : SUCCESS in 19m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e3789a77947846b9ba662c2dedb7cc15 : SUCCESS in 22m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/31a550812cc0477d8635b0e590f766f3 : SUCCESS in 45m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4486adc58fc402dac6002cf481b3e77 : TIMED_OUT in 1h 03m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dd50aafe77c54924ada151680aa36d23 : SUCCESS in 36m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59c48fc654ec4576961a7eca96d0dbda : SUCCESS in 38m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1827fa9022854b5196c796fc44d676ca : SUCCESS in 6m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd7d177ae75b45549102f323348fb518 : FAILURE in 11m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07a8156c8d0b4035a1edc4147950cba8 : SUCCESS in 1h 16m 37s","accounts_in_message":[],"_revision_number":6},{"id":"ae6c1bee426e39fabb6fd287dd8b41dd94ae7618","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-08-07 23:16:58.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"5a771b62d1add4e0d0f2a34ce65cc680be31df51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 00:38:08.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d92068fa9c7746bdbca0df755aa9565d : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/eb7040a667844900bb1d4f72c8ec1ad5 : SUCCESS in 1h 07m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/733edf960f8e4591a6d51c78affd1899 : SUCCESS in 50m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/29c92143c9ff4255a725ca67253ddfb1 : SUCCESS in 7m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fa0d83f486ae4e0da263084502b274cb : SUCCESS in 8m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b4478cc39e69441f8ed98704b14d2b15 : SUCCESS in 18m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/58db7c92f4c5470fa29e82ae0da9f125 : SUCCESS in 11m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/15435611abfa4519bbcc1392a01f4a6d : SUCCESS in 18m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/53caf34d78a34342ad15f43f27eb7a68 : SUCCESS in 37m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5d690b1eb9de42dfba7d74c698f67bcb : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f35ad16fad2745fe975860d2056793b3 : SUCCESS in 18m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a17f3a21354447fcbd9e59dd594e5996 : SUCCESS in 23m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/48209bf1a42d47159c856dfa1a990129 : SUCCESS in 14m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7ab4c8f3b9654a0d99a6a3a527fe83db : SUCCESS in 26m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d47fe764e0f4430cbb3d68f442e2cb76 : SUCCESS in 16m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5f9208c9ea2644d3b09bf517f72df605 : SUCCESS in 39m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/09c6e6c62ba7473d94b188215397c781 : FAILURE in 35m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ea0310f439b4828bc94383037587420 : SUCCESS in 52m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f54e06d94acd46b0b9b79c15281c99cc : SUCCESS in 41m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60db24bc65e74069933ea9588ba7365e : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37fb46cdbabb48fb846baad0d2234f80 : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/54da1dcbb4ee468986fd7dc20abfd6af : FAILURE in 7m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88d8573f640f47858f6ac72b6bcf1957 : SUCCESS in 53m 06s","accounts_in_message":[],"_revision_number":6},{"id":"16f112fc789a33a43ed35ea7d7f8d5c62a40645b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-08-08 04:32:59.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"988b4fb51281eb985894419d0544f6a42bce0eea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 05:31:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29376807bc3f45db8bd2dfb9aa6a6be0 : SUCCESS in 6m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/a379278d42cb491fb0b720be0b759b19 : SUCCESS in 57m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/25a72192bbc64f66b48262b14ec823a5 : SUCCESS in 44m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bddc447240f64bc1af9d754dc2426698 : SUCCESS in 9m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/353b2ce8e6d04e34b711dff2c6fe67ce : SUCCESS in 8m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a1f1462c4fd64d3fa5913f554bcbc20d : SUCCESS in 16m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/586072fca6714679a289d1acd3841119 : SUCCESS in 15m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1038774846a647ee8c3a4e199f7e7ea9 : SUCCESS in 16m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e6495176df0643fba86f97492f071a2c : SUCCESS in 15m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/efbf167e33be4c2e826982142f82fb29 : SUCCESS in 20m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fe9d4dbc91104d688da5ef0941ad713a : SUCCESS in 20m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7a6f1a954ed84e399aeb7c7bcae452a7 : SUCCESS in 20m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3d88b0ea123a406b9346a88922c7478c : SUCCESS in 22m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9ba7a28458b54a71a056d5cee95ec79a : SUCCESS in 17m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/51684ee4cb884e2aaad469ddc48961ba : SUCCESS in 24m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bf66b638c7d344dcad632df0d530fbbe : SUCCESS in 19m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bdac9de999f045d69317c15cacf0dbe8 : SUCCESS in 34m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/58df51a35fae440ab701cfb04dd5d02a : SUCCESS in 31m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/07224eeb8f8d4abfa6f5c655408fddea : SUCCESS in 31m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d6b6335ce9a4d0f833aabb8dc032fed : SUCCESS in 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/803701a993d84f2fbe0853719295a73c : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3938b21902c34f24b9ee0d65d61651dc : FAILURE in 7m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76f441fdec444db8b681928368dc11e7 : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":6},{"id":"5e2673de5f6f24d2e7b95df94ed3727a6fdb73fb","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-02 15:51:01.000000000","message":"Patch Set 6: Code-Review+2\n\nI\u0027m ready to merge this and \"Extend concurrent_gets to EC GET requests\" as soon as y\u0027all are!","accounts_in_message":[],"_revision_number":6},{"id":"252dae43baf9c2ae74fd3c19543a81432793ea33","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-09-03 00:55:38.000000000","message":"Patch Set 6: Code-Review+2\n\nYup, really like this clean separation and like we\u0027ve mentioned in the meetings. Duplication here is for ease of maintenance. So Love it!\n\nWill +A it as soon as I\u0027ve looked and some of the following patches :)","accounts_in_message":[],"_revision_number":6},{"id":"a1020077f714a0de49200f07865c0a4eef21d67a","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2020-09-04 12:18:34.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"3f086471c0edbb8fd578544e47ed946c09e88c7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 12:19:00.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"68de3bcbe6e08e71378cb4eec69f1ec781ab9562","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:22:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"b0fbf10dc4edaca88bc0fccd8fa4747d4babc471","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:22:17.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a31bf15bc334df9b4760d3b35403a10 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/14ac00ccb6f94dcb902c7f597275921b : SUCCESS in 57m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75487e827bdc4f0fad06b9a16f3d3a95 : SUCCESS in 43m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4cb7a75d4cc448bd800de333f6958bf2 : SUCCESS in 10m 35s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/54731a90df6d4968bacb3c05298b112f : SUCCESS in 12m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dbb712a4f0f1489ab85b1ae99cdf1f31 : SUCCESS in 13m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/65e6df167b02420da7cb5d65bab07fb9 : SUCCESS in 13m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7a06c4c6ae244d16911dc95f411ef4d0 : SUCCESS in 24m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ee9fabdf80ad4b179e14f257dc34a70c : SUCCESS in 13m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ae3d01a98aaa4aba8c4e9075d6b2fc89 : SUCCESS in 19m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/69a038807b7240439715b3bf6b5ba053 : SUCCESS in 20m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/febe496f66c448879fd5e768c4893360 : SUCCESS in 17m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f0102641527c4dbb905bf88b109f545b : SUCCESS in 16m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0bbe76e475a4453d92939b1e77ae8ab3 : SUCCESS in 18m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fc0c24337c2e48449f50b71687af6177 : SUCCESS in 46m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2048382d9b04401aaa28afc88b0a713 : SUCCESS in 41m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/02030c873fdd4119b64990e42d2d5e9b : SUCCESS in 49m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b3cbdde67054caf8d8bcaec1ad8ffe7 : SUCCESS in 14m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/573bfe10b90846aaa6fd744db0eabf1f : SUCCESS in 5m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9dcb0f61cfa049cfac62df89509d578a : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":6},{"id":"5ed0f9bf5809fbd963d130ce8c901920291a3545","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 14:24:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae3e551499e1488281adb9ccec901ddc : SUCCESS in 1m 27s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b1aa0d9c0d604ad688bef0b70bed0197 : SUCCESS in 1m 23s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d29de732dc184ca0b6b9c1b56bc64b6e : SUCCESS in 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"25eca6b2c52d6dd415e3bf67d4473660be8c9cf5","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-24 14:11:41.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5b2c846c69cba6dc707ff32318b715791711f34e","revisions":{"ffda7b149fb53ce88b6c92aad8f66b558967fbc6":{"kind":"REWORK","_number":1,"created":"2020-06-05 18:56:09.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/733911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/733911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/733911/1"}}},"commit":{"parents":[{"commit":"56278100a1f1201a3adf318f083f17f083db35ec","subject":"Merge \"relinker: Improve performance by limiting I/O\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56278100a1f1201a3adf318f083f17f083db35ec"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-05 14:55:05.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-05 14:55:05.000000000","tz":-300},"subject":"Start to decouple Object GET path","message":"Start to decouple Object GET path\n\nThese shared primitives are brittle and slowing us down.  To find a\nbetter way to share code we need to start by separating the very\ndifferent EC vs Replicated connection logic.  Eventually there it may\nbecome obvious that we can combine the common handling of all object\nresource download which is irrelevant to replicated database resources.\n\nChange-Id: I0dc5644a84ededee753e449e053e6b1786fdcf32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffda7b149fb53ce88b6c92aad8f66b558967fbc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffda7b149fb53ce88b6c92aad8f66b558967fbc6"}]},"branch":"refs/heads/master"},"2aa4fe6c6a23c7b1eafe211dcb7ba779a6f358b9":{"kind":"REWORK","_number":2,"created":"2020-06-19 22:13:30.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/733911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/733911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/733911/2"}}},"commit":{"parents":[{"commit":"5cf5548a85286b09a3441176dc773c2874203a39","subject":"Merge \"s3api: Add basic support for ?tagging requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cf5548a85286b09a3441176dc773c2874203a39"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-05 14:55:05.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-19 22:12:45.000000000","tz":-300},"subject":"Start to decouple Object GET path","message":"Start to decouple Object GET path\n\nThese shared primitives are brittle and slowing us down.  To find a\nbetter way to share code we need to start by separating the very\ndifferent EC vs Replicated connection logic.  Eventually there it may\nbecome obvious that we can combine the common handling of all object\nresource download which is irrelevant to replicated database resources.\n\nChange-Id: I0dc5644a84ededee753e449e053e6b1786fdcf32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2aa4fe6c6a23c7b1eafe211dcb7ba779a6f358b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2aa4fe6c6a23c7b1eafe211dcb7ba779a6f358b9"}]},"branch":"refs/heads/master"},"15107d01b990c7f63a3ff4746a85a655f59ec312":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-22 17:20:17.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/733911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/733911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/733911/3"}}},"commit":{"parents":[{"commit":"b1f38728963febfa409a803e5d44ff17a3e2226f","subject":"Merge \"py3: Stop munging RAW_PATH_INFO\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1f38728963febfa409a803e5d44ff17a3e2226f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-05 14:55:05.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-07-22 17:18:38.000000000","tz":-300},"subject":"Start to decouple Object GET path","message":"Start to decouple Object GET path\n\nThese shared primitives are brittle and slowing us down.  To find a\nbetter way to share code we need to start by separating the very\ndifferent EC vs Replicated connection logic.  Eventually there it may\nbecome obvious that we can combine the common handling of all object\nresource download which is irrelevant to replicated database resources.\n\nChange-Id: I0dc5644a84ededee753e449e053e6b1786fdcf32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15107d01b990c7f63a3ff4746a85a655f59ec312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15107d01b990c7f63a3ff4746a85a655f59ec312"}]},"branch":"refs/heads/master"},"0ad440525f7cf2ab87ee6d445d34a9d7ab64f7f8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-07-27 17:00:31.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/733911/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/733911/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/733911/4"}}},"commit":{"parents":[{"commit":"c613761b42c10639c4497e10f1466704261df574","subject":"Merge \"Treat add/remove/rebalance during part-power increase as error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c613761b42c10639c4497e10f1466704261df574"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-05 14:55:05.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-07-27 16:59:27.000000000","tz":-300},"subject":"Start to decouple Object GET path","message":"Start to decouple Object GET path\n\nThese shared primitives are brittle and slowing us down.  To find a\nbetter way to share code we need to start by separating the very\ndifferent EC vs Replicated connection logic.  Eventually there it may\nbecome obvious that we can combine the common handling of all object\nresource download which is irrelevant to replicated database resources.\n\nChange-Id: I0dc5644a84ededee753e449e053e6b1786fdcf32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ad440525f7cf2ab87ee6d445d34a9d7ab64f7f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ad440525f7cf2ab87ee6d445d34a9d7ab64f7f8"}]},"branch":"refs/heads/master"},"6598ebda8815248961b1cc35f9bb8865f715578a":{"kind":"REWORK","_number":5,"created":"2020-08-04 18:46:53.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/733911/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/733911/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/733911/5"}}},"commit":{"parents":[{"commit":"50800aba37947f6c1a88b8d6e0886049a9b8a86f","subject":"Merge \"Update SAIO \u0026 docker image to use 62xx ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50800aba37947f6c1a88b8d6e0886049a9b8a86f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-05 14:55:05.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-08-04 16:36:53.000000000","tz":-300},"subject":"Start to decouple Object GET path","message":"Start to decouple Object GET path\n\nThese shared primitives are brittle and slowing us down.  To find a\nbetter way to share code we need to start by separating the very\ndifferent EC vs Replicated connection logic.  Eventually there it may\nbecome obvious that we can combine the common handling of all object\nresource download which is irrelevant to replicated database resources.\n\nChange-Id: I0dc5644a84ededee753e449e053e6b1786fdcf32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6598ebda8815248961b1cc35f9bb8865f715578a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6598ebda8815248961b1cc35f9bb8865f715578a"}]},"branch":"refs/heads/master"},"5b2c846c69cba6dc707ff32318b715791711f34e":{"kind":"REWORK","_number":6,"created":"2020-08-04 21:57:01.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/733911/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/733911/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/733911/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/733911/6"}}},"commit":{"parents":[{"commit":"2a6dfae2f3d50750de2bdb5f31ea52070588e895","subject":"Allow direct and internal clients to use the replication network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a6dfae2f3d50750de2bdb5f31ea52070588e895"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-05 14:55:05.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-08-04 21:53:03.000000000","tz":-300},"subject":"Start to decouple Object GET path","message":"Start to decouple Object GET path\n\nThese shared primitives are brittle and slowing us down.  To find a\nbetter way to share code we need to start by separating the very\ndifferent EC vs Replicated connection logic.  Eventually there it may\nbecome obvious that we can combine the common handling of all object\nresource download which is irrelevant to replicated database resources.\n\nChange-Id: I0dc5644a84ededee753e449e053e6b1786fdcf32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b2c846c69cba6dc707ff32318b715791711f34e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b2c846c69cba6dc707ff32318b715791711f34e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
