)]}'
{"id":"openstack%2Fswift~735221","triplet_id":"openstack%2Fswift~master~Idbea6fadefb2061f83eed735ef198b88ba7aaf69","project":"openstack/swift","branch":"master","topic":"s3-logging-and-metrics","hashtags":[],"change_id":"Idbea6fadefb2061f83eed735ef198b88ba7aaf69","subject":"s3api: Use swift.backend_path to proxy-log s3api requests","status":"MERGED","created":"2020-06-11 19:39:57.000000000","updated":"2020-08-13 00:22:20.000000000","submitted":"2020-08-13 00:21:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735221-1597278066419-b7eb80e6","meta_rev_id":"7c5d62d5e70f6663f8b798ad6be273c1fb38be62","_number":735221,"virtual_id_number":735221,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-13 00:21:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-08-03 19:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-07 15:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-08-10 20:32:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-10 20:32:08.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-08-13 00:21:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c241158d7604d86444948dba25dbfcc4acb4acc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-11 19:39:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbe1b1a60d1db3695a6480f25ac18706fe66ddcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 20:44:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc2a7e3bf58f4afb94e803e74b43f785 : SUCCESS in 13m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/74740239a87c4a6884e317a69a54eff5 : SUCCESS in 1h 02m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5917fa73c624e59b68544bb2aaafd17 : SUCCESS in 50m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5cb2b02e71c44ea4bea9784fc386fc55 : SUCCESS in 9m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e5ccfc1988654f15bcfd7b56d710cd25 : SUCCESS in 9m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/110c846987754379902d56fb1e95d8ff : SUCCESS in 16m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/db46e8438a86483a869333f416208a46 : SUCCESS in 18m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/81ad9340aa4e4e7496f394a895524357 : SUCCESS in 20m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/885f0930700a4bcf8a948d4fb967527f : SUCCESS in 21m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7a9a151d9a794f21a913f4ce5235fd52 : SUCCESS in 24m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e6c30ddae9024d159d4f5e51baeb1f6d : SUCCESS in 19m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/38f18258f5f5461dbd84e2e9cab2b448 : SUCCESS in 24m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/82bfbb21b4d54b4b88e889a5957ced52 : SUCCESS in 19m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/98e0dfd71cb04640bd8a041e18d9c8b2 : SUCCESS in 18m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/02d650ab6e6b456a887375ebedce993a : SUCCESS in 17m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2495a0a0ef5143b9923410a375327b0a : SUCCESS in 23m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/78a18724e2ad434e8633a21090e1b570 : SUCCESS in 53m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/17a3ec6149b8477890d9522ae9a9d50c : SUCCESS in 47m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/194770785332433d95278373c9cff6bb : SUCCESS in 46m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/248b083e4d27472bbf907598c05c8a47 : SUCCESS in 14m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed4400aebfe41079ab773a03fe35ec5 : SUCCESS in 7m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/577e860564c54bad92ee962ca87f4e90 : FAILURE in 16m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b7d947b721a4659b1b2f1687b1fcb3a : SUCCESS in 58m 23s","accounts_in_message":[],"_revision_number":1},{"id":"bf80193d75809b9007dd1a8023c0557699d15e6c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-30 21:34:32.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nI\u0027m not so sure about this one... it\u0027s ... better i think","accounts_in_message":[],"_revision_number":1},{"id":"b7c5b47c4a43aadf2bcd0c9b20cf4cdbd80bd534","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-01 13:42:30.000000000","message":"Patch Set 1: Code-Review-1\n\nMaybe complete multipart isn\u0027t the best example:\n\nJul  1 13:37:03 saio proxy-server: 127.0.0.1 127.0.0.1 01/Jul/2020/13/37/03 POST /s3test/foo%3Fuploads HTTP/1.0 200 - Boto3/1.13.20%20Python/2.7.17%20Linux/4.15.0-108-generic%20Botocore/1.16.20 - - 256 - tx397b1ff5514a40ce859d0-005efc917f - 0.4783 - - 1593610623.420646906 1593610623.898972988 - AUTH_test s3test%2Bsegments foo/ZDBmMWFlMjctNTg0OC00NDRkLWJkNzMtNDBlMTYzNjVmNTYy\n\nHere s3 client did a POST to /s3test/foo?uploads\n\nBut our log \u0026 metrics parse the swift path with the marker because of the PUT:\n\nJul  1 13:37:03 saio proxy-server: 127.0.0.1 127.0.0.1 01/Jul/2020/13/37/03 PUT /v1/AUTH_test/s3test%252Bsegments/foo/ZDBmMWFlMjctNTg0OC00NDRkLWJkNzMtNDBlMTYzNjVmNTYy HTTP/1.0 201 - Boto3/1.13.20%20Python/2.7.17%20Linux/4.15.0-108-generic%20Botocore/1.16.20 - - - - tx397b1ff5514a40ce859d0-005efc917f - 0.2918 S3 - 1593610623.598799944 1593610623.890639067 0 AUTH_test s3test%252Bsegments foo/ZDBmMWFlMjctNTg0OC00NDRkLWJkNzMtNDBlMTYzNjVmNTYy\n\nPerhaps we can aim to parse both (to get the account) but then log/metric the s3api request based on the s3path components","accounts_in_message":[],"_revision_number":1},{"id":"27077c56a8e6609473444b39e3aa85efc3d6f6b7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-01 14:40:46.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nthis change is inexorably linked to https://review.opendev.org/#/c/728298/","accounts_in_message":[],"_revision_number":1},{"id":"685e3629b2dd6f4e2f0df097d7555d24fabe7db2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-01 15:56:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"549fe1ea393549100ccb21fed79166ef16d1a8ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 17:16:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/927b7e7df67d4564ae3ef01a2e295ab8 : SUCCESS in 11m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/e26e0c0b61c1433a9769af8e687d4382 : SUCCESS in 1h 13m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6cbb15a5bf264e78ab75096ca0c4400f : SUCCESS in 43m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fcc3efc747f0488c9679a24fb4fd52ff : SUCCESS in 10m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/93fcd7f2413b483ead99adfa1baa8c2c : SUCCESS in 11m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/26c192ef844f4bba95781523e400ca4c : SUCCESS in 15m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/097ecf008d0b4cfc8c308d1b961b4753 : SUCCESS in 15m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6473a86c1c3c4e37976d46236867e85c : SUCCESS in 21m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/05cf9454f9634605a5d6722809a8edfd : SUCCESS in 16m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/14ebb387c44145c3970cc918f7a168e1 : SUCCESS in 18m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/32b2f0f28aad4f209fef5d406e9d33ab : SUCCESS in 19m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6a62ea8453df42f18a30126524017bde : SUCCESS in 25m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f0f4236011b84ba3b153041d89b4fe48 : SUCCESS in 20m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b8dd4d5462c849f991982536afd8f15a : SUCCESS in 20m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/250f83e8b6214284a4ba0f461e65186a : SUCCESS in 20m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/60fe3afff47d418fa57eeb90e3ac24d1 : SUCCESS in 28m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5bbc7e6291d1466982775c848f569ac8 : SUCCESS in 44m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/01b40a93245d49bcb693a4daf55007b5 : SUCCESS in 51m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2134a2daac5943eda10f2baa9fc4ac42 : SUCCESS in 47m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe0aca85885347b981640471647d3179 : SUCCESS in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a90177ed2e34f6a8df469ff71f6fff0 : SUCCESS in 7m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ac5b8ae9381400293707005263b4dff : FAILURE in 13m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0aa42946c332471ead6f15f217fcf514 : SUCCESS in 55m 38s","accounts_in_message":[],"_revision_number":2},{"id":"cbc08619dbd374e2b70d8345e6c0f8daff97d93d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-08-03 19:38:56.000000000","message":"Patch Set 2: Code-Review+2\n\nI don\u0027t think there\u0027s anything else to address here; this should get merged","accounts_in_message":[],"_revision_number":2},{"id":"b3b0c0f161ac762200aebe859685717c2c517ec5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-08-07 15:57:28.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nI\u0027m still not sure I actually *needed* to introduce the s3api.backend_path key, but w/e -- we can refactor it away later if we really need to.","accounts_in_message":[],"_revision_number":2},{"id":"95c9ed168d27bb3e2e3866fe2847057ebebf9b3e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-08-10 20:32:08.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c07d8e01c813f67ee988752cdb6d90fea0f10f29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 20:32:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"54dbe2924faa2322de78343f30f268dc85c0bd07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 22:33:46.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b35d4944d20481490c3f038c5dc315e : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/92326986cc6444eaaaac04a208cee713 : SUCCESS in 1h 01m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/17bd2359a3e74501a5de4cd7e318d869 : FAILURE in 1h 06m 04s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bd8f51cbe89a4187be5f1cf2e876b7c1 : SUCCESS in 12m 05s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e5a8b1ae196b4d88bd7b637334bf9917 : SUCCESS in 8m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7652ebf1b7df4afa964934192cd82a41 : SUCCESS in 31m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/feac347642784b1698f34c5c882c0e5a : SUCCESS in 31m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b4521581db13448c913679f21e2d9bfb : SUCCESS in 30m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/52c092fe2dbf4ec0ab1b8c3fc02491a5 : SUCCESS in 15m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3caa1d051cb14245acbe7b5218388b31 : SUCCESS in 56m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/27724f62ca704ca98e388fd170debf0c : SUCCESS in 15m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/961ff9be33f746918196b34f396ed78a : SUCCESS in 19m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f5ba6ffb68614e369200603978364a88 : FAILURE in 58m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8708db701b614a988d361499deec2973 : SUCCESS in 17m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/94be969f1fb8475d9199a479d317fd75 : FAILURE in 1h 17m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8529f0ede2164b56b6179e8d858664ae : SUCCESS in 42m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b816577a4974468874f3e00a74702fb : TIMED_OUT in 1h 03m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f66933ad0fcf413da2ad6bde22d7afd6 : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad80588e93cb45868b729b9ad247042d : SUCCESS in 7m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f01a06f27a0341cfbd69998839cb64ab : SUCCESS in 1h 27m 44s","accounts_in_message":[],"_revision_number":2},{"id":"2adec08eae32d110323e4bd44a62b8c436e0ce77","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-08-12 18:22:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2ee577f9c00ca173cb49026afdd50d2da281467c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 19:55:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e5a7d24cbd41e088e327651c69887d : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/e80c7979025d466e97ca9085039947a8 : SUCCESS in 1h 03m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/33baf2b86c1a47d0bfeb98fdaa0477cf : SUCCESS in 50m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6ba12b1ffc3e4d9c83186b24fde20a90 : SUCCESS in 8m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e373e7e524e346f1bdc5ea0a348f8ebf : SUCCESS in 8m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d127ed6ce52b4fef89ae4fa5d32301d6 : SUCCESS in 12m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/26ad5503e46e46cdafc44d1d6b2b23cb : SUCCESS in 14m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/463ced79c4d243d1a852fe9fe7a3edad : SUCCESS in 12m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c7c0c97dcfad4f64aedb2abbbc431817 : SUCCESS in 24m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/41806851f5ff4d1392dc26786dd74212 : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7540330eba3248fca00a924860b31268 : SUCCESS in 45m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/607a78349c2745b08dad5f072f65458f : SUCCESS in 19m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4bc3eb7c63314701a643bcf7e40fd58d : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/15aec58599974b59ade1a32274afb4cb : TIMED_OUT in 1h 01m 10s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1fa075df55934c34b7365eb7b52399bc : SUCCESS in 20m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c191223fc7464c368d4f9049ad226fac : SUCCESS in 25m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/83aa22c1e6284179b87bc3474f3526c8 : SUCCESS in 48m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/275f1916db0646e1adb17192ded93f04 : SUCCESS in 41m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a0fe7aa63a3342399a2d24750bfca54b : SUCCESS in 51m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1457c137249549c0bc61cf160a41eb7d : SUCCESS in 24m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/215408291ac948e29362c734b1c00824 : SUCCESS in 5m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1dc0ede9f2b84634ad7612aacce844fb : FAILURE in 9m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bb739414ac146298a7be1f1b7bbb9b3 : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":2},{"id":"6630cced774d6bcf4f03359b54af1e2d6c0ac409","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-08-12 21:09:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1bf914a45e2eff569e045230d280514f9a9383bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 22:49:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7f095d316c044868a363c76e6280ecc : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/8bab48fff1ff46dbb340912182e4b2c6 : SUCCESS in 1h 23m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/451a3e8fad11492b9a8640304e71c7f6 : SUCCESS in 45m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4ea9673baa7348daa107adfd94925578 : SUCCESS in 8m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8eb8f32ed6fa423aae580604bb9adc9b : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/58f633a2a0c54206bc4cbd966e8b076a : SUCCESS in 12m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bac84b54adf7408f9e986fd5fa6be248 : SUCCESS in 12m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/51f032ff92e8492f9a34e4c6fdd9c648 : SUCCESS in 12m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fa1b24726ffb43f8809b29e9e9af3574 : SUCCESS in 13m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/28ef29dfbeec4e35a393490e8043c455 : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6981e57154944c6eb5fccf1562f4aad2 : SUCCESS in 16m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/46771051693540bcb2029d11db66dcc6 : SUCCESS in 16m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bcc20802b1394680b2d700b8b6149bdf : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ccd078901fd4444fa3d2279000401ed4 : SUCCESS in 19m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9bcae2908e654997ba3d716d7361fefe : SUCCESS in 18m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e7b5f3ce8a7b4b908f669317c5afbf45 : SUCCESS in 20m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e3de419c001d497998694c326b7f2eb7 : SUCCESS in 42m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/390ef8b8e73f48788b9d5ef5cb6a6396 : SUCCESS in 43m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5b2a142e086c4bb0bcbfc2e29274c36f : SUCCESS in 45m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f55a8932587a4bcb846350090c616c5f : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bde6cb4c80849179b1fdfafe23714c2 : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1dc2493be9a24b10998c03cfdad3e43e : FAILURE in 7m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66e850658f324693a98d97971927f64d : SUCCESS in 53m 54s","accounts_in_message":[],"_revision_number":2},{"id":"3374a7aea2f2897d5b3c72ff28fc6693e1e78f95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 22:49:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0bac7090b4fcfc9a6bd293e70943af92f5cf70ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 00:21:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d655963a5d44c2ab13b6567f65092d8 : SUCCESS in 10m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/1c2bbbbc1fce4404a99ff0fae1f68930 : SUCCESS in 59m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5c72b577e23149f89136d44e4511bf4d : SUCCESS in 46m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8ecd1dd34b2f48689d8322ec6febc22a : SUCCESS in 10m 20s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3d5c5070c3fc4dcb8e6714a0bfdaf268 : SUCCESS in 10m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0559eb2f13d5470b8a0af0c173977d4c : SUCCESS in 12m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ace46b3548154b8896b2957585ec2656 : SUCCESS in 12m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/284adc5060724dd49db28b621a0c5040 : SUCCESS in 13m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d3769674c9ce4b66b3a307b68e357b87 : SUCCESS in 13m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c320c726253a42c8972c07ddef471a09 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/33b7c992827f4b18999d42cd608f66b4 : SUCCESS in 16m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/add17224f808450eb6ce1d7934544999 : SUCCESS in 18m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b00eff27e45a436abbfba35cb8cf3a28 : SUCCESS in 19m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/58f29f025b6a4a12aed94341c4921989 : SUCCESS in 22m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9240a40dcd114ad6991a3cf5ea0d51f2 : SUCCESS in 47m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9550c8f7b2fc448e8d49e424801bb9c1 : SUCCESS in 37m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4dc2016836b74beca58c6e5df3837e54 : SUCCESS in 44m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cfd0975962246e6ab88e2df5df38a3b : SUCCESS in 16m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f2011e0097b4c329aa3eaa42e28c5db : SUCCESS in 5m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97a0d88499504c218baa619d2254d8d2 : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":2},{"id":"8d705c82117f707b052f4b6367db63009e70e8f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 00:21:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fe52429ac13b4ae2089548fa4e979c1e433dfb92","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 00:22:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b4baa65073642f5862fb700397be92d : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b8203fe8cb2a41efbc97edd5e5cd2172 : SUCCESS in 49s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/32c8d2c5183c48bd90b2674c12b43499 : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b2efd185ce0cc794a5d94880cb6b4a46075f8db7","revisions":{"16e85f6d7d11252fc2f3af97ef7c65bdeecb0656":{"kind":"REWORK","_number":1,"created":"2020-06-11 19:39:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/21/735221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/21/735221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/735221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/735221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/735221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/21/735221/1"}}},"commit":{"parents":[{"commit":"3c0a448f06fc4469a50cdc64f8b9a5d8f0d5ac5d","subject":"Merge \"memcached: Plumb logger into MemcacheRing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c0a448f06fc4469a50cdc64f8b9a5d8f0d5ac5d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-11 16:25:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-11 19:39:10.000000000","tz":-420},"subject":"s3api: Set swift.backend_path to the last-used path instead of the first-used","message":"s3api: Set swift.backend_path to the last-used path instead of the first-used\n\nChange-Id: Idbea6fadefb2061f83eed735ef198b88ba7aaf69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16e85f6d7d11252fc2f3af97ef7c65bdeecb0656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16e85f6d7d11252fc2f3af97ef7c65bdeecb0656"}]},"branch":"refs/heads/master"},"b2efd185ce0cc794a5d94880cb6b4a46075f8db7":{"kind":"REWORK","_number":2,"created":"2020-07-01 15:56:03.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/21/735221/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/21/735221/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/735221/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/735221/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/735221/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/21/735221/2"}}},"commit":{"parents":[{"commit":"2e001431fd0907c97d44b81dd1f3d2d14ae1208b","subject":"s3api: Don\u0027t do naive HEAD request for auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e001431fd0907c97d44b81dd1f3d2d14ae1208b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-11 16:25:48.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-07-01 15:54:08.000000000","tz":-300},"subject":"s3api: Use swift.backend_path to proxy-log s3api requests","message":"s3api: Use swift.backend_path to proxy-log s3api requests\n\n... and to determine {account}, {container}, and {object} template\nvalues, as well as statsd metric names.\n\nUpgradeImpact:\n--------------\nBe aware that this will cause an increase in the proxy-logging statsd\nmetrics emited for s3api responses.  However, this will more accurately\nreflect the state of the system.\n\nChange-Id: Idbea6fadefb2061f83eed735ef198b88ba7aaf69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2efd185ce0cc794a5d94880cb6b4a46075f8db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2efd185ce0cc794a5d94880cb6b4a46075f8db7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
