)]}'
{"id":"openstack%2Fswift~735738","triplet_id":"openstack%2Fswift~master~If0fc8ec4d8056afb741bf74b82598a26683dfcd7","project":"openstack/swift","branch":"master","topic":"s3-mpu-part-no-head","hashtags":[],"change_id":"If0fc8ec4d8056afb741bf74b82598a26683dfcd7","subject":"s3api: Don\u0027t do naive HEAD request for auth","status":"MERGED","created":"2020-06-15 21:19:55.000000000","updated":"2020-07-01 22:51:27.000000000","submitted":"2020-07-01 22:49:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735738-1593643760037-260f2999","meta_rev_id":"98f27fa471b64cf5a16ac91e4051f62f3f6f8e2a","_number":735738,"virtual_id_number":735738,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-01 22:49:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-01 16:36:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-01 16:36:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-01 16:36:54.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-07-01 22:49:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba7ff4e777a86e4972cf0751e1b4ad2a54370ea0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-15 21:19:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4822f8cf3cd123497e63dc1714df95564494e48","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-15 21:21:02.000000000","message":"Patch Set 1: Workflow-1\n\nTim said this approach seemed reasonable.  I\u0027ll open a bug for \"too many container HEAD requests with MPU uploads\" and write/fix tests at some point... hopefully.","accounts_in_message":[],"_revision_number":1},{"id":"fbd59ad3fe2a6440e5f9a43c146cdf8749a1bbe7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-15 22:21:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)\n\nMight want to get Romain\u0027s input -- sending the extra request down the pipeline might cause some minor headaches when using ceilomtermiddleware. On the whole, though, I think I like it.","accounts_in_message":[],"_revision_number":1},{"id":"d14db6413ecfdf900c5f54f3fa7f5440bd06d0a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 23:05:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d0319e789a4c4b97e01b294ec3d800 : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/5b61688659c3419e8ab75415f0bb4e1f : FAILURE in 19m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9bb0611f62b0433db41451131494f5df : SUCCESS in 50m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/16dd62f756344dbeb6df6151b7c87c9c : SUCCESS in 15m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/54ed647e5ab9485eaddc34a4d2f9b0e0 : SUCCESS in 16m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0a8abf984ce2445392a51238db3f7b7d : FAILURE in 18m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/275be45295644b93be0cd5c17b22f7bb : FAILURE in 18m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5df60cded977461b83526840996e2aca : FAILURE in 19m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/969e73d5b44a4e0790b9c0415b37c037 : FAILURE in 16m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8a3b12a2a868482dada85285de354c2f : SUCCESS in 21m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f12d7dd1b97c4ed48c4e1e6bd2e918bc : SUCCESS in 26m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4ac8aade1e214476b2d11c847b681088 : SUCCESS in 24m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/36346708cbb64927ada460ad935df4d8 : SUCCESS in 22m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bb0c7cbc2b3f46f89d44877853364d10 : SUCCESS in 23m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3ef652a95e2b45f891e3f1a19b5a138e : SUCCESS in 25m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f04502cd201343e0b33b941dc91df5ef : RETRY_LIMIT in 5m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/338270671d0b452e9acb635f4e0b0cad : RETRY_LIMIT in 4m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4b7a69d33de54aa9a971fc05559afd37 : SUCCESS in 42m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d19869e9cd014f24904323758dddfd59 : SUCCESS in 43m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d16aea074295478cad1d5a6668629bb2 : FAILURE in 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caeaa22775b94e13816bda9039aca2bc : SUCCESS in 11m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/887e0e3d108f4afd9b56f2f8dacc3f0f : RETRY_LIMIT in 10m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ec33d0da5e745db965976f5b230c1f7 : SUCCESS in 1h 01m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1b40e7e6f74455ac24559dc8646fba3f7960e9b7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-17 18:01:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d929ea0ac4feddf3b26281b427910c90dab69162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 00:49:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f71dc9aba4cd4ea7adf77bdc061f7a0a : SUCCESS in 7m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/3925e6e46eec43028f4916be51464eef : SUCCESS in 1h 12m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ecb134f94c6841768b48bb8043219973 : SUCCESS in 54m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/05239e05951e4cb8a7da0b9ec6c65ff4 : SUCCESS in 9m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/42a6e67730de4097a3e3b3325b07c4e7 : SUCCESS in 10m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8de66a681fa3463f93e3595da759cf51 : FAILURE in 15m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7d6ea35237584736a05991cc282f0955 : FAILURE in 20m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b96920df4bad4d318057abc003a985f1 : FAILURE in 12m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6aa4703e26294df68dfae5340a72fc71 : FAILURE in 12m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/15f8ede6924843f7ba6bfa311010d77f : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1e588cc9c404465aba44e8b9b85cb306 : SUCCESS in 21m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/678359e798e440539c206137310b3f48 : SUCCESS in 22m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a48f1697f2b74edcbbaae2a00553d932 : SUCCESS in 22m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7d8dad0518b243ea8d29293739dc7341 : SUCCESS in 22m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/33bfa1a81a87477f8c668c9a7b02fd77 : SUCCESS in 22m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0550f07a184b4a63a60310ff4a2d81ba : SUCCESS in 28m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7e690a23bd664cc9961049955ebb1ee8 : SUCCESS in 39m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1edc1b8e081b46e7a0ee18631527d289 : SUCCESS in 1h 03m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8521b918991e4b32905fd43e0111f394 : SUCCESS in 1h 04m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca3dc19fa71d4af781ea50eb6a0025ff : FAILURE in 27m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4d279e771ad487aaa26ee6d785f8149 : SUCCESS in 6m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b71b13a831304be689f37cc05591d0e4 : FAILURE in 8m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2abebf84beee495285e2a2d9962ba91d : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":1},{"id":"bec7b5aead6c5080ad01c0a9fdc77ad6fdfe418c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-25 15:58:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"460bc04b8ba4ff4023b5a3e8fe6c2f1ce773c8ef","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-25 16:10:04.000000000","message":"Patch Set 2:\n\n(9 comments)\n\nthere was a lot more test churn from this change than I might have liked - I think there\u0027s probably going to be more work here","accounts_in_message":[],"_revision_number":2},{"id":"c432b7753559605f4faa19f8fc2ad4dd7cfcaa0c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-25 17:30:20.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d40710db1b7fa8ac77c171e7ff8cc43f87ef3f60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 22:59:24.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ec52f695b9e47f1b7d4b5889353d077 : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/8e3ad6dc577640f58696163db9131e33 : SUCCESS in 55m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7c9676884dd344759f94b64eb249bc6e : SUCCESS in 41m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/65a6528ab24e490daef84ee98dd1d47f : SUCCESS in 9m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6af9570aa82f435fae93aea3af8c5b42 : SUCCESS in 11m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/39417536c9a4482195e5d38fa85f3a91 : SUCCESS in 13m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9c2709a44bc5491398f53a67896e9dde : SUCCESS in 13m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/862e3664e6ca465aa1285c05f7493776 : SUCCESS in 12m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e285dd8b59494710a98169c4e77bfc9f : SUCCESS in 14m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a5d9b705e7334a9e922f30065ceb87ba : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b0335ec86c144775b1ab1a854701c1f1 : SUCCESS in 16m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/083699accc0a4861ad02315b95fdba7a : SUCCESS in 18m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f9a5155183cb404387e64efb413d9a9d : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ad5207bf4d524dc2bd32f9bdba0dec10 : SUCCESS in 21m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f710a5124bb147158b2c4f7be0e4ddbb : SUCCESS in 22m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e02fe279e8834d4aa7dc2c74eb21afd7 : SUCCESS in 24m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b86bd0053a5d436199e2ca08fdd7d299 : FAILURE in 46m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/09d548f271e34135b760c1e0667aaebb : SUCCESS in 39m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c879a76da32e44199bd8b2a82790bc62 : SUCCESS in 37m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/639a71d92a5b46a5924ea5bc2d26a2e5 : SUCCESS in 13m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7dc5bca2932481db0fccb95dfa6baf9 : FAILURE in 5m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b7dbdd7b9d5b437db74162950f182765 : FAILURE in 10m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92cedc4ba86c4ce2a6b59f6e2adee811 : SUCCESS in 1h 07m 17s","accounts_in_message":[],"_revision_number":2},{"id":"143ff4eab2c18cab5cba7ada8f2735fdda219c71","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-25 23:42:29.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0e742cde78fb5ec5934a4ce7c3b680f8f3a75f3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 01:34:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f394d64dd0394e6fa248399a7aab49a9 : SUCCESS in 10m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/95f95a4af33545ac91144be12b3df29f : SUCCESS in 1h 06m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c2d120287c88431dba520c149d9faf66 : SUCCESS in 44m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cf35d523dcb5490fa97c0f7bc0fd9f96 : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0c8632523b594168b88bb94946aca8a9 : SUCCESS in 13m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ed20d8a9f7644e7fbae2e229b21781c3 : FAILURE in 16m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1d7849ddd971451a81b1cf4233bacafa : FAILURE in 12m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/694cf0b5b51a48a6b788c8c42c45b966 : FAILURE in 14m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2528494d12704aaa95859f318a260b96 : FAILURE in 16m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/70b9f07b3098450fa4a8cd8f4c436762 : SUCCESS in 24m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/960a23639208474ea10393393fbb444e : SUCCESS in 29m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/33cdc332620241edb45f12757257e164 : SUCCESS in 17m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b401421d062f4736ac0eea2c1fc02810 : SUCCESS in 20m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b67bf32fd9c649f19ed8050bbb34b704 : SUCCESS in 19m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ef050f0a6d554c76916d33057f9fb095 : SUCCESS in 17m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/59b3e40dd6c84de085e58f32d208f313 : SUCCESS in 25m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/71ebc48b92cf48abb63a8ad65e5451cb : SUCCESS in 46m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6f303e0027df43839a193525a92e926d : SUCCESS in 44m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eda98b7226024f12ae2a2e1a3c0a822f : SUCCESS in 44m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f17a7282eff44f3e884f5ac554a33eed : FAILURE in 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b486e49a3e5f4610b8d2f31c0cb11a79 : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7fe27a2af99d49c59366ea5740f1e94b : FAILURE in 8m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/448b85a089384387ba05ac6fed85d4ad : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":3},{"id":"41606c343948d5479a478700fe9680e35d25c8c6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-26 16:06:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3fb67422dd97509fd9e612843a3e363717ebace5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 20:12:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0dcec3e9c9b48e39b23355d2dccb41a : SUCCESS in 11m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/aedfc59e8f584fc2ae37b26af361ad80 : SUCCESS in 1h 17m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ee4b4e65dbda47a4989e90b3d1080528 : SUCCESS in 57m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a498e43cdee14a049e0ecab89da6849f : SUCCESS in 8m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/91bcff881eb74ed0b3bb6daf7609f66a : SUCCESS in 12m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e388d891b9614b74aa371a17878d6842 : SUCCESS in 14m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/97477317c9d34aaeb2ddeda3b40d3a21 : SUCCESS in 14m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0d9cfeabc20044a78b2326204a4de275 : SUCCESS in 13m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0bec3a54d1de45edbf53ef14d5eef380 : SUCCESS in 13m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c78ef2cfe4274e0ea2a93154b762af05 : SUCCESS in 21m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/df8abda787c14ce6bce578e549bbf844 : SUCCESS in 21m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0ef47a2a8a33486ea3e7e8c5882b2677 : SUCCESS in 20m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8ed39672cd0c4d76abdd917e50a66cd9 : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e6f1b833aa474a95b7acea5303aeb238 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d562bc865ff747128b39d4933a101021 : SUCCESS in 21m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/75e66a72a0ce4111a1f05d809c9a1086 : SUCCESS in 30m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ed50daf7ee3e44498346fbca7c516ecf : SUCCESS in 50m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/502c1aac84cf4192828d1fe264ef7d47 : SUCCESS in 47m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c7fb9a63d4aa49cd80d25a9d6182188c : SUCCESS in 47m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b8d3a06973b461ba4b4b1e84c9e5f0a : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea602578ff0d49a0b78020d4637e4cf2 : SUCCESS in 6m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b45aff0a9c44dde900fa6b4711797c4 : FAILURE in 10m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d72ad76cb67940ef9af4f1f68532eb0a : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":4},{"id":"e085fde205739e8d2e8149d82fd9b275927e6b7a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-01 16:36:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(3 comments)\n\nYeah, this seems loads better.","accounts_in_message":[],"_revision_number":4},{"id":"646dcba838bcbb112f0edcbf6b823719aa8798ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 16:37:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0d628878b10fd1eef1a9e31deca8201dce7f7ae7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 19:59:39.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae48c9e792914d4f816c793be0486e45 : SUCCESS in 11m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/bb571a343db849ccac939975ef20c8bf : SUCCESS in 1h 04m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6e4dfb2ccd89447eab9bfa690ddfdbc8 : SUCCESS in 50m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0f5e0cc68b344f35a171e3caa084e77a : SUCCESS in 10m 18s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b7c5dd43cdac420d964fa1b6e2526c69 : SUCCESS in 12m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9eaa0c711b6142c1b62310234e8b1a1d : SUCCESS in 14m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6ccb992094104378941f90e2865dc3b0 : SUCCESS in 18m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/68d21897bbd24d81ac97e0d270945af5 : SUCCESS in 21m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/78861cefcb624009b2702677c47f9d4b : SUCCESS in 20m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2c8870ccda3046c2a60737d0b199dee0 : SUCCESS in 19m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f3628f10e2bb4c508cc5e6d608eb700f : SUCCESS in 19m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a2f67c288f414e5bb4418b3c8acc0532 : SUCCESS in 20m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c334597c5f8544ea92d74f5dbcbc71e0 : SUCCESS in 19m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9709029a9e294a2298226a8a30bf94c4 : SUCCESS in 29m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e7e93fb92a774898891621b390e3e4f8 : FAILURE in 45m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d62a6a87cf44dbc8f432d8ffb04ef53 : SUCCESS in 41m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ded340819754e6da7bbc7cc784b8158 : SUCCESS in 43m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa9add944e6043f4a26e981fdacaaf11 : SUCCESS in 17m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/324d16cb0b53444b93e58c2a951528fe : SUCCESS in 7m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca4691c1ee294c639cafc30a4e8ee018 : SUCCESS in 54m 46s","accounts_in_message":[],"_revision_number":4},{"id":"dc2caa2464d305308000c42ee0ebfd5658726c55","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-01 20:19:44.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"59589f806bf91777b186a3754e931b2cccc8115e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 21:48:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02345cd7f2fd4a5480ca990836928286 : SUCCESS in 9m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/54a11c5c26994a539806f538b6813ee1 : SUCCESS in 1h 24m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5b01c7111e544b2cb16b074ec4c2db30 : SUCCESS in 47m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5cee3a0ac2f04c36ad50eabc4fdbf033 : SUCCESS in 8m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d74ae2df9cf34f8e9f3e37a8ebfe5dae : SUCCESS in 11m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83b91191f160407aadd9885580d3a530 : SUCCESS in 13m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c729e840deab49fa9e057215e0c9badf : SUCCESS in 13m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5718d164307b4e5ba0d8a7fc28fd0e71 : SUCCESS in 14m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/43c99f43b4724f82add2864dc4a8c014 : SUCCESS in 13m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9f3ef8bd716d4f8e8f0073455a127f28 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0fd6bf26be0b41988532ee92ff8e6c93 : SUCCESS in 21m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/59bca36d751948509169714f5cf22743 : SUCCESS in 22m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f8331cff78c34887b2479f1f0b9e8faa : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/889c7300f77f490b9f5335ccc0d8ce9d : SUCCESS in 19m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1868e45f41e347df888b9f6947287f38 : SUCCESS in 19m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a9ae321be233424dacb0bacb5a395a54 : SUCCESS in 26m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fb3151d46237420d83176c42ad2b5635 : SUCCESS in 42m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1e5f1195d5cb416989b1b1ca6d856e22 : SUCCESS in 1h 00m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9a4c6ede113c4743ba0df6cbed400cde : SUCCESS in 39m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7643f9f83049481792aee1d030529d7a : SUCCESS in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4292b826980943788ab6ebca1c9f0428 : SUCCESS in 6m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad808d5240224cff98f1d2458a8bf4b6 : FAILURE in 8m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75e341a904be416fb6832c9ce43627f8 : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":4},{"id":"5229a12b27882a47b1985c9bd0c37692e3b1deb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 21:48:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c3dd26d3c3360b3ce3c9192eeb3a8ec1ba97ebe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 22:49:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80c0c0c4d07b47138c10fc1b341132bf : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/b605acc771814048b9bac0f48d4bd33c : SUCCESS in 59m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/108207925c714863bdd58364f86e18f7 : SUCCESS in 43m 09s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2627dbc79b154a9291fadb1de67b4ddb : SUCCESS in 9m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c55b190e7c854f4d942066bd0667e930 : SUCCESS in 11m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1232f079b8be471aa3565f376d3a9fbb : SUCCESS in 13m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0ff79438675540aab9bb6dca833c0289 : SUCCESS in 13m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7ebcec8b565642cdb16fcbf6367697e0 : SUCCESS in 15m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f6d947e29ebc4572ba2f408ad3be3320 : SUCCESS in 13m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1d89e3d978bb41d28026a4fca58669e6 : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4373513100554796bc21903e407b8912 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e25893c08dc3465f92c035ed32bfd84f : SUCCESS in 18m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/08a050cc97dd46b598afa566635a31fa : SUCCESS in 17m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/da72e0c9e1f34afaa1e9234947187bae : SUCCESS in 18m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2d066bcb4eb04d0181af32c967456f0f : SUCCESS in 44m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c7dfa4c8ba3a4e9b910accd4c4934e26 : SUCCESS in 42m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4f79dcccb594432ea0f9107a41631c25 : SUCCESS in 55m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbafb79f81c640a483134ae99c335c2f : SUCCESS in 16m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b435319f49f441b5a3296d4a674f384d : SUCCESS in 7m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/508afacf06d74ba9a86ce0f89402fd33 : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":4},{"id":"e9ccfd87a24ad728d1ef61e3207497ab0aca3d3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 22:49:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"9b5b73fe24e62c87e48ee1211dfdb0bfd53ecbda","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 22:51:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b2ebb2c15f41b0bfd4787d436216f8 : SUCCESS in 1m 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/bbdfd37dd3a9473b8ef8583b5aabe4e4 : SUCCESS in 59s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/82242de8bfcb400abe1920286579432d : SUCCESS in 59s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2e001431fd0907c97d44b81dd1f3d2d14ae1208b","revisions":{"9f83b9c9d365d9ce6f7a6ce98a66d217c49464f7":{"kind":"REWORK","_number":1,"created":"2020-06-15 21:19:55.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/38/735738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/735738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/735738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/735738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/735738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/735738/1"}}},"commit":{"parents":[{"commit":"0bed59e3faae94b96f98bbbc61b09abe54453267","subject":"Merge \"Use ini_file to update timeout instead of crudini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bed59e3faae94b96f98bbbc61b09abe54453267"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-15 21:19:47.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-15 21:19:47.000000000","tz":-300},"subject":"s3api: Don\u0027t do naive HEAD request for auth","message":"s3api: Don\u0027t do naive HEAD request for auth\n\nChange-Id: If0fc8ec4d8056afb741bf74b82598a26683dfcd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f83b9c9d365d9ce6f7a6ce98a66d217c49464f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f83b9c9d365d9ce6f7a6ce98a66d217c49464f7"}]},"branch":"refs/heads/master"},"8698dcf0bf1e18fe4057b2f6f35ea4fd0973253b":{"kind":"REWORK","_number":2,"created":"2020-06-25 15:58:48.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/38/735738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/735738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/735738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/735738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/735738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/735738/2"}}},"commit":{"parents":[{"commit":"0bed59e3faae94b96f98bbbc61b09abe54453267","subject":"Merge \"Use ini_file to update timeout instead of crudini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bed59e3faae94b96f98bbbc61b09abe54453267"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-15 21:19:47.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-25 15:57:43.000000000","tz":-300},"subject":"s3api: Don\u0027t do naive HEAD request for auth","message":"s3api: Don\u0027t do naive HEAD request for auth\n\nChange-Id: If0fc8ec4d8056afb741bf74b82598a26683dfcd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8698dcf0bf1e18fe4057b2f6f35ea4fd0973253b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8698dcf0bf1e18fe4057b2f6f35ea4fd0973253b"}]},"branch":"refs/heads/master"},"a6bd9dc2e8f2f88ab6279dfd11590f55a986acba":{"kind":"REWORK","_number":3,"created":"2020-06-25 23:42:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/38/735738/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/735738/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/735738/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/735738/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/735738/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/735738/3"}}},"commit":{"parents":[{"commit":"0bed59e3faae94b96f98bbbc61b09abe54453267","subject":"Merge \"Use ini_file to update timeout instead of crudini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bed59e3faae94b96f98bbbc61b09abe54453267"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-15 21:19:47.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2020-06-25 23:42:24.000000000","tz":0},"subject":"s3api: Don\u0027t do naive HEAD request for auth","message":"s3api: Don\u0027t do naive HEAD request for auth\n\nChange-Id: If0fc8ec4d8056afb741bf74b82598a26683dfcd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6bd9dc2e8f2f88ab6279dfd11590f55a986acba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6bd9dc2e8f2f88ab6279dfd11590f55a986acba"}]},"branch":"refs/heads/master"},"2e001431fd0907c97d44b81dd1f3d2d14ae1208b":{"kind":"REWORK","_number":4,"created":"2020-06-26 16:06:34.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/38/735738/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/735738/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/735738/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/735738/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/735738/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/735738/4"}}},"commit":{"parents":[{"commit":"5c087ad7b93cd52d2ba8c34db5ac0fc06db93f08","subject":"Fix up curl commands when object name has unix dir characters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c087ad7b93cd52d2ba8c34db5ac0fc06db93f08"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-15 21:19:47.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-26 16:05:07.000000000","tz":-300},"subject":"s3api: Don\u0027t do naive HEAD request for auth","message":"s3api: Don\u0027t do naive HEAD request for auth\n\nChange-Id: If0fc8ec4d8056afb741bf74b82598a26683dfcd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e001431fd0907c97d44b81dd1f3d2d14ae1208b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e001431fd0907c97d44b81dd1f3d2d14ae1208b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
