)]}'
{"id":"openstack%2Fswift~737056","triplet_id":"openstack%2Fswift~master~I0a7d65f64540f99514c52a70f9157ef060a8a892","project":"openstack/swift","branch":"master","topic":"737056","hashtags":[],"change_id":"I0a7d65f64540f99514c52a70f9157ef060a8a892","subject":"swift-container-info: Show shard ranges summary","status":"MERGED","created":"2020-06-19 18:12:23.000000000","updated":"2020-07-24 21:41:31.000000000","submitted":"2020-07-24 21:39:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737056-1595626792762-81340386","meta_rev_id":"b1e1b59e8bc618e6f4464e7cde46a7ad4a5bff68","_number":737056,"virtual_id_number":737056,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 21:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-07-22 17:31:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-23 23:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-23 23:07:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 17:31:25.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-07-24 21:39:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"870c2d107d3a3b4e4e97578b0e4de8099fe160ba","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-19 18:12:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31f554c7557b3a733aca5d95ce71c8b14129cd60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 20:08:08.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6fd4195f8c74820bbca59e316d6d67c : SUCCESS in 10m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/e08ecbc158fe4fbbb881ec14bd225941 : SUCCESS in 1h 02m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bbfed64cd8214fbb8e273428bafda59a : SUCCESS in 57m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1decc07bb0c04df4bd1410a860b7406e : SUCCESS in 15m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/092c0c80552049b9b20a27c2e56aa5ee : SUCCESS in 9m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cf406e8fbe7d439590164c80c6f2e6ef : SUCCESS in 14m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/220d4d8e551f471eb64605f785409cfe : SUCCESS in 14m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dc0680b179c84987b17b5ac65b896848 : SUCCESS in 13m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5d6bb3a972a64edb904f1ed52e08a605 : SUCCESS in 13m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ecaad6a1e35f4778aa82321c70ddd0cd : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c96873db342e454bb284315b4930266a : SUCCESS in 19m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/519d8712ae394b45ab62be49c0e037c7 : SUCCESS in 20m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/476b51d6f7584cada4f0b9dc19db5934 : SUCCESS in 20m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d7d13a88cd91427fb2d67c47783028ac : SUCCESS in 23m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/68cdc8872a864b8fb6c38f7faa5c3af6 : SUCCESS in 17m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/62a5c01c9eda4b6f95054ca495f003ad : SUCCESS in 25m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1b3e3c1a7fca491ca43304e81c81434b : SUCCESS in 47m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c93525af40a94b9ba14916e623ba978e : SUCCESS in 42m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c2f7cbf73cd463abe00439226046d09 : SUCCESS in 42m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2621773704f94284a4703f8c5ff6177e : SUCCESS in 22m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a071ef74dd948c19375369f0bd87a11 : FAILURE in 7m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a6b43e655d4d4c3bae62328a3b10e0a2 : FAILURE in 13m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/423a758b471a4f46bd6b26145151c656 : SUCCESS in 50m 48s","accounts_in_message":[],"_revision_number":1},{"id":"cbc4f9a61d3dcb16c926757283774faa51176610","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-19 20:12:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"db6e033c4accf441c63e05496323e0b1cf64add9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 21:27:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb62726c47ba49f8841190b9f6475a0b : SUCCESS in 8m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/9b713c7f35ba4f9b8fd57d6371459925 : SUCCESS in 1h 00m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2d1a7eca1ff3469e8ebfdb2d7ac3f6b1 : SUCCESS in 45m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4ef00d6382f54091b3b6320f44ef09b7 : SUCCESS in 8m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bfbf1a4c8fbd4e27893b5b775af8a90f : SUCCESS in 17m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e236f498538d43598ef99fcc552b2561 : SUCCESS in 36m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a753c5ee2d444fe3add42b448d8f4a4f : SUCCESS in 28m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1bc3e1cc52e1464d985824a34c1e010e : SUCCESS in 25m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e6f960ed80014b509d314d3327f555e7 : SUCCESS in 30m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/55c8858905484f04bf7aeef51f363105 : SUCCESS in 56m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4ba50856e78441768224ccdeb4fe3459 : SUCCESS in 41m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5fe9066abc974e4db5379783fd5f63f0 : SUCCESS in 1h 00m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/59d09a64b18847edbd5b98768a2cef43 : SUCCESS in 57m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/41288a8ce3f14e33a93d4d979af83453 : SUCCESS in 18m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/46c6c257b48645f88dde6c43f2860e76 : SUCCESS in 59m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aabbfe3f3f4f4346a0146b14ad0e42d3 : TIMED_OUT in 1h 01m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1ac9ef446c2b439889d29b2a3997ae6c : SUCCESS in 40m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/330fe352f5ad4b52bd5f950f7b480f4a : TIMED_OUT in 1h 05m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77edff839efb4933a7c5695038dad0d6 : TIMED_OUT in 1h 05m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c518929e01f4dc1af16c6533302f7b5 : TIMED_OUT in 42m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373cbf55f8d34d8a8f4352d77945d1b6 : SUCCESS in 6m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a89b6b53c84a4589b0d4ec990ca97f4f : FAILURE in 13m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05341b24a38445b5a783b58d19f6cf33 : SUCCESS in 1h 13m 33s","accounts_in_message":[],"_revision_number":2},{"id":"692806851d61d3fdcd4e8248a7b8ba3268348b8c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-20 00:21:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4479fc2593833ea203edc2adce377d5386e5d5f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 01:25:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8246f246242d4fb68f6013b9d2936ec7 : SUCCESS in 7m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/84ac9020a1014bfca466976b1e506468 : SUCCESS in 59m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a655cc9e3f9f429888727d7357f8ddcd : SUCCESS in 42m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b8e09a3aa3aa42908cb911a2752cb915 : SUCCESS in 10m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/05d0aab0e8f44fbcbe864936a4f6e10c : SUCCESS in 10m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4f1f193333c24c10beac905040dd3b76 : SUCCESS in 25m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9319795aa3ea461382177ab69335d4ae : SUCCESS in 30m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/61217dda61994743813737d5d2504ef3 : SUCCESS in 31m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1d574dcec09d433dafa86074cfd67667 : SUCCESS in 34m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8dabe0309b8a4aa4987d05b70b071a0d : SUCCESS in 38m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e9bfb4d2e5a040d8aaef040ff82b2c69 : SUCCESS in 45m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c14f387021c4390aadb8df5c7a10782 : SUCCESS in 39m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/10b5d6b4b35d4b3ca4742f8550795ec3 : SUCCESS in 45m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9166cb9f34794995b16d7589e5b3b0ed : SUCCESS in 14m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/70d6cc210d7945dea26fa8669b394109 : SUCCESS in 40m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/04d89678f56443adaec9bd8e21b67683 : SUCCESS in 51m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/29ebc23e4dc1410a83b2b03a7682e984 : SUCCESS in 47m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9ce641d753584f5ea0d6095b15fa1074 : SUCCESS in 54m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/43defd6f504e43268967381d630fee2b : SUCCESS in 1h 03m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8e8afc23b5f4ac8bd4f6518d73ca185 : SUCCESS in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d501f5846474dfca0dc077184338bb2 : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb0896c895d147d5921bc4f6061c88b4 : FAILURE in 7m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94eefb9490084df9a9bfa358c9bc5a0d : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":2},{"id":"1eeb50234ea42f9112d5ca2230a3a412872f7ae8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-30 20:34:42.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)\n\nthis works fine, but it seems like we could do better\n\n1000% agree listing all the shard ranges is TMI - what we really want from this tool is some set of information that provides key info about the roots shards \"at a glance\"\n\nMaybe squash?  https://review.opendev.org/738692","accounts_in_message":[],"_revision_number":2},{"id":"ab7834f1f3fdc8f9b16573c746c73660568c6819","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-06 19:03:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"70d74818b9141435550ef0e798c02a9fb5e6ff18","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-22 17:29:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fbcae25f28d32b2b2dfbbb894a7ff27fd4931f22","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-22 17:31:25.000000000","message":"Patch Set 3: Code-Review+2\n\nI think Tim and I at least have some consensus on a compromise here","accounts_in_message":[],"_revision_number":3},{"id":"0ad1b538c29f502ab9d7b5be0cc5dbe5b314bda5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:52:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35fe2cd50ab4d8eb6ea4dc95b57b033 : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/2eeb94a5f7a7480897fdc244a6bc1622 : SUCCESS in 1h 01m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5017b3dc2eb41d49fec0497aa06ee69 : SUCCESS in 43m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8ce003ca44b740f5abd3df865ead1c2d : SUCCESS in 10m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aea6c538b26d4018a20cac60f2cd06f3 : SUCCESS in 9m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/81f9c6e567384cdfa1e76dfb487a7440 : SUCCESS in 14m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/47e8d4323dc147f680bb00d6263abe20 : SUCCESS in 15m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ac9ed55522f84c2fbb6e35d2e7c36d60 : SUCCESS in 15m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/796763f8ab8848a482eb5bb0c4e9385a : SUCCESS in 15m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/13aec345c23c42cebfd216857f749628 : SUCCESS in 19m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5138ffc1ebe545b982877c4e3b90fb48 : SUCCESS in 21m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e4c1a6672d1845d9ac3a8b0766394ce4 : SUCCESS in 19m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dbcdfece761e49ff9ab3c0eff58b12d1 : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6ef609df07b94dc18bdb8218e41c5f25 : SUCCESS in 18m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/483c357d3e024640a0190981c74c35e8 : SUCCESS in 19m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ef3aa5e456304bfda58231cb34bfc462 : SUCCESS in 32m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b517765cb6a04a349e42c2425f6414bd : FAILURE in 1h 16m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0dda3df72d2f4f878e833e7a333609f6 : SUCCESS in 56m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/90b928775b0a4f638df9794fa7b6486a : SUCCESS in 1h 00m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/901c395bd76d478ab4904a4a1b4dda63 : SUCCESS in 23m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/868ff23c838b4a2d8cd31117b3789b31 : SUCCESS in 8m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0e0e647295b443ecb277537bd123be58 : FAILURE in 12m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff1c18ec5d20413e862ef1c95c099a6d : SUCCESS in 52m 31s","accounts_in_message":[],"_revision_number":3},{"id":"e7da791f99af57eef1ffe855f4b7c0661692ac5b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-23 17:06:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ec53b7a505b6c4ce625a0611abc499dfb3c53942","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-23 17:11:12.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)\n\nJust checking in on whether we want to include the numeric state. Otherwise, lookin\u0027 pretty good!","accounts_in_message":[],"_revision_number":3},{"id":"2bf1dc52b632d733729a8c6412467c54ec045e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 18:17:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9505bfdcc993499d916695b6f0161087 : SUCCESS in 10m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/c3465fa7c249450ba5bafec60c5d7c3a : SUCCESS in 43m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1c009685a49e4e9aa254d0e9bc8eab09 : SUCCESS in 31m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d70b04a7a23d4bc2b682675ab4eb7dc7 : SUCCESS in 7m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/65143f78bbe34df08734528273c4103b : SUCCESS in 8m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1e8af04655a14c84983f3220c05d05e3 : SUCCESS in 13m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0a5c5ad95b4840bbaeb2313f7d75836b : SUCCESS in 14m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d41e4e1992664e3f8e5ae13ad1c0b14d : SUCCESS in 16m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3060fad4a3e14093813a414cececcc86 : SUCCESS in 13m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6d9225f3018d4377b01fce84ac513819 : SUCCESS in 19m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e6ad2f9fb7e9480599aa35544b623420 : SUCCESS in 26m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/409351b1a2a848e7934fcf49fe245b19 : SUCCESS in 25m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f9a5156b0ca74e29bead2d08c5e19448 : SUCCESS in 20m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9d8350de7f2246ada69a71d1e05d9ea4 : SUCCESS in 37m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/14b461ea4e704b679bfa9a01c86e4a78 : SUCCESS in 24m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/26c8f5267a1049ebb43091e5f193ac7b : SUCCESS in 24m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a4dd1f476ca1415cbd6e6578934299d1 : SUCCESS in 45m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2009670570e04589a4e9ac6c89a42e52 : SUCCESS in 45m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/59a21cd544e0430796dcf0446507c6cc : SUCCESS in 46m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/771e1a257b724d04b81b4a0023ccc002 : SUCCESS in 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd9e801b32cc4d8e95516ca40ebad95e : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a73befb4f46145198ff6e9f660023fb9 : FAILURE in 15m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b47c1bc077c4dce89ec12b91f994bfa : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":3},{"id":"424ae525768888ef2608328a3641833c2f009484","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-23 18:28:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ba970e07e3bf1526e8c8a6c795583b3e52658330","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-23 23:07:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82eb4acd2960cb9ba8a5952f39b6d9037938305d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 23:07:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2c0814f25fd13067f8f7a904153ae45de392f1d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 02:30:19.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13e94cb3a5f94c1f8e45d4d2718a951b : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/eb9610f4297c47e99a671417b4119db1 : SUCCESS in 1h 04m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4c9e73ea59084ec0944c59e6c3fac864 : FAILURE in 35m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8bec3108a5744f7491c5028f2eced387 : SUCCESS in 10m 02s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/23dfde21caf24f04b5835da90ecb1b4f : SUCCESS in 11m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/87de8aa5133b4e679f186b3232f98628 : SUCCESS in 16m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1c50134a57c340a6a1a6cbf5a0fe09b7 : SUCCESS in 12m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6223e00ea95c46e28a5c8b0d1a0b97ff : SUCCESS in 12m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7ca11da340434f59baf9c573b4c1b612 : SUCCESS in 13m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6ef2d4ee156349c0ae19707d4226fb5f : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a28ce92709ed4dcdb657c852bf00ef60 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1a4073cb70f649d8b10f7f75ea60f917 : SUCCESS in 21m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/028322711dba43a2aaa049257b4efc8a : SUCCESS in 18m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ee3804699b4b4e5ea8db605d036be8a4 : SUCCESS in 19m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/93c3e0abc7494351a8d2057b761da5cc : FAILURE in 49m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eca4a85682d5436db3c5e380e007d7da : SUCCESS in 40m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7d64ef7da7044c4da6bcbe37498e7638 : SUCCESS in 40m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0ac2453296043f78b64b52cfabaa5e0 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51be98dfb1b94da98ac0d56a6b643b82 : SUCCESS in 6m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bac1e08d0994dc595014fb4bc4e3564 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":3},{"id":"f647034b1785e7f2a68726347b5cbd6180556b1a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-24 04:37:28.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d53f64ee01fb441d8ffc10c24d7a922e7cfaeb82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 05:31:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6b5c03d283e47849060cb5456b75e03 : SUCCESS in 7m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/1c87bfa0ce344b60b5f62f2a07dd5142 : SUCCESS in 40m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ed3b1302f4ea482fba02504785ee1bd6 : SUCCESS in 31m 48s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fec2ed95bf1e499691fc8804afa6fbbb : SUCCESS in 8m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/17fe4b1bbd3942078463eee98c149395 : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/555dda070060467bb38ea7f3300f0acf : SUCCESS in 14m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3ba4ca7d43ab4f9593b52bfa5a74a0d3 : SUCCESS in 16m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/329e9ce8c40d4431aa7d519aed629707 : SUCCESS in 10m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/beb4bafc1cfa4c91b6e5895525fb8643 : SUCCESS in 16m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aa689049db004992a093c15ebf7aa9c2 : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c3e3e590740248bc9ef5888724c6b63d : SUCCESS in 17m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3f4967dcad114a0e87de7fe439b4b6ec : SUCCESS in 16m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f52c3b366994488f8a3a4354319ec2b3 : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/643d17b6d7cf4caa8b8320d38582226d : SUCCESS in 16m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f2b28353f9b34e82b6e07fc1b854caf6 : SUCCESS in 19m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/85fdcf8a4edc4277b9c03b38863bbaa1 : SUCCESS in 20m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/186b87e6ea0a4f2c87212c4e0f73c19a : FAILURE in 45m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26b97d5f135e44abb5c017a002c5ae33 : SUCCESS in 47m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/80f95e200b8a4c8bba24e9be653f9c97 : SUCCESS in 43m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9135c301ae8a4965a10f02340ebebd0b : SUCCESS in 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0224f39e56b74241aafeb98fde311c1d : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/66ca8e6f08c24cf8a1b61d451229ff11 : FAILURE in 7m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9ecc4261bb94fd1aa93aa1a46ec10ea : SUCCESS in 49m 23s","accounts_in_message":[],"_revision_number":3},{"id":"ad5cc58640209152163ed4c5ce3b7d1fcb304e4d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-24 17:11:06.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1977f2b788d7566eb25b07d370b1d51a71db5aea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 18:13:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4795c07be74c928250100f669b85df : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/685364d663794bc19d8812c0b2535f4d : SUCCESS in 1h 00m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/de3a63b38d8644028069f0d59cfa6b8d : SUCCESS in 39m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6beba6f919e94221bb8261ad35418337 : SUCCESS in 10m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e1f9690fa19846cb925b6e3d1a308f9b : SUCCESS in 10m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/81e697ee274340a493e1c3bfe8e39dc4 : SUCCESS in 13m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0b62bc472b484eff9abe66661729eeff : SUCCESS in 26m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/650dc66f39674811854dd721d1c690b0 : SUCCESS in 26m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d72f9b86952249999895b92edd6d72a1 : SUCCESS in 13m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5dbc66fd4a864704bd6d2b623dc3d72c : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b658b74074d84711bf202e9099aaed6d : SUCCESS in 17m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b675d1b8597842f2acea08c019b96f80 : SUCCESS in 18m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/16150c6fe9614db6bfb52591e989bac5 : SUCCESS in 25m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/38d41680e3eb486aae2d3504a71f39a0 : SUCCESS in 20m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0692eda7c5d848f4abc79da83e991f34 : SUCCESS in 23m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8073ed23355f42d1969e4f83e886cdce : SUCCESS in 36m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9c937d181460414d90b59ea0fccb922f : SUCCESS in 40m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/16e1cae78c50441eb703cdbac1c7204d : SUCCESS in 41m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f40787d8486426f9c41bf7f12c63734 : SUCCESS in 49m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ae43ca7eac24016aaf6a6ed2e1b6852 : SUCCESS in 16m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95f08ce52ab84be99576f499a6f19970 : SUCCESS in 6m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/64fab42c0af343c7a8a7dd7e1cce0e47 : FAILURE in 10m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6242c1504c9b453b929bc8d2bf83c1ab : SUCCESS in 55m 10s","accounts_in_message":[],"_revision_number":3},{"id":"9c8cd52de0f614304f74fb6620a4c65d792ac456","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 18:13:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"abbdb60dabf1ab62a992cd902422faf2ec5c9bf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 21:39:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9877c0eaec644d48f48cbc60dab46e5 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/39af32efab3b472e96520e715525ace5 : SUCCESS in 54m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/70ad988d1aba4aed9810fa0af7b48e2c : SUCCESS in 44m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/01201426d0a14b4483a7c1fc86ad864c : SUCCESS in 10m 20s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/cef0d7be7506414fb920d6bc81a2858f : SUCCESS in 11m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7f4c8bd32a1f4e02a6a4628d763e53a5 : SUCCESS in 12m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8118cba0561647368e2e368c8dd2215a : SUCCESS in 16m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/06926f9eda214e079ed19e41fe64ea20 : SUCCESS in 13m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8ff54dddd8f74d079499f5a42ab490b6 : SUCCESS in 26m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4b8423ceb01b4bff87d7e12ac8724949 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d47edab62aaa4a67b0b70e4e468723ed : SUCCESS in 36m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3e3c537516b94e8fb680d2e4f0f93fa9 : SUCCESS in 16m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c5505fd1edfd45abaed039563090885d : SUCCESS in 21m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f148e4732f764a87ad39276a77ccc6a5 : SUCCESS in 39m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/90e41e69295f4227829f10b90741fb04 : SUCCESS in 44m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/418de200eb6b45c8a32ec2fd8d222f61 : SUCCESS in 42m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/593777723ee64116a026932cda4cfeda : SUCCESS in 57m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68d4500a77eb487fb26e515f096bc65a : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/071e9c045d714ff198c8d4823c14b771 : SUCCESS in 6m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23d20a25b1ee41509261e556f8c826d0 : SUCCESS in 52m 39s","accounts_in_message":[],"_revision_number":3},{"id":"aa2ff36b18d9e3bf00907bc844a1537f0031aae3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 21:39:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"82a48861844bb51aaa157e388927d23f3540160c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 21:41:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a82266a0be754bba9f9ef5d7a808a037 : SUCCESS in 1m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e8d918165af3401f98d275ffc1696901 : SUCCESS in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c1b5cec69fb34a13b4e5fcd5bd910b71 : SUCCESS in 59s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"67e3830ab92ff32d16d32320cf30b67dca671515","revisions":{"258a6bd3ff1edeffb9c31b2c5c8ab7c484313ccd":{"kind":"REWORK","_number":1,"created":"2020-06-19 18:12:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/737056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/737056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/737056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/737056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/737056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/737056/1"}}},"commit":{"parents":[{"commit":"5cf5548a85286b09a3441176dc773c2874203a39","subject":"Merge \"s3api: Add basic support for ?tagging requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cf5548a85286b09a3441176dc773c2874203a39"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-19 18:09:11.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-19 18:09:11.000000000","tz":-420},"subject":"swift-container-info: Don\u0027t show all shard ranges when there are many","message":"swift-container-info: Don\u0027t show all shard ranges when there are many\n\nAt least, not by default. Add a -v/--verbose flag if you really want to\nsee them.\n\nThe current behavior is really painful when you\u0027ve got hundreds of shard\nranges in a DB.\n\nChange-Id: I0a7d65f64540f99514c52a70f9157ef060a8a892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/258a6bd3ff1edeffb9c31b2c5c8ab7c484313ccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/258a6bd3ff1edeffb9c31b2c5c8ab7c484313ccd"}]},"branch":"refs/heads/master"},"01aef1ed4fa05a89b752990a4c033ebd59b7e4b3":{"kind":"REWORK","_number":2,"created":"2020-06-19 20:12:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/737056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/737056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/737056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/737056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/737056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/737056/2"}}},"commit":{"parents":[{"commit":"5cf5548a85286b09a3441176dc773c2874203a39","subject":"Merge \"s3api: Add basic support for ?tagging requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cf5548a85286b09a3441176dc773c2874203a39"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-19 18:09:11.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-19 20:12:32.000000000","tz":-420},"subject":"swift-container-info: Don\u0027t show all shard ranges when there are many","message":"swift-container-info: Don\u0027t show all shard ranges when there are many\n\nAt least, not by default. Add a -v/--verbose flag if you really want to\nsee them.\n\nThe current behavior is really painful when you\u0027ve got hundreds of shard\nranges in a DB.\n\nChange-Id: I0a7d65f64540f99514c52a70f9157ef060a8a892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01aef1ed4fa05a89b752990a4c033ebd59b7e4b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01aef1ed4fa05a89b752990a4c033ebd59b7e4b3"}]},"branch":"refs/heads/master"},"67e3830ab92ff32d16d32320cf30b67dca671515":{"kind":"REWORK","_number":3,"created":"2020-07-22 17:29:57.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/56/737056/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/737056/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/737056/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/737056/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/737056/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/737056/3"}}},"commit":{"parents":[{"commit":"5cf5548a85286b09a3441176dc773c2874203a39","subject":"Merge \"s3api: Add basic support for ?tagging requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cf5548a85286b09a3441176dc773c2874203a39"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-19 18:09:11.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-07-22 17:29:53.000000000","tz":-300},"subject":"swift-container-info: Show shard ranges summary","message":"swift-container-info: Show shard ranges summary\n\nThe current behavior is really painful when you\u0027ve got hundreds of shard\nranges in a DB.  The new summary with the states is default.  Users can\nadd a -v/--verbose flag to see the old full detail view.\n\nChange-Id: I0a7d65f64540f99514c52a70f9157ef060a8a892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67e3830ab92ff32d16d32320cf30b67dca671515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67e3830ab92ff32d16d32320cf30b67dca671515"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
