)]}'
{"id":"openstack%2Fswift~737341","triplet_id":"openstack%2Fswift~master~Ib457be89990163c4bae38f74eaae45eacba33691","project":"openstack/swift","branch":"master","topic":"dnm-probe-failure","hashtags":[],"change_id":"Ib457be89990163c4bae38f74eaae45eacba33691","subject":"DNM: probetest fail device busy","status":"ABANDONED","created":"2020-06-22 17:58:41.000000000","updated":"2020-06-25 21:18:57.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"97b06b17286278804cd963cf0e10938d66575a16","_number":737341,"virtual_id_number":737341,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-06-23 14:41:14.000000000","_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-22 19:53:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-06-23 00:21:57.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 19:53:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-23 00:21:57.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"f28da5527be1daf7fee3814c1aa6c615ba252f34","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-22 17:58:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"624b5c655ad3fa27977a77a25f99c1bf98ce55ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 19:53:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f61dc5fca7a4acd83f5b89fff70edea : SUCCESS in 9m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ccc308ec4bf642b5b6aca3e59ff50723 : SUCCESS in 26m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/85fafb07fdc847868bd7dda5f327c3a1 : FAILURE in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"87b0392dc7bc9be94110f40e12005e806d19b439","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-23 00:21:57.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nYeah, I know I\u0027ve seen issues with that umount before...","accounts_in_message":[],"_revision_number":1},{"id":"e6a68c230bd2d46fbaf2bc50692446af5996bdaa","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-23 14:41:14.000000000","message":"Patch Set 1:\n\nthe failure I saw showed no swift services running according to swift-init; but oldies or orphans is a good idea, even some pkill\u0027s (it could be rsync)\n\nI\u0027m disappointed lsof was not found on these images","accounts_in_message":[],"_revision_number":1},{"id":"449a203462de30122d226ffcd7ac73bda9ad4dc7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-06-23 14:41:28.000000000","message":"Abandoned\n\nlsof was not found","accounts_in_message":[],"_revision_number":1},{"id":"242c801471be7b10c32e78fde881a09aa4efd1cb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-25 21:18:57.000000000","message":"Patch Set 1:\n\nFWIW we\u0027ve already got some orphan-killing behavior from https://review.opendev.org/#/c/542513/ -- but I don\u0027t think that can help us avoid failures to unmount as-is, since the umount is part of resetswift :-/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2082c423defff4cdae27f18d89a53aa7a6e8e24","revisions":{"e2082c423defff4cdae27f18d89a53aa7a6e8e24":{"kind":"REWORK","_number":1,"created":"2020-06-22 17:58:41.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/41/737341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/737341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/737341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/737341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/737341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/737341/1"}}},"commit":{"parents":[{"commit":"a47d5a11521f0aca5aab4cdfc48a9a4be3eb4ead","subject":"Merge \"Clean up some proxy tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a47d5a11521f0aca5aab4cdfc48a9a4be3eb4ead"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-22 17:58:12.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-06-22 17:58:20.000000000","tz":-300},"subject":"DNM: probetest fail device busy","message":"DNM: probetest fail device busy\n\nChange-Id: Ib457be89990163c4bae38f74eaae45eacba33691\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2082c423defff4cdae27f18d89a53aa7a6e8e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2082c423defff4cdae27f18d89a53aa7a6e8e24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
