)]}'
{"id":"openstack%2Fswift~738114","triplet_id":"openstack%2Fswift~master~If08bccf753490157f27c95b4038f3dd33d3d7f8c","project":"openstack/swift","branch":"master","topic":"sharding-split-brain","hashtags":[],"change_id":"If08bccf753490157f27c95b4038f3dd33d3d7f8c","subject":"Address a sharder/replicator race","status":"MERGED","created":"2020-06-25 21:30:02.000000000","updated":"2020-08-07 07:49:53.000000000","submitted":"2020-08-07 07:46:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738114-1596786379871-1ab9881c","meta_rev_id":"03c593f42795817a68b4b89accfb4ce67706bf48","_number":738114,"virtual_id_number":738114,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-07 07:46:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-07 05:11:24.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-08-03 16:47:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-08-03 16:47:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 16:47:14.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-08-07 07:46:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e46469cecaf5a43946e935f1a4e123708bda34bf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-25 21:30:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0a23967b1096767bee06b2158b235dd77c594a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 01:19:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae62c828faf84be78efdae4e8514ec78 : SUCCESS in 9m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/62b3de0ad48e463d81cf7d8991ab24ef : SUCCESS in 42m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e5cdb4acc2204f23beff7455abf71c04 : SUCCESS in 40m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca1937bcfd0d47f5ad0ec31748034313 : SUCCESS in 8m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dc9d156b88e3465694a5996c7febf5b1 : SUCCESS in 10m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c4934f0bbec548c4b8776e5a3aa8a8a4 : SUCCESS in 17m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5828a2380d954ca59e932fdd2a5f3d47 : SUCCESS in 13m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/157dc886ade0450398ceedc60c1f53dd : SUCCESS in 17m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3af408462e57469498cc7832760da8d0 : SUCCESS in 15m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2881369ce695489c8213eed01e3ea02d : SUCCESS in 21m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4ea4676448ea4d369ba6ff80a16cbbf3 : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/759c154ca14344558be4eebac3ddcaaa : SUCCESS in 23m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f5c54c05bbe34a13b094405d3dec9819 : SUCCESS in 22m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/47868aeae1184c4c963e1fbb3be8a6da : SUCCESS in 20m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fccfcaa196144ef1888a0cb3b2111c6e : SUCCESS in 18m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3be93f727fb14cf1b83b1e6745f974b1 : SUCCESS in 22m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2acc4a12fa454c0e9e642dd52ae9426f : SUCCESS in 37m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aaf9771b56e14cf798adf2db64000df9 : SUCCESS in 34m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/364ef58c79c54c49a1283c0c3795c47f : SUCCESS in 46m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61404a0f65f54ee485f2f3a395347018 : SUCCESS in 14m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22534421e29849f5834bfc4ea58cea54 : FAILURE in 7m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01c28a451ce14560ae34830fe3499d6a : FAILURE in 8m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/907170ed30c4492ea518e14b7a6a74ce : SUCCESS in 56m 34s\n\nWarning:\n  Comments left for invalid file swift/container/backend.py","accounts_in_message":[],"_revision_number":1},{"id":"1413be9bac11839c775dcfae18ffc42e7e1eb5cd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-26 04:33:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"12b56425f2c793b78a27b9c3cc6a0520a9606a1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 05:28:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9af5caf56eb4479a55007ab72127b52 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/cc561dc6a50d483babc308cef8602c41 : SUCCESS in 52m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/11873ec9ff73413bb36fe58e0bb5639d : SUCCESS in 34m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/70794950ab184e3185eb0b45bb3b7fc9 : SUCCESS in 7m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3aa5a28b7c86437b9e0fd4420507526a : SUCCESS in 9m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fa012bc0a7f44ef8ae1cd1842d44dc99 : SUCCESS in 16m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7a2abd47e04b4578a404363b847d6811 : SUCCESS in 16m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a877968f58354be7918bc7c53f3c5dcc : SUCCESS in 16m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a8d390c97542424cac79a11c818cc6c4 : SUCCESS in 16m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b195bd3420b24a15b29666255c47df34 : SUCCESS in 21m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e8ab71549c014fc6a8aee8fe2bfe044b : SUCCESS in 19m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7385e4b9431d441189bbbd692f0d5db9 : SUCCESS in 20m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3bce2d8f77f64d1d8ce371507a0af761 : SUCCESS in 19m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/afe7d2dc403a4dc880de6d5497e6fa70 : SUCCESS in 22m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/afc11e5ead4146959dd14b69149887e3 : SUCCESS in 20m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f6efac2bde8e4091b684a81adeba8821 : SUCCESS in 26m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/70571087804a42f5b027fca81ce1baa8 : SUCCESS in 52m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e1a043a9cf149289ea96dd010b56f9c : SUCCESS in 47m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0f60e552d5c547fb939c705b99311300 : SUCCESS in 42m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f098f77a11f541a6bc54d1265a001773 : SUCCESS in 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c17de53b42ff416fad40b79c188791df : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/991dc052b6c440cf9c17c361f67c4ad2 : RETRY_LIMIT in 6m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ab9581621ed46fd89ae4ab71bcaaee6 : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":2},{"id":"4da7eceb703705f70a9f03e03e88a07a79abf445","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-08 05:01:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ddf4e303092427f377a61c81a4a4abb7258a8fb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 06:11:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/467c64cfacad475ba783d60d02e400a0 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/c44d434c40d643a9bc47ffe5bff9adc8 : SUCCESS in 1h 09m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8dc1c9f859124d2db2bf954cd49b46c9 : SUCCESS in 51m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/38eee09ccf46486e8df864c97c609ac6 : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/69331c006d4a42c3883cbca0d8da4201 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cff79e63db48448bb92ea75bab9c2bbf : FAILURE in 12m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/deb5c8a783224652b7f6f3b21c0272fe : SUCCESS in 20m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/be66e440b35345ff94741eff269b48af : SUCCESS in 12m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0522747a94f340cdb89091db73826a04 : SUCCESS in 16m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5de5dffbdac14d5682c71c05d003958a : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/631ab06c1853470c9bb9427dabcea9b2 : SUCCESS in 17m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/926fe98599a54f269f493bf6d01a7ea7 : SUCCESS in 16m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/55c23d711b334d0ebee4386c0dde9b0a : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/26140951856d4dd8a932d408f14bde94 : SUCCESS in 17m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3b96886e9f174ee58968317a213ae516 : SUCCESS in 18m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/deda1b1c1a6645a1bcb32ed6f024f86e : SUCCESS in 20m 27s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e49ba2ccfdae44858348f9aa6215ebdc : SUCCESS in 43m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac8083060c3848e9904a897841559c79 : SUCCESS in 42m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/54b2701519f54665812a4c7526deb40e : SUCCESS in 42m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73984d788ae84e0ba4d1df06191c5fcb : FAILURE in 13m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/623b1574983f44bd85a6e782ef565c93 : SUCCESS in 6m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb3202ea4d0a4b3ca5f2d631271f4056 : FAILURE in 11m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46c72368db4d4681bb4b0e8d9236e4c3 : SUCCESS in 36m 36s","accounts_in_message":[],"_revision_number":3},{"id":"fe44a8580450862a9648ca5ace31d3e566f88ef0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-10 21:51:24.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nI guess I need to work my way further down the chain to get the full picture...","accounts_in_message":[],"_revision_number":3},{"id":"f9368bb2de1ff8ec7015d851b6bf186da8bf56ad","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-13 20:42:09.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nUnit test failure is easy to fix -- sorry, been playing with probe tests some more.","accounts_in_message":[],"_revision_number":3},{"id":"1991c8def49ffa5303765cf1de63df22c90010ec","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-13 21:57:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"27a158b340723a6ac9a0d1be00b457463c9ba699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 23:06:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94ebddba009548b8af89c6f1503c4b56 : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/383d89749ae04dbbaeda7aa53876e36b : SUCCESS in 41m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4a9351b2259e437c8ee540b25a4ed38e : SUCCESS in 31m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c1134122cfaa48b6ae144f9d1ec4bd50 : SUCCESS in 7m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a352709d31714aa880e3a582480e8e94 : SUCCESS in 8m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e5ba2f1ddce455db4b8640ba2a1b53c : SUCCESS in 12m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/61058f1065e446a89e791c4f73ed295c : SUCCESS in 11m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/01cd87f57cb148699b7f5563e19c86ef : SUCCESS in 13m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/98a4ab8bebc2450e871e52f312961805 : SUCCESS in 11m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/517bd7e6b61b47a598f048312650e734 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/76ee37421c654fa5ad4ec331b1dd8f63 : SUCCESS in 19m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6e1b532518514f6b8a36d2fccc47f2db : SUCCESS in 17m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3ca1cb355fec40abbd62b45896fbe910 : SUCCESS in 27m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/335fbf5158ad4b3caa884552de469f16 : SUCCESS in 15m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a12a33815fdb48678c6b9b6ddc797703 : SUCCESS in 16m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3440adfce929406cbbf9841c46da7ff9 : SUCCESS in 25m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d871b6528e884ac8b7a574c655362acc : SUCCESS in 37m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ffaa3aabfca14139b558ca3f985906b6 : SUCCESS in 41m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/841cd9422a6e4d85b982dd6d76f194bb : TIMED_OUT in 1h 04m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37bb37d0a0c3453ca1a37753b6f4a49b : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dfb98c1a7974745992a70c1cb700be7 : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a512a9107cf44fb95174682c9bc84e0 : FAILURE in 8m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9fe974d791f488484d4c74a0cd376f4 : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":4},{"id":"c1ef433bf4151051abf00c5ce18bbb0247c44d6d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-13 23:19:34.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9cb02302fc6e4423db1d79d9e442310ba20c1825","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 00:30:55.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4efa942935c343fe9c3fffb2f01ff122 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/93b2f1f03c1245b7bd9ee1397bab12d4 : SUCCESS in 52m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e06bc746af846009f3e137f54cd0496 : SUCCESS in 48m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/373aa988da7d4f1b8653e71f33ea2e2f : SUCCESS in 8m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4825800c48854cd3868d2247680064b2 : SUCCESS in 9m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/111f34416ce94a3ea048599a6c1d12f3 : SUCCESS in 11m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/25101f70e7b64e3c9c4f1c1cfc86e91c : SUCCESS in 12m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ef424081bb164dc197377dd12897254d : SUCCESS in 12m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/42ff6ddae0d84619b49bec25333d3af9 : SUCCESS in 13m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f10ade3c294d4b4e881d1a01d3a2a7c7 : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7f464b668fac4294bd959634055e862d : SUCCESS in 15m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aee0b10dcf484e75bcc2c89b55378325 : SUCCESS in 17m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2daa16e1d5e7419384e453b2ccb72ccd : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bbe4e5b904ae41c898c3da902cb13601 : SUCCESS in 16m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/84fdc29effff4a8aab7d39acc3752526 : SUCCESS in 35m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/568ab40225e948a6a752551dab31c5d2 : SUCCESS in 24m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4b5ceb799db0488f98dd332e19243e17 : SUCCESS in 39m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eba7c05253ea4a529b152ab599df6c0c : TIMED_OUT in 1h 09m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/68532548e9134107b414a0c573b35b93 : SUCCESS in 40m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4ee4198d1a54626a25bf51d392451ae : SUCCESS in 25m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45cc7f011d5544efaee35462e9d369b9 : SUCCESS in 6m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e17d3cb4393440729c9e73808bc23871 : FAILURE in 9m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32cd8fc209b1446bafe9d34872791bba : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":4},{"id":"37034f94f7628b407b722cd873492945d1075b95","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-14 04:38:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f8f2548e16c48ab531853c5d67062ef1a9f74953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 05:37:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deadc2f969d0471c9c9d7648ec389e83 : SUCCESS in 11m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/6402836d03eb4a56bcbf7162af9fb4c4 : SUCCESS in 42m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fdd07e63bc53416ab818051b620f0b01 : SUCCESS in 50m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e2abfd2ef27f43678905c7093d836c69 : SUCCESS in 7m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/00327c676ed94e4a909924e943aea9d2 : SUCCESS in 9m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/11d04bf1b20b4ed5a346fbf5f15d7b65 : SUCCESS in 12m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4080fa35301d41b3898baba6a7afba5d : SUCCESS in 12m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ca4531cd81fa46348ea53a2eca71c91f : SUCCESS in 12m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/84b2c91a63b244a5b6d0994e078de3b5 : SUCCESS in 13m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/653d962d437f4200b1e08be73e593e65 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d08da7fddb2a4e8c85bd9cea3ef46edf : SUCCESS in 16m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5416a3fb0b844a51b100d693e178e163 : SUCCESS in 16m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fdaa6b729fb5483192367d17928157ca : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/482d49db61cb47c4832cc686fdbe53eb : SUCCESS in 18m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/72724d006253417e99cd84dece5809aa : SUCCESS in 17m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fa4a8999caae42a38204af2b8094ad06 : SUCCESS in 20m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/99af5a5451c341b18e83d22bc95e7bb4 : SUCCESS in 41m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a99acf6f884d472596f810a4ef4934ea : SUCCESS in 35m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fbbd12cf25494a859b01ce9458106e55 : SUCCESS in 45m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23bcd001f56a4d88981800fd17cfe372 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03845cbdddee4c36b85785587aa5d90e : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60e87ef84df14e62bc3810209893c74c : FAILURE in 7m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/111483f9a3844dbcb59ed35dddee1167 : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":4},{"id":"e16a229a2e73c6320794933b0915795bdab4ade6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-28 19:09:58.000000000","message":"Patch Set 4:\n\n(5 comments)\n\nthis change seems reasonable to me, but is lacking unittests; I\u0027d need verify current coverage to +A \n\n... but I\u0027ll put it on the list!","accounts_in_message":[],"_revision_number":4},{"id":"c9508715b0702b7098762c706c3a0115d7f2cbfb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-28 21:46:22.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"79e4497ebe3de1a9bb00d8bc06fc0c788385697f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-08-03 16:47:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9b4d25ee21f6488bbced0f85f082f31c37757df1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 16:47:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3fe2374fe040284fec5736b2eac1956a6b82dd25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 18:17:01.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4370174b83f2443aaeb2e710808f7a3a : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/5cce9a1eae9c498ea3809518dc4cd76f : SUCCESS in 1h 09m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e109749e111843e58cae316a58738a6f : FAILURE in 45m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4f16d5aa8a2a4710933b3d313cc6d5e1 : SUCCESS in 9m 27s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/60e25fc88cd54a17b26c0c244277c68e : SUCCESS in 10m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5700e9f814a7483490ccac802de63827 : SUCCESS in 13m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bd906dc0b08344bbb5ff41c712f0f0b2 : SUCCESS in 14m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e2a9a6d0ae794bdf8b9fae1af09d85a0 : SUCCESS in 15m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b10179c03e734384922c69f32a6da659 : SUCCESS in 12m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ad1aa9d05b549b6b3d0b55bda8ad7f8 : SUCCESS in 28m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/17fa54e3d62645f59d99840280f22c39 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/230e0f7da4a84f739a516157b8c3ce9f : SUCCESS in 24m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c976349bdcb6465199952761a36b20bd : SUCCESS in 17m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b848064764834cac93ada846f68d7b85 : SUCCESS in 21m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2a7c6160b4f841c39cbf24a2e3d558a3 : SUCCESS in 43m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1f0d45ed97ed4931acf256b2424d35df : TIMED_OUT in 1h 10m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/feb13c2fb26046a98eb252c316283f20 : SUCCESS in 43m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2915cbbb36a48d39e356a4ff0840665 : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d71e5c215364be7b6319e78a3ad9cc9 : SUCCESS in 5m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fe2c76123e84115b26d82d1203b1da2 : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":4},{"id":"1673766cd4efee54a5b4232e87ed94e2e00b4d40","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-08-04 19:52:43.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ca9e982a846ea03aeab64f8a80561f097aa93463","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 20:53:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e5d7e46e4a74295a9b7039be664b141 : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/2172eb6f6b4743f8987eb3f3d60e637d : FAILURE in 53m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6db714eb8b1d4c81948b032ce5ac60c2 : SUCCESS in 47m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fd96aaac673f494287daaebaee312200 : SUCCESS in 8m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e157af9606c94c478c2789d280f10e52 : SUCCESS in 9m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/710c645ff9dc461da34f86e13c3c5d54 : SUCCESS in 12m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c1d50580657a47fc92a8275d82c49d7b : SUCCESS in 12m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b30156422d644928a7e5b6218c976b57 : SUCCESS in 12m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f0990dffd0b142f5b61f27b518c867f6 : SUCCESS in 12m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/362dad85d8aa468bb3bbbb11f264c2c6 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7f821619de15480684e740755b81b01a : SUCCESS in 15m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/19111aea17a948a8a36a32c63ee97ecc : SUCCESS in 16m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8c64645f4e1940b98711e06767bf7533 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/837288db854746e99ea04eb5635f3566 : SUCCESS in 17m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/035494d272a24e9cbf97d14cce605389 : SUCCESS in 17m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3d71853b280445d2b6d6d86e549d7f74 : SUCCESS in 20m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dfc5eb0270d144a7a575e030e1974b5f : SUCCESS in 45m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/375a469461de46708e544efae96fd1d1 : SUCCESS in 39m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/56569af30c234474bf1b27697db0e377 : SUCCESS in 39m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ded1231311a448c5b55f93712f8225c2 : SUCCESS in 12m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5239405e40754986ba356e067da71a4c : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d9bd4e548a5b43e7856e478cb5aed1c3 : FAILURE in 7m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d05532251d446c2b7b3b7d220f45719 : SUCCESS in 52m 37s","accounts_in_message":[],"_revision_number":4},{"id":"431dc6ceca407ae10cc945f7e70f5fc75b0a3c0b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-08-05 16:44:46.000000000","message":"Patch Set 4:\n\nreverify","accounts_in_message":[],"_revision_number":4},{"id":"466fdbc7b477c448f53d7d72a2b372481155e88c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 20:00:00.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b242ad479fc94017832d05ae5f528826 : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/8a2a15030d3a49b596040b1568389955 : SUCCESS in 43m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/55b17f3681184624895a9cc29c612bfb : SUCCESS in 52m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/657b6282d58446d5a9dcd18f36ad54e5 : SUCCESS in 11m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8832817368164992bfebe87a9b131ed1 : SUCCESS in 11m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2dafb7a82bed4433bcc2f19469cfd2a3 : SUCCESS in 12m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f82cff1ae09943779a81afb2c86557d1 : SUCCESS in 13m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a12d2fe2200f4088a0efa5401ffa8293 : SUCCESS in 21m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c43f354aeaf247b2919d23e45aea5a8c : SUCCESS in 13m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/33218fc27cd74c6d9d6af7ec352f9a76 : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5320fa3c37e54872994516179a7309c4 : SUCCESS in 19m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/462639d652ca4da3abb8cf7780ccf95e : SUCCESS in 17m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8912eb6694674359be64341f0c001453 : SUCCESS in 21m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/69b804cc274a489f8e1798d993aeabb2 : SUCCESS in 19m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bd43e03c8b4848629e2b1ac7b1107f39 : SUCCESS in 16m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3663b01960e247c585c65330d988d10a : SUCCESS in 32m 04s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f50c8cce12c8479294adee9460df427a : FAILURE in 45m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e48c553cbc249dda229ef586678b49c : SUCCESS in 45m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0de9199e0d0042a48bd52dc35861b1ec : SUCCESS in 43m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e0589c8afdf4604bf3829e526123232 : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be2e269ea65f494190cab3f03cb5b2ba : SUCCESS in 7m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/acceb3e4aa634d9c8f40622e8fa9fcf0 : FAILURE in 9m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54dfe40e4e7c424ca2dbd59e490ea53b : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":4},{"id":"24b1019d7f17f2c55621215e0253b466bcda7294","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-08-07 05:11:24.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1cd49dbab8290a45028da30fc7583da2a06d97e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 06:32:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c78e158f9941508f9ce9a4ae36cbb6 : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/1acf9932436943d785fa8877052f39ad : SUCCESS in 1h 10m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0544ca97ff74423ba4ae555bef24e102 : SUCCESS in 35m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3ada7a4c18c04beea31895b4f17b2b08 : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bf30dff99b4d4b6ca397d11e8c1052dc : SUCCESS in 8m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5f67a99f16c34ab3810005722c73787f : SUCCESS in 13m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c22daeebfd04452c82316fe90b886ff2 : SUCCESS in 18m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a82dd2fe37824bb3874de82683cadd47 : SUCCESS in 13m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a8a34a5ab6844f9a8f4832982f582a55 : SUCCESS in 12m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dcef88bd149a43e6991b48de6042ceb1 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d43de228bb994fd39cc8a348d09b8914 : SUCCESS in 15m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/70caf313d7c14a959e468bdd85dbfe6f : SUCCESS in 16m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/42f910bc6bbd45a8bb41f90c68e5e40a : SUCCESS in 13m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d15ac00b4dfe4288b5f0888fedd9ecb9 : SUCCESS in 17m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/47ebf5844db84578bc3ed051bab1af38 : SUCCESS in 15m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1f6f121102ba4a17b6b5c6f5b84061f2 : SUCCESS in 20m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f1aa1c4cb6fc49c89453c0c4535a6387 : SUCCESS in 47m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b211b3cbda640bfa4b13508b90c8862 : SUCCESS in 41m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4de776361c374d50b9bbb2f4e5ac43a5 : SUCCESS in 37m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8f9b4a97cea4d4081c5f5d1e16a46fa : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae2af30eaa2b463090b7aff81ea3136b : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/772cd61e02974c57a3a3b71acdaa9351 : FAILURE in 7m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5dcd3af826cf4159acfb23bd0fb19be4 : SUCCESS in 58m 57s","accounts_in_message":[],"_revision_number":4},{"id":"373adb954a42ecb6b457c7064848c3c893e78d09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 06:32:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"87a4879289c1c9dd066a9d73fe29d7827724a470","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 07:46:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"506f03155b706efe7c65b957e480bd7071b71335","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 07:46:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8471b1c38e4b40d09e4bddf361821a11 : SUCCESS in 7m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/08c07a9422b245b193033948eb6147d3 : SUCCESS in 1h 10m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1493a5a93fc44d35b31b988c059b247a : SUCCESS in 40m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/99ca0c662e614c38b08e8e0226e8c452 : SUCCESS in 9m 22s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8e68cb9a90694c56b3a0092e0c4b11c6 : SUCCESS in 12m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/04257974b25148abaefa786903ca9e7e : SUCCESS in 14m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c00be2fe1a444a1696c1a43a60384109 : SUCCESS in 16m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2e2e85b4333f484898aec7f23d8cb35d : SUCCESS in 15m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2672ed28b6eb44809e3099febc363aac : SUCCESS in 13m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/55a3249bdcd34e03bd058a258e38de52 : SUCCESS in 20m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f5c5c2816f664ceb8739b3350f5ac5a5 : SUCCESS in 23m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1a384f9e9ea94398bf5fe26a60b3bb8c : SUCCESS in 19m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a79d872a526649d4b32981063065a5bd : SUCCESS in 16m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/61ef0d0b33f44859b741e6d48916e58b : SUCCESS in 29m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8090c78b68b249bfba5f3b304e334878 : SUCCESS in 1h 01m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6ea03f12e59b4a498a2dcd9a01180b1f : SUCCESS in 49m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/34fac8a8535c452485d21f42a7c1b3e2 : SUCCESS in 47m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d7ef0833254459f86b155bfc054041f : SUCCESS in 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef291fe4a884b58aa447241f2203803 : SUCCESS in 7m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef3cf73081974120bae7cc2bb17ba82b : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":4},{"id":"a3067dbeef2b76cc4b643dbf8f72d03bba00885a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 07:49:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4d372deccbd4ee5a65a4e53f0bfec1e : SUCCESS in 3m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/189b9bb66435429794bccc20f69ed5fb : SUCCESS in 1m 44s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c67d376bbf2940f0b04906e07ef04113 : SUCCESS in 1m 45s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"36bd21488e2b1c2fce667412c05ca439c3a17cde","revisions":{"75b5aad4ccdaac8fa8f218b776f6d2e67c04ff3c":{"kind":"REWORK","_number":1,"created":"2020-06-25 21:30:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/738114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/738114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/738114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/738114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/738114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/738114/1"}}},"commit":{"parents":[{"commit":"fbee9f0615334203aabefed8e3d7fba58047936d","subject":"Identify deleted shards as shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fbee9f0615334203aabefed8e3d7fba58047936d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-25 21:08:34.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-25 21:08:34.000000000","tz":-420},"subject":"Address a sharder/replicator race","message":"Address a sharder/replicator race\n\nIn the previous patch, we could clean up all container DBs, but only if\nthe daemons went in a specific order (which cannot be guaranteed in a\nproduction system).\n\nOnce a reclaim age passes, there\u0027s a race: If the container-replicator\nprocesses the root container before the container-sharder processes the\nshards, the deleted shards would get reaped from the root so they won\u0027t\nbe available for the sharder. The shard containers then hang around\nindefinitely.\n\nNow, be willing to mark shard DBs as deleted even when we can\u0027t find our\nown shard range in the root. Fortunately, the shard already knows that\nits range has been deleted; we don\u0027t need to get that info from the root.\n\nChange-Id: If08bccf753490157f27c95b4038f3dd33d3d7f8c\nRelated-Change: Icba98f1c9e17e8ade3f0e1b9a23360cf5ab8c86b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75b5aad4ccdaac8fa8f218b776f6d2e67c04ff3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75b5aad4ccdaac8fa8f218b776f6d2e67c04ff3c"}]},"branch":"refs/heads/master"},"046e970db6c624cf54dbbf888a35f1466dba6b2d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-26 04:33:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/738114/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/738114/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/738114/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/738114/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/738114/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/738114/2"}}},"commit":{"parents":[{"commit":"1af55c0dbf4b89446c99d1c2deafc84ebb9463ef","subject":"Identify deleted shards as shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1af55c0dbf4b89446c99d1c2deafc84ebb9463ef"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-25 21:08:34.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-26 04:32:11.000000000","tz":-420},"subject":"Address a sharder/replicator race","message":"Address a sharder/replicator race\n\nIn the previous patch, we could clean up all container DBs, but only if\nthe daemons went in a specific order (which cannot be guaranteed in a\nproduction system).\n\nOnce a reclaim age passes, there\u0027s a race: If the container-replicator\nprocesses the root container before the container-sharder processes the\nshards, the deleted shards would get reaped from the root so they won\u0027t\nbe available for the sharder. The shard containers then hang around\nindefinitely.\n\nNow, be willing to mark shard DBs as deleted even when we can\u0027t find our\nown shard range in the root. Fortunately, the shard already knows that\nits range has been deleted; we don\u0027t need to get that info from the root.\n\nChange-Id: If08bccf753490157f27c95b4038f3dd33d3d7f8c\nRelated-Change: Icba98f1c9e17e8ade3f0e1b9a23360cf5ab8c86b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/046e970db6c624cf54dbbf888a35f1466dba6b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/046e970db6c624cf54dbbf888a35f1466dba6b2d"}]},"branch":"refs/heads/master"},"eab83c68b44368720e4a7c439cc3948e6a022cec":{"kind":"REWORK","_number":3,"created":"2020-07-08 05:01:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/738114/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/738114/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/738114/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/738114/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/738114/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/738114/3"}}},"commit":{"parents":[{"commit":"a4cf508a3859f6b37ac4d9767de38fc98cbea068","subject":"Identify deleted shards as shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4cf508a3859f6b37ac4d9767de38fc98cbea068"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-25 21:08:34.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-07 23:06:28.000000000","tz":-420},"subject":"Address a sharder/replicator race","message":"Address a sharder/replicator race\n\nIn the previous patch, we could clean up all container DBs, but only if\nthe daemons went in a specific order (which cannot be guaranteed in a\nproduction system).\n\nOnce a reclaim age passes, there\u0027s a race: If the container-replicator\nprocesses the root container before the container-sharder processes the\nshards, the deleted shards would get reaped from the root so they won\u0027t\nbe available for the sharder. The shard containers then hang around\nindefinitely.\n\nNow, be willing to mark shard DBs as deleted even when we can\u0027t find our\nown shard range in the root. Fortunately, the shard already knows that\nits range has been deleted; we don\u0027t need to get that info from the root.\n\nChange-Id: If08bccf753490157f27c95b4038f3dd33d3d7f8c\nRelated-Change: Icba98f1c9e17e8ade3f0e1b9a23360cf5ab8c86b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eab83c68b44368720e4a7c439cc3948e6a022cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eab83c68b44368720e4a7c439cc3948e6a022cec"}]},"branch":"refs/heads/master"},"36bd21488e2b1c2fce667412c05ca439c3a17cde":{"kind":"REWORK","_number":4,"created":"2020-07-13 21:57:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/738114/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/738114/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/738114/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/738114/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/738114/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/738114/4"}}},"commit":{"parents":[{"commit":"a4cf508a3859f6b37ac4d9767de38fc98cbea068","subject":"Identify deleted shards as shards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4cf508a3859f6b37ac4d9767de38fc98cbea068"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-25 21:08:34.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-13 21:57:25.000000000","tz":-420},"subject":"Address a sharder/replicator race","message":"Address a sharder/replicator race\n\nIn the previous patch, we could clean up all container DBs, but only if\nthe daemons went in a specific order (which cannot be guaranteed in a\nproduction system).\n\nOnce a reclaim age passes, there\u0027s a race: If the container-replicator\nprocesses the root container before the container-sharder processes the\nshards, the deleted shards would get reaped from the root so they won\u0027t\nbe available for the sharder. The shard containers then hang around\nindefinitely.\n\nNow, be willing to mark shard DBs as deleted even when we can\u0027t find our\nown shard range in the root. Fortunately, the shard already knows that\nits range has been deleted; we don\u0027t need to get that info from the root.\n\nChange-Id: If08bccf753490157f27c95b4038f3dd33d3d7f8c\nRelated-Change: Icba98f1c9e17e8ade3f0e1b9a23360cf5ab8c86b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36bd21488e2b1c2fce667412c05ca439c3a17cde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36bd21488e2b1c2fce667412c05ca439c3a17cde"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
