)]}'
{"id":"openstack%2Fswift~738269","triplet_id":"openstack%2Fswift~master~I8e5878e27c9773f952861298f3f4d325d6d38a04","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I8e5878e27c9773f952861298f3f4d325d6d38a04","subject":"Move jobs from bionic to focal","status":"ABANDONED","created":"2020-06-26 19:45:18.000000000","updated":"2025-02-27 00:58:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"43c0ec5df786c38709dbb726a93f750e593a5f86","_number":738269,"virtual_id_number":738269,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-26 21:19:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 21:19:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da93ca7e674aff64d62a375bcbad6a5afe36e793","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-06-26 19:45:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c82b2a43824097baebd92eb36def027d9642fbd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 21:19:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/022784e39550478285d4ca599f381577 : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/8e2ddff6f0894998a36955cc699e9235 : SUCCESS in 1h 10m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f227079e53e24249a6031b3b8f37fc71 : SUCCESS in 51m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6161883c7e1a41b786586a7154870bb9 : SUCCESS in 10m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/96b2d2b6ebc64497878860817b6d2dbe : SUCCESS in 10m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b23a40ce389e4b5ca6ef3f2d5750fa87 : SUCCESS in 23m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/58c4d2c0996e4e9296c7c918bf5207b4 : FAILURE in 6m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d83f6364d57840a2ad7b8c69933ea56e : RETRY_LIMIT in 4m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e65a7828f08244c8a7b471bc4e92bf70 : SUCCESS in 13m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b2a73c9c6f824dd491b282867156d471 : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f73f907af1a6443c8795102f9344fffc : SUCCESS in 32m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/87c5d6ea8ed24d41b5f3c5783e0fba9e : SUCCESS in 23m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e0ed2fb0bc834a7ba3e04af4a4f8ae43 : RETRY_LIMIT in 3m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/daf5a4620a914dbb92fd6dc25218f0bd : RETRY_LIMIT in 3m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3628656ba9a24cd38cd60c1d8a3f025b : RETRY_LIMIT in 3m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d999a575ff814194971645f0eef9d13e : SUCCESS in 30m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/74cbacfd38e341039bf9c65f89ab1d58 : SUCCESS in 58m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5a4856b486f84ff1bc2da397fd16890c : SUCCESS in 50m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3ba4b4fc2f554e7ea7e3008652f69761 : SUCCESS in 51m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e03329bc9c94ff1b0bbc4257fc29bf4 : SUCCESS in 18m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c690d908c94a426b926e404652b0d1a9 : SUCCESS in 7m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c057a2f8061d46bb8e3c6bb17b6ab53b : FAILURE in 11m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e2716f41c2e4e53951c9d05f542e199 : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":1},{"id":"43c0ec5df786c38709dbb726a93f750e593a5f86","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-27 00:58:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11393a539c441c9d03c94748deb67376428d1b50","revisions":{"11393a539c441c9d03c94748deb67376428d1b50":{"kind":"REWORK","_number":1,"created":"2020-06-26 19:45:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/69/738269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/738269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/738269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/738269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/738269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/738269/1"}}},"commit":{"parents":[{"commit":"5c087ad7b93cd52d2ba8c34db5ac0fc06db93f08","subject":"Fix up curl commands when object name has unix dir characters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c087ad7b93cd52d2ba8c34db5ac0fc06db93f08"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-26 19:42:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-06-26 19:42:16.000000000","tz":-420},"subject":"Move jobs from bionic to focal","message":"Move jobs from bionic to focal\n\nDepends-On: https://review.opendev.org/#/c/734700/\nDepends-On: https://review.opendev.org/#/c/731207/\nChange-Id: I8e5878e27c9773f952861298f3f4d325d6d38a04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11393a539c441c9d03c94748deb67376428d1b50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11393a539c441c9d03c94748deb67376428d1b50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
