)]}'
{"id":"openstack%2Fswift~740866","triplet_id":"openstack%2Fswift~master~I81c66c57362c9a08b55bdb7925f567318d50d02f","project":"openstack/swift","branch":"master","topic":"low-replica-rings","hashtags":[],"change_id":"I81c66c57362c9a08b55bdb7925f567318d50d02f","subject":"ring: Be willing to query nodes with no assignments","status":"ABANDONED","created":"2020-07-14 04:47:59.000000000","updated":"2021-03-30 22:33:55.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7fd73d2ac1a45e37ce006583acfa69af05e9166b","_number":740866,"virtual_id_number":740866,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-15 21:53:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-14 15:06:15.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-07-15 21:53:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7519d3c08245e139340481b51588eec046957617","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-14 04:47:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3e2e8ed97d978c80efd5361b7a3b2ea937b67fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 06:09:55.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1024c9113644b75962356c5541d1ea1 : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/8904c8f581f34b6ba53a681fbd94c4fe : SUCCESS in 1h 01m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0c68cffea1a54a7d85c43b26b23c21d2 : SUCCESS in 43m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/464e40de24e74600b25e8285ae9a3081 : SUCCESS in 8m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f6494c120b62417db0236e6013659b68 : SUCCESS in 8m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/853b7ea19cbf4035aece6b34e0c676e4 : SUCCESS in 29m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b928ddf961594a90a6c323129dc636e8 : SUCCESS in 35m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0958dcdc83444705ab2ea5b919d9f10c : SUCCESS in 29m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/72457b9c581a445baff38ec9ff38da48 : SUCCESS in 12m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f2519c855abd44529b2a73beec5cc56e : SUCCESS in 50m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/76932f8074ff436799f6048d8034486d : SUCCESS in 38m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/32f456df488840df8205625c32a10306 : SUCCESS in 53m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fbe0570706d5440e8a63d64a0a407687 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e877c38f59d34a4791ca1156a94ec78b : SUCCESS in 16m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/132a2f426444413bad8c6c74d11023f9 : SUCCESS in 17m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/313ca3606a2c49898675812295b6e80c : SUCCESS in 21m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/469af7fdc09f4429bedbeae6d9084469 : FAILURE in 1h 20m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6081d9e835b3418c8ba668962a92c362 : TIMED_OUT in 1h 04m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0788bdc53b6348779bb8bc3868bbb15a : SUCCESS in 1h 01m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcc1cdbf0d7f49ad9b988ccd81debe05 : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2030dad98d049edbd7b7f49a5f4aecc : FAILURE in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cc92d88f5e724ba29dcfa62981490a00 : FAILURE in 9m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a417e2bf81814783b5a31dfd7c68a064 : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f7481e9cad7c9da6cebfc08bd63ef9b1ccb9211c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-14 15:06:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\ni\u0027m sure this is related to something you\u0027re working on in a saio maybe?\n\nSeems like the tail wagging the dog.","accounts_in_message":[],"_revision_number":1},{"id":"73b647f48d64c59cdd572c68d97317a31416b4cd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-14 16:39:18.000000000","message":"Patch Set 1:\n\nSomething from me tinkering with a home cluster, actually -- though I remembered someone else asking about this sort of thing before, too.\n\nBasically, I started out with just a couple big spinners, and put A/C on them knowing full well that I\u0027d regret it later... then did. After a a few too many\n\n blk_update_request: I/O error, ...\n\nthat required a cold boot to resolve, I finally got around to adding an SSD, adjusting weights, rebalancing... and of course creating an availability issue :P\n\nFair enough, I did something dumb -- but then I was sad to see that I couldn\u0027t work around it by just increasing request_node_count -- I\u0027d have to add a tiny bit of weight to the other disks and rebalance again (with an explicit goal that none of the assignments should actually be used)... which *also* feels like the tail wagging the dog. This at least has the benefit that in anything approaching a reasonably-sized, well-configured and maintained cluster, this change has no impact -- request_node_count will ensure we stop digging before we ever hit the if block. (At least, until we do something like https://review.opendev.org/#/c/704435/ that exhaustively checks every handoff.)","accounts_in_message":[],"_revision_number":1},{"id":"ccb9800d771c8a02435717c061d942f79996a651","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-14 23:13:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e630b9f1c5f2b156628067e9ee4b4f67ad747ae8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-15 18:09:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3714b7537b42178c09c722bee8413399ab8551d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 21:53:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f6636bb91ff4718897c855a55258357 : SUCCESS in 10m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/c23d5feb2c5a4b9995bec81610345895 : SUCCESS in 1h 01m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5e124b9b65cc4c3397daee4f35983fcb : SUCCESS in 52m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0e05ff34610242329e7265ae1124f325 : SUCCESS in 10m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4592e0a6172b49a697b1cda1e4457516 : SUCCESS in 11m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dfd083fe34ec485a931b3438795728ee : SUCCESS in 14m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5f55e8da112d4ce0b3960cf88c2932f2 : SUCCESS in 14m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3ab2825ea77649409587c7735dcdbc45 : SUCCESS in 15m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4fb0f11d652445d78c70b7994a495627 : SUCCESS in 14m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dc2f454cf6c74007ae72667b693debf5 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b62e843369d54151a78c1e896e1fb02f : SUCCESS in 19m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b51485de746048369d5d458f476109e1 : SUCCESS in 18m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/03592b57a19648dca860141c0c4cb1b3 : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/94bdd9ac9fc349299032d81761aa42d2 : SUCCESS in 34m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/34b8cf49213a4969ae9ec21af973aa16 : SUCCESS in 18m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6c525ed10ae044ad917ffb5fc8af7e18 : SUCCESS in 45m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c69a471c90eb4007ba1935a2d5ba439f : SUCCESS in 47m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/894179d2bd0841f688b77a6ae78bb4fa : SUCCESS in 45m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30db0bbbfbc44ba490dea5f677dd7b41 : SUCCESS in 43m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ae05823343d45bbb160a18e274f539b : SUCCESS in 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d86b91df78294193b94873deaeb7361a : SUCCESS in 7m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5c7b45bad94d4abdb91e95aa4d09f8f5 : FAILURE in 11m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6e0e11d640f464e8e85e16208ce2f12 : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":2},{"id":"7fd73d2ac1a45e37ce006583acfa69af05e9166b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-30 22:33:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"abb4ba7c0049c5d2ffa0bd8083575495ff0e8a74","revisions":{"6ede7645e0f4313f7526d24321ad3300acae1b8b":{"kind":"REWORK","_number":1,"created":"2020-07-14 04:47:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/740866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/740866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/740866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/740866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/740866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/740866/1"}}},"commit":{"parents":[{"commit":"2e001431fd0907c97d44b81dd1f3d2d14ae1208b","subject":"s3api: Don\u0027t do naive HEAD request for auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e001431fd0907c97d44b81dd1f3d2d14ae1208b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-13 21:27:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-14 04:46:58.000000000","tz":-420},"subject":"ring: Be willing to query nodes with no assignments","message":"ring: Be willing to query nodes with no assignments\n\nThis is helpful when trying to remove devices in very small clusters\nwith a low number of replicas.\n\nChange-Id: I81c66c57362c9a08b55bdb7925f567318d50d02f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ede7645e0f4313f7526d24321ad3300acae1b8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ede7645e0f4313f7526d24321ad3300acae1b8b"}]},"branch":"refs/heads/master"},"abb4ba7c0049c5d2ffa0bd8083575495ff0e8a74":{"kind":"REWORK","_number":2,"created":"2020-07-15 18:09:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/740866/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/740866/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/740866/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/740866/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/740866/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/740866/2"}}},"commit":{"parents":[{"commit":"2e001431fd0907c97d44b81dd1f3d2d14ae1208b","subject":"s3api: Don\u0027t do naive HEAD request for auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e001431fd0907c97d44b81dd1f3d2d14ae1208b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-13 21:27:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2020-07-15 18:08:51.000000000","tz":0},"subject":"ring: Be willing to query nodes with no assignments","message":"ring: Be willing to query nodes with no assignments\n\nThis is helpful when trying to remove devices in very small clusters\nwith a low number of replicas.\n\nChange-Id: I81c66c57362c9a08b55bdb7925f567318d50d02f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/abb4ba7c0049c5d2ffa0bd8083575495ff0e8a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/abb4ba7c0049c5d2ffa0bd8083575495ff0e8a74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
