)]}'
{"id":"openstack%2Fswift~740867","triplet_id":"openstack%2Fswift~master~Ica2b41a52d11cb10c94fa8ad780a201318c4fc87","project":"openstack/swift","branch":"master","topic":"replication-network","hashtags":[],"change_id":"Ica2b41a52d11cb10c94fa8ad780a201318c4fc87","subject":"Allow replication servers to handle all request methods","status":"MERGED","created":"2020-07-14 04:51:52.000000000","updated":"2020-07-23 21:24:43.000000000","submitted":"2020-07-23 21:23:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740867-1595539387660-1a0e6018","meta_rev_id":"7bcbc353d2729ac4af96c0d0789843f146a0fca0","_number":740867,"virtual_id_number":740867,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-23 21:23:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-07-23 18:38:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-07-23 18:38:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-23 18:38:41.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-07-23 21:23:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd6d58c7773d5c1e51be47b124b82f994e5015f0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-14 04:51:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca8e4105ca2cb4ca25a13d6cf3859bba6935d010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 05:56:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf744c079194567838fc8ae58ac4aab : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/0bd5fb210110492cbe964c2e4ffd45f9 : SUCCESS in 57m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e838c65f84a14d74afab43322c9d0f5a : SUCCESS in 43m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6ed59b021084409db3b62d1c14f178ca : SUCCESS in 9m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a6fdcd99518a4e3e8c80cbfdb3265394 : SUCCESS in 10m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a0d299ee773a4f7bbba24dfc9d6daab6 : SUCCESS in 16m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d335e0ec6d274bf5aab68266f89d71db : SUCCESS in 12m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ec206eeb229c4e049a628e55d10c49bb : SUCCESS in 12m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f9a61aec7978416a970a817d2aff6955 : SUCCESS in 24m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e4811b5864ed4e81ac6e037d2f42cb1e : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2de41889eaef4d6f8d928380dc4d4c64 : SUCCESS in 19m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2b860ba86d6c4a26b5c80ff8a86f6083 : SUCCESS in 36m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/57b8988a4b4540d5a8def4da31af12b2 : SUCCESS in 39m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/44b664f731d942c0998ab36c0975c8cf : SUCCESS in 17m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/77cae7f0838a45d780a39f90af6a4c74 : SUCCESS in 19m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1c9f44014f4a4b1b9104eb93d394a6da : SUCCESS in 19m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5d2080c8ca334fbba89c46387f446d3e : SUCCESS in 47m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6974ae8613594660a12288e10eda30ee : SUCCESS in 39m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/59f957853bfc49e8851393c9b5459471 : SUCCESS in 41m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26268c9923e64ca8acc6c141789c79f4 : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f289f274774560b2277f4f29a5118c : SUCCESS in 6m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a0927c6519d94da18128bf90e6a26875 : FAILURE in 7m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cea200698a204c41a71a510cdd164bad : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":1},{"id":"6d6f474d940d6765e4fa08eb52a92571f20f96d2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-14 15:03:54.000000000","message":"Patch Set 1:\n\nI\u0027m pretty sure if this does what it says it also closes this bug:","accounts_in_message":[],"_revision_number":1},{"id":"6abd14075ae786e0746fe8c6fe1ffbeffb0c49b2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-14 15:04:01.000000000","message":"Patch Set 1:\n\nhttps://bugs.launchpad.net/swift/+bug/1446873","accounts_in_message":[],"_revision_number":1},{"id":"07bc9c4a60f44c02776ac135264800ff00c7b3f5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-14 17:38:39.000000000","message":"Patch Set 1:\n\nNo, not yet -- maybe this was premature -- long term, we almost certainly want replication_server\u003dtrue to have the blank/unset behavior. I was just irked the only way to get the \"do all the things\" behavior on master is to leave the option off entirely.","accounts_in_message":[],"_revision_number":1},{"id":"67a8676f00823640520ecd37bbe6010b6aad2667","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-17 21:47:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2a7bd578a17c5ceb25f8a6256b1ff5fffbb572e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 02:14:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf81d4b700ff4e2f920ebaaf107b1dfc : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/f4be30cac02f45f1acd5a38be9fb2c6b : SUCCESS in 1h 03m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1c02c0bc07944c3bb6daae271035972d : SUCCESS in 42m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c8dc71ad12694ea4b0e7b161f7e07573 : SUCCESS in 9m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e22a7a43557a4e729d395da6260ed6e4 : SUCCESS in 10m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae76c4270d1a4ea3a52014956a1ea0aa : FAILURE in 12m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0e4ff1b72bc94ad38c4a7e88235f202c : FAILURE in 16m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4e1ce516288941df82990b9b91f1ec30 : FAILURE in 13m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fbcc70f12c3f48ba82fe950383b8ec4a : TIMED_OUT in 1h 01m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/56247db77ece4cf7bd0f62460d6a61b4 : SUCCESS in 18m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/57a73ca361244664b25fe19de9aac7c4 : SUCCESS in 28m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/25ce52573fb247ef9134a2f468ba219f : SUCCESS in 18m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/af44410be3234433b4220034e2116e4b : SUCCESS in 18m 32s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bfba323d24194014a989e1d5145217ea : SUCCESS in 19m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f0ea2ac7069c4fc2a043d944125b4cfe : SUCCESS in 18m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/295e110b68c14301bcd8f81a20837cb0 : RETRY_LIMIT in 7m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4ccf366b4ed244879fd1bb2c7df35ea6 : RETRY_LIMIT in 5m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8d1d1fdb500b40b188fc17296131736b : SUCCESS in 45m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/448121af2ce2480fa026095f85d5580c : SUCCESS in 43m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55e87347eac04d3e94f3bf40e9c30310 : FAILURE in 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ba0a1b08ddb43068d7a32d4e5fac04e : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0a4aa6237b7c46cdbda912d325f2e715 : RETRY_LIMIT in 8m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20afc976c8144ecbb974ea1101da7bcb : SUCCESS in 59m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4d7f44ced9089745408f67458e541d6b4cc20251","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-21 18:02:17.000000000","message":"Patch Set 2: Code-Review+2\n\nthis is pretty great!","accounts_in_message":[],"_revision_number":2},{"id":"92943a960b710d497e05ad5924513fcedc668eb9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-22 03:39:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e95b7efd123fed2b9f9b2bce3a17f8e0aba1ebc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:48:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07602b09228e4b94b434134f89471463 : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/765890b5c49344038668152b70959856 : SUCCESS in 1h 03m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/79de2cb8cfe94dc0a80ee4045ee224f2 : SUCCESS in 48m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e14e26677d34107bdb20d719c76ca5e : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/095e6dd5bd1749139700018a1fac39bb : SUCCESS in 9m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d4da13d0b3b042bb947f8c7d377d1f0a : FAILURE in 13m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2c5c6a62b34446b5a972d4c8a1397f54 : FAILURE in 12m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e2c05f449b6340e2859796feb79daf0b : FAILURE in 13m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/465df0c68e174fbaafb50c6ce4efaaf1 : FAILURE in 15m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2260afd80971471bac69260c326923f2 : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b0fc55671a1c49a9a109b9113dfbf33b : SUCCESS in 17m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e5cd3405c01d4005af9287429288a0b8 : SUCCESS in 21m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a8cdabd1228648bb87be3049fdeef2b1 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b2e9f88f165d483a905c20504f73f986 : SUCCESS in 22m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/749238d1d34c48bf94ae7abd3524c47d : SUCCESS in 22m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8419e761596144e7acc8128a33267e51 : SUCCESS in 22m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ebc09248640c4270b66d473219a1bc44 : SUCCESS in 47m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/626cf83f35c442bcae6f69f991eb53df : SUCCESS in 45m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c407f09aff9d4ce3917150fccb04e4f9 : SUCCESS in 42m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03ffc94d175c4cc4b8d7ceb92853a3a5 : FAILURE in 28m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c420f1b6f58847cf84b1ddbe4b54bb1b : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f733357c9d704d6eb37da3ccaf666dc0 : FAILURE in 8m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46d04e411608414d937d983cc8c3134f : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":2},{"id":"4ae15e63f51c2654a1efb45149d8f0bcf9a893ea","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-22 20:59:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"37128b73e5ec49f1229a6e6ef1c17f1be06fc50a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 22:13:03.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55882c2787f74278b263d42440a6cc4f : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/aa6b447fab5b4adfae238d9d4625ede0 : SUCCESS in 56m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/76395f80f1224e57b3cf0f7e5987e24a : SUCCESS in 37m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a18fd8ed83904bde83d1328ac00459b0 : SUCCESS in 10m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/df9f15ab0bad48bba39cb4beda0d786c : SUCCESS in 9m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f95abd65acac4659bb9562f1415a2ba6 : FAILURE in 14m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/83fb35fa44304c08835dd953c19702c1 : FAILURE in 16m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2be68378ff4c4fc0b81cd43e9093672d : FAILURE in 13m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8a0d379d994f4ccbbbb369ffa2b822ea : FAILURE in 13m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e1fbf48c34af43339ae5caa5ed4b8422 : SUCCESS in 20m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ac079020c4864dfbbc3e2b3056b72783 : SUCCESS in 22m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8372a26d5f7a457894b3c06732c8c8c4 : SUCCESS in 18m 13s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cdd5c42dd8b4494db405a11ac6d57cd6 : SUCCESS in 19m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/be373ed6175d4a22b8557c018773d198 : SUCCESS in 22m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5ebb5d81bbd64874bd52650e6f0af7bb : SUCCESS in 25m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4e473d6d9e2c4836b4f35a8eaedfde6d : SUCCESS in 25m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a26ce6e73b7644cfbba7e78af508a3c3 : SUCCESS in 46m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1e437b4ff5c940c1888f35c4f36d5b68 : SUCCESS in 53m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff5cc53902dd4517b0a28b47b89dec46 : SUCCESS in 54m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85d1b15bb7a4486e8e5b83b5ad2270f9 : FAILURE in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29bb7b0653df4ab6969e48475feb8dc9 : FAILURE in 10m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e34667b70ed34b6b96aa1eacb61ff1de : FAILURE in 9m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/baaa76c281224db3aebdfdda4a7ed098 : SUCCESS in 45m 01s","accounts_in_message":[],"_revision_number":3},{"id":"f8b76df3436217513918dfb10dc0d77d3ffdc1e1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-23 16:11:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fd4b6d66a2e637e53ba1aaa8eb313824ac7a6e30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 17:34:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2372dbfbe572463aa733be01d2e4b15c : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/78f535003ae94b75a3e62737a22075ea : SUCCESS in 1h 07m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b13fd87bf64546f49424bbc297144c04 : SUCCESS in 51m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4116d28ca4f340c8bcd5cb4c4e15fa89 : SUCCESS in 10m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/785c12a992d94d1f841091ef9440b75c : SUCCESS in 12m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4bcd43043bef42fdaefad234937e1e71 : SUCCESS in 13m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/736610c9f66d43938f7746a1f237d98a : SUCCESS in 14m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/50232c00d7194df29b8efea41379f0a9 : SUCCESS in 15m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1b8aa2cb2bb24706b427475c4a6b66b1 : SUCCESS in 15m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b2ce039a017741c8889ddc48c33deb43 : SUCCESS in 26m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/acaa749c46264122a82c20fd476f0f62 : SUCCESS in 25m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6a0212d915534bf9b28a377c9b3b7e2c : SUCCESS in 23m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ef0dc5c9860f4b55b0c7c73c2726e6fd : SUCCESS in 21m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e560f89ab0ff4dd584d279e14dd3a7c5 : SUCCESS in 24m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9de90b39a9fd44ce96b01a0ccdf1bc60 : SUCCESS in 21m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e0b0adc7cb60428cbcf11df25e727748 : SUCCESS in 32m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/425b2a87e5314631ba946af3d7978373 : SUCCESS in 48m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d0640d4a0ef44dbbff1df0a31a805d4 : SUCCESS in 47m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/263ede3dc73a4d1eae947672180f9aa1 : SUCCESS in 51m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1eb3e9a8ca0a449e9cd9b546d787f605 : SUCCESS in 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af995675769b416a8824017f3afc73b0 : SUCCESS in 7m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/616fe949dd524eb9a25a7d357c736988 : FAILURE in 10m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eda9dbaf5e9f43e78d8d201fc544c3e9 : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":4},{"id":"e85f6eb66ed1b1c88c17346d2f46d39122a055bf","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-23 18:38:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(3 comments)\n\neverything seems in order!","accounts_in_message":[],"_revision_number":4},{"id":"1aaa77e1b713c6df8e2b6d8f3dee3129bcae4b99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 18:39:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8870a71a11b7f078077835052ccc89edc4c65c81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 21:23:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b6f8e6476434c0ca977296b821a4f25 : SUCCESS in 9m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/76f61937bc194e6486b21741bd4a8242 : SUCCESS in 46m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3cbf08afccde46e9812312ede9d87b3c : SUCCESS in 42m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8292ce13d95742aa8c709911d0b3305a : SUCCESS in 11m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/31ff38fb1b3d4056b272de42c6af530c : SUCCESS in 9m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/775c1e878bea42a5b03daa2e80b7acba : SUCCESS in 12m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/53d7d2b5583f4e2c81f24d2951b39739 : SUCCESS in 18m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e18ec216d19a4534ba4483fa5d1298d4 : SUCCESS in 15m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c842781ecdf04cb1b487605fb0a33287 : SUCCESS in 15m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ee2188f0eedb4358a44669931243d919 : SUCCESS in 18m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eec0ca806b144b07b5bc4a3b83efad8e : SUCCESS in 22m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4f4a38e7acb4dc9bb40e4b0e81583e0 : SUCCESS in 18m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bfd98aab84884f05bbaf67c0dee86e32 : SUCCESS in 21m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4b4f663e6a404bde90f13ff7860e5329 : SUCCESS in 18m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b08d9a1524374b1a9c9973538c639404 : SUCCESS in 49m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/011b6e4dd9fa4a099c6471dfc90846e9 : SUCCESS in 41m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/18ad8323e2b34dbcbf10b83a0e9ccdb4 : SUCCESS in 41m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1f9913cc14a4d62abfae437cc86b823 : SUCCESS in 27m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8a7bbe9ef6a4283b11cf91e72f7e7d9 : SUCCESS in 7m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61919aeb99c54733b5ff2229e37ac11b : SUCCESS in 51m 25s","accounts_in_message":[],"_revision_number":4},{"id":"bbfaeef22d8c5d54ba98c6ef5cb59813aafa73be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 21:23:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"296d1271a30fe1688a71496d1f3516b043f4fa79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 21:24:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a51f62c2cf484521a139c60b6de1dea8 : SUCCESS in 1m 11s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a5014828d3194f55b710907dded7c597 : SUCCESS in 42s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b9d31d0e4eb74dc6834db5012c8b6a7d : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9eb81f6e698ef8ac8855c413b0b1bf4e030cb687","revisions":{"cefddda9a7404e8a73ea6a95672eb62556e050c7":{"kind":"REWORK","_number":1,"created":"2020-07-14 04:51:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/67/740867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/740867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/740867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/740867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/740867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/740867/1"}}},"commit":{"parents":[{"commit":"2e001431fd0907c97d44b81dd1f3d2d14ae1208b","subject":"s3api: Don\u0027t do naive HEAD request for auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e001431fd0907c97d44b81dd1f3d2d14ae1208b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-08 04:28:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-14 04:49:39.000000000","tz":-420},"subject":"Make blank replication_server behave same as missing","message":"Make blank replication_server behave same as missing\n\nLeaving the option blank seems to signal very different intent than\nsetting it explicitly true or false.\n\nFWIW, this behavior seems to go back to the original replication server\nimplementation. It seems very wrong, though, that there\u0027s no value you\ncan set to get the default behavior.\n\nChange-Id: Ica2b41a52d11cb10c94fa8ad780a201318c4fc87\nRelated-Change: Ie9af5bdcdf9241c355e36053ca4adfe49dc35bd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cefddda9a7404e8a73ea6a95672eb62556e050c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cefddda9a7404e8a73ea6a95672eb62556e050c7"}]},"branch":"refs/heads/master"},"bda3c4cec0218cbbacc4338bab1931c2a9336611":{"kind":"REWORK","_number":2,"created":"2020-07-17 21:47:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/67/740867/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/740867/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/740867/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/740867/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/740867/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/740867/2"}}},"commit":{"parents":[{"commit":"2e001431fd0907c97d44b81dd1f3d2d14ae1208b","subject":"s3api: Don\u0027t do naive HEAD request for auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e001431fd0907c97d44b81dd1f3d2d14ae1208b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-08 04:28:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-15 22:36:07.000000000","tz":-420},"subject":"Allow replication servers to handle all request methods","message":"Allow replication servers to handle all request methods\n\nPreviously, the replication_server setting could take one of three\nstates:\n\n * If unspecified, the server would handle all available methods.\n * If \"true\", \"yes\", \"on\", etc. it would only handle replication\n   methods (REPLICATE, SSYNC).\n * If any other value (including blank), it would only handle\n   non-replication methods.\n\nHowever, because SSYNC tunnels PUTs, POSTs, and DELETEs through\nthe same object-server app that\u0027s responding to SSYNC, setting\n`replication_server \u003d true` would break the protocol. This has\nbeen the case ever since ssync was introduced.\n\nNow, get rid of that second state -- operators can still set\n`replication_server \u003d false` as a principle-of-least-privilege guard\nto ensure proxy-servers can\u0027t make replication requests, but replication\nservers will be able to serve all traffic. This will allow replication\nservers to be used as general internal-to-the-cluster endpoints, leaving\nnon-replication servers to handle client-driven traffic.\n\nCloses-Bug: #1446873\nChange-Id: Ica2b41a52d11cb10c94fa8ad780a201318c4fc87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bda3c4cec0218cbbacc4338bab1931c2a9336611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bda3c4cec0218cbbacc4338bab1931c2a9336611"}]},"branch":"refs/heads/master"},"347d5ffc5e2a610cb52a6c072a3d1738ce5f68ba":{"kind":"REWORK","_number":3,"created":"2020-07-22 20:59:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/67/740867/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/740867/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/740867/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/740867/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/740867/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/740867/3"}}},"commit":{"parents":[{"commit":"b1f38728963febfa409a803e5d44ff17a3e2226f","subject":"Merge \"py3: Stop munging RAW_PATH_INFO\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1f38728963febfa409a803e5d44ff17a3e2226f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-08 04:28:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-22 20:58:29.000000000","tz":-420},"subject":"Allow replication servers to handle all request methods","message":"Allow replication servers to handle all request methods\n\nPreviously, the replication_server setting could take one of three\nstates:\n\n * If unspecified, the server would handle all available methods.\n * If \"true\", \"yes\", \"on\", etc. it would only handle replication\n   methods (REPLICATE, SSYNC).\n * If any other value (including blank), it would only handle\n   non-replication methods.\n\nHowever, because SSYNC tunnels PUTs, POSTs, and DELETEs through\nthe same object-server app that\u0027s responding to SSYNC, setting\n`replication_server \u003d true` would break the protocol. This has\nbeen the case ever since ssync was introduced.\n\nNow, get rid of that second state -- operators can still set\n`replication_server \u003d false` as a principle-of-least-privilege guard\nto ensure proxy-servers can\u0027t make replication requests, but replication\nservers will be able to serve all traffic. This will allow replication\nservers to be used as general internal-to-the-cluster endpoints, leaving\nnon-replication servers to handle client-driven traffic.\n\nCloses-Bug: #1446873\nChange-Id: Ica2b41a52d11cb10c94fa8ad780a201318c4fc87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/347d5ffc5e2a610cb52a6c072a3d1738ce5f68ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/347d5ffc5e2a610cb52a6c072a3d1738ce5f68ba"}]},"branch":"refs/heads/master"},"9eb81f6e698ef8ac8855c413b0b1bf4e030cb687":{"kind":"REWORK","_number":4,"created":"2020-07-23 16:11:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/67/740867/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/740867/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/740867/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/740867/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/740867/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/740867/4"}}},"commit":{"parents":[{"commit":"0dbf3d0a954522e5f6cc529c6da961ef2f87991d","subject":"Actually interpolate error message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dbf3d0a954522e5f6cc529c6da961ef2f87991d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-08 04:28:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-23 16:11:07.000000000","tz":-420},"subject":"Allow replication servers to handle all request methods","message":"Allow replication servers to handle all request methods\n\nPreviously, the replication_server setting could take one of three\nstates:\n\n * If unspecified, the server would handle all available methods.\n * If \"true\", \"yes\", \"on\", etc. it would only handle replication\n   methods (REPLICATE, SSYNC).\n * If any other value (including blank), it would only handle\n   non-replication methods.\n\nHowever, because SSYNC tunnels PUTs, POSTs, and DELETEs through\nthe same object-server app that\u0027s responding to SSYNC, setting\n`replication_server \u003d true` would break the protocol. This has\nbeen the case ever since ssync was introduced.\n\nNow, get rid of that second state -- operators can still set\n`replication_server \u003d false` as a principle-of-least-privilege guard\nto ensure proxy-servers can\u0027t make replication requests, but replication\nservers will be able to serve all traffic. This will allow replication\nservers to be used as general internal-to-the-cluster endpoints, leaving\nnon-replication servers to handle client-driven traffic.\n\nCloses-Bug: #1446873\nChange-Id: Ica2b41a52d11cb10c94fa8ad780a201318c4fc87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9eb81f6e698ef8ac8855c413b0b1bf4e030cb687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9eb81f6e698ef8ac8855c413b0b1bf4e030cb687"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
