)]}'
{"id":"openstack%2Fswift~741538","triplet_id":"openstack%2Fswift~master~I918ab4ccbf4d081b26f4117937410cdad1caf8d3","project":"openstack/swift","branch":"master","topic":"bug/1862645","hashtags":[],"change_id":"I918ab4ccbf4d081b26f4117937410cdad1caf8d3","subject":"Stop syncing empty suffixes list","status":"MERGED","created":"2020-07-16 20:13:18.000000000","updated":"2020-07-27 16:46:07.000000000","submitted":"2020-07-22 04:22:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741538-1595391752331-b2062de1","meta_rev_id":"7ddf0ad313d7089217bfcf2668e89e7a48de0e73","_number":741538,"virtual_id_number":741538,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-22 04:22:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-21 22:53:09.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-07-20 16:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-07-20 16:07:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 16:07:22.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-07-22 04:22:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab0e3c0a04c1086417d6e11c756ef5d878acb5fa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-16 20:13:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"249c6d756ba680de0cd0c31c3ff9ae1d432f6027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 21:30:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa3703f3a7e43fb956589586f12bf78 : SUCCESS in 13m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/f32e385eb1a84677b9e96329347d96a5 : SUCCESS in 1h 00m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/616c9e5579e2434cb0b4e4b723e1e806 : SUCCESS in 46m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dee6a6084c4e4df1b2509397ee126e37 : SUCCESS in 9m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2f42eafc6f7b4977b57aa06ee5ae281f : SUCCESS in 11m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e50e9a1a9d724ad0be30f289970ede4e : SUCCESS in 18m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2acf800a6d904919a859268e093ce561 : SUCCESS in 23m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b7de9a3282db4a82a88f107d7767ae22 : SUCCESS in 11m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4e9336fbc2bb44b2897f289d714a6fef : SUCCESS in 14m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/12adc767a84544029ee59fb0401d78c1 : SUCCESS in 20m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/922e44bfc4224ed683a570d6cb47e393 : SUCCESS in 22m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eaeef04e925b4d00bb41dddc5b634b25 : SUCCESS in 20m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/46a6d75bea314e0d85283954ede0effc : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/06eab377fc904b4ba75bac7c76202de5 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3d7db7aa67aa4b709942d8c89675fd26 : SUCCESS in 22m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/168a9f929e8d4205a4e27e2074f21244 : SUCCESS in 21m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d938a16d7a2d47b3a52f7e02f891a58d : SUCCESS in 42m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c97a1cf4b394f76b51fd8e2e7bb8394 : SUCCESS in 40m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5377c92a54d0424daa339ab44aabf7aa : TIMED_OUT in 1h 04m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3761efb41fc4c58bd4de90d75fc8edc : SUCCESS in 15m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92203cdd0d594df8914d581dad7f8872 : SUCCESS in 7m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f3d3a91134fc4825a720332c3525cb2f : FAILURE in 8m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8de08d3b940047c2b7bc2937297ffea8 : SUCCESS in 1h 08m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4bef62343338f50abcbae9b195a8ca348f8fc993","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-16 23:00:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e0ffb6cd042ecaaf4b2111f051b1d2dbb39d7e0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 00:05:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1047379f7a3432ab21cd65b20c43b6c : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/0d60db59be77406a83a07e5a322869ab : SUCCESS in 1h 01m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/252b7e1614ae486c8acc852beff1d456 : SUCCESS in 45m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f1e75e9872fb4fda83a69ef95f18f353 : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c9e396b90f244e10ad7cc2c373a55bad : SUCCESS in 10m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/677d927d00884f7f8be233b9376fd33f : SUCCESS in 12m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fce58d93392b474fbbd02c0394a0ab15 : SUCCESS in 13m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4a5410cb58a545b2ba8fae0891b23577 : SUCCESS in 13m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b62c612ec40e4006b359e936dd654a37 : SUCCESS in 11m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/29557f4a674b40e286b0f110a92cf4d6 : SUCCESS in 18m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/67e50e6635134e9a84683348a48259f3 : SUCCESS in 16m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bb7b96ee3de74feca2c6d870324a747c : SUCCESS in 48m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/41d0f87f021748edb760c3c78ff76d2e : SUCCESS in 21m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2f4e813ab8f948cea64dfe5c78fbcbda : SUCCESS in 18m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/00bbdc43d0b6496e800c78987a288121 : SUCCESS in 18m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ddd34637b0f942eeb76aab9927c0b499 : SUCCESS in 20m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eeeb6abeb10046c79166c13d558ada75 : SUCCESS in 41m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/358841cc119248d0bf04f366c6a1b7ee : SUCCESS in 43m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/64701993220742f9a75ff2367f67d259 : SUCCESS in 48m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5136844c4cb64abf8f5e92477a49a13d : SUCCESS in 18m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c22d266ebb19410fb1529ab9162284dd : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd8ecbb5e59a4c118ade2817077d457c : FAILURE in 9m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b8ed8a1ab6e4f94a52b4f3a7e6fdc1c : SUCCESS in 58m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1312e2616fd44df29dd45d60191e91d70905186b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-20 16:07:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(4 comments)\n\ni think this is ok, and look at all those closes-bugs!!!\n\ni wish i understood the motivation for the bug reports... does an empty suffix sync *break* something?  Is the race somehow more common that we expect it to be?  Maybe we\u0027ll get some more info after this fix gets out there...","accounts_in_message":[],"_revision_number":1},{"id":"adcdc56928c64c264ae9eabb058012436555d6e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 16:08:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5cf67bc4d3393d0b744ce4ecf7cff73df69961dc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-20 18:56:07.000000000","message":"Patch Set 1:\n\nI don\u0027t think it *breaks* -- next round of replication should come up as hashmatch -- but it caused us to report failures when nothing was wrong, which makes it hard to tell how healthy your cluster is.","accounts_in_message":[],"_revision_number":1},{"id":"1dd9202050b4edc1ebff9dc5287914118bae66ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 19:00:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c65fca0350b64a90b7d2d9df9ccb32ca : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/9fc88b97a5e444fa97c48bd5d2fb642a : SUCCESS in 52m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b647217049a848b38bc8268f3673500c : SUCCESS in 39m 56s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/92d9495b72804f0d90ae482e80791ad4 : SUCCESS in 8m 35s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/6ef08e37af6c430cb88db2122cab324c : SUCCESS in 11m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c31f1e7d9a4b4bbd8925686b69eddbca : SUCCESS in 12m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/250b6ac4b4ef42e190dd4c4865357bf4 : SUCCESS in 21m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fc126fb8f52b4b3c956e0689f8bf607c : SUCCESS in 13m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a31eab2a80344b779ebbb8223328874a : SUCCESS in 13m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6fda1c37d50b430091cede86c9d604de : SUCCESS in 22m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/832634f972e04dddb32e2119badcf42c : SUCCESS in 20m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bbcf6a0be67d44058ce9d825669b6d91 : SUCCESS in 18m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/78031c8f4de047c181cf87d962d625fb : FAILURE in 16m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8458bcbb27e342dca2f4b7a5633c486c : FAILURE in 19m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ffbf413f544f47aaab174ee2e3b83576 : RETRY_LIMIT in 5m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/925902503cb84e579517ceb0e478a83e : SUCCESS in 45m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c944353b2134656829760291917eddb : SUCCESS in 57m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d85e90b4dce240589b26fb23feacf418 : SUCCESS in 21m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfba24a3baea4c4d80eb07e28820e0f8 : SUCCESS in 6m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77f6d71662b84bb18b4e387d10c44878 : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b52c82592c4e7b0c26d5ec9d8b4e0f0860ce415a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-21 22:53:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b26f82f35da8f42d03e3b9685747add551a24734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 03:02:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6470e96179004986937770a50f6680a5 : SUCCESS in 10m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/392e05ce995a4346bbb5b0176209acf8 : SUCCESS in 54m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3321a29510134018b6dfea3cf5625977 : SUCCESS in 42m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/db21393733d047c6aefa035835d35ef4 : SUCCESS in 10m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dbf7d9d84a16465ca6108bcbf4d53ec8 : SUCCESS in 9m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c6a19badc633409b931131b6a1607a28 : SUCCESS in 15m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a944296875ae47b8adf1fdbd42d5741c : SUCCESS in 14m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9bd999269b5448beb6527344a5dc06bb : SUCCESS in 22m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/59b6336a845a471599c8d195cbabee75 : SUCCESS in 26m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3533ace24bf34768a97336d6bd98522d : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6f38f5e01b9046aaa89c0ff53a92e60d : SUCCESS in 18m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/25e8f4db8f6b465cbe89fb37f6da2458 : SUCCESS in 19m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3ea772eedef64296ad00414c3a13da33 : SUCCESS in 18m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/541c32986fcd479a819cddb17c96ded6 : SUCCESS in 19m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4ff6ffc022a24ee6b88dfd5b24812bc3 : SUCCESS in 21m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/16a7014c25134deab4ab0231b1f873b4 : SUCCESS in 20m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fcb5b08940a546d8b6276dba9a93bced : SUCCESS in 44m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08a4cdfd51134212b46c3c0b79a82888 : SUCCESS in 45m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1add7155e7eb4313be0be8b10c5452e5 : SUCCESS in 42m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e396325fbe2246a3982610ce2c8ffc8c : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a22d52f0b33488eb32343330a4ae7a6 : SUCCESS in 6m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/183fdf63e0934bb6a32dfdc19372ecb3 : FAILURE in 9m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a8461035eae455c9837497061c141f3 : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":1},{"id":"bbc9eab5b227b6920a38b60a0f26905c89bac86a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 03:02:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aeca4b538f2ea21d764ad96346645f459bcbc77a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:22:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bebdade2d55312b542f8a1ef29d625cee9b71422","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:22:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7883e9cc192446b195de624efd386d1e : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/22ed79cc512c47a885ffb49804b3f56f : SUCCESS in 51m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3d9063bd2dd4a3b88a5a02bf57dd6e8 : SUCCESS in 36m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/60684bc77f3e44a7b2dcfa789d2fa11b : SUCCESS in 10m 21s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a00a6c594e334cef8778e964ce07f2c0 : SUCCESS in 10m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/301e360f8bd849b6bb191a51897dcfb0 : SUCCESS in 12m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/251fbdc5c4ec4b95b1ac9faac4abe8ea : SUCCESS in 21m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f0c52a52db404b18a7e6f841158248d5 : SUCCESS in 29m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5fff629d4a884350a6dcdeb3109041ef : SUCCESS in 27m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/836b53d29a8f4b7b9c81504518892290 : SUCCESS in 26m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/52377687660b4cfba8ac01219f9c82a9 : SUCCESS in 29m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/38e65603280d47b7822b8a6aa94cc205 : SUCCESS in 17m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/438dc9e85cbe4bc88daf4e2acea64302 : SUCCESS in 28m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b9b0a63bb537421085c378631c2a8205 : SUCCESS in 20m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9a083c6bc4fd4364acae8e4822007e66 : SUCCESS in 40m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/47cca4d3cbd741b3adf468a9c712e256 : SUCCESS in 43m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ac572265f2df489aba6e19aa652d5201 : SUCCESS in 48m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14717fd871ee445fb02c78b96cf9acd7 : SUCCESS in 14m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d32da4c3bf714014a0fea918b76c1e39 : SUCCESS in 6m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90ba173010e24e0084cb2e3aa22a4968 : SUCCESS in 1h 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ce58043b08f8a854b4a419c3fc95a93956e63ba7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:24:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63fab72af8d54a28a37086408ac0f71a : SUCCESS in 1m 18s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f235abe8de39447ab947775b9be76acd : FAILURE in 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/82363c0209d54948b2376ff50388187a : FAILURE in 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5255b2981168dcc1c243c4b805b0d964639078b1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-27 16:46:07.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 0bdbbc4a655fb4f74cd087636fc136c0f9d57754","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"907942eb4743e5a4213b1f84fd57a58386594bb6","revisions":{"907942eb4743e5a4213b1f84fd57a58386594bb6":{"kind":"REWORK","_number":1,"created":"2020-07-16 20:13:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/38/741538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/741538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/741538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/741538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/741538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/741538/1"}}},"commit":{"parents":[{"commit":"2e001431fd0907c97d44b81dd1f3d2d14ae1208b","subject":"s3api: Don\u0027t do naive HEAD request for auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e001431fd0907c97d44b81dd1f3d2d14ae1208b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-16 20:04:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-16 20:13:04.000000000","tz":-420},"subject":"Stop syncing empty suffixes list","message":"Stop syncing empty suffixes list\n\nChange-Id: I918ab4ccbf4d081b26f4117937410cdad1caf8d3\nCloses-Bug: #1862645\nCloses-Bug: #1886782\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/907942eb4743e5a4213b1f84fd57a58386594bb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/907942eb4743e5a4213b1f84fd57a58386594bb6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
