)]}'
{"id":"openstack%2Fswift~741559","triplet_id":"openstack%2Fswift~master~I13de27674c81977c2470d43bbb2126ecc4bdd85a","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I13de27674c81977c2470d43bbb2126ecc4bdd85a","subject":"docs: Improve replication-network remakerings","status":"MERGED","created":"2020-07-16 23:38:59.000000000","updated":"2020-07-20 18:57:19.000000000","submitted":"2020-07-20 18:54:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741559-1595271275851-56289871","meta_rev_id":"fc749d15d1daf1008eeae28c8fa36118aebe144a","_number":741559,"virtual_id_number":741559,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-20 18:54:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-07-20 15:55:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-07-20 15:55:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 15:55:47.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-07-20 18:54:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f505af471bcba63d6d8f3d13a1d57224f6783a1e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-16 23:38:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"032946955be90824c1af4a711202d038c14a7e54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 23:52:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6328952d55454ea08561583ece730027 : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6933be4f4ccccdbd44e050c4c859dcbd6fbf4b91","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-20 15:55:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nawesome","accounts_in_message":[],"_revision_number":1},{"id":"454ad9bf5b5eaced1dfb9e78088d8a9d2f87b721","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 15:56:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03c7e061dd80d79497832ff5cc05a77df99694bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 18:54:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaf40bba15a242648ee9ab63ae936f8c : SUCCESS in 8m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/afb82fe6b7734806837c06f0d8e72007 : SUCCESS in 10m 17s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/9292cc64ef164470839036e0568d14f5 : SUCCESS in 10m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52b1942201671bbced41c92b711ac8743c539aa6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 18:54:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fc749d15d1daf1008eeae28c8fa36118aebe144a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 18:57:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2654648cdfd44f8994fc4a0647bc7a2 : SUCCESS in 1m 29s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b537073f53c04bcb92baf8ce80570bc7 : SUCCESS in 2m 14s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/32f1120c732e4c8592063602938738bf : SUCCESS in 58s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c31e30ec2f1abd7b21d2e94b3bcf21d9f84888b0","revisions":{"c31e30ec2f1abd7b21d2e94b3bcf21d9f84888b0":{"kind":"REWORK","_number":1,"created":"2020-07-16 23:38:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/741559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/741559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/741559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/741559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/741559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/741559/1"}}},"commit":{"parents":[{"commit":"2e001431fd0907c97d44b81dd1f3d2d14ae1208b","subject":"s3api: Don\u0027t do naive HEAD request for auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e001431fd0907c97d44b81dd1f3d2d14ae1208b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-16 23:22:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-16 23:34:32.000000000","tz":-420},"subject":"docs: Improve replication-network remakerings","message":"docs: Improve replication-network remakerings\n\nLower the part-power -- 18 is way higher than is needed for a dev\nenvironment.\n\nAdd commands for reduced-redundancy and erasure-coded storage policies.\n\nRelated-Change: Ibe46011d8e6a6482d39b3a20ac9c091d9fbc6ef7\nRelated-Change: I6f11f7a1bdaa6f3defb3baa56a820050e5f727f1\nRelated-Change: I0403016a4bb7dad9535891632753b0e5e9d402eb\nChange-Id: I13de27674c81977c2470d43bbb2126ecc4bdd85a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c31e30ec2f1abd7b21d2e94b3bcf21d9f84888b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c31e30ec2f1abd7b21d2e94b3bcf21d9f84888b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
