)]}'
{"id":"openstack%2Fswift~741746","triplet_id":"openstack%2Fswift~master~If5dd79f926fa51a58b3a732b212b484a7e9f00db","project":"openstack/swift","branch":"master","topic":"saio-replication-net","hashtags":[],"change_id":"If5dd79f926fa51a58b3a732b212b484a7e9f00db","subject":"Update SAIO \u0026 docker image to use 62xx ports","status":"MERGED","created":"2020-07-17 22:12:47.000000000","updated":"2020-08-01 02:40:10.000000000","submitted":"2020-08-01 02:39:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741746-1596249540390-b464fa81","meta_rev_id":"36af9f820e4c759164f7a67a8097db007d20ae4d","_number":741746,"virtual_id_number":741746,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-01 02:39:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-01 00:31:48.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2020-07-31 02:11:40.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2020-07-31 16:21:11.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 15:46:28.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-07-31 16:21:11.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2020-08-01 02:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5d0260c6b3a9e2493f608bfb6ed36456eee3212","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-17 22:12:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2585cdd0cfc250838ffae1ccd3b28443b7f729d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 02:31:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01bf3145bb974bd09279738c743fd7a8 : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/3c70735037b5448a8e0909a4c3e92db5 : SUCCESS in 56m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20a1c82e98ec484ab7767777cc2146f5 : SUCCESS in 44m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/86301bf321c043dfbf09aaf771b3289f : SUCCESS in 8m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0ef0d0a0a99e4e9ebe285c51d2196fa9 : SUCCESS in 10m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/51258b9f52a442aebc88cfcbdb9563a0 : SUCCESS in 15m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/73d08ea199b64513b9627441d15a13ce : SUCCESS in 13m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ee01ea2891424cf6b611539cdeac1904 : SUCCESS in 14m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/44dba93acff242d680b5e3887f52dd92 : SUCCESS in 14m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e5614df3ffd64074a4407337c25cd68e : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9fa25a39cef54be6bfcd6fac0d3e1d07 : SUCCESS in 21m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/81a498fa53804180847e4cfdd15ee510 : SUCCESS in 25m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/98165d18f35342b2888e5b16504ca545 : SUCCESS in 25m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6ace8f921145405ab4ed67c912e80014 : SUCCESS in 25m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/aa9ea4686d364f0db4502db35db6547d : SUCCESS in 22m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1f34d056dcac4bae82cfd8f59134a53b : RETRY_LIMIT in 4m 04s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d350af2b0b0f4469aec45c8cd236a09e : RETRY_LIMIT in 4m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/18041c29d4df4c7d9144dcae1fc25807 : SUCCESS in 44m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/71245615a12345c2a85dc5e55ab3622d : SUCCESS in 42m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef1bfe00ab0d47f09f88d05092f1adb1 : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/511c33e8b61a489cb837898b23cb44f6 : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0f58790b2ab0499788fe4fa830acfd35 : RETRY_LIMIT in 7m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e985ae46408447febc98acb8344d4da3 : SUCCESS in 1h 10m 17s","accounts_in_message":[],"_revision_number":1},{"id":"623dc08422407d1acd5d796a2592eb0e66b57bcb","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-20 15:46:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nseems reasonable to me","accounts_in_message":[],"_revision_number":1},{"id":"9a01707703acc4e80658edf834634106a24e7166","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-20 18:49:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e154ddb74609554dc33fce3b7ac03279b4b39026","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-20 22:17:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f52899b884e1ab3eddc5a7e7715cfe91a573ed43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 23:38:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e20036621147c1920a21ec91a6f843 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/f4a2ff41bf6f4fcb98053fd81dce3646 : SUCCESS in 41m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cfa77e645dfe44759340f1997f1e1771 : SUCCESS in 46m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1033284b690a4834aab4985c13d165e8 : SUCCESS in 6m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ce00ba39e572473ba68c8a3572e5bbf5 : SUCCESS in 32m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6e851dd1c7a344de94bee6671b1d208c : SUCCESS in 12m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/85e46dcb45fb452b9c3e0e0b2fc921bf : SUCCESS in 10m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/936451c5f72f47a19802c49765acc0fa : SUCCESS in 16m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0accc256690a4e1fa75757cb55a1ac1e : SUCCESS in 16m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3eed634cc17344ec86dd5cdb6abf3fc3 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7fc2a8d2e5e2483a99bf6d32a9abd960 : SUCCESS in 23m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8268ff61a5e14845ad1c3178d393e3a2 : SUCCESS in 14m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e78975b3ede746e6b818f55aea59fb52 : FAILURE in 18m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/206f983f2e734608bbc4809b01271fcb : FAILURE in 46m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/08d64a1ba17d47f6a03005cb0d34b250 : FAILURE in 13m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/45ad65f73b574f4d93be42456a91d49b : RETRY_LIMIT in 3m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f3464b74788a44528dac35efe3fa8926 : RETRY_LIMIT in 8m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e5d88e00d19545aa87a5cc6086b2893d : SUCCESS in 43m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de367a0a692f40148a796a0e50255177 : SUCCESS in 47m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b18cb89baa8440ee9b830891fcf791c5 : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81edec33ef174c3c9cd0cd2c0e77260e : FAILURE in 9m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/287da9b7fca749899e9a0e768b86e7c0 : RETRY_LIMIT in 6m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64ca8c1058ec463280b80d600b96f583 : SUCCESS in 44m 09s","accounts_in_message":[],"_revision_number":2},{"id":"56fee271581426b84e9f766be216c9573a8bae5d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-22 03:39:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"179c5de7da32b24825f421b1560e986c4fae6a24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:48:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f065602b5c94cb499e4027c9f3f0c0b : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/31f4b9376c5e435f9624e87deb4c345a : SUCCESS in 50m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/138670a36cb74166aa3f5f677e364395 : SUCCESS in 36m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/da5bdbbebfd14272b635d50079864f4e : SUCCESS in 8m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cef25ffaac044ae7bd4f76742eb10544 : SUCCESS in 10m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d5229b22878e441d945e00d922263706 : SUCCESS in 16m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6d6cee061af24745830b464d6c878272 : SUCCESS in 12m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/178875679999451898112a8602e8785b : SUCCESS in 10m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/85420c722c3248e88fe9b06f54263f31 : SUCCESS in 15m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2d09369b4e7243e58cca5da634de8509 : SUCCESS in 24m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e32a2d69e4c04be4904dc5cafdcbd1df : SUCCESS in 16m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6316e43838f34a1abd98742bf2912900 : SUCCESS in 18m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2474799cc8a14d7a857b04c99d1e0605 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7e2f814496b34f54abedb3e291511930 : SUCCESS in 15m 21s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/06eb49bde084460f8d193a05bc49a6b0 : SUCCESS in 16m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f89ee4ee7bb44180b97a73048f63bcbd : SUCCESS in 28m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2efb0c7c0ecc45d99aada86828bb96ac : SUCCESS in 38m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a82dcee071f64e648c41019edc19bc0f : SUCCESS in 43m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/390cf21192e54d55afb1c732cad7f2ef : SUCCESS in 43m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dc658c8c6154ca5a3ed53b00e9e6386 : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bde7085abac4067bef5be6e07138654 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e3f6bf9bbd7c4a94aa459f1efe63ff4b : FAILURE in 8m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c5069019e414582af04674b34485c40 : SUCCESS in 52m 57s","accounts_in_message":[],"_revision_number":2},{"id":"5629abb7b9a405e633467d1f3225a22781179367","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-07-31 02:11:40.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a3da87f02711a4886bce32dc6c983db82051db4f","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-07-31 02:13:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d71f8e36273f015ccbb2fa086f9c990a5730bee2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-31 06:01:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"14dddebeec7595f54aa29c8f93090448ad8a4535","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2020-07-31 16:21:11.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bdbae44e068f6f73b481829e79a6d6937e4ef86c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 16:21:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f554a5136fc6ce98e960a42dc6efe2be7aa01ffc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 17:43:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdf913e351e4435e867259f0a95c23db : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/ae9181f9caab4226a26c563a0d139dd5 : SUCCESS in 1h 08m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d0050292ce4496f962749086b891911 : SUCCESS in 43m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/61bace582d304ddeb1e74e5485a13fb5 : SUCCESS in 10m 07s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/22f93ee287dd4b6f9a3f9d76d26ef1e8 : SUCCESS in 8m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7f3987e52762436f9f3134fee5f43bed : SUCCESS in 22m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2adf086097934447a04222c372e0a8c9 : SUCCESS in 17m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2f0841d4d5eb4571857c5301a0034014 : SUCCESS in 13m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6ab4b94a10024358b1b74577302a126f : SUCCESS in 13m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b97def57dc094c158e7aa70bdc5a4e3e : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5941bb03453641ecb02317b211c293d0 : SUCCESS in 29m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/225d17ae1f4249c8991ec3e93dbebc46 : SUCCESS in 21m 13s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c780cbf4d9d44873a6693f0e7789f193 : SUCCESS in 20m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f8a1be350f524fee8a1213ef081457b7 : SUCCESS in 17m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4137eef0964642aa80bb6090e1bba7c6 : FAILURE in 36m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e307ac28746a4cea8aa1a3169365c053 : SUCCESS in 44m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6a43da1e8e044380881c9849a9102c7a : SUCCESS in 53m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0f354f7e80549838ab07682c459d0a2 : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/807dc443dd7848889c25390574f741a9 : SUCCESS in 6m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6bfefc517a4464589a92195234d8212 : SUCCESS in 1h 11m 27s","accounts_in_message":[],"_revision_number":2},{"id":"86a04925bc5ddc133d6177b41d80466a510581db","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-31 17:59:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fe11cda5caa784a2731a8bea257143f8ae6c7be1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 00:00:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a608f28a6a65480f93cba2ffea3f14e1 : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/5d0cc1469ffa4421b0323423f933b80d : SUCCESS in 1h 19m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ec7a80515b2d4bc78525fc2a5c21a5e6 : SUCCESS in 49m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/27547ae6d0eb49c3929483db0555a728 : SUCCESS in 9m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/abc3bf1a33ab4590aefacaa8022e8139 : SUCCESS in 9m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aab581b44ab34b949043fddcc3ad76ec : SUCCESS in 13m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e6da2e5c32184237af0f69c033a4f133 : SUCCESS in 12m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d3db15af0a2d4258903f456a085c2afa : SUCCESS in 10m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6858d4ee63ba49ffac7310cc7072e56c : SUCCESS in 13m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ab6b0798bbd34c52a47aa3423c5c15e1 : SUCCESS in 31m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0ab869b5f9434569a14ce6afceddd70e : SUCCESS in 17m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d57d320b609142f6a077f867afbdac5c : SUCCESS in 18m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/73c9382a57694aaa963385d88aa7eeb5 : SUCCESS in 22m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/da4187a563e24f1e8dd797c2a71e857d : SUCCESS in 17m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6c2fbfea839c412ab25bb409343255ce : SUCCESS in 31m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f6b08b84a4494b26a7e950d011560379 : SUCCESS in 20m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c2ffde8dd12d4e989b3d4e218e8823ff : FAILURE in 42m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6033d433e65f46afa0eddb6afd33b68c : SUCCESS in 41m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9a53d46f27c540f580c7c33000641e71 : SUCCESS in 59m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ddf6015543541ce9da7ef22d15f2535 : SUCCESS in 12m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f441786c582431c804593d3b3c60ae5 : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76bb313b583c470992740a1ac017b6d5 : FAILURE in 7m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9570cf71f873419a84f01ff8752cbe8d : SUCCESS in 1h 08m 19s","accounts_in_message":[],"_revision_number":2},{"id":"d979b191107be8656041ad9cfb018b3d0f6fb374","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-08-01 00:31:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8a941fd1e429624e0b4c4904a1d70f1f373d2bb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 01:34:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb9677277b840f38a131e076772fd48 : SUCCESS in 7m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/34c039d3a3064f2a80ca30978dfde7df : SUCCESS in 59m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0b003547237348af83cef9c0d951896c : SUCCESS in 41m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7107753d9f8e4c0890a38c01f86dc9df : SUCCESS in 10m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2fdef3b261c048bb81428a0713601133 : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/408cced4702440a99fe6939ae154b22e : SUCCESS in 23m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aa0e45237c2947c38d5f9e6805a9f1f4 : SUCCESS in 14m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/44e47cf5b8204ffd870f8349b4d969f9 : SUCCESS in 12m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/462a173dd1d84c989b7a13391a7358ab : SUCCESS in 20m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b2b2819c27f84207b883661353528492 : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e44fc0b9946045a4830dea4de12040ed : SUCCESS in 29m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dc555846d4ae4e12a0b97c546420b715 : SUCCESS in 29m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a673acf35f6f48978cf6c95f93d934a7 : SUCCESS in 26m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8ed0948dfa17481d91691127242e40a1 : SUCCESS in 18m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b1323786d8bb4a2eb8ed3a04e8ca6add : SUCCESS in 18m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/867c25f57e994c4a99772330a8f59b87 : SUCCESS in 27m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/419fb012006641289e99174632baf9b4 : SUCCESS in 45m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5aa0cf063ff94ee5be65a123f1fc1698 : SUCCESS in 31m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ed146a9f9a2940a59ce4e427593a14db : SUCCESS in 49m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11eba3fcd47c465094286d3f9ce2406a : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd7961691224b3daa86b58efe07525d : SUCCESS in 6m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6bc8de52d2594862b45dc7ff72bca5cf : FAILURE in 7m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7991b65709c446fe8269e339b5a3ee14 : SUCCESS in 52m 26s","accounts_in_message":[],"_revision_number":2},{"id":"c160520c49a0ac0d03d59c47512a3d960796607f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 01:35:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6cbfbd269d33e4b832dc6349cf24afcd50d488bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 02:39:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2745d2fe6c9f16d465ecab3da5679cdf4e674356","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 02:39:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27893219fe9c402b98683a36a65b2923 : SUCCESS in 9m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/c2624e01588442ecaa62a0c7b81362fb : SUCCESS in 58m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5a2a35546ba9437895dc9452c02f435a : SUCCESS in 32m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/fa2552e2847b4d8b8adcf1887354418d : SUCCESS in 14m 51s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/4111c93764ae4784821417f85e6edb98 : SUCCESS in 9m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b510e5f7b7f0400881e99e57975a026c : SUCCESS in 10m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/573393f51c2a42d88978b08be37ca83c : SUCCESS in 17m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bcc86201a34d43b7983c9a71ce9b82f9 : SUCCESS in 11m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7658eca34e23436283dd69a7179111fe : SUCCESS in 12m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/501d601dc7114ab8a2add40fb3bc8e39 : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5cf2ec3b076b4637942c95d6817f3bc6 : SUCCESS in 30m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b7cfbdd9136f4a9dbc665253df9b22d4 : SUCCESS in 13m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8471dd265cd64f48931daba43e647cce : SUCCESS in 17m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8e972b1e885e46b69709baf8f47f2ee5 : SUCCESS in 22m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/12faf29dec184288a86f349778b349e0 : SUCCESS in 50m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee3e7b399d704e17972cc394600acb92 : SUCCESS in 32m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/89d4ea55bebf4e6294fb6ed784e4f873 : SUCCESS in 35m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c99bcecae5e348b1add2c1328e4842c9 : SUCCESS in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a811fe4496b4b1cb6ed1d1af1b5f8c9 : SUCCESS in 5m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3248095d77a0437da80ce7ce3d8b3fa6 : SUCCESS in 54m 12s","accounts_in_message":[],"_revision_number":2},{"id":"cccae3a79a8d48305148c06d681e792158328b64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 02:40:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920bd9312e0c443c8f6c55c367362c43 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1476a30d013049f28179011488310854 : SUCCESS in 37s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f6074933d50348509bd779889bad43f2 : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"314347a3cb3770523b6324f4c4d57f63d89c5936","revisions":{"0dc523c75d15c084df0e552f214c08a5a8c186ce":{"kind":"REWORK","_number":1,"created":"2020-07-17 22:12:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/46/741746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/741746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/741746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/741746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/741746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/741746/1"}}},"commit":{"parents":[{"commit":"2e001431fd0907c97d44b81dd1f3d2d14ae1208b","subject":"s3api: Don\u0027t do naive HEAD request for auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e001431fd0907c97d44b81dd1f3d2d14ae1208b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-17 21:59:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-17 22:12:35.000000000","tz":-420},"subject":"Update SAIO \u0026 docker image to use 62xx ports","message":"Update SAIO \u0026 docker image to use 62xx ports\n\nNote that existing SAIOs with 60xx ports should still work fine.\n\nChange-Id: If5dd79f926fa51a58b3a732b212b484a7e9f00db\nRelated-Change: Ie1c778b159792c8e259e2a54cb86051686ac9d18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dc523c75d15c084df0e552f214c08a5a8c186ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dc523c75d15c084df0e552f214c08a5a8c186ce"}]},"branch":"refs/heads/master"},"314347a3cb3770523b6324f4c4d57f63d89c5936":{"kind":"REWORK","_number":2,"created":"2020-07-20 22:17:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/46/741746/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/46/741746/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/741746/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/741746/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/46/741746/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/46/741746/2"}}},"commit":{"parents":[{"commit":"0eee3b5bcd6bfbfacc6bc354e84c4b52973a1c03","subject":"Merge \"docs: Improve replication-network remakerings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0eee3b5bcd6bfbfacc6bc354e84c4b52973a1c03"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-17 21:59:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-20 22:17:12.000000000","tz":-420},"subject":"Update SAIO \u0026 docker image to use 62xx ports","message":"Update SAIO \u0026 docker image to use 62xx ports\n\nNote that existing SAIOs with 60xx ports should still work fine.\n\nChange-Id: If5dd79f926fa51a58b3a732b212b484a7e9f00db\nRelated-Change: Ie1c778b159792c8e259e2a54cb86051686ac9d18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/314347a3cb3770523b6324f4c4d57f63d89c5936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/314347a3cb3770523b6324f4c4d57f63d89c5936"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
