)]}'
{"id":"openstack%2Fswift~741755","triplet_id":"openstack%2Fswift~stable%2Ftrain~I3ccd145c14a9b68ff8a9da61f79034549c9bc127","project":"openstack/swift","branch":"stable/train","topic":"batched-reclaim-stable/ussuri-stable/train","hashtags":[],"change_id":"I3ccd145c14a9b68ff8a9da61f79034549c9bc127","subject":"Breakup reclaim into batches","status":"MERGED","created":"2020-07-17 22:17:33.000000000","updated":"2020-07-22 20:42:44.000000000","submitted":"2020-07-22 20:40:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741755-1595450405506-a26e969d","meta_rev_id":"7835892241149fb41d5b54e2a81a72b2da0a9932","_number":741755,"virtual_id_number":741755,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-22 20:40:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-22 19:16:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"date":"2020-07-17 22:17:33.000000000","_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-22 19:16:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-17 22:17:33.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-07-22 20:40:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9b4cddfe3add3205d30cbbe7a3c8a96c120dec5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-17 22:17:33.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"8ba726b13699d30f83072beb68042bfd2d3c93bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 02:43:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a88c1abe36fa4925b1766f536357b8ae : SUCCESS in 6m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/874a736aacff4797aeb8757b2bb26d83 : SUCCESS in 1h 00m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6ab1c507061f4a229b428bc70adfa12a : SUCCESS in 48m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cfcc7950dc5e4702881606c3d2e2dd56 : SUCCESS in 9m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6e0a1eab8cbc4b628302f8703bd511d7 : SUCCESS in 12m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f3a2bda6ff61418e9872a422e2e830f4 : SUCCESS in 13m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5959de68871142529797521b0fdb2fd3 : SUCCESS in 12m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/83d1df90c2d54467a7df8ac20a24a4a7 : SUCCESS in 11m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4c94dc2e8bc74aa8a52e89f1b4bb0197 : SUCCESS in 13m 30s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/8213c95da02a495284240ba8236c8149 : SUCCESS in 12m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/77a2bf79fd894d12b226f2c7c94d61fc : SUCCESS in 12m 32s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/114556ccf980434199259b9682de334c : SUCCESS in 12m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cb57dded53ff477fbbb34b29ce5355c2 : SUCCESS in 11m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a50fbd330d244d5da8e400b172bb24f6 : SUCCESS in 22m 49s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/113a88c7a664437f8dc88f53a51764f8 : SUCCESS in 17m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/242d425045b4428d8ecdfe863a2369bf : SUCCESS in 11m 39s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/41e9ad91d4224cfbb44e69cca28dbcaa : SUCCESS in 15m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9cdd1a101ea84b459f11003f54da4ce6 : RETRY_LIMIT in 4m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c2abe2665cd649fd9c5b220705ba6ca2 : RETRY_LIMIT in 4m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62e30764b0104f7aa40723e4bf958e92 : SUCCESS in 34m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/63c78c5c7b644dc1a06b29857caa0216 : SUCCESS in 39m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8aa26390fa1948feb1f5a18a32816068 : SUCCESS in 31m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57f0786332164799901dc51878203d80 : SUCCESS in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b96c2859d334088916b0117a5e0b700 : SUCCESS in 6m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/be556b08c7654eaf85f885090278e70e : RETRY_LIMIT in 7m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36b7252bdf674da191275b4455c74bed : SUCCESS in 53m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4d498ef1cbb94111a299776bee51b4e2 : SUCCESS in 1h 00m 02s","accounts_in_message":[],"_revision_number":1},{"id":"1bfdf403ff701a1384abf921f39b33ba1d12cf9e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-22 15:09:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c3141f908dff2a1522320e80bde4554dea356363","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 19:12:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc842a2125064e11b1a45ece29f2664c : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/dbff784f77e8494198ba61c085c19923 : SUCCESS in 59m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/18772c90298d484bbfd4c470001af6be : SUCCESS in 40m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/26aec852da4c4ef5a3e169cca3309c85 : SUCCESS in 12m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cbeef7bd1431400a923f6228143c99fb : SUCCESS in 13m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/93d224d4833641008372eb38c734b9b7 : SUCCESS in 18m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c4523a2b51a848ac9805e7fc714ac1e0 : SUCCESS in 13m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/06fd1ce85c604fb2bbf4c2356e873dce : SUCCESS in 11m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/076a6f51a301410395e5cd19945c5b55 : SUCCESS in 12m 05s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/bfa2246fb07c46ddbd3b83f7c9149157 : SUCCESS in 10m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/619dcf1990db40a5b495080e8b5e3435 : SUCCESS in 12m 11s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/23064d84657c485fa5ffb441a9fada5d : SUCCESS in 13m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cc0a5425351c4e959ef858ba7105a097 : SUCCESS in 10m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/955aa1a6efc148388965e9b2abf392b2 : SUCCESS in 11m 18s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/17242914b6ff4e1090ed5fb0d0e93d9b : SUCCESS in 14m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/29ccb5562f7d4dcbb5e44dc3fa28ef4d : SUCCESS in 12m 42s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b1561149ec874a2da380f2e35e3505f2 : SUCCESS in 13m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd0d5bf900584280a7c50447d18c4488 : SUCCESS in 18m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eef43eab3ea643639bd2072854ba552a : SUCCESS in 32m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2fc52c345ac45bebe7cba9895fe04d5 : SUCCESS in 39m 27s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1dc4be0974ce4e32ab5fc57443660947 : SUCCESS in 43m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5d58c429804d4d3cb46c422ea2782962 : SUCCESS in 42m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/636e61a6743c4224a7abac55162410f2 : SUCCESS in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb6ebe7257064c8d9845252883726e3b : SUCCESS in 6m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10fae2b4a11f488092fc035eacb03508 : SUCCESS in 25m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb7001bb8d514e97add6e4296b007727 : SUCCESS in 58m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bd2dce58643e4d54bbb9149cf1041128 : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b7f47ce8d660dd62d665c09c836b00da053257ef","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-22 19:16:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bea68ef633ee11f0861cb3e852572e07b49c4ad8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 19:16:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83fda09c5df3a41cfae6731538f8ab1716236539","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:40:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e5365cac6ad955029204500fe8057805ec4cd5bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:40:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9592068933e140938f8da7709c747580 : SUCCESS in 6m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/ac6aa41b95f7411ea0ce1200c826673a : SUCCESS in 1h 15m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/726df2cec019426da499c3b452022245 : SUCCESS in 50m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4d8e7c07affb4b5895a4892e138fdef2 : SUCCESS in 10m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/19e3648b0e104820830f8014d55e490a : SUCCESS in 14m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a63513227c7146f2833a32f0c6b52aa8 : SUCCESS in 15m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/adb04acd07a54ad88e47a85500623e67 : SUCCESS in 19m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/254792bb5473412e913b99b8fef96ec4 : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/44c4cb9d260e481587e03260bfc1a16c : SUCCESS in 16m 49s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/85247ebdc564467985a5e211598227d5 : SUCCESS in 15m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/17a59d7d35ae44e0a8b6c43bb299e4c0 : SUCCESS in 13m 03s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/727fb72ebefd4fdf9aaafcca25d269ab : SUCCESS in 13m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a90ca99d70244ff4bedd8b80f7c7f465 : SUCCESS in 11m 02s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/fd70bdaccb314e9d8096241e7273a535 : SUCCESS in 14m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a7ab2c9778804df0b5a9a68f985d29b9 : SUCCESS in 14m 36s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8bd9092bedbf4b19ac7c954d0d1e69a8 : SUCCESS in 17m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/71e29e34bd464237b254e7f89166e7df : SUCCESS in 35m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c8fb578d792c41438d78a7c1196cb077 : SUCCESS in 28m 40s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ad34b89e7915440cb3dd2d5caece6805 : SUCCESS in 28m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e77ff8b58aa437793953f02ca12b462 : SUCCESS in 35m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa4f9c8836ca41cbae4badfbfe812daf : SUCCESS in 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c42e412b06149deb6b10272b80d47d3 : SUCCESS in 6m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11096c58e2c241c0846373d1296aa61b : SUCCESS in 56m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/115835f89c4f4d61a68eb8c90486ec17 : SUCCESS in 1h 05m 07s","accounts_in_message":[],"_revision_number":1},{"id":"dd55a46382dc9af542c6644e00bb87e211394603","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:42:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/817aa531a8b649cfafab629c25a0a467 : SUCCESS in 1m 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/fc62c43cced44d4a84e9b97dba91d7be : SUCCESS in 1m 31s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"170dddaacc609493d4b3e7ba1f6a9168e7633216","revisions":{"170dddaacc609493d4b3e7ba1f6a9168e7633216":{"kind":"REWORK","_number":1,"created":"2020-07-17 22:17:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/741755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/741755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/741755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/741755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/741755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/741755/1"}}},"commit":{"parents":[{"commit":"635c71c59e2add58f5cb2ac63aabb7c9228fdf12","subject":"Merge \"py3: (Better) fix percentages in configs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/635c71c59e2add58f5cb2ac63aabb7c9228fdf12"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-05-13 18:32:18.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2020-07-17 22:17:33.000000000","tz":0},"subject":"Breakup reclaim into batches","message":"Breakup reclaim into batches\n\nWe want to do the table scan without locking and group the locking\ndeletes into small indexed operations to minimize the impact of\nbackground processes calling reclaim each cycle.\n\nChange-Id: I3ccd145c14a9b68ff8a9da61f79034549c9bc127\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCloses-Bug: #1877651\n(cherry picked from commit aab45880f8aaddfed15fe641939aac0d2eccc465)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/170dddaacc609493d4b3e7ba1f6a9168e7633216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/170dddaacc609493d4b3e7ba1f6a9168e7633216"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
