)]}'
{"id":"openstack%2Fswift~742304","triplet_id":"openstack%2Fswift~stable%2Fussuri~I6eda54ccd2792effadb334ce9324887132b62b6f","project":"openstack/swift","branch":"stable/ussuri","hashtags":[],"change_id":"I6eda54ccd2792effadb334ce9324887132b62b6f","subject":"Use latest pip in gate jobs","status":"MERGED","created":"2020-07-21 22:53:41.000000000","updated":"2020-07-22 04:49:33.000000000","submitted":"2020-07-22 04:48:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742304-1595393305042-b2062de1","meta_rev_id":"5dce9435fd669abdce203b53c845e18d93d41cff","_number":742304,"virtual_id_number":742304,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-22 04:48:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-22 03:49:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2020-07-22 03:33:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2020-07-22 03:33:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 03:49:09.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-07-22 04:48:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"514a443c7a9921d54032d656133ff16a8c0fc8fe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-21 22:53:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"271b28672b60dc782bab7b0fe07e11dd9002322c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-21 22:53:47.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 498454a13931f93c8f42d1df149d15bb1d22141f","accounts_in_message":[],"_revision_number":1},{"id":"97724868d1eadab1e31e49013098eb5024f8ccf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 03:01:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96063d8870634b53875af973004477cc : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/f4d66d4bffc346c692e40b54b724fe6e : SUCCESS in 1h 10m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/69305e904303481eb7ad47edceaf7ec0 : SUCCESS in 38m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dcf88211c89240ac90a9c2e96c71bfa4 : SUCCESS in 9m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/36ca63ed2b364dc7b593ed8dd666dc82 : SUCCESS in 10m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d5656c68a40544feb2d7f8161c9cb025 : SUCCESS in 12m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0dd32a14b7d245a59e4f1e06fb056240 : SUCCESS in 12m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/af8be9e7a12c42229775ffe9ed0ca696 : SUCCESS in 12m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2ea17089e34e4c1f892566863ef5b88e : SUCCESS in 13m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/860e977f88e54d09a93b8bf83fe607d8 : SUCCESS in 11m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3baf9947364c49d9bc52a89d6a9cd607 : SUCCESS in 13m 56s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/50d9d6a34add405eb46cd1333af95d88 : SUCCESS in 10m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/77b60999cd664a6f941dc57f71adae02 : SUCCESS in 11m 58s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/515d2423400d4240b652ec79eee034c4 : SUCCESS in 17m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/86af1cd06ed441a7a05db0b1710136b6 : SUCCESS in 22m 32s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/451da36285a7429f81620a0c627cbeca : SUCCESS in 14m 28s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/faf20805933a4bf3967801461137e573 : SUCCESS in 10m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9da3b97aed7742099c700a98f1bb6377 : SUCCESS in 12m 09s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/21720eaf0e89445fab2fd98fcfd7e33c : SUCCESS in 17m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4f2ac83117654888af5e2ae6850874e6 : SUCCESS in 23m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/90b848209ada401fba9c9f01a3ebc6c0 : SUCCESS in 44m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/906c6c1be94e44969776f6cb961cc9ea : SUCCESS in 30m 53s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/17daa086485f4cbca3548be36b4055e0 : SUCCESS in 38m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/378d87f5042748568ac67efbbb620007 : SUCCESS in 24m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d956c877330429db437a34cb4029dbf : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6590b523024940c1a4fb8971c8eb1aa7 : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cbdaa2bf8ac84d21a20036e981baf86a : FAILURE in 9m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ed52ea2eda343b7abf8278d6b2fd2ba : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":1},{"id":"53bd3ed4464b6fdc829b4e71150224208aceb832","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-22 03:33:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nGate\u0027s broken without it.","accounts_in_message":[],"_revision_number":1},{"id":"046d713e6ba74d2a329c668402ffc6ceac3643f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 03:34:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7aa505cc7145636d837a85604b066401b56fd7e","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-07-22 03:49:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5c58bd02a6d1ecf3ac7c540b92ccf6c81ac0768a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:48:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/236a7ed55c574c30bcad92d707b9b34d : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/959f7a7559164bfaabb180566d584a95 : SUCCESS in 1h 13m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/654767dbc2d14b8dbb481f2c8eec6aa2 : SUCCESS in 41m 19s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/aeb5943bd3314893aec3e7394c648b20 : SUCCESS in 8m 40s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ce12e726fae54615a2f39afea6a497ae : SUCCESS in 8m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a40281ad767445282e66148ece428ce : SUCCESS in 15m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b1c9d9b2bdf342a6b1b15be8c56258d2 : SUCCESS in 15m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fbe0062f76134e8fbf25c8a601e8fd21 : SUCCESS in 15m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/abb04d2e4efc46d1a071f94d29bf2c56 : SUCCESS in 15m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3584f81a93bc4fb388dbcd7723cf054c : SUCCESS in 12m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e6a8fac77916422ab6052ea833a874ff : SUCCESS in 15m 39s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/e50219778d0c4b74a51cb90e8b74cc2b : SUCCESS in 12m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/29225e76ec354ca6b2fcde0817d1ae92 : SUCCESS in 11m 34s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/2ab8d57974b140d6bff4e6fffcd86106 : SUCCESS in 15m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dbb45f024fd14cc099966cb10847e977 : SUCCESS in 10m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/505ddb55b24044c4811ccc4a59a5447b : SUCCESS in 12m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1df716fa057845efbaa85854d1e9e9dc : SUCCESS in 12m 07s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1ec123f3bcd94e18b615df623d2f7ec8 : SUCCESS in 19m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e2038034385849c589a7b920a421e277 : SUCCESS in 40m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5526930cf13146999b43e700a5ead2c6 : SUCCESS in 26m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/22b3791fb8114a8e8c70828caf603bf1 : SUCCESS in 30m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6cb93cd6ffb744caac17e86ab9ba991a : SUCCESS in 35m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8d06c63ecd246b288c2706f3c9c6930 : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e03a809b589841f392482e18a860db6f : SUCCESS in 5m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea15e3a80b2a4adeb488cd7c82ac858c : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":1},{"id":"9504b82c353915cc932449c43eb65410f824d761","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:48:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5dce9435fd669abdce203b53c845e18d93d41cff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:49:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d6d3e1c8b941c4b334af0167d12921 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3463291b1a2643c18e471f47959c4ad1 : FAILURE in 38s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/fa10989cd20348dd8407598bf23791e4 : FAILURE in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e53fa8ed30a227c9ea1f60bc8893ea7ae9fab3d","revisions":{"2e53fa8ed30a227c9ea1f60bc8893ea7ae9fab3d":{"kind":"REWORK","_number":1,"created":"2020-07-21 22:53:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/04/742304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/742304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/742304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/742304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/742304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/742304/1"}}},"commit":{"parents":[{"commit":"40978e1b347010f24241322e54961ebc2f744e98","subject":"Merge \"py3: (Better) fix percentages in configs\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40978e1b347010f24241322e54961ebc2f744e98"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-20 19:33:26.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2020-07-21 22:53:41.000000000","tz":0},"subject":"Use latest pip in gate jobs","message":"Use latest pip in gate jobs\n\nWe\u0027ve seen failures with probe tests lately where dnspython 2.0.0 is\ngetting installed even though it doesn\u0027t support py2 anymore. I think\nusing latest pip should be better about noticing that and installing the\nlast 1.x release intead?\n\nChange-Id: I6eda54ccd2792effadb334ce9324887132b62b6f\n(cherry picked from commit 1c91547643f9c85f152d1d8c15431fec8334e214)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e53fa8ed30a227c9ea1f60bc8893ea7ae9fab3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e53fa8ed30a227c9ea1f60bc8893ea7ae9fab3d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
