)]}'
{"id":"openstack%2Fswift~742305","triplet_id":"openstack%2Fswift~stable%2Ftrain~I6eda54ccd2792effadb334ce9324887132b62b6f","project":"openstack/swift","branch":"stable/train","hashtags":[],"change_id":"I6eda54ccd2792effadb334ce9324887132b62b6f","subject":"Use latest pip in gate jobs","status":"MERGED","created":"2020-07-21 22:53:47.000000000","updated":"2020-07-22 04:49:52.000000000","submitted":"2020-07-22 04:48:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742305-1595393307639-b2062de1","meta_rev_id":"f7abf99edad725e9264f39cd8e04f2cf971c3b23","_number":742305,"virtual_id_number":742305,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-22 04:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-22 03:48:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2020-07-22 03:34:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2020-07-22 03:34:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 03:48:58.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-07-22 04:48:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5a1188895fa92ed8f4ce2f584fc1be8317c31ec","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-21 22:53:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"63d29305f70c65714bdde5752933aaf9d86b7595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 03:00:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f475cea77db4877a3071efe4606d4a4 : SUCCESS in 5m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/af68230bfa484c5c9bdb4f6e62b9aff3 : SUCCESS in 1h 01m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b2d744a849514640a2bb463f74a1ba4a : SUCCESS in 48m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cc36cd6c54374427af3dc8f94c86efd3 : SUCCESS in 9m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f55741f21b934b9690a4043b753f308f : SUCCESS in 11m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ad45584a3b0b42af956e532fdaf6fca4 : SUCCESS in 13m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9a817397e7e34775ae811b81c5db6d4a : SUCCESS in 13m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a0b0f2a1488a4553b9b4c343f238759e : SUCCESS in 10m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/38716fc9f13644b6b4360fc5a66b20bb : SUCCESS in 13m 04s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/3af2e636b7054308b5b9b5e05bdf07d1 : SUCCESS in 13m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5dcaee26ed434aa9a4d4cd8fd1ac76b8 : SUCCESS in 14m 37s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/5e2caf7946fe41cb870b6f9e57e69219 : SUCCESS in 15m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2fcc3deeef694f598a1db1e962d96167 : SUCCESS in 10m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b2f33875ed5640f58df7d95f847580cc : SUCCESS in 11m 59s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/67df6029ad55467582b6bf7d05a15a2c : SUCCESS in 10m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c0f18e135c384bf1b1d4253bd42b8bef : SUCCESS in 12m 17s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/aebe8b2e1ae34699a8e4ef866b387f43 : SUCCESS in 15m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b95ab317438f486eb9840128988c901b : SUCCESS in 21m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9dc99a98f9224f7bb4ed863e961fd216 : SUCCESS in 34m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b87527094e374ffd8f314090340659fd : SUCCESS in 26m 46s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b3a092556dec40949ac2f5d9a334d089 : SUCCESS in 27m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4f97584f259e4c9d99646b9d151417a4 : SUCCESS in 25m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aca89519200c4399b92d3a7e0e7b80b5 : SUCCESS in 22m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adb04099db2d4123acca88a492483afa : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c10f90481f1641acad39e8e4fb6a4989 : SUCCESS in 18m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d764330e05b44e11b042093ef45f489c : SUCCESS in 1h 01m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ba8a68df02cb47649229dce89de88bc3 : SUCCESS in 45m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e70f1a83d869ab71a8ffff87a63c2000396ff9d6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-22 03:34:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nGate\u0027s broken without it.","accounts_in_message":[],"_revision_number":1},{"id":"9d5ca7f7371ebd5a2b5b6d2308be2a496196fba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 03:34:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3a79ed3b2b7b6140ecca0024c45c44985479dff","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-07-22 03:48:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"42dcd24226fbe63370ae4ba820701e2fa8999605","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:48:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f70e100560443aebe189ba553c6c92e : SUCCESS in 5m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/b90678e2c4664e33a33438c354c9676a : SUCCESS in 1h 08m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/457f5aba6ceb4616a6dbbf6a3329ffdf : SUCCESS in 48m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2a1113e19ffa45abaaca324197f436da : SUCCESS in 9m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d151ae12973b4ab48c9b88bbdf48481b : SUCCESS in 12m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/af1de82428e447b39feed7c6969d35f8 : SUCCESS in 12m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/73bdceaafcdf4e7790e9992a797c6983 : SUCCESS in 15m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3517b8ea481040b0a863468b5181fdb6 : SUCCESS in 12m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5752e6d1500f47d582aafa4a9d6ebe86 : SUCCESS in 13m 11s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/5b9ebb3aed5c453f94984487145ed6d7 : SUCCESS in 9m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a9fcf618fbad4559a4c1dbea96bb7e84 : SUCCESS in 10m 02s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/b9858955fa7b477bb0b0b3244d3992d2 : SUCCESS in 12m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a815fabe4b55435499c30162f11449a8 : SUCCESS in 10m 35s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e0b10c5305bb43469e08544e9f7f55a9 : SUCCESS in 10m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c55280fcd9414e3cafc5f9c198375406 : SUCCESS in 10m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b00f8e2059a6432fb9c5184a9797689b : SUCCESS in 13m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f7061ef7c3094690b7e3ab32ba5e53c1 : SUCCESS in 30m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7e95ad75552a4ca0b91b76bc8f1fb782 : SUCCESS in 33m 18s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/76e7311e12584f67abd5ffc659856551 : SUCCESS in 32m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/12416c9ec4cd46e5b7288bc373c6afbf : SUCCESS in 30m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75718c288ccd4fd1a19cbaf24e660b09 : SUCCESS in 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8575ab534b614c1ca1f56d9a20f25c07 : SUCCESS in 6m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff648ccbf53342fabad727778111cf40 : SUCCESS in 58m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3780ae1acf8943fc9fe36853320e4d5f : SUCCESS in 1h 05m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7885737173f9e929a6eaf9888e03acaade0159bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:48:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f7abf99edad725e9264f39cd8e04f2cf971c3b23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:49:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de212c9b401f44928831d1798b871558 : SUCCESS in 1m 05s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/dbcd1877834e4ad88d04a3842633d047 : FAILURE in 38s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"498454a13931f93c8f42d1df149d15bb1d22141f","revisions":{"498454a13931f93c8f42d1df149d15bb1d22141f":{"kind":"REWORK","_number":1,"created":"2020-07-21 22:53:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/742305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/742305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/742305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/742305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/742305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/742305/1"}}},"commit":{"parents":[{"commit":"635c71c59e2add58f5cb2ac63aabb7c9228fdf12","subject":"Merge \"py3: (Better) fix percentages in configs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/635c71c59e2add58f5cb2ac63aabb7c9228fdf12"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-20 19:33:26.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2020-07-21 22:53:47.000000000","tz":0},"subject":"Use latest pip in gate jobs","message":"Use latest pip in gate jobs\n\nWe\u0027ve seen failures with probe tests lately where dnspython 2.0.0 is\ngetting installed even though it doesn\u0027t support py2 anymore. I think\nusing latest pip should be better about noticing that and installing the\nlast 1.x release intead?\n\nChange-Id: I6eda54ccd2792effadb334ce9324887132b62b6f\n(cherry picked from commit 1c91547643f9c85f152d1d8c15431fec8334e214)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/498454a13931f93c8f42d1df149d15bb1d22141f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/498454a13931f93c8f42d1df149d15bb1d22141f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
