)]}'
{"id":"openstack%2Fswift~742342","triplet_id":"openstack%2Fswift~master~I8b5b9b08087c045445d6fadf337ea0b206e2b755","project":"openstack/swift","branch":"master","topic":"742197","hashtags":[],"change_id":"I8b5b9b08087c045445d6fadf337ea0b206e2b755","subject":"skip same suffixes sync when update_delete","status":"NEW","created":"2020-07-22 05:06:02.000000000","updated":"2020-07-22 06:10:27.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b31241070cdc63624b7e2985d0560e99ac062041","_number":742342,"virtual_id_number":742342,"owner":{"_account_id":32225,"name":"zhouyunyang","email":"zhouyunyang-123@163.com","username":"mjzhou"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-22 06:10:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 06:10:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63ed5ebd978f3a3355027f14ea26470d0adb2ebf","author":{"_account_id":32225,"name":"zhouyunyang","email":"zhouyunyang-123@163.com","username":"mjzhou"},"date":"2020-07-22 05:06:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ca774fd2102964591374c4b51c608e3ce5018b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 06:10:27.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b5466a636f48bc9fb1a50af854eee1 : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/16e713a14e464fd78b1e5c1c2cb56caa : SUCCESS in 41m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/65d5fbfb3c3a4ba9b61420410b12b5ad : SUCCESS in 30m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a4d43da036340bbb200c78bad58b4ee : SUCCESS in 6m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/27885703d867457d9c02741cdd2f570a : SUCCESS in 7m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c238ed9e7c78406cb5a950fa9ddd438e : FAILURE in 10m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a9313f68a59e42cdbe0cbe1197f8af8c : FAILURE in 11m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/954707c0859d480aa434c33bd2506cdf : FAILURE in 11m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b63d076765b6433dabcca2bab8beda7f : FAILURE in 9m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/27b9e856e9a44fe481b01931d773c1ae : SUCCESS in 33m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cba7cd9dbb7644458b70719dc464156f : SUCCESS in 14m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1eea781c39c0416c97a59a160c239378 : SUCCESS in 15m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8220f169425f4b2e9b5fc092f7887dc6 : SUCCESS in 34m 32s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ec4019da4695461fa356ec96c482b020 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b75f35b05b63487aad897f349f23c587 : SUCCESS in 35m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/34a3d88abcff4e05bcb0f4bc2a791fda : SUCCESS in 24m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/51f4ee1e0260471bbc409777ace622b1 : FAILURE in 35m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b948a57416aa464699949eafd0fd2667 : SUCCESS in 41m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/11a58830433c4f9f9174574cf1c85589 : SUCCESS in 44m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd214b75ecb7446195454b707df2c572 : FAILURE in 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/676731f3c6384036b27fa3214dc8a21a : FAILURE in 7m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ab59100c30744f6a800050c0bad897b6 : FAILURE in 7m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6ce1ee21a144c5d8d062e9c9d0d64ab : SUCCESS in 49m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"580bbfc7b8a79ec18b9282aab559815375e2012a","revisions":{"580bbfc7b8a79ec18b9282aab559815375e2012a":{"kind":"REWORK","_number":1,"created":"2020-07-22 05:06:02.000000000","uploader":{"_account_id":32225,"name":"zhouyunyang","email":"zhouyunyang-123@163.com","username":"mjzhou"},"ref":"refs/changes/42/742342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/742342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/742342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/742342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/742342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/742342/1"}}},"commit":{"parents":[{"commit":"e1c71fd61b46e4ba12dee0799366792c00abd21d","subject":"Related-Bug: #742197","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1c71fd61b46e4ba12dee0799366792c00abd21d"}]}],"author":{"name":"mjzhou","email":"zhouyunyang-123@163.com","date":"2020-07-22 04:53:29.000000000","tz":480},"committer":{"name":"mjzhou","email":"zhouyunyang-123@163.com","date":"2020-07-22 05:05:10.000000000","tz":480},"subject":"skip same suffixes sync when update_delete","message":"skip same suffixes sync when update_delete\n\nOur Cluster have PB data and data increase fast,when scale node the rebalance is slow ,also have effect in network and disk io and cpu,rsync once maybe failure,we hope rebalance fast and have less load to Cluster.\n\nFor example,when three replicates cluter scale,this fix just sync more data to new node,the other two nodes just compare suffix hash and sync less rsync check，save bandwidth and lower effect to disk io and cpu.\n\nThe other situation is PB data cluster is not rebalance,we have to rebalance cluster many times when cluster level is 80%,we want reblance fast take time\n\nChange-Id: I8b5b9b08087c045445d6fadf337ea0b206e2b755\nCloses-Bug: #742197\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/580bbfc7b8a79ec18b9282aab559815375e2012a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/580bbfc7b8a79ec18b9282aab559815375e2012a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
