)]}'
{"id":"openstack%2Fswift~742360","triplet_id":"openstack%2Fswift~master~Ib711418d02d2ab41867fde8d55de3c26e5afafea","project":"openstack/swift","branch":"master","topic":"742197","hashtags":[],"change_id":"Ib711418d02d2ab41867fde8d55de3c26e5afafea","subject":"skip same suffixes sync when update_delete","status":"NEW","created":"2020-07-22 07:43:39.000000000","updated":"2020-07-22 09:30:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ea64ddac4f7b78c10c7cf703b167d08e36ddb0f0","_number":742360,"virtual_id_number":742360,"owner":{"_account_id":32225,"name":"zhouyunyang","email":"zhouyunyang-123@163.com","username":"mjzhou"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-22 09:30:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 09:30:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3397969b73bea9eab0a994ee2896396de887491","author":{"_account_id":32225,"name":"zhouyunyang","email":"zhouyunyang-123@163.com","username":"mjzhou"},"date":"2020-07-22 07:43:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b40a7c687a600e1a6fc9032db5dc97680344369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 09:30:10.000000000","message":"Patch Set 1: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/985538866508494b893f3a21b88685da : SUCCESS in 8m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/cef5533e58654a1da92603fa452bb47f : SUCCESS in 57m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/016851e8a37643bfb9f1b06dedf5699b : FAILURE in 45m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1ebbc6fbae2f4061bc0cf675cfe8d5e2 : SUCCESS in 8m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/60bd4430231541efb22f9222a3a2570b : SUCCESS in 10m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/06567518fd1342199b77c86e402ce31f : FAILURE in 11m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/40d3941a01e74f7f8ac31168f0724788 : FAILURE in 16m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3e73e8a0aa5b45d6a6b69e9d24f1ac8a : FAILURE in 14m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/67d995e88f23431d88fc5e372bd24054 : FAILURE in 14m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ecaea2b5dda24a2fb2df4fe876b40368 : SUCCESS in 19m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c3d95e50463d4f0884b4ce92150327dc : SUCCESS in 38m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7ec0b0f6494b48459258ad7290e4540e : SUCCESS in 38m 24s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8c78b639125c4a288a31b90e2440ba3f : SUCCESS in 28m 26s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c495e94b78d4439dae8f24fe371c8dd3 : SUCCESS in 31m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/aae093ac11884df39f9f7f57cdcf2e52 : SUCCESS in 19m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2e360ed9c5b64a838d0db0879616737d : SUCCESS in 28m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c69dfb154bb4412e8a8627fa7f2e48ee : FAILURE in 49m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f191bfe9183140c19122f408b2f72a73 : SUCCESS in 41m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/42449ea64a1946a88c18aed2171a7800 : SUCCESS in 48m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b30b9705ece44ab1a69651eb2b0d3009 : FAILURE in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2bc6044ac2b4d318774e5ede00a400a : FAILURE in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2aa8929c2de14716b70b57e3ae8ae0ea : FAILURE in 8m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e45f4f245a544788c7ceed823445fa5 : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ab7c20b493bf2531d77a7b49d4ed235728c9d36","revisions":{"7ab7c20b493bf2531d77a7b49d4ed235728c9d36":{"kind":"REWORK","_number":1,"created":"2020-07-22 07:43:39.000000000","uploader":{"_account_id":32225,"name":"zhouyunyang","email":"zhouyunyang-123@163.com","username":"mjzhou"},"ref":"refs/changes/60/742360/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/742360/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/742360/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/742360/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/742360/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/742360/1"}}},"commit":{"parents":[{"commit":"e1c71fd61b46e4ba12dee0799366792c00abd21d","subject":"Related-Bug: #742197","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1c71fd61b46e4ba12dee0799366792c00abd21d"}]}],"author":{"name":"mjzhou","email":"zhouyunyang-123@163.com","date":"2020-07-22 04:53:29.000000000","tz":480},"committer":{"name":"mjzhou","email":"zhouyunyang-123@163.com","date":"2020-07-22 07:43:18.000000000","tz":480},"subject":"skip same suffixes sync when update_delete","message":"skip same suffixes sync when update_delete\n\nOur Cluster have PB data and data increase fast,when scale node the rebalance is slow ,also have effect in network and disk io and cpu,rsync once maybe failure,we hope rebalance fast and have less load to Cluster.\n\nFor example,when three replicates cluter scale,this fix just sync more data to new node,the other two nodes just compare suffix hash and sync less rsync check，save bandwidth and lower effect to disk io and cpu.\n\nThe other situation is PB data cluster is not rebalance,we have to rebalance cluster many times when cluster level is 80%,we want reblance fast take time\n\nChange-Id: Ib711418d02d2ab41867fde8d55de3c26e5afafea\nCloses-Bug: #742197\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ab7c20b493bf2531d77a7b49d4ed235728c9d36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ab7c20b493bf2531d77a7b49d4ed235728c9d36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
