)]}'
{"id":"openstack%2Fswift~742756","triplet_id":"openstack%2Fswift~master~Ib2fd1ef3fa94d3dd26e62c01728fc94e765e9392","project":"openstack/swift","branch":"master","topic":"bug/1888037","hashtags":[],"change_id":"Ib2fd1ef3fa94d3dd26e62c01728fc94e765e9392","subject":"crypto: Add config option to support rolling upgrades","status":"ABANDONED","created":"2020-07-23 21:57:24.000000000","updated":"2020-07-30 05:06:54.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"65c302e31f459ec6a42412ef41e4854c34b59cfe","_number":742756,"virtual_id_number":742756,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-24 00:38:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 00:38:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3eaae19643c6dfa1c2ab3f3f5ab966e71c834877","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-23 21:57:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93e070da01cc44ce706a74c7a34491279db64f3f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-23 22:57:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a748e8498b954a6ef99ad47292dd06e1023d553a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-23 23:00:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1e81c725e578170ac6976c38f51620f60a3037f4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-23 23:28:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"180c0fecad04caf49b446c6bfaacbf5d132620ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 00:38:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f64343f0564641eba5309428b6219cac : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/d9f4d47423e641e3bc763c983f925b17 : SUCCESS in 1h 02m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/77dc9301f33043499112e4f9866e9c2f : SUCCESS in 43m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e6fd04e7d54748718653bb3eae7e2d40 : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/15e232cb40974f93a6464d2f69cae090 : SUCCESS in 10m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1f4cf9ea81b14cf19314953e43c566f8 : SUCCESS in 24m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b8d5c80dea684cfd8457958eb8e7ba0c : SUCCESS in 21m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a33d5c2738934c218dc54f9e873361d0 : SUCCESS in 12m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a1d2666414274b3cb0a18078e9e193da : SUCCESS in 21m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/76a5741d54b64afdaa59ba80b8f59f03 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/38c8453f082442239e798bd124e2c0c4 : SUCCESS in 19m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cd91b52036cd43fc9251006d3eed751e : SUCCESS in 17m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0ac7e35fee1e48ff96438f7acbff22a2 : SUCCESS in 19m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bee143f43f94421da99e205b89c61c89 : SUCCESS in 17m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d262c31a7ab04acd84279f161ad7b08c : SUCCESS in 20m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b1238e35916e438089e25867dd80b53b : SUCCESS in 20m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f6d668ad298f40708bf2887403052ca7 : SUCCESS in 1h 04m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b1602ab59b2247f4b1e7138b1f94021b : SUCCESS in 43m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7c75ac8107f6463a8ba56d39abc14e90 : SUCCESS in 50m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b98ea7f8e5e40c0ab50f8c76bb82e2f : SUCCESS in 13m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3574d43816fb4fcd96175d77851a31ca : SUCCESS in 6m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e9d12f1c98d742f39c377c4cf32cdc5f : FAILURE in 8m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/215fea42d3c747318f7f158cedf0bbf3 : SUCCESS in 51m 33s","accounts_in_message":[],"_revision_number":3},{"id":"c719b4fa61359b32da1eb7d2869d7bde2c56fff2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-30 05:06:54.000000000","message":"Abandoned\n\nSquashed into parent patch, which probably shouldn\u0027t land without this functionality.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d8487fa3900c7f9e5aa9a16d800d9ebf5d2e8ed7","revisions":{"dbbede949ccfd893f67d262025ee83aaccd29d08":{"kind":"REWORK","_number":1,"created":"2020-07-23 21:57:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/742756/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/742756/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/742756/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/742756/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/742756/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/742756/1"}}},"commit":{"parents":[{"commit":"802ec5ee83f7df06f688914714a943a39c3d67ed","subject":"py3: Work with proper native string paths in crypto meta","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/802ec5ee83f7df06f688914714a943a39c3d67ed"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-23 21:57:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-23 21:57:00.000000000","tz":-420},"subject":"crypto: Add config option to support rolling upgrades","message":"crypto: Add config option to support rolling upgrades\n\nChange-Id: Ib2fd1ef3fa94d3dd26e62c01728fc94e765e9392\nRelated-Bug: 1813725\nRelated-Bug: 1888037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbbede949ccfd893f67d262025ee83aaccd29d08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbbede949ccfd893f67d262025ee83aaccd29d08"}]},"branch":"refs/heads/master"},"48ad37a022571f63bce014c8f0dd50329b39fbf6":{"kind":"REWORK","_number":2,"created":"2020-07-23 22:57:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/742756/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/742756/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/742756/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/742756/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/742756/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/742756/2"}}},"commit":{"parents":[{"commit":"802ec5ee83f7df06f688914714a943a39c3d67ed","subject":"py3: Work with proper native string paths in crypto meta","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/802ec5ee83f7df06f688914714a943a39c3d67ed"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-23 21:57:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-23 22:56:45.000000000","tz":-420},"subject":"crypto: Add config option to support rolling upgrades","message":"crypto: Add config option to support rolling upgrades\n\nChange-Id: Ib2fd1ef3fa94d3dd26e62c01728fc94e765e9392\nRelated-Bug: 1813725\nRelated-Bug: 1888037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48ad37a022571f63bce014c8f0dd50329b39fbf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48ad37a022571f63bce014c8f0dd50329b39fbf6"}]},"branch":"refs/heads/master"},"d8487fa3900c7f9e5aa9a16d800d9ebf5d2e8ed7":{"kind":"REWORK","_number":3,"created":"2020-07-23 23:28:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/742756/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/742756/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/742756/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/742756/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/742756/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/742756/3"}}},"commit":{"parents":[{"commit":"802ec5ee83f7df06f688914714a943a39c3d67ed","subject":"py3: Work with proper native string paths in crypto meta","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/802ec5ee83f7df06f688914714a943a39c3d67ed"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-23 21:57:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-23 23:28:27.000000000","tz":-420},"subject":"crypto: Add config option to support rolling upgrades","message":"crypto: Add config option to support rolling upgrades\n\nChange-Id: Ib2fd1ef3fa94d3dd26e62c01728fc94e765e9392\nRelated-Bug: 1813725\nRelated-Bug: 1888037\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8487fa3900c7f9e5aa9a16d800d9ebf5d2e8ed7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8487fa3900c7f9e5aa9a16d800d9ebf5d2e8ed7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
