)]}'
{"id":"openstack%2Fswift~742768","triplet_id":"openstack%2Fswift~master~I033f6fba4ddb3dd502dfd5393dd2e9a2115672e0","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I033f6fba4ddb3dd502dfd5393dd2e9a2115672e0","subject":"Treat add/remove/rebalance during part-power increase as error","status":"MERGED","created":"2020-07-24 00:00:28.000000000","updated":"2020-07-25 06:51:13.000000000","submitted":"2020-07-25 06:48:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742768-1595659685051-8d15fb70","meta_rev_id":"6dba82fa5553ea3b01087feaae980408906692a3","_number":742768,"virtual_id_number":742768,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-25 06:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-25 04:40:34.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-07-24 18:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-07-24 18:22:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 18:22:24.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-07-25 06:48:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38d9cfdde2853d0f09a9d1f007950c2e3ce3f2e7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-24 00:00:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef362a58b9647dae49fec51adfdec2ca9ab6a931","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 01:16:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b5d989ccf0a44608068765edd6f9e52 : SUCCESS in 12m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/6a1d960a6b244ef399606400cb07abef : SUCCESS in 1h 14m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6f2125e7bc4346b1a5f68e2b45067c4d : FAILURE in 55m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec1f88c38b6f4e0bab98cba2ebcc06a4 : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/67736a8ace7a472f9d0b01419c8f5745 : SUCCESS in 10m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a8e0f1b19ac144f1bb8f8ad319b2e9f5 : SUCCESS in 22m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5feab644f7344b2f950fad3f5187fe2c : SUCCESS in 22m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1a8d249af2694b35a529853b1abd45de : SUCCESS in 21m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1ac96046d4304490bc20b1e6e73fdcce : SUCCESS in 15m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/297efa2075264018ac8cee1c3e65a81b : SUCCESS in 27m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/607b3cb0736442cabb04d60b04de1b70 : SUCCESS in 20m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/024cdf60d8a347389d0de3135faa21a0 : SUCCESS in 30m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/354b668fa965405093a4e567b765b507 : SUCCESS in 38m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3399c0a336f24cb996a1024588b741be : SUCCESS in 18m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bbee13b948af45ec9ec80b9a428423d3 : SUCCESS in 41m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b29ec9c865874784a955e4bd34904196 : SUCCESS in 25m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c70f534944754c31afd75dbce01602c3 : SUCCESS in 47m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cea243177cfe4169a466e44fc39da6ed : SUCCESS in 58m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4fbb344e5f684b9992577f8eb232e8e6 : SUCCESS in 59m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44424e93f3034dc7ab6b5adfb6e25b71 : SUCCESS in 28m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/280f56e3d2114c2a86c2727169cf509b : SUCCESS in 7m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9bdeca6f509d48c8a77ff023b309f873 : FAILURE in 8m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6963512c2be4fdda5a18735301fca3c : SUCCESS in 45m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c4c144dc60f3c6a1d1e278ed0a40c688bfb4dc64","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-24 04:38:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1a82b055196a0416d52a056ab3937ee6cabcaee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 05:54:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d24e632fe94711b840a25713de3d67 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/04197c98be2a4d5ca5509ebe5690d4b0 : SUCCESS in 1h 01m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b9e5d6c8ea8d40b985f460f09da947cd : SUCCESS in 47m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c0c47fa4466e4b809a92aade737eb612 : SUCCESS in 7m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d8af853ec88d48909aeb60a7a1834ac0 : SUCCESS in 10m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/26274c5c33a34dcbb3d700cd0345a4d5 : SUCCESS in 15m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/544f9968b3ea4a58871ec8ae9e46b6c2 : SUCCESS in 14m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ddcc3adb5d7d49a385c7e1d948dc5090 : SUCCESS in 13m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7b150257883b480f814bbde15fb5b335 : SUCCESS in 20m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/64eaf00776254fd7abb62503fd9f7de4 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2532a6f713434e1aa422cd8511dc4605 : SUCCESS in 17m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8e01182749c846b1b24721f3f127a233 : SUCCESS in 19m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/20abb3cab4a549408142bf6601d2863e : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bf05af366da14f7b8fa7bffd54d3ce43 : SUCCESS in 17m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a6a0f2d154a94de68169c698d2c2e62a : SUCCESS in 17m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cd54e4d970104185bb03b1b8c724ff3b : SUCCESS in 19m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7995c81b48cc4edea43a32a48e47daef : SUCCESS in 42m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c4a4fe6c2818410aae8bfa41b44bf353 : SUCCESS in 45m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c6655b5583af45e1b87f3f865b3bd1db : SUCCESS in 47m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2960aa5a0f40410aa59823d65b2e0fc8 : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad5bed97a9b47b4991e29f50efe1dd7 : SUCCESS in 7m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/972a6b9515b54f31a72129776d843820 : FAILURE in 8m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93a178d2c55a44f3b236eeb11171d9ee : SUCCESS in 1h 06m 17s","accounts_in_message":[],"_revision_number":1},{"id":"eb76923183528d16093a2068d77f46f449fb3355","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-24 14:20:26.000000000","message":"Patch Set 1: Code-Review+1\n\ni\u0027m fine with this; but it\u0027d be helpful to understand the activating force - as long as the exit is non-zero; what does warning vs error really matter?","accounts_in_message":[],"_revision_number":1},{"id":"15829db54ee9723af6d95075e354a17f3420a657","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-24 16:13:26.000000000","message":"Patch Set 1:\n\n\u003e [A]s long as the exit is non-zero; what does warning vs error really matter?\n\nThings like https://github.com/swiftstack/vagrant-swift-all-in-one/pull/102/files#diff-a5095946784cfeab17ed11ceb2126922R54 -- where you can say \"it might exit 0 or 1, but anything else, we should bomb out\"","accounts_in_message":[],"_revision_number":1},{"id":"79aeb6578a1816fc342808968ebf3e525ac1b95b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-07-24 18:22:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nok, yeah other warnings (esp. in the context of the referenced build script) seem related to \"parts may not be exactly where you want\" as opposed to \"you\u0027re in a wonky state and can\u0027t do what you\u0027re asking\"","accounts_in_message":[],"_revision_number":1},{"id":"3bf1df8448ec097e3d6eb280b8c132e7582c7a09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 18:22:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81205feccc88820dc0bbb7c53f89959f691f5ed3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 21:40:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb9fb66841d94e01b29352160d6e4d4b : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/4a740bc0f64d4a81899324edbb89d3a6 : SUCCESS in 1h 05m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e7793a66b41f46e6bbf46f17483d12d9 : SUCCESS in 54m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8f19802ce5574d60ae3abb8719fd7761 : SUCCESS in 10m 37s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d1a22182a01845ebb4563919fe9f4d57 : SUCCESS in 9m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/03a59fe60a0644b087c75e161f107fad : SUCCESS in 14m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/28a93edb27974bf889d732cfb1df9397 : SUCCESS in 13m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6d42be1ab11045629d3701d131c17594 : SUCCESS in 13m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8d667d9a76224bc59c384c32610caa97 : SUCCESS in 13m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7e2dfd6a776c4de192d1857b4b30f9dc : SUCCESS in 27m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3ba3045bbe8d44fb84f6cf4103f00911 : SUCCESS in 23m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0021d492c9254965bac9f0467e5fecd2 : SUCCESS in 19m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/80899a1aae43463b8dfebe7360165e47 : SUCCESS in 24m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/aaa2b58c69234bc29e7142bed3a7f57e : SUCCESS in 33m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/036f5801233b4a3caec97d7ec1cf29de : FAILURE in 54m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/32f8346a9982476daf6ced794bde173f : SUCCESS in 39m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c88cebf36a4446f1870cd94385090a6d : SUCCESS in 1h 03m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8923de5e47ef4c18b293a1fceb0d9d8e : SUCCESS in 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/075e8cd3715a4da999458c082b4d4574 : SUCCESS in 6m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/904e19de79e14ce78f8568f2d1856406 : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a524b9442cc3a1b6fdc2a6bea464f324cf2f0c36","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-24 22:18:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4c98e1047566d62fdaa9336295178f2f5d1d36a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 23:24:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c984aac3d4a946cfbcd0703bb29b2648 : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/a647cddfd4e344a6b36bcd44885ada96 : SUCCESS in 1h 00m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/58a5cdfb62dc4b0a8f85db4a1e443f90 : SUCCESS in 1h 03m 48s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5685a835225b40fd9f73505f05a1059b : SUCCESS in 14m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/88a3bc5340864c3bbe85117f8b4999ec : SUCCESS in 10m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1a6cc5a64caa4a22abb4ddae3a74f8d0 : SUCCESS in 15m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2e8179dcf40f40bba6863c57e44ee191 : SUCCESS in 20m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/02a2d959c17c4f038603a17c28e5f87c : SUCCESS in 41m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6f34c06740b14a63b25dad1acc430528 : SUCCESS in 41m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fb4934023f304e24bbae0003bb445722 : SUCCESS in 50m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7918b62d29074123a950c2da674c2d51 : SUCCESS in 25m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cb3c760ed34f465781bb5bd6367793ba : SUCCESS in 51m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7f24d062fc85409581e5b0138f657c07 : SUCCESS in 32m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/586f3e1584724dd8b9622329c05aa42b : SUCCESS in 24m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/53cf0444b9ea405789e05552ca0896b4 : SUCCESS in 18m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d57341718fee48648a4ae0e7c4a92ed3 : SUCCESS in 25m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/646ba598adb44f09a0f2678f7965c840 : SUCCESS in 55m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86c3086248034f09b656e6aa2e0a5d29 : SUCCESS in 46m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/75a27518d94b4b0db2a3b97fc80c3f03 : TIMED_OUT in 1h 03m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99bc9c45f89b4918a8c5829873ec9358 : SUCCESS in 12m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f85597bcb8542adaadb5d1acc138dd3 : SUCCESS in 7m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/180959ac45164442885d85ccdf0af509 : FAILURE in 10m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f9d41403b684f51a903ac0778a5e6f1 : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d3e54836dd727857f2753dc704e71c42480f06d0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-24 23:53:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ecac79f09315ac6335ae7cfd31fd8bfcd319c742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 01:25:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95a7f3848c4148a9bf6a954480c72387 : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/fab5421ad2e64fc1b63ff9fea097c3c3 : SUCCESS in 59m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5947ddcd8fdf48c89a7d4f99b6a31b02 : SUCCESS in 1h 21m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/49571ae55cf6461f9e7b8d5301c3f5ff : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b24a6ccfb75d45efa37eda0af3d37f51 : SUCCESS in 9m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6181f25b81914bf4bdc0d8f9b29b1144 : SUCCESS in 12m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/36a675c48b3c487ba72eb2f35c0d1ba9 : SUCCESS in 16m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/de20cea24a2a46ddbf30221d17b09ab8 : SUCCESS in 17m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/319559f083f042459ed7cb895d3473e7 : SUCCESS in 14m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6643086d9d19456c97d3a768a4a5b3d7 : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ddb9a8b720f04bb2a04dccfd0134595f : SUCCESS in 18m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e1e0ab01cff844c8b72fa040c7c5fad7 : SUCCESS in 18m 03s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6eb46a576d1141b091e9358b7e6fe140 : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a7f05569b83c42438eadd0d76b2ffab5 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/245513e583dc43b19ae1ba75bdf4dfa9 : SUCCESS in 18m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/96b47249a11c4622b67444f9a10a91a9 : SUCCESS in 21m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eddc12c0c64248b4a46adf00f6a28e34 : SUCCESS in 39m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ee41a7e493d435db095c049c6193a24 : SUCCESS in 42m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e35a76f545da42a2a0aee8e50555a9b5 : SUCCESS in 48m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56a0e2cecbf14878b84fd9dd3c965e86 : SUCCESS in 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebf4d098efb141cbb9a337cf0a30ca6f : SUCCESS in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/68f27f4f4ec44c0788ed8baac1d050bb : FAILURE in 9m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65c68e8fe2d84d929167f866f579d5f1 : SUCCESS in 1h 14m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0b12d58a624f71ef21ea4faaac09f090b775a919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 01:25:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5fbcffaa465f9337769bc857e093202c0146716","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 02:27:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4aeb90672f649d8bc72970acdc144a3 : SUCCESS in 7m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/e13eaff280794d8da1c68a9e30fb7068 : SUCCESS in 1h 00m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3c2fe1f7510f4f269b52a45ba71065e5 : FAILURE in 54m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0c4984917d5c4c6fb76d261b80b05561 : SUCCESS in 9m 04s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2b259025edef44598d3202af14831c99 : SUCCESS in 9m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4ca0b8b5ea004f1ab65fb382cec4b74f : SUCCESS in 11m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6be35e3c88ba487184b0be38ffedaa30 : SUCCESS in 11m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/18a0ef066df14d2690fc490e5d52605f : SUCCESS in 15m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/96bbe6eb3e4f4909932f998bfc383bcb : SUCCESS in 15m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/962a8d28cf5f4357a6c7309e49b75ef4 : SUCCESS in 25m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2c2e1ae4d8184d469c538ab5aa26ac8f : SUCCESS in 17m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bb066a1778a14438bf9520dd5dc7105d : SUCCESS in 18m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a574daf7158044c6b12aba4628aff2e7 : SUCCESS in 17m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5e8bb760393c4effaac15beeca1fbab1 : SUCCESS in 30m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c23d767499a844dcbdd5af3186a14d5e : SUCCESS in 45m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/837e44a7916145f8a3c23cca2ae0b1de : SUCCESS in 37m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7141a328a36146409f7ba34a130e190e : SUCCESS in 38m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d9edaa0998149f5aafe77e4eb34a42d : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb81dafef67e462fa82aadbb2892bec6 : SUCCESS in 6m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2220daf4b5b46c69a3daefc790a7d0a : SUCCESS in 43m 23s","accounts_in_message":[],"_revision_number":1},{"id":"010c1fae0741129c3b68925c6733c7a74600d6f0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-25 04:40:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6213fd45a6092d1a71b0ca4d938acc70a0649190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 05:37:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a100c14a28241b3b85d0b568ba50027 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/ad14ace3341944d0af5cd9d1eeff72fd : SUCCESS in 51m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d8eedffa26d944efb15f46b1dd166fee : SUCCESS in 43m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eb9c92ec5a164c86a7245069dfab29c5 : SUCCESS in 8m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ade6939182d04d5893bd64c6ee2084fe : SUCCESS in 9m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/404da0a81e2f4128830acb01202abda9 : SUCCESS in 14m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/640bde95112b402784380c471e5f3530 : SUCCESS in 13m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9d2e1948bcb1461cb41d137ad37b046d : SUCCESS in 16m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e0dd940aeee8420d846b66019cb642ea : SUCCESS in 13m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b672c3ea8c594869b2e8d4a6eb0d4be8 : SUCCESS in 21m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/70cfec22877c43b28d34aa7374d094ac : SUCCESS in 20m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/327473d318af4ef6a4db7ff06a145c41 : SUCCESS in 21m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b32ea51d65a1442191e8d5ada3636bf5 : SUCCESS in 21m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c34bfa81770447a39fae035b4f2d1f76 : SUCCESS in 26m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e1d020359bb9442a8eefcb3c96a7ce05 : SUCCESS in 22m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f9128f9b633648928da7a23d29566a31 : SUCCESS in 28m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4a6f2f4fd9b14964a86fcaf37760e8da : SUCCESS in 49m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a42b601bac5a425f9f151566fe36b104 : SUCCESS in 41m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd2348459a554836905338652f8e8acd : SUCCESS in 42m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/393d7f58b9d54a62bbcb4e137e3a0a95 : SUCCESS in 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4bf92aea9d4432da15a4b9ae02306d4 : SUCCESS in 5m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10ed41453e2142429f54824b58f8c9f6 : FAILURE in 7m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/477262cd6d7b4cba911089b05bf92899 : SUCCESS in 54m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1658002d0d6eefab4cdef0b15d35c18a62bfabfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 05:37:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"527909fafb46f90a3992bbceb0f41fd6fbcf4c04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 06:48:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bb7bc213b1a475981324339d705f065 : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/a7b7cf31b72546c29fe7762bc6ded535 : SUCCESS in 1h 02m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6f1e39323cbe4a03adab371bb6e35817 : SUCCESS in 45m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1ce6a7e1e5d44b169cb3c74e9b3761c3 : SUCCESS in 10m 11s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/85129480721e4ce8a106dc3de064c96f : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/47187133e6f54cd29513500b5464d310 : SUCCESS in 45m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/73bb4323564649e284bb94d5f25a1d21 : SUCCESS in 11m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/172da5af126c430981a2919117e0f518 : SUCCESS in 47m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3efdf91c9c084f31a95502387075ee57 : SUCCESS in 16m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8f64575e54854f06a1e6960d7a738c49 : SUCCESS in 33m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5b08d2add2b744b9825f8999b9a9ea70 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/abc9a53a95334235abb73edb5342c5d2 : SUCCESS in 37m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/46f61f790de649fb9ccace6f59e9712e : SUCCESS in 35m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4fc809f64de047ca991e789022fdffba : SUCCESS in 20m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d226780cfaf448bf958780c89d4a25ef : SUCCESS in 52m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/205a7bd5deb14a00ada5fc6978b842c7 : SUCCESS in 37m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e58ed2fc4e44da0800c268a91f8b6a4 : SUCCESS in 41m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcba9715bb0a47d2a08877df740304cb : SUCCESS in 27m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ee9ef0be24242d5813aaffb21012a09 : SUCCESS in 6m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75f3401e97694b328595f151e67d9022 : SUCCESS in 1h 05m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2d928e5c4a156e2de6708d277d7a3c61f99140ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 06:48:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6dba82fa5553ea3b01087feaae980408906692a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 06:51:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5f4771c3404061a7ef175edda5e04b : SUCCESS in 2m 31s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7e04c33ee0bd474e9ba8907b35fd02f0 : SUCCESS in 2m 06s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/7787bd3e23fa4d3cb8ab63e1f215dbac : SUCCESS in 2m 04s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"770cc287aa340b5cae0d8750a864b332beaf52ec","revisions":{"770cc287aa340b5cae0d8750a864b332beaf52ec":{"kind":"REWORK","_number":1,"created":"2020-07-24 00:00:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/742768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/742768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/742768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/742768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/742768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/742768/1"}}},"commit":{"parents":[{"commit":"9eb81f6e698ef8ac8855c413b0b1bf4e030cb687","subject":"Allow replication servers to handle all request methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9eb81f6e698ef8ac8855c413b0b1bf4e030cb687"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-24 00:00:11.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-24 00:00:16.000000000","tz":-420},"subject":"Treat add/remove/rebalance during part-power increase as error","message":"Treat add/remove/rebalance during part-power increase as error\n\nWe weren\u0027t actually willing to change anything; sure sounds like an\nerror to me.\n\nChange-Id: I033f6fba4ddb3dd502dfd5393dd2e9a2115672e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/770cc287aa340b5cae0d8750a864b332beaf52ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/770cc287aa340b5cae0d8750a864b332beaf52ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
