)]}'
{"id":"openstack%2Fswift~743614","triplet_id":"openstack%2Fswift~stable%2Ftrain~I918ab4ccbf4d081b26f4117937410cdad1caf8d3","project":"openstack/swift","branch":"stable/train","topic":"bug/1862645-stable/ussuri-stable/train","hashtags":[],"change_id":"I918ab4ccbf4d081b26f4117937410cdad1caf8d3","subject":"Stop syncing empty suffixes list","status":"MERGED","created":"2020-07-28 18:37:56.000000000","updated":"2020-07-29 00:07:48.000000000","submitted":"2020-07-29 00:06:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743614-1595981164633-4e4c0980","meta_rev_id":"a8329c67e6dcd0b2cf7b0d298714df9c205ce686","_number":743614,"virtual_id_number":743614,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-29 00:06:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-28 18:38:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-28 18:38:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-29 00:06:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4fc99db16e5281c512c46c30ff7723306db60634","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-28 18:37:56.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"16266074bd2e52ef0cfd37f4cc84d59d7caaabb7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-28 18:38:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eb1d6bf2a2db8ccb1a102f0bdedbab306e63f5a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 22:31:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77298119259947fb81f4becb60f53a42 : SUCCESS in 6m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/b034d18b78ee4c7b9bca14ad736fe988 : SUCCESS in 57m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bc0ed476d40f48b29f34854792d7bd72 : SUCCESS in 49m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/184139749e004da0b58d65b251826825 : SUCCESS in 7m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/32c7f727412b4323a88c2717961c93c9 : SUCCESS in 12m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/70ec5fc7dfbd4da4819978a51112e85e : SUCCESS in 14m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7f0f18ca5c47430caf7c192eaac7200d : SUCCESS in 20m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/925df693b0f54810bcb8f28b6dc5566b : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/25eb8606c4294b0788cfd9270fb95337 : SUCCESS in 10m 04s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/6de3280e9c98452aaf65b37cef424255 : SUCCESS in 10m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fcfff020476d4d26a8bc300853248f3f : SUCCESS in 23m 29s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/c91f293fe24a43e193a19e1f7751a1e3 : SUCCESS in 17m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9b2dff14fdd24268a22223e8bf9c57dc : SUCCESS in 21m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/72ecb26006504aad93885e1b7ed5d7fd : SUCCESS in 11m 20s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0df14ce5d2e34613b44830c92865468f : SUCCESS in 13m 21s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fbcb1686adde4928bba008e83460ebbb : SUCCESS in 10m 09s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/0d0d5b746a5c42b7a2085a8bd8f7823f : SUCCESS in 14m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/63abb148c77f493380ae30de357c6552 : SUCCESS in 29m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/caab3ef99427483093845d54210e97b2 : SUCCESS in 31m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7c6ff6256df74f05bc0748112733f65e : SUCCESS in 34m 15s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/400e475f7ebf4d188d6c0725b5d01abc : SUCCESS in 33m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1e5e331b024345e89c8b1a513cbe3d80 : SUCCESS in 29m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8400f16d042b41c6a190bb629475d350 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf4eb2f9319b41c8a6a4be340fb275ca : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0f85d0bc256d47b4a1d35c83ba2c6c97 : SUCCESS in 21m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3628dfd7bbbc44b6bcb4c29c4ea7a49d : SUCCESS in 1h 02m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/77fc0b0d26dc4926af3b1ef49ec3bcec : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":1},{"id":"086f9b2516e982feea86028579f2eff728607ab6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 22:31:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2226da2d0514135efbe63680909dc5402513914","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 00:06:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/422ac40f4f004f1994c6ea95a459e2b2 : SUCCESS in 6m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/20a1e8dcf74c4dcfb26869fe25be806c : SUCCESS in 1h 05m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aedff5be2110489f81d7ad7be8ebc26a : SUCCESS in 48m 31s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d2c05e4aef224c4994dc07eda363f09c : SUCCESS in 10m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eff32bfb967b4ebc9666afa2e5b8bd45 : SUCCESS in 15m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b76480f737c14bbb858663ed87985165 : SUCCESS in 9m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2c6b0d965f934a528f2b3823a9b34556 : SUCCESS in 15m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/076ffcdce2e941a7ad81c910a60cb885 : SUCCESS in 28m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fec058e85f094b46bad756f6430ca043 : SUCCESS in 11m 17s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/6e0f2c22d4b349ae9081947791bb7c34 : SUCCESS in 15m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ce2dc01a16e14e478e25529cb07be1af : SUCCESS in 11m 05s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/eae665557f8041c8b2880df3abab9f5b : SUCCESS in 16m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/17218ddc05cd43ce8dc4dfb4fe31fddf : SUCCESS in 15m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f3fbc3629d7049c082140a9abe874d0a : SUCCESS in 11m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/708cd02cad714281a11e09d912e145e2 : SUCCESS in 13m 16s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5d6f1acd9fd1460eb03b63f2b2abc0b9 : SUCCESS in 13m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/279ab77dcf98446fa0a9f16d55d32f0f : SUCCESS in 33m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6ccb77123fb64d82b967915b34ed970b : SUCCESS in 33m 48s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/11104c0a928d4ca1967e2e0fe9b248b4 : SUCCESS in 30m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0a0a3f60e8d64411b1fb8f8964c2ebf1 : SUCCESS in 32m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a199fd7d59246ebbbe0943339951dc6 : SUCCESS in 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1529baa6c2d94f9cbafe141ee932c314 : SUCCESS in 6m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b35e00afc1a4687bf5be88ba5c33c2f : SUCCESS in 52m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3f58e3acf84941c1896dc8b714c19e1b : SUCCESS in 1h 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"de184959529a65e8b4aa772c625dbdf36ec65673","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 00:06:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"59efbf0b0e6eddf524316056b9698c63b902a532","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 00:07:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed91e1aafa24ee7a0ff55d6825c88c4 : SUCCESS in 1m 20s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/65eb337a355e4139af8df7ff0724a8fc : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69225ada606c00e5a68d82c297c7c5c100bf33ab","revisions":{"69225ada606c00e5a68d82c297c7c5c100bf33ab":{"kind":"REWORK","_number":1,"created":"2020-07-28 18:37:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/743614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/743614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/743614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/743614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/743614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/743614/1"}}},"commit":{"parents":[{"commit":"0ed9a23d59e0dd73a235f96d65b7f73b169db357","subject":"Merge \"py3: Stop munging RAW_PATH_INFO\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ed9a23d59e0dd73a235f96d65b7f73b169db357"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-16 20:04:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2020-07-28 18:37:56.000000000","tz":0},"subject":"Stop syncing empty suffixes list","message":"Stop syncing empty suffixes list\n\nChange-Id: I918ab4ccbf4d081b26f4117937410cdad1caf8d3\nCloses-Bug: #1862645\nCloses-Bug: #1886782\n(cherry picked from commit 907942eb4743e5a4213b1f84fd57a58386594bb6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69225ada606c00e5a68d82c297c7c5c100bf33ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69225ada606c00e5a68d82c297c7c5c100bf33ab"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
