)]}'
{"id":"openstack%2Fswift~744256","triplet_id":"openstack%2Fswift~master~Ib225abd17666e8c966ebb4a559cef3851bcab844","project":"openstack/swift","branch":"master","topic":"simple-manual-shrink","hashtags":[],"change_id":"Ib225abd17666e8c966ebb4a559cef3851bcab844","subject":"Add swift-manage-shard-ranges shrink command","status":"ABANDONED","created":"2020-07-31 21:28:14.000000000","updated":"2021-01-21 17:30:04.000000000","total_comment_count":35,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0557ad0f978b0a435eb5cc27714023965b68b5fc","_number":744256,"virtual_id_number":744256,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-11 03:08:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-19 20:08:12.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-11-11 03:08:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5cbf1d99ad2cb0d4d819a6565269dc15875e0944","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-07-31 21:28:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8385f978e6e2781f91fe226828aebe2f2d2d6e64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 00:22:29.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e73f61a28a9d44f6b5bbee268a79101f : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/b3fa728cdafd4cf9a19c5728f5e93e4e : SUCCESS in 59m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/57eb3656a0114e55b8db7a664b10a680 : SUCCESS in 41m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d4ce6199f190491498c9ab7a5d913384 : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e4077b21d2034a7ca10b05ed61af7b85 : SUCCESS in 10m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9fad59729ba746debca1b18f499d6167 : SUCCESS in 12m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f48f70df10014b3795520a61e500a181 : SUCCESS in 12m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a542bceed45346f4b9511ace90441936 : SUCCESS in 28m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b69248f26c874899bdcda21979a584af : SUCCESS in 11m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ef549e11f7ce4b41bb69ecdc9413aa75 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6739bf2569174147a404a0c850fae887 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c80a58db091e4c4c895556fb15b30879 : SUCCESS in 14m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/538a453fd0894836810f3889816ba657 : SUCCESS in 19m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/683a61baf19d4e31bdd4b3b5717f1523 : SUCCESS in 27m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/64d97a81cedb46db9ab5fc9e5f8318d6 : SUCCESS in 14m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bead761fb21c4fce90ce3c3aad2129b3 : SUCCESS in 21m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7781f303f8fc4702a3d0103d7ee79ba4 : FAILURE in 40m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3ccfe0b4ff14e4e8ae303bc6d61bbfe : SUCCESS in 44m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/155499350c7c48f5a18b321b8d3f3be8 : SUCCESS in 55m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b1a567b7d1e4d59b998e514995e387f : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c13b3d902c324e06b6e6cbc354ef1a59 : FAILURE in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e83899cc47d54d44a8b91b414b600772 : FAILURE in 10m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/956459155e534750ab4f03a0b22fb5f0 : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6508b328237599948004b053c763f783c8d985db","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-02 00:36:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"caf3aae4902e9fbb0555e888c439c0995cdec9c3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-02 00:49:52.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ee49b2ae1087ea4ab7fe0593143207bcb3c20dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 02:54:37.000000000","message":"Patch Set 2: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d69fac7b6bfc461f94273b1e2f646dd8 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/47994df35cb34c4fac194672f593468c : SUCCESS in 59m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6086d5521bb64791a6c24b6994e9f490 : SUCCESS in 49m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94d105357a27489da2f48b858db2d7b0 : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9174d251e01e41588d900ff4d226ea31 : SUCCESS in 8m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5dd164bbfd8043668685f610824ed348 : FAILURE in 14m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3bc81ed3d11a4985b90a280e08e249b7 : FAILURE in 15m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f5910a6e1e5c4202b70bf8a7c2657c0e : FAILURE in 12m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/45365eb1bc454a49ba57fcdaaa156a53 : FAILURE in 12m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bf5d88ac2adf4f499406aea14f9ae3e0 : FAILURE in 4m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ab253ac092c04fd2bd877fa05d5a4588 : SUCCESS in 16m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/974d9b2f82d14527822440936b55055f : SUCCESS in 16m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/16a921f0d42849759fbe56eed4c68207 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b24f5de066404aa9b176f9cd61025842 : SUCCESS in 17m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9915ec63711a4cc281da289eda73a435 : SUCCESS in 19m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1dfe5b93602443a87ccd0cc67e1a0a7 : SUCCESS in 23m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/30e4ebf6b4524daeb9d7ffa077a49677 : FAILURE in 44m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dc8a9904bab34bd38d0af5a8def4aa80 : FAILURE in 13m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/95b932164e574867ab1f73cd0a6bbf6a : FAILURE in 15m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e01706914914293b23ff51925c7d0a9 : FAILURE in 23m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5039ce6bbb8e4f3699d1a49f9a628032 : FAILURE in 8m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7d189e8489744eca8064523d73c5f43c : SUCCESS in 20m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83d3ffe6e113498195a82fe1ded6082c : SUCCESS in 53m 53s","accounts_in_message":[],"_revision_number":2},{"id":"a6079430aebe074f9f702b8f44c3eafab3027b4b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-10-12 20:22:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dd19e52d54308eb25953b3052182f26b18a90bef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 04:29:59.000000000","message":"Patch Set 3: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/185502a64816482bb1eb447d8dbfb655 : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/d45ca94e110a48aaa00114420aaf883d : SUCCESS in 1h 01m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e2ca13b12198419caf4bd43889534bfc : SUCCESS in 49m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e176015a8f364128a6ea3c0a6dc89f53 : SUCCESS in 9m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3e00094d5f724302a56ab6e5ac7f9dd6 : SUCCESS in 9m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae946b88dcfa4b7bafd8031dc881cfff : SUCCESS in 13m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0cc295ece779411d922ac77c27cf6f17 : SUCCESS in 13m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/39312a5a55c04f66b3073b215cd86d16 : SUCCESS in 12m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3815e3cb50054489a04c1552ede8f74b : SUCCESS in 13m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4e2de57a047a404b8143090f6f6a80ea : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/872d06d6ca904905a5b63c5780aef95f : SUCCESS in 28m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8da678a936dd4e7aaba617248d64155a : SUCCESS in 25m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cfb5908752d34f168b9be9ac531d68b5 : SUCCESS in 36m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/08620dc8cebb438eab1f75674ab90406 : SUCCESS in 20m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a2e4ca83baa24339a0d02bfbd56ce664 : SUCCESS in 28m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d35528c1a47f47a7a665e465c51924d1 : SUCCESS in 29m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f478262aedb2406faa387b8390de3ab1 : FAILURE in 45m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7123f7a420e84fa0a30a88d53491862e : SUCCESS in 38m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1bad3f11124a4e87904f02ba34b93736 : SUCCESS in 44m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/159af1c6bf5745989932184bb86be586 : SUCCESS in 14m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e77da6c99a0d47b29af2f1d89400773e : FAILURE in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36cf89fcfd4848ae9963f040c95df521 : SUCCESS in 20m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36af4fb00c0a44cda6a7df1c7529a32f : SUCCESS in 1h 03m 34s","accounts_in_message":[],"_revision_number":3},{"id":"4cd104d0721969a997de9f2019caef21b4801844","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-10-19 20:08:12.000000000","message":"Patch Set 3: Code-Review+1\n\n(5 comments)\n\nsome of the state cleanups seem quite reasonable!","accounts_in_message":[],"_revision_number":3},{"id":"2d255ed963c913176a02a5472473e6ddacc47e07","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-11-09 22:23:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ed6c087cebf46766b08e6259ffcd3f34eac332fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 01:52:41.000000000","message":"Patch Set 4: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7635fb46359a460b873528213ed9752d : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/1298510e60a9457f9db046c32a9af13c : SUCCESS in 1h 03m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b4b49d9e37fe43c0a5b8f00a57e36f3f : SUCCESS in 52m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c6b7b3450ebc473f89a5b946cab1a637 : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2afdfe1a76974102bca972eb0d1e4be3 : SUCCESS in 9m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/70c0f2aca94149ada0c140a3aabe73c1 : SUCCESS in 12m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/13168def70de4cd294066e8d8ec26987 : SUCCESS in 22m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cee260da98f04d95bf2de6ac2529a04a : SUCCESS in 12m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/65dc92facd994be493259bd04babe3d8 : SUCCESS in 12m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9220986ba0da42608045fdb2a8bff9f8 : SUCCESS in 21m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/58def7b965ac46ef8a40fa00a916aaa7 : SUCCESS in 19m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a7cadd15444044da9c8edeb0ae1f774f : SUCCESS in 19m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fc2a183dec7b43efad2d34d0cabb68cb : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3f9e25c15d0e467f80ebbd3c3146919b : SUCCESS in 17m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9df0893dc6b44d4a98b5ff05eeebe846 : SUCCESS in 23m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/be454ab183ba4fdaa6ca06aaf510d92b : SUCCESS in 23m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c1113b0ffcca4afbacfb3675d36043f5 : FAILURE in 48m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9072a39040114cce829a899f1d3257ea : SUCCESS in 40m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a25f1286583049c3b99e0d8298a4b40b : SUCCESS in 41m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d38f33efbcaf47c1a9d361201209e36b : SUCCESS in 12m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92a81bbb05d24a6e863d73fc94197994 : FAILURE in 5m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c48bf0a39235489bb763d04e2a157441 : TIMED_OUT in 1h 01m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/136c7d6a8d89451eaa9c702e5e930d6e : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":4},{"id":"da60ec4e8ecfcb16426a95501ae5ea13db6be111","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-11-10 14:41:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"133ab4c17d4b4e988d161d930b37b5e69c2a63e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 03:08:43.000000000","message":"Patch Set 5: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d3ba874e5d44475add02373eafbf1ff : SUCCESS in 14m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/3274647299944034945d4e7d3721f5eb : SUCCESS in 1h 02m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd3f451cc0764853a4e1cb5f3b444efb : SUCCESS in 50m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e66ca05a20504d34baad983296f7a167 : SUCCESS in 11m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2ccbd53fc3ba495795a91123348bfe59 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4bf5c38db855463c83aae117fd367a40 : SUCCESS in 17m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/65f579d07e3e4ce49089b65dbe1b6619 : SUCCESS in 13m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8d559260bff84c57872f3189da260c6d : SUCCESS in 12m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c676dc5fb48f42d597bcba7b97246dee : SUCCESS in 13m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5e7fa2a45b554036acdb8fe476401860 : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ffa50f8e854849178190aed5e82c5376 : SUCCESS in 20m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5071c1ca1dc14fcbbc183fbb0805c8b0 : SUCCESS in 33m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/08ec166d718b4aa787345f66ab653f17 : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/157a57029303447ba84d40dde5318ceb : FAILURE in 18m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a5db609afcb449f4973d2897ea3ce6f2 : SUCCESS in 18m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/52be1ca1e293402ba6a83cfa80665c8b : SUCCESS in 23m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bbc9af91148e47969a95b450e470c129 : FAILURE in 48m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b57eb6fb409a41ec944b52b5be1cf7e9 : SUCCESS in 39m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cc7d313560b24172ab921b3c8b705b00 : SUCCESS in 40m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b6fbad8d33f48b29c4cf5b197c592eb : SUCCESS in 12m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c4b03c142e14ab88666da899122a819 : FAILURE in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f7fcef70871147db98e85840ec304236 : SUCCESS in 21m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30fae045e55249ef8eedd99e242796c6 : SUCCESS in 1h 01m 48s","accounts_in_message":[],"_revision_number":5},{"id":"0557ad0f978b0a435eb5cc27714023965b68b5fc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-01-21 17:30:04.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/swift/+/765623 will be better","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9793b7604aaabcfae6b9418c6a64c332d6caed22","revisions":{"fe837dfe53293c2e31bf7857d2a970bcb1216548":{"kind":"REWORK","_number":1,"created":"2020-07-31 21:28:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/744256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/744256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/744256/1"}}},"commit":{"parents":[{"commit":"fffbb9cffd70361e4dcd6cdd2260dec2be5b3e31","subject":"add swift-manage-shard-ranges shink command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fffbb9cffd70361e4dcd6cdd2260dec2be5b3e31"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-31 21:27:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-31 21:27:58.000000000","tz":-420},"subject":"sharding: probe test to exercise manual shrinking","message":"sharding: probe test to exercise manual shrinking\n\nChange-Id: Ib225abd17666e8c966ebb4a559cef3851bcab844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe837dfe53293c2e31bf7857d2a970bcb1216548"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe837dfe53293c2e31bf7857d2a970bcb1216548"}]},"branch":"refs/heads/master"},"e2242201749e10a483b232319cc572559cda4c22":{"kind":"REWORK","_number":2,"created":"2020-10-02 00:36:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/744256/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/744256/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/744256/2"}}},"commit":{"parents":[{"commit":"c06fc4f85d97acb9f77682f03866e4b1ec94c395","subject":"add swift-manage-shard-ranges shrink command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c06fc4f85d97acb9f77682f03866e4b1ec94c395"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-31 21:27:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-02 00:29:20.000000000","tz":-420},"subject":"sharding: probe test to exercise manual shrinking","message":"sharding: probe test to exercise manual shrinking\n\nChange-Id: Ib225abd17666e8c966ebb4a559cef3851bcab844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2242201749e10a483b232319cc572559cda4c22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2242201749e10a483b232319cc572559cda4c22"}]},"branch":"refs/heads/master"},"a1019f1fad1cd5217313f7aa4f057a34fd5f0c0d":{"kind":"REWORK","_number":3,"created":"2020-10-12 20:22:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/744256/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/744256/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/744256/3"}}},"commit":{"parents":[{"commit":"c06fc4f85d97acb9f77682f03866e4b1ec94c395","subject":"add swift-manage-shard-ranges shrink command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c06fc4f85d97acb9f77682f03866e4b1ec94c395"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-07-31 21:27:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-10-12 20:22:43.000000000","tz":-420},"subject":"sharding: probe test to exercise manual shrinking","message":"sharding: probe test to exercise manual shrinking\n\nChange-Id: Ib225abd17666e8c966ebb4a559cef3851bcab844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1019f1fad1cd5217313f7aa4f057a34fd5f0c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1019f1fad1cd5217313f7aa4f057a34fd5f0c0d"}]},"branch":"refs/heads/master"},"003c2c494ec7c6f3926f674a7de3f71651fcf7cc":{"kind":"REWORK","_number":4,"created":"2020-11-09 22:23:29.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/56/744256/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/744256/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/744256/4"}}},"commit":{"parents":[{"commit":"d45a674004b109c840feacae4f640101d397a141","subject":"WIP: Fix sharder consistency issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d45a674004b109c840feacae4f640101d397a141"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-09 21:46:45.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-09 22:20:10.000000000","tz":-360},"subject":"Add swift-manage-shard-ranges shrink command","message":"Add swift-manage-shard-ranges shrink command\n\n... and some new probe tests and code for finding acceptors\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ib225abd17666e8c966ebb4a559cef3851bcab844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/003c2c494ec7c6f3926f674a7de3f71651fcf7cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/003c2c494ec7c6f3926f674a7de3f71651fcf7cc"}]},"branch":"refs/heads/master"},"9793b7604aaabcfae6b9418c6a64c332d6caed22":{"kind":"REWORK","_number":5,"created":"2020-11-10 14:41:11.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/56/744256/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/744256/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/744256/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/744256/5"}}},"commit":{"parents":[{"commit":"5e0ad7119910aa655a736d2e7447b7b08f19590d","subject":"WIP: Fix sharder consistency issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e0ad7119910aa655a736d2e7447b7b08f19590d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-09 21:46:45.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-11-10 14:41:05.000000000","tz":-360},"subject":"Add swift-manage-shard-ranges shrink command","message":"Add swift-manage-shard-ranges shrink command\n\n... and some new probe tests and code for finding acceptors\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: Ib225abd17666e8c966ebb4a559cef3851bcab844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9793b7604aaabcfae6b9418c6a64c332d6caed22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9793b7604aaabcfae6b9418c6a64c332d6caed22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
