)]}'
{"id":"openstack%2Fswift~744525","triplet_id":"openstack%2Fswift~master~Ie52932d8d4a66b11c295d5568aa3a60895b84f3b","project":"openstack/swift","branch":"master","topic":"sharding-split-brain","hashtags":[],"change_id":"Ie52932d8d4a66b11c295d5568aa3a60895b84f3b","subject":"Refactor audit shard tests","status":"MERGED","created":"2020-08-03 16:47:09.000000000","updated":"2020-11-06 02:38:53.000000000","submitted":"2020-11-06 02:37:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744525-1604630235927-294064b8","meta_rev_id":"72d96af296aaad3c0f36f0aa34a29934ae11a05e","_number":744525,"virtual_id_number":744525,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-06 02:37:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-06 01:56:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-06 01:56:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-06 01:56:38.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-11-06 02:37:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a00b53c66fe8d5dd070fa9fef1014445a6a8f78c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-08-03 16:47:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9429366eb6d76302e7fc311302ee6a7ecbf10fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 19:17:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4921de094a9c45338e05553d69123802 : SUCCESS in 14m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f7abad10c5c94a7e9c02a5ef8010566f : SUCCESS in 11m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a2118ed6f0e34e488727db08a8e453e8 : SUCCESS in 12m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/590d3d41b3fd43ec8a3038c5db24296a : SUCCESS in 15m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/df108450e2834459ab78c11cec43f437 : SUCCESS in 10m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eacb214d599d4cfdbc4ff67a8ed4ea29 : SUCCESS in 15m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1785fde6e6ce49849878ae1678ca777d : SUCCESS in 16m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5c6c87676bd54006bd10670e6065ba61 : SUCCESS in 19m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0db553ac014d4fbb8e660ca1298e23d3 : SUCCESS in 18m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e35ea6da759f45198aed9318fa8fb0a3 : SUCCESS in 23m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/02b3e54486474b09a526505691af5a8f : SUCCESS in 21m 20s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dc1bdb82a36e486c9b264afaa4bd8fb0 : SUCCESS in 25m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/322755c1121c406eade5a065b2d26e0e : SUCCESS in 24m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f3e810fb6bf3429bad8d587387d6e199 : SUCCESS in 53m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/adebb12129aa4360b31d5eb224447c3b : SUCCESS in 56m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/950a9a7690c14391907ed4970c8e4e1d : SUCCESS in 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b114397d6c48418994e20dce40f7e815 : SUCCESS in 7m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/80444bb0a7b24b118dbc1449d3918e5f : FAILURE in 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce689f4f311fce293371b3d18a472c28dba9ee93","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-11-06 01:56:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a020a4916db13a2108192295ddb3d60cd19bdf1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 01:56:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2620a194bb6a8c9e7f2c24f69e2203e65243185","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 02:37:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a042d8a2fb376178498859bb8af8d4a0d7ab6162","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 02:37:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3ac565e2d9a469f9906da6093d727e0 : SUCCESS in 9m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d2897b1c09394234b5c078d527cebb93 : SUCCESS in 10m 34s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/80909e2c56274c0f90723ec14ec31ff1 : SUCCESS in 9m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/622a082efc7e4f19b40f013c52aa10cc : SUCCESS in 12m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0f2ee3d8a7d64769ba30c9ec340eea32 : SUCCESS in 12m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e0b06f1122a04688ba3df78b72ca805a : SUCCESS in 11m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/398e082beb9a4401abfebd5856a1c5d5 : SUCCESS in 13m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25407fc977cf4a56a020a8e9757a5b30 : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/652fcb435f2b49b58b485295612f19e9 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/69195a97ef4f45ce9077b9351d6a2175 : SUCCESS in 16m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e3237b93c1bd46e6a3db39b1c8e53f27 : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a7047a1cbb184ec6b49a908c49efc5dc : SUCCESS in 15m 42s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/34036b4750eb492d8aa232906d74c68d : SUCCESS in 17m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/37d80deacae24966bc4f7eff658a34b6 : SUCCESS in 37m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/da329de5301b4173bd2f858901cd8f32 : SUCCESS in 38m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4705100ea7b64bfba99a0e42014f292b : SUCCESS in 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b160f231b49b4e00b661804f6b0a5ed8 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b3564703f534eb793fda1112360457bee72b1001","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 02:38:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b9b70f8eefb4ce49209685655fbb3fc : SUCCESS in 1m 12s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/656754af9b39400e9217db0d9462bb45 : FAILURE in 45s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2c8061698dde4bb5b6e0c620fdb4b507 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34dd24e8046f9eff7f3041ffd2e1d519367c2344","revisions":{"34dd24e8046f9eff7f3041ffd2e1d519367c2344":{"kind":"REWORK","_number":1,"created":"2020-08-03 16:47:09.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/25/744525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/744525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/744525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/744525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/744525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/744525/1"}}},"commit":{"parents":[{"commit":"36bd21488e2b1c2fce667412c05ca439c3a17cde","subject":"Address a sharder/replicator race","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36bd21488e2b1c2fce667412c05ca439c3a17cde"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-08-03 16:40:13.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-08-03 16:40:13.000000000","tz":-300},"subject":"Refactor audit shard tests","message":"Refactor audit shard tests\n\nThe existing tests cover a lot of behaviors and carry around a lot of\nstate that makes them hard to extend in a descriptive mannor to cover\nnew or changed behaviors.\n\nChange-Id: Ie52932d8d4a66b11c295d5568aa3a60895b84f3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34dd24e8046f9eff7f3041ffd2e1d519367c2344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34dd24e8046f9eff7f3041ffd2e1d519367c2344"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
