)]}'
{"id":"openstack%2Fswift~747311","triplet_id":"openstack%2Fswift~master~I4daa18c36235e6df65e8b1c00a12dbf10677ca61","project":"openstack/swift","branch":"master","topic":"404-error","hashtags":[],"change_id":"I4daa18c36235e6df65e8b1c00a12dbf10677ca61","subject":"ec: Close down some unused responses more quickly","status":"MERGED","created":"2020-08-20 23:14:12.000000000","updated":"2020-09-08 21:29:59.000000000","submitted":"2020-09-08 21:27:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747311-1599600443224-0b22d573","meta_rev_id":"4ec1ec23054413c0a9e15fbe967d1597ddfcf0cc","_number":747311,"virtual_id_number":747311,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-08 21:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-04 00:29:54.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2020-09-08 19:52:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2020-09-08 19:52:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 19:52:23.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-09-08 21:27:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58fc965cfb19815f154ae0f6ec727b632e6ac7ab","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-08-20 23:14:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6908c4e9a9ca54e94d069922a60b400d2253641a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 01:40:27.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38e068a03539456ba0a90ac8f2e82a23 : SUCCESS in 9m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/731f9a2902be4d9cad4ecd2f84c6e78b : SUCCESS in 1h 15m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1ba50336c24a41fc990ecce1f6212a81 : SUCCESS in 33m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/04b1e67bc54c4356ad6ba3de89548a24 : SUCCESS in 11m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aa565bc0c9fb47af9e46f9e8423f416d : SUCCESS in 11m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a57a6faa17db4015817c179a818e4b24 : FAILURE in 16m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e102ebca71c748c4b24e9010c2f4eea0 : FAILURE in 15m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/26f9aeaf7dec4dc5a52ef3ec431ffa2e : FAILURE in 15m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/50a39f5f73764912b186c5591ae04fb0 : FAILURE in 16m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5edd439e80624d2f9943d17ee674cb5d : SUCCESS in 21m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3c143daef31945faa4ad65c2fc77f360 : SUCCESS in 23m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e014523611c84c94a65cf29f5d87a120 : SUCCESS in 20m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9f52c9cdc10e4649a74df267fdc813a8 : SUCCESS in 24m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dd73fe2472514d2a9d1f42694c2208c7 : SUCCESS in 21m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8da73e833534498f9a5c72c94c2197e8 : SUCCESS in 20m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eea4ca01ffdc46558caad02c890e5122 : SUCCESS in 22m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1bfc13b636bc475c87a676c63ba279da : SUCCESS in 45m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d1d09be756e343fab22da363e196fd48 : SUCCESS in 42m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/82363bf6cde143c0a3cc6c99bd72aacc : SUCCESS in 41m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/deb4800a20344c0099531f7bcb967856 : FAILURE in 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6daa07d921c44ab9a158e8033f3a066 : FAILURE in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b62776f335a043148ea462116a2a7c4f : FAILURE in 9m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46c8b9fb0e9d45fb8c132d24aa07f22b : SUCCESS in 50m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3af10096d03c19f3f4b13c8e0117c94234404546","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-08-21 16:23:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fa93378062ab3ee91f15a1b591a8f093c74133d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 21:05:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb86cc7fb1bb46daa4667bbe524e9e03 : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/3d71ea11e75945089b15da3f804c8e67 : SUCCESS in 1h 14m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a99b28d619ec4bab871d9ffe89d141f1 : SUCCESS in 39m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/21e65270881948cba42ac2596dda0b31 : SUCCESS in 10m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0adfaa3abc8441dd978a3fcc14f05b4c : SUCCESS in 11m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/88338489fa684cc4970cdc8a121ef5d3 : SUCCESS in 12m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2b78f3983eb14cc681843f561db27d8a : SUCCESS in 15m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cf1a501e83ef4f899fc5ddd1eab95a1f : SUCCESS in 14m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/322977f46a9745f6b2ce98f123e724a9 : SUCCESS in 12m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/431793b249ac4e55993fcd4cf2159003 : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d056aecaa95746a7a6de300594eeace7 : SUCCESS in 42m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8886f7ca09124384bca383589922d9fb : SUCCESS in 18m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/01e38e12bf2e40deb2a6e46dfb39a01a : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/390c850b3f0749188e0fe01819deecb2 : SUCCESS in 16m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/81a4ee23f2c840c2b33ae19daf51cc64 : SUCCESS in 17m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/afb6a6a3cfda419d91ae179ab259ab7b : SUCCESS in 20m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9b94a0f9994a4e76924b1ed427280de8 : SUCCESS in 44m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c68e90fb94d41429609483bf264991f : SUCCESS in 32m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e49742416ca94b8fb9e16fc11bf04f21 : SUCCESS in 41m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cd49172fd464dc2b57c299c944b3491 : SUCCESS in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1f960190090459dba4cf8d1c1a7b8b0 : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/079711d3508c4140ab0bb062289656a6 : FAILURE in 8m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c079706e4cf14b91b2c5edfe9b825ec9 : SUCCESS in 1h 12m 02s","accounts_in_message":[],"_revision_number":2},{"id":"013cb6712d07367e422ddc77b4e1c18f3ebbc584","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-08-21 22:17:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6980b7450b043159c02c4aa44638eec276cd6536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 23:27:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f3fe3b3b894eafb4269e47123d120c : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/45703a93c08844a39196e9263263e010 : SUCCESS in 56m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a5d87310c808436abb59dd85572bb95b : SUCCESS in 43m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/df904a3aef5f422185d8dc8f22c1d808 : SUCCESS in 9m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c482eea6de024fafa209fbf5e03d2c5d : SUCCESS in 10m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/63267f4c38b64e47b81320ec2e2715f5 : SUCCESS in 13m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aa1ee62214ad45f4a0a1cc4c690fc89a : SUCCESS in 13m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/62d5098c7bd14504907e9454ca2eea9c : SUCCESS in 16m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b96a8c5837d54e098ca1e08f3061b59e : SUCCESS in 14m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/20fe7217ab074a45a3463005c2e6ed3c : SUCCESS in 19m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dadedfba05f34f7d9f8c47ac6be17810 : SUCCESS in 19m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b02a7dfe9c4499c994ac4ba328efad7 : SUCCESS in 20m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8c2b082ba3f04cb29489c6ca80c0ab9c : SUCCESS in 19m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4a27f63f01ce4e5187be29060f343fb2 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c21d49ad15d54c4e8bc1d123ecfbe233 : SUCCESS in 23m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8cf86fda16684b2f8220484369276eff : SUCCESS in 20m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/daf07b9b1ee446a4a45bda9cb046a6d6 : SUCCESS in 40m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a806d868f6b430c9dd40263a419bebe : SUCCESS in 42m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3630cde6c84d4f2493175f06644b863c : SUCCESS in 40m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a12c7a91cc54403781d49255dae67e2c : SUCCESS in 25m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/642a653e80f0419690ec066eb90ed718 : SUCCESS in 6m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f29dfd576e664eceb8bc1ae432fa8fa5 : FAILURE in 8m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55f84a1a4b7f49b8b1b5d76435a413b6 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":3},{"id":"c4eace1d13a7eef751bcd14d3ff11682ce804bd1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-08-21 23:54:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1f8aca7298e478a633bdcbd1ae04c121d55c9ef1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-08-21 23:55:46.000000000","message":"Patch Set 4: Code-Review+1\n\nThanks for the tests! You\u0027re right; it was lame that it left best_bucket open when we wouldn\u0027t reconstruct.","accounts_in_message":[],"_revision_number":4},{"id":"4c4f140c5ba0a6d9510cbd31016b2730d0c37e98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 00:48:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b4af95cea7f4952b00ec93c732a30ce : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/79d03350eff04eb8a92720be8383892d : SUCCESS in 52m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f73ab705a69e4c73b8be76f6c686ddab : SUCCESS in 42m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d9f140f73ed9424584cb9db81a87596d : SUCCESS in 9m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e2c456f317f444e4b1cea4c6ce941b4e : SUCCESS in 10m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/de1b6e4ff4404f238e1e68aebad76369 : SUCCESS in 15m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fb1e8ac2694541049d1ba2b69c3b4431 : SUCCESS in 12m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ce2f40e054244c018945167064cbae03 : SUCCESS in 15m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/17a93ba07db742fdbfce2f9dc26d2a51 : SUCCESS in 10m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/29ec43d931684dfe8a6360b0e9b412f7 : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ba4292ce1ab04dfea9f8959639c21f13 : SUCCESS in 26m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/06a1ebae74cf4602b7e6feaa73732d74 : SUCCESS in 20m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c0b539527731446eb112da7f70fbd6a0 : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c4f13b88f7f049ce917968d36c663f68 : SUCCESS in 17m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/55ed8673c2e240e3a59991f147802a9d : SUCCESS in 28m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f075f5913d5d414fbbae1946914da289 : SUCCESS in 25m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2dd2ee7003d54361b0cfa76bc5257a58 : SUCCESS in 46m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/29315ff57f47451dbbe18cfe96a25f0a : SUCCESS in 38m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/10cc20084a4542acb125ca97ed7fd670 : SUCCESS in 38m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1130e1cb84fb489c95efd54f2dba75a9 : SUCCESS in 22m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe3f25a07ad542e5acbb8ff1e375b1f1 : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a78534fb6bc943dbac0f8d94a36135cd : FAILURE in 10m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57949d6032834bcc99f7537e3dabd34c : SUCCESS in 44m 52s","accounts_in_message":[],"_revision_number":4},{"id":"77c63479188bb37d7847bd49c4964be76cf2b086","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-08-24 18:35:57.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)\n\nyeah, I think sooner is *probably* better than later","accounts_in_message":[],"_revision_number":4},{"id":"e6d73705031efa5e1cbbfccd0ec64137ada61c29","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-03 18:22:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"60436d7f0722de31d0aece7a9e6974bd32e4c62b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-03 18:24:06.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e57094bf149d3e2fde72f9f18c8e1fb361dd5748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 21:46:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab90585beab4151b3033b20529f50af : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/484804998d1a4a2dae916d68dc4b115e : SUCCESS in 1h 27m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2d2591bc8fc6496287288316382f75fb : SUCCESS in 31m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7a8ac35e38054644bb778568fb0bc90b : SUCCESS in 8m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8f55120ef5484ad39dd8e08b40853c22 : SUCCESS in 10m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/15247af1071846b78b039fec615729e5 : SUCCESS in 13m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0d1538440e8a496eb1fd05da7839dc67 : SUCCESS in 15m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/77521975ec724cccb02ad3a387e3d7a9 : SUCCESS in 18m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c8ca3206ebb5478c820a609032e23f80 : SUCCESS in 18m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/59121b49666143b6af548b401144ff03 : SUCCESS in 23m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c3a724e7f6844304846eccb0417f5c28 : SUCCESS in 20m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5d03276891f246e69cd6287814146d94 : SUCCESS in 21m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ef69df2548b74beeaad43af5d5890113 : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cf67ff4846f04995ba5940c64c4b2e69 : SUCCESS in 18m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d71dbdc5a9dd40abad1b12f837cb6cb0 : SUCCESS in 19m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/727ee234050e48dab00bbac5a2648294 : SUCCESS in 20m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/42e26b79a794432ca66a2fb5652dd526 : FAILURE in 45m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c24f1f8258b408c8ed3c5da9dfc25bb : SUCCESS in 40m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2672376ebab047e8b1c3b5443c14476e : SUCCESS in 41m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9430c9474f5f4e3095b78473600ce3cd : SUCCESS in 18m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35764152c6524784896fd8be7d86c744 : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e2296bece644e938d0898d8f75c8263 : FAILURE in 9m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8e2133fc8f048b7a0e03f04ffc33e5b : SUCCESS in 1h 07m 33s","accounts_in_message":[],"_revision_number":5},{"id":"e78dea888cdc15b7681b724e55f2193f806c24e4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-09-04 00:29:54.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"decf52b8a1e62649f5a16ade5242a4acc2ed57fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 01:37:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b783eb1ad5549028f9a9d6bb71174fa : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/ba05db63d9e844458021fe5905dd15e9 : SUCCESS in 1h 04m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6e2e27a59e8a4163916c1dd9f3a7b515 : SUCCESS in 38m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ecbf73b4305843c8b364cf8978b58c17 : SUCCESS in 9m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8cae8a2bbf0b4194a9f0b8f57e85f016 : SUCCESS in 10m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eee07e87b607496b9b8f5d666a71aa5c : SUCCESS in 16m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cdf045a2908c40ff96326fba3138b44a : SUCCESS in 16m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ebe2de2aac8649769ae4a9ef0a6c0509 : SUCCESS in 15m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ec03596ada1d48a6a9a069318beca5b5 : SUCCESS in 13m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/523820133a12430fa810312d3f2bba24 : SUCCESS in 19m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ec91812b6bcf4aca8802b6833a28c692 : SUCCESS in 17m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8f4f4691e3cc4fa6960958921a6cb92f : SUCCESS in 20m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3204c86435e147e08244a596f7758fcf : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3a4ab9d2d35a475da14bfe8bb75754fc : SUCCESS in 20m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ec83c2bfdef74d908244423633210346 : SUCCESS in 18m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d780f2e5f94f4a99a26caa6cdc283ab1 : SUCCESS in 25m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7c87affc6a564c38a5c6304c2ebfbd8b : SUCCESS in 44m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/38eb35f1227044ea9890880e1feb7517 : SUCCESS in 38m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0e5917284f25432e900e5e74a79b7816 : SUCCESS in 43m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2671cf78982f4a84897c0aaff3f86625 : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/053e94a1178d4f209b17653e175c68ba : SUCCESS in 6m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2cef8203f9ef4507a89ec76f94875cad : FAILURE in 10m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/887ecaf7b02b462ca085f5031431431e : SUCCESS in 51m 44s","accounts_in_message":[],"_revision_number":5},{"id":"eaef68aadfff6362d37aafd72c69a1febeb3c1b7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2020-09-08 19:52:23.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)\n\nthis looks like a nice improvement to me","accounts_in_message":[],"_revision_number":5},{"id":"02ebec2c24744256be722bbb1542b2d68b3873af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 19:52:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0c969352d4e00933cada4b398520947ed9bbf005","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 21:27:22.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3057a98d01564de1ad947fdfb108e940 : SUCCESS in 22m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/1667f02e09e9473eb4be58e62242f0e8 : SUCCESS in 1h 10m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3b43dd5825334c6c91e3497e404b0a81 : SUCCESS in 50m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3009614335d44581b7a4d430f06994e5 : SUCCESS in 10m 21s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d1e29a08e7e048a9a31bda9d1dc8eafd : SUCCESS in 11m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6aff08173555461ab46a166846f40a99 : SUCCESS in 15m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7b73c28b96e04c7c92102ff012efea4d : SUCCESS in 12m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/05bb916b43124f08b7040b9c16a4268a : SUCCESS in 14m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5a4eac37eaba4d71b5f26436434c1acd : SUCCESS in 15m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/46b4daf126df47b4a3c79de22baced80 : SUCCESS in 17m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5631ca5977a34b48a7c789c9844e92e2 : SUCCESS in 19m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/67873e8f6e5c4bb5b33d515f6fd4e097 : SUCCESS in 24m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bbf5be3b0f1349e1938e914f86e87520 : SUCCESS in 18m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e21cd428930647b89750acc1f13a75f7 : SUCCESS in 18m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/02fb865ae5d5457a8b17c39874d76b72 : SUCCESS in 41m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d3485e319a04973bb6aa722d9c741b9 : SUCCESS in 36m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a84ad6964d234a38b5eea789e3728122 : SUCCESS in 45m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9cdec3916374f4b8a4ffb0e9fef8c8f : SUCCESS in 15m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0388b05fed5648e3acb39ac68e51eed3 : SUCCESS in 5m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc16aa27d9d749fe869d0a14fcf42d5e : SUCCESS in 54m 33s","accounts_in_message":[],"_revision_number":5},{"id":"f1f5666d6d3da5b12dbd656465625359f5052e96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 21:27:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"6c82d1412167be4d42eb463f752890ef1b74730c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 21:29:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/387e1084282e4bec8b02613b26a89228 : SUCCESS in 1m 25s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e6b46efb08e747419564e73162aa16d2 : SUCCESS in 1m 21s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8d79f749a13e4ceeac22901a83651450 : SUCCESS in 1m 21s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b7b45eadcd3b948e3313ce88e34ed5a7152eaa7b","revisions":{"bd7a6c81c35a19fe5b6d31da475ea65a9169bf86":{"kind":"REWORK","_number":1,"created":"2020-08-20 23:14:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/747311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/747311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/747311/1"}}},"commit":{"parents":[{"commit":"fc4b33c35dd4fbe4069582f19c40a24aef628ce4","subject":"Add sorting_method\u003dnone to get a consistent node order","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc4b33c35dd4fbe4069582f19c40a24aef628ce4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-08-20 22:13:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-08-20 22:15:33.000000000","tz":-420},"subject":"ec: Close down unused responses quickly","message":"ec: Close down unused responses quickly\n\nThese should get GC\u0027ed eventually, but sooner is probably better than later.\n\nChange-Id: I4daa18c36235e6df65e8b1c00a12dbf10677ca61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd7a6c81c35a19fe5b6d31da475ea65a9169bf86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd7a6c81c35a19fe5b6d31da475ea65a9169bf86"}]},"branch":"refs/heads/master"},"2348405b58b04c347d99c8021fc3034d0c5d9ea2":{"kind":"REWORK","_number":2,"created":"2020-08-21 16:23:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/747311/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/747311/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/747311/2"}}},"commit":{"parents":[{"commit":"666262976ea4c550a7eab5849ae911fbca769156","subject":"Add sorting_method\u003dnone to get a consistent node order","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/666262976ea4c550a7eab5849ae911fbca769156"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-08-20 22:13:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-08-21 16:20:42.000000000","tz":-420},"subject":"ec: Close down unused responses quickly","message":"ec: Close down unused responses quickly\n\nThese should get GC\u0027ed eventually, but sooner is probably better than later.\n\nChange-Id: I4daa18c36235e6df65e8b1c00a12dbf10677ca61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2348405b58b04c347d99c8021fc3034d0c5d9ea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2348405b58b04c347d99c8021fc3034d0c5d9ea2"}]},"branch":"refs/heads/master"},"43e3dc9d8b1fc2db5051cc9d700716d1fe412019":{"kind":"REWORK","_number":3,"created":"2020-08-21 22:17:22.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/747311/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/747311/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/747311/3"}}},"commit":{"parents":[{"commit":"1394bbd3134c22207437e73d9c727c7320dde645","subject":"Extend concurrent_gets to EC GET requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1394bbd3134c22207437e73d9c727c7320dde645"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-08-20 22:13:41.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2020-08-21 22:16:34.000000000","tz":-300},"subject":"ec: Close down some unused responses more quickly","message":"ec: Close down some unused responses more quickly\n\nThese should get GC\u0027ed eventually, but sooner is probably better than later.\n\nChange-Id: I4daa18c36235e6df65e8b1c00a12dbf10677ca61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43e3dc9d8b1fc2db5051cc9d700716d1fe412019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43e3dc9d8b1fc2db5051cc9d700716d1fe412019"}]},"branch":"refs/heads/master"},"cbd235e5e6569e9179ec0a85b6b6473db187332e":{"kind":"REWORK","_number":4,"created":"2020-08-21 23:54:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/747311/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/747311/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/747311/4"}}},"commit":{"parents":[{"commit":"1394bbd3134c22207437e73d9c727c7320dde645","subject":"Extend concurrent_gets to EC GET requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1394bbd3134c22207437e73d9c727c7320dde645"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-08-20 22:13:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-08-21 23:54:18.000000000","tz":-420},"subject":"ec: Close down some unused responses more quickly","message":"ec: Close down some unused responses more quickly\n\nThese should get GC\u0027ed eventually, but sooner is probably better than later.\n\nChange-Id: I4daa18c36235e6df65e8b1c00a12dbf10677ca61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbd235e5e6569e9179ec0a85b6b6473db187332e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbd235e5e6569e9179ec0a85b6b6473db187332e"}]},"branch":"refs/heads/master"},"b7b45eadcd3b948e3313ce88e34ed5a7152eaa7b":{"kind":"REWORK","_number":5,"created":"2020-09-03 18:22:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/747311/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/747311/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/747311/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/747311/5"}}},"commit":{"parents":[{"commit":"8f60e0a2607514f05fb873e4a313ab4a93df7601","subject":"Extend concurrent_gets to EC GET requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f60e0a2607514f05fb873e4a313ab4a93df7601"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-08-20 22:13:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-09-03 18:20:07.000000000","tz":-420},"subject":"ec: Close down some unused responses more quickly","message":"ec: Close down some unused responses more quickly\n\nThese should get GC\u0027ed eventually, but sooner is probably better than later.\n\nChange-Id: I4daa18c36235e6df65e8b1c00a12dbf10677ca61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7b45eadcd3b948e3313ce88e34ed5a7152eaa7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7b45eadcd3b948e3313ce88e34ed5a7152eaa7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
